)]}'
{"id":"openstack%2Fkolla~822323","triplet_id":"openstack%2Fkolla~master~I112c0ea510aad80f28fc3f8a9f04cdf09c924b18","project":"openstack/kolla","branch":"master","topic":"sledgehammer-setuptools","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-12-20 19:58:49.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-12-20 19:58:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I112c0ea510aad80f28fc3f8a9f04cdf09c924b18","subject":"Workaround setuptools\u003d\u003d60.0.0 breakages","status":"MERGED","created":"2021-12-20 13:48:19.000000000","updated":"2021-12-21 05:18:48.000000000","submitted":"2021-12-20 19:58:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"822323-sledgehammer-setuptools","meta_rev_id":"0e2d6af09258d176f3d3000136804c4799c2abc6","_number":822323,"virtual_id_number":822323,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 19:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2021-12-20 18:40:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2021-12-20 18:40:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-20 14:22:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 16:55:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-12-20 18:40:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"adbb17b84f406f6a497073dde88349be7a953153","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-20 13:48:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b16022621be45cba30dcbe17e779be4433de61c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 14:22:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3322888bfa2047fdb64d80df4f6bb41b : FAILURE in 29m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bf4cdbd2948f4f588ceff9c91d11301b : FAILURE in 15m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b6453b9e57da428492d16faa5e77e122 : FAILURE in 21m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a1466170ed4e4f75b079077e9ca85f5c : FAILURE in 26m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5d28e185e75c43c6b581e3c80d9417f4 : FAILURE in 20m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/df8f807b36414161acfef1da37de3e5e : FAILURE in 14m 01s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e1e429daca2d4e59be940a8d5edf91d7 : FAILURE in 17m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/beb3de4b015f409f8032802b20ca5c52 : FAILURE in 18m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5efe89bfddbe40b99f81343b2fbd554c : FAILURE in 26m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db1a2cbec60c4adeb9888eae13593fc2 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/833e3548c17a45a588d51cf236804bb5 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81dbcff259f644388bf238a2163244a1 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08efcf5bd17043f5b1efaecb60040a33 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0072f716099f4c4bae654f1a4e984513 : FAILURE in 18m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b8930c817d649c38bf12bfdc9014183 : FAILURE in 12m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/370a55c3f9ed41bc9c1eb6251922abd2 : FAILURE in 15m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b7582b2acd8e416d84feeb7f3ab5d022 : FAILURE in 21m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f448997cd6e646c2a23421f8e00db73b : FAILURE in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dbd7d47d1e44ea15c0f97b5d2587946b893a86e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 14:32:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d7f503d24e6c401690420bf552fd927b : FAILURE in 21m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc6393cfafcb4fe38cb6d6cc7aa7721a : FAILURE in 17m 31s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/912ed673eb3445f1a651dbef5895665b : FAILURE in 42m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dab5b0aa10de4221897727741477a178 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebd38467b47440f8a09bdd87d8139b9a : SUCCESS in 8m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d855daaea96941609f37a95bf6bd226e : FAILURE in 31m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a9cdda85d339c1f300d5fae1fff07b424fda685","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-20 16:55:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"deb2e43480472b1f65c492c033785fff1311976f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 17:02:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0e24dbddd6e13d6b43aa9df43ee3727d4dd45d8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 17:03:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"03c340ce841797021f4bc40cb1d74cb4e05b3e91","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-20 17:17:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"547ae0606636615978e4b89ee8280f3163a71eac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 18:35:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d250ddb785aa46c9af430a0b5c79225e : SUCCESS in 26m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1e6d88c788914e80afd4b1f55f8c2481 : SUCCESS in 33m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3aee8a469d0c4be28eed43275b72a0ce : SUCCESS in 1h 09m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/38df2e86b21743caa76c183cd356952c : SUCCESS in 1h 07m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c0b8ae3803cd4918a08352ebec28b1f7 : SUCCESS in 1h 20m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e1c004e24c934240862ca6cc6de1a8eb : SUCCESS in 34m 22s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c1860b9addd84c53aceb88628072bfe2 : SUCCESS in 18m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d766477d5ef144ee833d04d5927b5b60 : SUCCESS in 55m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4e7965aaf96541a28b8c3f8c1d9ac430 : SUCCESS in 1h 05m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4ec289541746ebab2dc952026aa8b5 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f43637ecd3a411ba646e0230cd825b0 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3fc667a0b424d7088e4bfa6c19eaa44 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963d7b2243304e9d8b1a3641881e936a : SUCCESS in 6m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/749cdd41bc3b4e5cac5aa77f758f169b : SUCCESS in 20m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5687882d8ed146b49063cfe99ae28f43 : SUCCESS in 35m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/44de43be7bcf444ea71c47932889a816 : SUCCESS in 51m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/38f80e8279524abba9dd00ef75ff382a : SUCCESS in 52m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f35963f48b9e4766b7acb4122fa687fb : SUCCESS in 1h 15m 02s","accounts_in_message":[],"_revision_number":3},{"id":"96e7de3d8f579d26e075566cc920ea7fb9f672e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-20 18:40:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7876ecbd211a4ad86c82f70d2f24aaffa2b424d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 18:40:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1987c0e36ecae6227e42152240bff2b4255524da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:06:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c2fbb40c83f4162b0d4e4dd7839f908 : SUCCESS in 1h 51m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/87ee8d2668a04b95a9d714639b5c03da : SUCCESS in 1h 33m 56s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fb7e000b951e468480f70e1034d23df3 : SUCCESS in 1h 55m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c35627d77692451abc9aabcad1623934 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52a598dcf40e433e927afdbee27139c5 : SUCCESS in 8m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/856e15c77e8b4a6cbec401d17f694298 : SUCCESS in 1h 35m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b80dfd6cd34471f4563760c9d4c5e00ad300a03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:58:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/abb348d3d61d46ccac6b590dbe8f1f71 : SUCCESS in 35m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4fe8a5b7549640e48c5f9cbcbaabb6e1 : SUCCESS in 1h 06m 11s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e31b6b57394a4705ab9d62461b3c8dde : SUCCESS in 1h 16m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/808f0477d7e740d398b0577fd5ebb341 : SUCCESS in 28m 46s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/27bb405688674350bc2a212b727e3231 : SUCCESS in 1h 01m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5d74bf13ab574e91aec7778657f73b8a : SUCCESS in 1h 06m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78dde808c1344188aabdcea1abcf21b : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e84ff2020a1c4be99e6b2fe2db86d34d : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5355b202fe44355a9869d966d6e85b9 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dde66c41a924e1087ec5a2b5a086d69 : SUCCESS in 5m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/30525a6179d241f8980caf1e8854ec9c : SUCCESS in 30m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/62c270e328b048d39d12d4f0c6a7f625 : SUCCESS in 42m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5ae5f5d894fd424a9982d82eed1bbd43 : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":3},{"id":"0f7a662f83aec6d63aed120d5eba0d4ef5d9d106","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:58:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5e6da78f8f2a1dde59d459a6e5052adeb9086695","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 20:00:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63cb7f0792284edfa37ec3628b658fb9 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3},{"id":"0e2d6af09258d176f3d3000136804c4799c2abc6","tag":"autogenerated:gerrit:revert","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-21 05:18:48.000000000","message":"Created a revert of this change as Iff79de9dabcd4bba795f3c44d3017baa5a860306","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b69c58971895a6ab8a23bcdc4b7b56d01f26e2f5","revisions":{"3eeb6bab4d69976ce4915d4796630cbeed054aec":{"kind":"REWORK","_number":1,"created":"2021-12-20 13:48:19.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/23/822323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/822323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/822323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/822323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/822323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/822323/1"}}},"commit":{"parents":[{"commit":"5dede104bcdfa72a8da32c46fe0bd27f54221d4f","subject":"Merge \"horizon: add various dashboard plugins for ubuntu/binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dede104bcdfa72a8da32c46fe0bd27f54221d4f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-20 13:44:58.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-20 13:47:45.000000000","tz":60},"subject":"Workaround setuptools\u003d\u003d60.0.0 breakages","message":"Workaround setuptools\u003d\u003d60.0.0 breakages\n\nSetuptools 60.0.0 is trying to use it\u0027s own distutils[0]. Once again.\nUse the workaround they propose until all pkgs we need have fixed their\nbuild processes.\n\n[0] https://github.com/pypa/setuptools/blob/main/CHANGES.rst#v6000\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I112c0ea510aad80f28fc3f8a9f04cdf09c924b18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3eeb6bab4d69976ce4915d4796630cbeed054aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3eeb6bab4d69976ce4915d4796630cbeed054aec"}]},"branch":"refs/heads/master"},"07f7396a13cfd3e01eafc002653fc3a0f05889ad":{"kind":"REWORK","_number":2,"created":"2021-12-20 17:02:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/822323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/822323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/822323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/822323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/822323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/822323/2"}}},"commit":{"parents":[{"commit":"5dede104bcdfa72a8da32c46fe0bd27f54221d4f","subject":"Merge \"horizon: add various dashboard plugins for ubuntu/binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dede104bcdfa72a8da32c46fe0bd27f54221d4f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-20 13:44:58.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-20 17:02:38.000000000","tz":0},"subject":"Workaround setuptools\u003d\u003d60.0.0 breakages","message":"Workaround setuptools\u003d\u003d60.0.0 breakages\n\nSetuptools 60.0.0 is trying to use it\u0027s own distutils[0]. Once again.\nUse the workaround they propose until all pkgs we need have fixed their\nbuild processes.\n\n[0] https://github.com/pypa/setuptools/blob/main/CHANGES.rst#v6000\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I112c0ea510aad80f28fc3f8a9f04cdf09c924b18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07f7396a13cfd3e01eafc002653fc3a0f05889ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07f7396a13cfd3e01eafc002653fc3a0f05889ad"}]},"branch":"refs/heads/master"},"b69c58971895a6ab8a23bcdc4b7b56d01f26e2f5":{"kind":"REWORK","_number":3,"created":"2021-12-20 17:03:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/822323/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/822323/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/822323/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/822323/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/822323/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/822323/3"}}},"commit":{"parents":[{"commit":"5dede104bcdfa72a8da32c46fe0bd27f54221d4f","subject":"Merge \"horizon: add various dashboard plugins for ubuntu/binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dede104bcdfa72a8da32c46fe0bd27f54221d4f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-20 13:44:58.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-20 17:03:41.000000000","tz":0},"subject":"Workaround setuptools\u003d\u003d60.0.0 breakages","message":"Workaround setuptools\u003d\u003d60.0.0 breakages\n\nSetuptools 60.0.0 is trying to use it\u0027s own distutils[0]. Once again.\nUse the workaround they propose until all pkgs we need have fixed their\nbuild processes.\n\n[0] https://github.com/pypa/setuptools/blob/main/CHANGES.rst#v6000\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I112c0ea510aad80f28fc3f8a9f04cdf09c924b18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b69c58971895a6ab8a23bcdc4b7b56d01f26e2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b69c58971895a6ab8a23bcdc4b7b56d01f26e2f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
