)]}'
{"id":"openstack%2Fkolla~822351","triplet_id":"openstack%2Fkolla~master~I1a8363da02d98255fce31749f4c9838088443a67","project":"openstack/kolla","branch":"master","topic":"drop-vmtp","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-12-21 12:16:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a8363da02d98255fce31749f4c9838088443a67","subject":"Drop vmtp","status":"MERGED","created":"2021-12-20 18:30:44.000000000","updated":"2021-12-21 12:19:28.000000000","submitted":"2021-12-21 12:16:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822351-drop-vmtp","meta_rev_id":"8fa29978f7867c65dd35262cdad709b20f104547","_number":822351,"virtual_id_number":822351,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-21 12:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-12-21 09:42:00.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-12-21 10:21:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-12-21 10:21:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-20 19:46:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-20 19:47:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-21 09:42:00.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-12-21 10:21:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6968f21fd2777b949313348df53f1d6ce4671b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-20 18:30:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30ae997b341cc621ec1ed8704e51870010f302e9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-20 18:34:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed1ce06f024a6032c7ebe76b69d317c7d107fb9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:46:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/583b2c62a1024d3a8295ed23037735c1 : SUCCESS in 59m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/163e89ba2da5426bacf1750e1e4260e8 : FAILURE in 1h 07m 19s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/449b3c0e3ff84c89992af8d20be3eea9 : FAILURE in 35m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e32de0d10494500ba2faf6afef88224 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11a5d99bdebc4a57a39bd05bf830fb97 : SUCCESS in 8m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/887bb9b5f78b4fed811f9d0347b69e6f : FAILURE in 1h 06m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f87b305344623fe9937fc2658518c19d23beec3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:47:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5dd9a7b6cbf34422bd36e6fea7e4014a : SUCCESS in 2m 53s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bd12570a65604c1f9fb896b53ab7eb41 : SUCCESS in 28m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c8c0859f81424c178c787f47f95c396e : SUCCESS in 39m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d9384d454ecd4cdeb6fd7c76ed210a95 : SUCCESS in 1h 05m 59s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f337f8d9cb5b40e4a2eef5ce77a38416 : SUCCESS in 1h 09m 13s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2bfd7b905166474293c622636bbd94bf : SUCCESS in 1h 14m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0c5f1fd101fe427ea3c62e3b0e452d23 : FAILURE in 35m 05s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/af0647a6c05d4c1abdb8fa474832d2b2 : SUCCESS in 20m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/13feda779da14cfd8db3c7718854c34e : FAILURE in 20m 48s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cc5cfc38da1b4645b883719393cea13f : FAILURE in 19m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eeb8da1a3e74415bb2fd382166d7928 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d51e2d4e78b4c0885113cfe81ab86a9 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a69ba2bb7a6241a791c4f1a857a8f479 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4393618cdd4b4a6da32a060f5099a39b : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11ea8312a18d4981a9fa1d92d595670a : SUCCESS in 3m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/683be393a11e4672a14e7b69ded0c9e7 : SUCCESS in 18m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6832429838124a9280275da6488aa721 : FAILURE in 36m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c20e10aa2334e67b4c11f8ce764f382 : FAILURE in 16m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6104176dae2e4b6aaaa61de83a03476d : SUCCESS in 51m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f88445dbb1994f54a924fd5069e1c4d8 : FAILURE in 16m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4166bf934073a49bcbe68e2132df3c2497e5684a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-21 07:29:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d28d100269b9c4d6cbf65e46c2c6305e2785625d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 09:02:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/caceb7d6ceaa40b996d72682701d46f9 : SUCCESS in 3m 15s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/91cc0cbdfd4d4ada933f4fca994a02da : SUCCESS in 32m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/eac4916cb7874c94a514da8185e93be6 : SUCCESS in 40m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/61f603e74ca44e4aa8563b51cad7d887 : SUCCESS in 1h 22m 07s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9d9d89a0c6184a8b82faf30188ca26be : SUCCESS in 1h 14m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5526c1d57b6946bfb9fd65efbadf3abf : SUCCESS in 1h 30m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1e36b2da9a2748ebbb26efe7a6e0c849 : SUCCESS in 31m 18s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/22aa1418972a4f71bca84d439247361e : SUCCESS in 24m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/08dcaaa39df549f9897efd27d26534d3 : SUCCESS in 1h 12m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3a2f0439d55e4363a4a96e97d7bc103d : SUCCESS in 1h 08m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca24459556fc4392bdfe27ebb51ddcbe : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b0eba87117644c9ab38c46f1f849956 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21763f55eac546c08fcb3f5410fb524c : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2549404c632a4b178de189ba8e84c472 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/621aef8580e742108ce3d5e2fd4a1b2f : SUCCESS in 4m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ba22aa5060fe47658694eb2a7f3da5ce : SUCCESS in 22m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c7465d3e99349fa9759de0174c42fe2 : SUCCESS in 32m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2269e458440b4820b9526758f397f886 : SUCCESS in 57m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d89a39b53a441c19d8bcb072c551b41 : SUCCESS in 56m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f9c25616074244a0a9ad588fe43c25a6 : SUCCESS in 1h 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"048bc1d22352e74884384dcd95296bb381c92668","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 09:18:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/84a10dd79cf94ffe9469fc33dfa642e9 : SUCCESS in 1h 14m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b766d79ae3364d16916574ad87753934 : SUCCESS in 1h 09m 07s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8ff5a1c009c42d3ba02176055ad396d : SUCCESS in 1h 38m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b015bf1d0f5d405aa69e86ecbc5f6996 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6632507dbc2d47ac84121487aa4b5c54 : SUCCESS in 4m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4bbac695db714eeca93c048deeae36c5 : SUCCESS in 1h 08m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12ec519750c5e2efeeed8b9bfdb076059640310e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-21 09:42:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0e2f1701f28fb9e0ecb6c69b4f086ffc7bf3591","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-21 10:21:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"55df9827b92e5d0be71f6888e04502d2ef7d79f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 10:22:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4e1013ded464c4ec9afa78d814a9107717aaf8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 12:16:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/038f043bf1834b94a78bb5c2d96b42fe : SUCCESS in 3m 18s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/82b2461a0b1c4d6db0ef80f0cb409755 : SUCCESS in 39m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/059874d4c2404722abc1025dd7dbf247 : SUCCESS in 1h 10m 25s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2fb040beb9614a99a4e711cc82e232c5 : SUCCESS in 1h 48m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8f613aebbbbb43f69906bcc959f0c8ea : SUCCESS in 30m 27s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7b653a916b4c48bbb3a4519b3ce59556 : SUCCESS in 1h 05m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8355d0747b7745f89a69c154dfd9bcd7 : SUCCESS in 1h 08m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9f4c8994c845c296227db9de35c024 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e757f8ac93c4c9bb60e349e6fbd1d63 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b38450dbe57d4e41913b7f90b0b8de86 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/403169c0d1604b67a04a1a1791becead : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00c8bc96e45f4d99a9d60259ee7a0bfc : SUCCESS in 4m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a78d803891124272bdec08c89e10371d : SUCCESS in 46m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/889bc777f3564ffe888591e504f565c1 : SUCCESS in 1h 02m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/62fac9e2edb74214ad2edeef1156845c : SUCCESS in 1h 08m 21s","accounts_in_message":[],"_revision_number":1},{"id":"88a691fdb3903fe1b8ff9cf6a37ae994ad29ff0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 12:16:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8fa29978f7867c65dd35262cdad709b20f104547","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 12:19:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/948d457222304d7581ff0de9cfb73a38 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5326b85f74f24ae79741a256cdd37b24 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36a1286bbafa94daf3f91932cd0137bb0494fdf0","revisions":{"36a1286bbafa94daf3f91932cd0137bb0494fdf0":{"kind":"REWORK","_number":1,"created":"2021-12-20 18:30:44.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/51/822351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/822351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/822351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/822351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/822351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/822351/1"}}},"commit":{"parents":[{"commit":"5dede104bcdfa72a8da32c46fe0bd27f54221d4f","subject":"Merge \"horizon: add various dashboard plugins for ubuntu/binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dede104bcdfa72a8da32c46fe0bd27f54221d4f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-20 18:30:24.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-20 18:30:24.000000000","tz":0},"subject":"Drop vmtp","message":"Drop vmtp\n\nDetails in the attached reno.\n\nChange-Id: I1a8363da02d98255fce31749f4c9838088443a67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36a1286bbafa94daf3f91932cd0137bb0494fdf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36a1286bbafa94daf3f91932cd0137bb0494fdf0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
