)]}'
{"id":"openstack%2Fkolla~822676","triplet_id":"openstack%2Fkolla~stable%2Fxena~I112c0ea510aad80f28fc3f8a9f04cdf09c924b18","project":"openstack/kolla","branch":"stable/xena","topic":"sledgehammer-setuptools-stable/xena","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-01-09 11:03:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I112c0ea510aad80f28fc3f8a9f04cdf09c924b18","subject":"Workaround setuptools\u003d\u003d60.0.0 breakages","status":"ABANDONED","created":"2021-12-22 14:21:31.000000000","updated":"2022-01-09 11:03:04.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":822323,"cherry_pick_of_patch_set":3,"meta_rev_id":"5aaa98a5b822ba40ee3798eec9df578c13e0a527","_number":822676,"virtual_id_number":822676,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-22 16:01:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2021-12-22 14:22:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 14:47:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-12-22 16:01:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07233a5512099eb98f781b10c4183f825d4e6a09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-22 14:21:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f649c8b35289795c92eda30c6c1b4dd4ebb3f36","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-22 14:22:00.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a963be1a67a6912b6997fd4f3e1f6695e169e80","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-22 14:47:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af2cf44fb405edb4bf41ee57131ea413799979d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 16:01:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9b51f062601d4dbaa59f951c4539a4c7 : SUCCESS in 25m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c848f8d8024540f4abe5fb600779def0 : SUCCESS in 37m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6ef19d7d65f246eaa193abea99b72b2c : SUCCESS in 1h 15m 28s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3ff075fb59844b358e61c75575b6108a : SUCCESS in 1h 15m 14s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8228e3536d1e4efbb2002ee187e93fd8 : SUCCESS in 1h 35m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/83ed2627160e40d2a316c0dc346afc21 : SUCCESS in 46m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d489cf56dc294aae9d20e73c15b29076 : FAILURE in 25m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/347130e223f845d186cd072f994f07cb : SUCCESS in 1h 04m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9852759276d34959aff46e701f699541 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/386f5eaf2e68412bb78f3ae6348cf925 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa4b3247c6824100801fefb90def8d85 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2228a91cf124f66b76440211dacf097 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5861cc89859347e8bbc07d7a0dbffc94 : SUCCESS in 7m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1a23a41481d140c7b5a60878f8eab8e8 : SUCCESS in 26m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba3dd26872434f8f9b2509fa78ffdacd : SUCCESS in 45m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/15ed03af799f43f08282d255488fed7a : SUCCESS in 58m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2d23e22fe89b4546b54adf236634b84a : SUCCESS in 55m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ebb5e47e2c442288abf03bbf265a04b : SUCCESS in 1h 29m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c1e0cdc41e0223d864d43458c0b92a68b618aacb","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 16:09:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/b6fff018881040deb0186dbcf6c0da40 : SUCCESS in 44m 10s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/136855f1488b45c2bfc7380f5bba26da : SUCCESS in 37m 26s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/310a63f544b143d296872356febc3ff9 : SUCCESS in 1h 37m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5b078bf2384849c69013b6cae2fc65de : FAILURE in 38m 00s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/3d5987b7b68f4e7cbd91b3d8b36f10f5 : SUCCESS in 1h 45m 21s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/2e166017a13540cb99fc1dffc0d0d73b : SUCCESS in 1h 37m 48s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/a14149b34aaf4c1599625a30d279da31 : SUCCESS in 1h 28m 25s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/a761a9fc017c4fc1a1e06b987ee519d3 : SUCCESS in 1h 23m 37s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/0e4af66920924fecb66406d66ed386e4 : SUCCESS in 37m 05s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/6041b0689db040b3b31e3ed9a32eeff4 : SUCCESS in 1h 05m 37s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/a39a71a1b3824e65a3a3a4bd005ff09b : SUCCESS in 1h 27m 56s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/5d4347def96c4f448cba055ac7740cf8 : SUCCESS in 36m 17s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/f2b0003349074e4fbf9fd913c6e4bc66 : SUCCESS in 38m 02s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/93f8fdd4d33045e89128a8fb3391babb : SUCCESS in 28m 52s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/72f5c5f25b7e4dbd8d56ec53712f3358 : SUCCESS in 1h 16m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2185159bb2eb4b2e93946b0b887dacd8 : SUCCESS in 1h 34m 43s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/98573ce2d4764f89a8f6e07c27e88f32 : SUCCESS in 1h 14m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ec61fc4ba0a34560803cbd34e20bf73b : SUCCESS in 54m 52s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/d27a9a76341b4a94a58753a11ca56823 : SUCCESS in 1h 04m 12s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/35a3d70af2ad4a43be5fccd1f20e8c29 : SUCCESS in 27m 52s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/da4684b2ba1c4a13aa20d4ba66d83453 : SUCCESS in 44m 42s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/51cc27e8e6f64629bbf7be9d33b10589 : SUCCESS in 51m 56s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/85bda7b3d6e9424f97ee95ebebc93fc9 : SUCCESS in 1h 07m 02s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/a87785f190684972992b9cfb538bbb22 : SUCCESS in 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d0e5d2faf88b11dea18319b71ebed9afaa69620","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 16:10:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b54f87f618614232b391b889d364135c : SUCCESS in 1h 46m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ced579c66804246b8fdd14bb8eaf82a : SUCCESS in 45m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e6bd3966bc8940ae895360d8b4a64592 : SUCCESS in 1h 29m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c1e99963e9f46e687c66fa2db46f1b7 : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d58443767eb348fab884eaaccc546caa : SUCCESS in 12m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8b0211dae7e43fca0d40b8262dae7bc : SUCCESS in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a101b92af54c7726436626a869e0663c96d4e16","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-22 18:26:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"502be5b01c291ba77907660b76fb2085ba579367","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-07 17:28:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5aaa98a5b822ba40ee3798eec9df578c13e0a527","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-09 11:03:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"602578976e0da1377e71dfe62083a2f727a8c0a3","revisions":{"602578976e0da1377e71dfe62083a2f727a8c0a3":{"kind":"REWORK","_number":1,"created":"2021-12-22 14:21:31.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/76/822676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/822676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/822676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/822676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/822676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/822676/1"}}},"commit":{"parents":[{"commit":"24fba48e8dd2a705a5c3a14e9715938338097476","subject":"nova: drop pypowervm dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24fba48e8dd2a705a5c3a14e9715938338097476"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-20 13:44:58.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2021-12-22 14:21:31.000000000","tz":0},"subject":"Workaround setuptools\u003d\u003d60.0.0 breakages","message":"Workaround setuptools\u003d\u003d60.0.0 breakages\n\nSetuptools 60.0.0 is trying to use it\u0027s own distutils[0]. Once again.\nUse the workaround they propose until all pkgs we need have fixed their\nbuild processes.\n\n[0] https://github.com/pypa/setuptools/blob/main/CHANGES.rst#v6000\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I112c0ea510aad80f28fc3f8a9f04cdf09c924b18\n(cherry picked from commit b69c58971895a6ab8a23bcdc4b7b56d01f26e2f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/602578976e0da1377e71dfe62083a2f727a8c0a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/602578976e0da1377e71dfe62083a2f727a8c0a3"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
