)]}'
{"id":"openstack%2Fkolla~822816","triplet_id":"openstack%2Fkolla~master~Ia37792ca6e5b40156ebd57b6c290d1ee9d4ff87a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-12-23 15:14:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia37792ca6e5b40156ebd57b6c290d1ee9d4ff87a","subject":"nova: remove EPEL repository from nova-spicehtml5proxy image","status":"MERGED","created":"2021-12-23 10:56:29.000000000","updated":"2021-12-23 15:16:22.000000000","submitted":"2021-12-23 15:14:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"822816","meta_rev_id":"51f7fe4874c92d8dc9d77fcb7729144811108f75","_number":822816,"virtual_id_number":822816,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-23 15:14:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-12-23 11:54:39.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-12-23 13:42:42.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-12-23 13:42:42.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-23 11:54:39.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-12-23 12:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-23 13:42:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"ec895356583730cc3e7565f0633d05bbf317b770","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-23 10:56:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc5695e49230e86ec46b58493c9b05ac72455361","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-23 11:54:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9172b383b31d947e87c5ec660c56c74d51e866c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 12:21:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f1a44c3b1e7f40969c5ca9515d738df7 : SUCCESS in 24m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a9c8e02225a14552bc841353d23f3021 : SUCCESS in 42m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a8f0fc5be2384bd5878037d04ab9fd56 : SUCCESS in 1h 11m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a2811fa05c244d67baa2a21fc1b398ad : SUCCESS in 1h 12m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a2016b5a940d46e8887696699bfaa943 : SUCCESS in 1h 20m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/72e950b33cba405ea1b3ff24527a6786 : SUCCESS in 31m 34s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/eeb41cb971824559bd481cd6405ba48d : FAILURE in 14m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5775fca81c954b3bb432947e3c74b39b : SUCCESS in 1h 02m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ed3e358ea42b4b1e8c64498d16189506 : SUCCESS in 1h 08m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7abbdd9fc8343eeb3db5e1cf71f629f : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d406501c54f4452bc02142e70d23775 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b30dbe10e2324c88868c3c0b10addc8c : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d820669246fa464a82377cf45ba03f74 : SUCCESS in 5m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/66de891f8a8b4c31956e79425955ecf5 : SUCCESS in 33m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/547b256b832249bba9c4f9ec66daba57 : SUCCESS in 35m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/04872fa6a6ca4eb6bce03b9e6496bef6 : SUCCESS in 50m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f8f9793e3935433f8b1e79b5e51d3d5b : SUCCESS in 49m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/27a32a6626be44afaf38993b3ad31306 : SUCCESS in 1h 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d0d14188df3f5c7a81915bfe939a7f162d7c8828","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 12:27:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5deb297a42b740758066a3f4e928aa7a : SUCCESS in 1h 27m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2703857cbab14831871376e3ec40edee : SUCCESS in 1h 04m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/db6f2e5ed8e742359f03b82ac9592adc : SUCCESS in 1h 24m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e39d032234394322830f5a4985056969 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/17f56e82090e4b328d6f22a2eaea265b : SUCCESS in 4m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/89e0120591654284a4eeed020294bf17 : SUCCESS in 33m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae8e12a03c9c16acf6b2a714f37615247f7b1543","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-23 13:42:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d067ab238e36dd001853e1c1a7fca7b82b5836c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 13:43:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"218ad34203d7fe3e0ac277293a210a7223f7bbe0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 15:14:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/764874bcfe5346cea9fbdaad4a02d1e4 : SUCCESS in 40m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1583b7de0818424cbca615c230eb32bb : SUCCESS in 1h 00m 13s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/61f552d734924dd58968e69a6fc759dd : SUCCESS in 1h 26m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a67237ba2581438386ea501b3e0f9554 : SUCCESS in 29m 11s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/33b7fa03acea4722ae34453813210618 : SUCCESS in 54m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/267341b879c84e41ac8aaf841d8e5195 : SUCCESS in 1h 03m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0f67c5a65fb438d84617a0509ac8e25 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e52a346bc3848ea870034eca735e387 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/163bf35e53014692acacc5733bb4380a : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acff1ec51724406686991b7364800637 : SUCCESS in 7m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf3745f4276d41c68f38cfeb3d69a71b : SUCCESS in 35m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3882f32365fa4d9a9862055f4a6dbda2 : SUCCESS in 50m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/01e587d5975e45fcbe0af83036a6b92f : SUCCESS in 1h 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e7cd79aac5516a573ebe1fb80513983008891790","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 15:14:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"51f7fe4874c92d8dc9d77fcb7729144811108f75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 15:16:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1cc42b44013409588d6f17d9bcfe810 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7893306addf97084250923ea88cd4b4be62677b8","revisions":{"7893306addf97084250923ea88cd4b4be62677b8":{"kind":"REWORK","_number":1,"created":"2021-12-23 10:56:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/822816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/822816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/822816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/822816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/822816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/822816/1"}}},"commit":{"parents":[{"commit":"c8da82967946d5252c557c20423684bed383c228","subject":"Merge \"Make cron\u0027s logrotate schedule configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8da82967946d5252c557c20423684bed383c228"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-23 10:54:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-23 10:55:04.000000000","tz":0},"subject":"nova: remove EPEL repository from nova-spicehtml5proxy image","message":"nova: remove EPEL repository from nova-spicehtml5proxy image\n\nThe EPEL8 repository doesn\u0027t provide the necessary spice-html5 package,\nand the image is marked unbuildable for CentOS. Let\u0027s not make EPEL look\nmore useful than it is.\n\nTrivialFix\n\nChange-Id: Ia37792ca6e5b40156ebd57b6c290d1ee9d4ff87a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7893306addf97084250923ea88cd4b4be62677b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7893306addf97084250923ea88cd4b4be62677b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
