)]}'
{"id":"openstack%2Fkolla~822832","triplet_id":"openstack%2Fkolla~master~If70e2a3949699caa9191dcc646856889e6315c3a","project":"openstack/kolla","branch":"master","topic":"bindep-drop-dbus","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-12-30 15:57:52.000000000","reason":"Change was submitted"}},"hashtags":["trivialfix"],"change_id":"If70e2a3949699caa9191dcc646856889e6315c3a","subject":"bindep: drop dbus","status":"MERGED","created":"2021-12-23 13:38:14.000000000","updated":"2021-12-30 15:59:11.000000000","submitted":"2021-12-30 15:57:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822832-bindep-drop-dbus","meta_rev_id":"5c5bba8a96c20a5fe20854a9270643b096870396","_number":822832,"virtual_id_number":822832,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-30 15:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-12-30 11:33:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-12-30 14:19:37.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-27 13:32:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-12-30 14:19:37.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-23 15:08:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-27 13:32:24.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2021-12-30 11:33:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-12-30 14:19:37.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"}],"messages":[{"id":"a1258c607978a7fb205bead32cbaeab8485999fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-23 13:38:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5936b0cc4e5576bf772b8934f31e428569ae6c5b","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-23 13:38:26.000000000","message":"Hashtag added: trivialfix","accounts_in_message":[],"_revision_number":1},{"id":"3c4111324c9ced7e4e45bcce87dbc5399b4c6e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 15:08:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5b2118a0db5b4e0fb45ff35685c62d79 : SUCCESS in 26m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/56cc97893ed74d0bb9f9ab373bfd1375 : SUCCESS in 39m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0debc580087b4caeb5543cdbc93f3bcf : SUCCESS in 1h 13m 58s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ef91ae87862b4d909d3f7ccac1de01c5 : SUCCESS in 1h 27m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1ae6114f6a7b401d8490b1325bca12e2 : SUCCESS in 1h 13m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7907f641c5054a92a74e1e59f7be303c : SUCCESS in 41m 28s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/64ffc0006b564404b34b64ed4da8a869 : FAILURE in 15m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b2334942c8f046d280dca77b3222386b : SUCCESS in 56m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9d860faac01f4d59882ccde44294544b : SUCCESS in 1h 06m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d8319f553374050b2f348a944f16276 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32622ccc78dc4b1ca610d515526373e9 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3f8e4ddb4f64c509fac43b45c7388ac : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5627ea6f811343728c4f6fc8cbc736a4 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d495ce6e23d48798b90a57b465afa27 : SUCCESS in 3m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/506b010ac47c450f8c4962ffd6b4e1d2 : SUCCESS in 21m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3a4673aab694c31982907029e042a0f : SUCCESS in 34m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e650584ad0004ff78f1f70ca183c8eac : SUCCESS in 47m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8729f07092444224b36afa859ef559a0 : SUCCESS in 1h 00m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9e23ed5b9c98488d96e43dbc302c16a5 : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":1},{"id":"138a0648aeaba267bebe1f88e263bb8d7f5a2de5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 15:09:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/16a83f195a9c4f72a0af88b10903803d : SUCCESS in 1h 18m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bd0cac8ba2a94dc9ab435c1bfa890ea7 : SUCCESS in 37m 05s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ceca1d3cf5f648cd9e806f2d0ddfca76 : SUCCESS in 1h 24m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42a26a4ff6a840738469caca04528b8f : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3162a20fc0bc4b1983b708f460c2ae19 : SUCCESS in 4m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f3136af59e124cbb899cfbfbecb9f592 : SUCCESS in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"282bd7ef7180939dfae16cd6a3c6f4119108b87f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-12-27 13:32:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3e432c7e28aba59a6e7a6f69efd695f3eaa2e10f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-30 11:33:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"42aae83fb62679fa82281f80e10b612e84a3bae4","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-12-30 14:19:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5316fc2a9bf8dc3b0abfb0d2b9042b604d334332","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 14:20:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ab7f0a2594be70d461db85ba4546a9d47791077","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 15:57:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3d54d111e4c741938efd4ef1dfe40db9 : SUCCESS in 39m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d62ab444b7544baca25e43fa355e7d43 : SUCCESS in 1h 11m 17s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/63212917e69e46f5bbc2a0dfec8b58d6 : SUCCESS in 1h 32m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4800d56e57cb4f9c9843482c7dd42097 : SUCCESS in 31m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/927d2f8136aa41dbb724251ab6f655f8 : SUCCESS in 1h 00m 30s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/68156b3953394aea8e9423bf85c20b81 : SUCCESS in 1h 09m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1f6791935a44f9aac0efd6955236652 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52f4ec987ce44626bf32ad08ee3bbc75 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbb941d9a85b4653a65af8c30ef274bf : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d9f003236545d8a75c36588aadc6e0 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6db70d53d21a4817a62de40a09c5fe15 : SUCCESS in 3m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/68962a85b3ec4d58930925c891ee404d : SUCCESS in 30m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2f60d259dbe426998c457562ee0878d : SUCCESS in 38m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c838eeaeff374c5d80d0bc08d5b180da : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":1},{"id":"189e5a86b11a62a0a97bdcd7dc730b22df1a2c07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 15:57:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5c5bba8a96c20a5fe20854a9270643b096870396","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 15:59:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ecd1cb017c41cab74e3ec41f11ee83 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37fa1d66aa714ed2ae017e2a3af7c94d : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7306428284cf5c6abf50b579edc7b6e597d33b7f","revisions":{"7306428284cf5c6abf50b579edc7b6e597d33b7f":{"kind":"REWORK","_number":1,"created":"2021-12-23 13:38:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/32/822832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/822832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/822832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/822832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/822832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/822832/1"}}},"commit":{"parents":[{"commit":"c8da82967946d5252c557c20423684bed383c228","subject":"Merge \"Make cron\u0027s logrotate schedule configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8da82967946d5252c557c20423684bed383c228"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-23 13:37:06.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-12-23 13:37:06.000000000","tz":0},"subject":"bindep: drop dbus","message":"bindep: drop dbus\n\nIt\u0027s an artifact from \"times long gone\". (TM)\n\nChange-Id: If70e2a3949699caa9191dcc646856889e6315c3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7306428284cf5c6abf50b579edc7b6e597d33b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7306428284cf5c6abf50b579edc7b6e597d33b7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
