)]}'
{"id":"openstack%2Fkolla~823105","triplet_id":"openstack%2Fkolla~stable%2Fxena~I423db39e77f337a73d3d88e3dbd56469be45110b","project":"openstack/kolla","branch":"stable/xena","topic":"customizable-variable-fix","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2021-12-28 21:18:30.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-12-28 21:18:30.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-12-28 19:53:49.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I423db39e77f337a73d3d88e3dbd56469be45110b","subject":"Fix variable name","status":"MERGED","created":"2021-12-28 17:59:48.000000000","updated":"2021-12-28 21:19:52.000000000","submitted":"2021-12-28 21:18:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823105-customizable-variable-fix","meta_rev_id":"6c664e46966a3768c7a4d8deb83e3d708d76b742","_number":823105,"virtual_id_number":823105,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-28 21:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2021-12-28 19:53:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2021-12-28 19:53:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-28 18:00:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-28 18:00:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-12-28 19:23:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"597cf8a727d463ca5f96e2ce5ca9616f15af4b26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-28 17:59:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d9c2e6262f73a821ff850e65904db1287d7b114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 19:23:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a07bba0172404de8ad954b57dcc19a9b : SUCCESS in 27m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fc0da898b1fa471abd3b3afe059042be : SUCCESS in 32m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/06a4353e10074c028be9edc64291feab : SUCCESS in 1h 07m 06s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2b47f3645223431c8c43ed1e90e295c7 : FAILURE in 17m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7fc5dac3be9b4c52a3af1f23d9dcbd78 : SUCCESS in 1h 19m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bb6a21112d434b5a90bab09f7304a0a7 : SUCCESS in 22m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/037efce40ac949329399e3e7ae87134e : SUCCESS in 19m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/312dd7be80d1440ba01f2211e2c1efe9 : SUCCESS in 55m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed2898c9c049475db0a4737d8dcb0913 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b34ec175b83b429fadb4619ae037fdb2 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c394d67f134046e6875dcbc6dcb96dda : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4823c50ab8ae479484e4d6223a8267b8 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c27da77c644c4ebeb731288502a69b32 : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b4c8282f62224e71a448778fce673e19 : SUCCESS in 18m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/619d34e3fe69479b8b562ea764552d8e : SUCCESS in 24m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/443a8705f2fe4f7997e00d415104bafc : SUCCESS in 48m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aea574d15fe3418fb32557d438a8d906 : SUCCESS in 1h 10m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/39a474deca8d41eea88822fe8ce9a8f8 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5102b5e6a38e37de43f6007140c4b2d37d3e48ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 19:40:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8f16dc4fd8f4d50abea3f7a0752576c : SUCCESS in 1h 27m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/de6fc5232ded45179cbb3fb003dce846 : SUCCESS in 41m 54s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/20f1f914da154cd9907e76b5e701d22d : SUCCESS in 1h 34m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab7f9b7e32954b45b400438fdc67898c : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/692f6d89aa064595810fb1c12af78a78 : SUCCESS in 4m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/69f38ea6b0f84fccacc1df025d766a89 : SUCCESS in 46m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd9413b0868127983922b6361a8817cf2ad6fe28","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-12-28 19:53:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fadafc4dc1eaaf0909ccbef592b14d5fb5e02203","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 19:54:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"485e27764eb26a1dfb3ce344677ffafa7c7fd072","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 21:18:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a5f2e32a8b0f459c8ff50b3321c88b23 : SUCCESS in 32m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f1ab8642d7f246d093ab3ef0ab7e300e : SUCCESS in 57m 35s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/79dd2946ee3e4e398841c8aea821fcb3 : SUCCESS in 1h 23m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312ad177b079413590cc8430a2d05879 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cba19888db6343689eba0ff9bff234c6 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86f99214d331406e8ccbf9dc60b76829 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50997e922034f49882e8636815d437b : SUCCESS in 5m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/04ade8d805604dfbb22395fa2a179069 : SUCCESS in 26m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e49c2d5bab3847d389f12c9f074da5c7 : SUCCESS in 40m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/995143f8335246f784a7c5b90bef3886 : SUCCESS in 52m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e5a533550b05aa84e84d900021c44d2fa560e23c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 21:18:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c664e46966a3768c7a4d8deb83e3d708d76b742","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 21:19:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/670bd2e23fe44b18a78d193566078993 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8d3709436c6bc0f9d6bc8c038f3f3fb64dd6cb1","revisions":{"c8d3709436c6bc0f9d6bc8c038f3f3fb64dd6cb1":{"kind":"REWORK","_number":1,"created":"2021-12-28 17:59:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/05/823105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/823105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/823105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/823105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/823105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/823105/1"}}},"commit":{"parents":[{"commit":"24fba48e8dd2a705a5c3a14e9715938338097476","subject":"nova: drop pypowervm dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24fba48e8dd2a705a5c3a14e9715938338097476"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-22 07:45:58.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-28 17:59:16.000000000","tz":180},"subject":"Fix variable name","message":"Fix variable name\n\nTrivialFix\n\nThe customization format of the variable name should contain the\n\u003cimage_name\u003e [1]. This change fixes the name of the \u0027remote_apt_keys\u0027\nvariable in base Dockerfile.\n\n1. https://docs.openstack.org/kolla/latest/admin/image-building.html#dockerfile-customisation\n\nChange-Id: I423db39e77f337a73d3d88e3dbd56469be45110b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit cb99f46929298f8e1d939cfca2a1d4831cd1efca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8d3709436c6bc0f9d6bc8c038f3f3fb64dd6cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8d3709436c6bc0f9d6bc8c038f3f3fb64dd6cb1"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
