)]}'
{"id":"openstack%2Fkolla~823320","triplet_id":"openstack%2Fkolla~master~I3a98f1e34100112ba37f8ed8f980aa35d60d8071","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-01-04 14:10:47.000000000","reason":"Michal Nasiadka replied on the change"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-08-18 13:02:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3a98f1e34100112ba37f8ed8f980aa35d60d8071","subject":"Move to Ceph upstream repos","status":"ABANDONED","created":"2022-01-03 16:22:54.000000000","updated":"2022-08-18 13:02:01.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"329e5f1dd6bf1bc0fa8c5240c2ea5cfb70e687d4","_number":823320,"virtual_id_number":823320,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-04 13:29:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-03 18:51:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 13:28:10.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"}],"messages":[{"id":"431e3572ee9e7c50f96088dcd004f035f6c9a64e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-03 16:22:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc4d455ca24fc577ca462457449944b4c3a185ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-03 17:20:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7c95335ac18e26d75d2b71739997db59b7745afc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 18:51:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f3dd1841765045449f9b4b380c3ebda9 : SUCCESS in 32m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4bae0cec63f84b0389c9ba3d134c9dbb : SUCCESS in 49m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8cd194c0df814ef68c1babe46b30784d : SUCCESS in 1h 15m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fa33dce97c754bc3a2954a605fd6d4c4 : SUCCESS in 1h 13m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5254354510a04769bb254b2330806f26 : SUCCESS in 1h 28m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c9721239296f4fa9a5db4570e1f5c200 : SUCCESS in 25m 52s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3b4ae5ac3b034e539e15ee982755c161 : SUCCESS in 22m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3854c6531f064f28af54c15c159796ec : SUCCESS in 1h 07m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7674cb59fde8470eaa76e935798aea36 : SUCCESS in 1h 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d25997617e4a898c929574958b20aa : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b8b98d94d6d4164b95b4515da4f69a7 : FAILURE in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfd050b84a15407bbe0ce22c4786b412 : FAILURE in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2fc83f7369d4893b8cee5552be4e9ed : SUCCESS in 7m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/635e23c9a39c4808bc2b93c098d9dfb9 : SUCCESS in 20m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/76647580576d484f82e9e6c74bf05a0f : SUCCESS in 30m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e712e8df9544713b86db10136769b13 : SUCCESS in 54m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f35b4d3edb8d4001b8cbaabecd716d18 : SUCCESS in 57m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d5467a7a7083424c925861f901b9adb8 : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c29598c13cf7f44773f09d4b61dfb579bbde5e1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 19:16:57.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8bc327e6fe84d2397d8956dfc61de80 : FAILURE in 1h 54m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ce95e6db0fc341f1af861385574c4812 : FAILURE in 52m 41s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9b66ae6731c4ddc9f4225a73d4212c4 : FAILURE in 31m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8efed3534aeb47e59644dbaf32f52bed : FAILURE in 6m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db74fef437334bc296173537aa8b603c : FAILURE in 6m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/60ddf50eb134408aa0875e547e04ff6f : SUCCESS in 1h 49m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c1328e5aafed73c21ba880b2efca4a8d16c2b9dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-04 08:09:19.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d13ee55cbfe4841b277c2c77d5959d5aec9f5d62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-04 08:11:36.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"dfac09b00c107bb20210e46586bf70705b6d0137","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 09:12:00.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/aca0a5ca50524d9a8b044fff49ef755a : SUCCESS in 50m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/225970c2c0794306ad4893542a15635f : FAILURE in 19m 32s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/335f85169f8647788058e0da4c6caf76 : FAILURE in 27m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd4c3837448b49de88874acd89870c71 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aca59341e4bd4a71af375c0be58ed032 : SUCCESS in 4m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/337d734b9f5e4eaea71760177f744178 : SUCCESS in 40m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"52c3c84e288ff3a475fa25167628f7c711e4df3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 09:35:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/982eb3b2164949e4b36351a4ed8e67c7 : SUCCESS in 30m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5c762d8d4a2b4721981b756b34bcda33 : SUCCESS in 43m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8cfe55bbc5e8418798a71d23def381b7 : SUCCESS in 1h 11m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f0cabc2b5a5043fbbcc24140a8a9e0a6 : SUCCESS in 1h 05m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4f99debcd3094318a1e72e7cd2cf00d5 : SUCCESS in 1h 19m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0736fd17a4af4595b2de8c4a01c4b53c : SUCCESS in 32m 22s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/785925cf409245779983456bde67f46c : SUCCESS in 16m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ff3a3915de7e455ea0ca0244d0929b85 : SUCCESS in 57m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8b66ca70be61431ebb1f8a6ff3448833 : SUCCESS in 1h 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4c299e0fcf4f90ba5fb7ef415f9fe3 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/202929c28a064c3a990f217c36acec8d : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/307c1e0587ce440d8ed15fc4d424769b : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1424a07c0bf4737b38f9d1e8d4aed5e : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/505e47556a7648a5b9c5b708777498bd : SUCCESS in 18m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ab19d24b9194467fad0705dee62b24ff : SUCCESS in 26m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/19b0f33d883f480d9de9cdba17a295a8 : SUCCESS in 49m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/64eb3af5ff6a49ff805cca1093bfda6c : SUCCESS in 49m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/946714b075b54da28bb640179896b32a : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":4},{"id":"e31eed99810fb1e87be987a066e23b97614808b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-04 12:01:44.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"bd2aef9a1b6233258d1b13fb1521712b209da4f6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-04 13:28:10.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b326c366e89be9036e47bd68f020be7cf3f6227d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 13:29:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bb4652d3da0b4b79ba7b914f17bdd772 : SUCCESS in 27m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ba074b8570344503ad1e60ac92ec1f8f : SUCCESS in 44m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/22b50d040c9d401e868505661f9bdff6 : SUCCESS in 1h 08m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fea69e05b052439f9e7f0d9d44c076fd : SUCCESS in 1h 18m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/55b4f37e0ae240238bfdb0e1e0f020f2 : SUCCESS in 1h 23m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2cdee230044d45e2a6b64add2c7e06a9 : SUCCESS in 32m 18s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3c1a5d5c0d8e4cb3bb62638728754e5f : SUCCESS in 15m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/50f84397444d4fb9ada5561ccd59acbe : SUCCESS in 1h 00m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ccecc658d0664240bd2b884fd3a35e7d : SUCCESS in 1h 09m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19d383a5c73649819063a1dece2b54ad : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d82ac22b0284863ab49ad64879a940a : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/850027a534c340299fb423c7421f07e9 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668fb031ae9f4c59ade4640224c4e215 : SUCCESS in 5m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/abaa6ae8bd934058a21fa90ddd1e0661 : SUCCESS in 22m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/17583cdf36e84e37a2265d5543d5d224 : SUCCESS in 39m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f11515c333084d15867d79721f6bb7b2 : SUCCESS in 52m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/485f1eed101542b18bea02edcf122b31 : SUCCESS in 43m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3ad1583528874078af377fece211bd7f : SUCCESS in 1h 10m 49s","accounts_in_message":[],"_revision_number":5},{"id":"b9d1dcaa35e2f8bcb765542c3660bf4a0a6bfc4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 13:47:31.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/af268630f018454d8ccf4190453fea24 : SUCCESS in 1h 43m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/24a504e7406346acbb28d6ad123e414f : FAILURE in 28m 42s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/29ae74a1e98c421fb997c45db1254528 : FAILURE in 32m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fef7b9ccdfc7456e8ae6c545182d0f9b : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31e4493b2fd74c53ac7473b6a1e5e220 : SUCCESS in 6m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9aac9cf4260a4eb0947b6c2f7e706d67 : SUCCESS in 1h 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eaca3764e37c6507d3b1ab50a6d6c91b4db94b05","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-04 14:10:47.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"329e5f1dd6bf1bc0fa8c5240c2ea5cfb70e687d4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-18 13:02:01.000000000","message":"Abandoned\n\nnot pursuing this time (if ever)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fa0693e2ba56ad4534b9683d56889c4b6a99ac4f","revisions":{"02b8307b9af218f34b532ee2c40072b7212ba85e":{"kind":"REWORK","_number":1,"created":"2022-01-03 16:22:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/20/823320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/823320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/823320/1"}}},"commit":{"parents":[{"commit":"21c761260a7c641020c101a47b08c77cbc4f94fc","subject":"Merge \"bindep: drop dbus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21c761260a7c641020c101a47b08c77cbc4f94fc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-03 16:20:52.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-03 16:22:48.000000000","tz":60},"subject":"WIP: Move to Ceph repos","message":"WIP: Move to Ceph repos\n\nChange-Id: I3a98f1e34100112ba37f8ed8f980aa35d60d8071\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02b8307b9af218f34b532ee2c40072b7212ba85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02b8307b9af218f34b532ee2c40072b7212ba85e"}]},"branch":"refs/heads/master"},"c789178f2e0b5acd3c2a90a270f70a2b4a4e8028":{"kind":"REWORK","_number":2,"created":"2022-01-03 17:20:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/20/823320/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/823320/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/823320/2"}}},"commit":{"parents":[{"commit":"21c761260a7c641020c101a47b08c77cbc4f94fc","subject":"Merge \"bindep: drop dbus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21c761260a7c641020c101a47b08c77cbc4f94fc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-03 16:20:52.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-03 17:20:20.000000000","tz":0},"subject":"WIP: Move to Ceph repos","message":"WIP: Move to Ceph repos\n\nChange-Id: I3a98f1e34100112ba37f8ed8f980aa35d60d8071\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c789178f2e0b5acd3c2a90a270f70a2b4a4e8028"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c789178f2e0b5acd3c2a90a270f70a2b4a4e8028"}]},"branch":"refs/heads/master"},"7f002f836f0715361284e9bfcf7026e30bfd464e":{"kind":"REWORK","_number":3,"created":"2022-01-04 08:09:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/20/823320/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/823320/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/823320/3"}}},"commit":{"parents":[{"commit":"21c761260a7c641020c101a47b08c77cbc4f94fc","subject":"Merge \"bindep: drop dbus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21c761260a7c641020c101a47b08c77cbc4f94fc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-03 16:20:52.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-04 08:09:17.000000000","tz":0},"subject":"WIP: Move to Ceph repos","message":"WIP: Move to Ceph repos\n\nChange-Id: I3a98f1e34100112ba37f8ed8f980aa35d60d8071\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f002f836f0715361284e9bfcf7026e30bfd464e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f002f836f0715361284e9bfcf7026e30bfd464e"}]},"branch":"refs/heads/master"},"9b6972b9e541d1e73976fea155678a78020a6cc5":{"kind":"REWORK","_number":4,"created":"2022-01-04 08:11:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/20/823320/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/823320/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/823320/4"}}},"commit":{"parents":[{"commit":"21c761260a7c641020c101a47b08c77cbc4f94fc","subject":"Merge \"bindep: drop dbus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21c761260a7c641020c101a47b08c77cbc4f94fc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-03 16:20:52.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-04 08:11:35.000000000","tz":0},"subject":"WIP: Move to Ceph repos","message":"WIP: Move to Ceph repos\n\nChange-Id: I3a98f1e34100112ba37f8ed8f980aa35d60d8071\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b6972b9e541d1e73976fea155678a78020a6cc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b6972b9e541d1e73976fea155678a78020a6cc5"}]},"branch":"refs/heads/master"},"fa0693e2ba56ad4534b9683d56889c4b6a99ac4f":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-01-04 12:01:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/20/823320/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/823320/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/823320/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/823320/5"}}},"commit":{"parents":[{"commit":"21c761260a7c641020c101a47b08c77cbc4f94fc","subject":"Merge \"bindep: drop dbus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21c761260a7c641020c101a47b08c77cbc4f94fc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-03 16:20:52.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-04 12:01:44.000000000","tz":0},"subject":"Move to Ceph upstream repos","message":"Move to Ceph upstream repos\n\nUnfortunately that means we need to enable EPEL.\n\nChange-Id: I3a98f1e34100112ba37f8ed8f980aa35d60d8071\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa0693e2ba56ad4534b9683d56889c4b6a99ac4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa0693e2ba56ad4534b9683d56889c4b6a99ac4f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
