)]}'
{"id":"openstack%2Fkolla~823455","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I423db39e77f337a73d3d88e3dbd56469be45110b","project":"openstack/kolla","branch":"stable/victoria","topic":"customizable-variable-fix-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-01-05 16:15:50.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-05 16:15:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I423db39e77f337a73d3d88e3dbd56469be45110b","subject":"Fix variable name","status":"MERGED","created":"2022-01-05 10:12:01.000000000","updated":"2022-01-05 16:17:49.000000000","submitted":"2022-01-05 16:15:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823455-customizable-variable-fix-stable/wallaby-stable/victoria","cherry_pick_of_change":823454,"cherry_pick_of_patch_set":2,"meta_rev_id":"08c968f88de5806094841d1f58b982026346b23d","_number":823455,"virtual_id_number":823455,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-05 16:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-01-05 10:20:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-05 14:52:39.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-05 14:52:39.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 10:12:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-01-05 10:20:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-01-05 12:00:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-05 14:52:39.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"e9bf3c3c95e5c4534a58fd47a765517f3124e4cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-05 10:12:01.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"3432f843c84529a761d89ccf91eab29f4264668f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-05 10:20:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f157d41de47d5b9b16ba9e93671a533097aa4240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 12:00:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4764585f5a504e649d2beaa8b3f71853 : SUCCESS in 42m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f63d32cc6ee849a49a80c22f7d5fdb2d : SUCCESS in 41m 28s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7d933597b8984f91996d0242586b7162 : SUCCESS in 28m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a8915c79dc1e4c999808364ad44be635 : SUCCESS in 49m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e7efb0ff1ac34ba4a44e1ac6749b6851 : SUCCESS in 1h 23m 49s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/dc7fb3910834463b98843f81b6f11b3b : SUCCESS in 1h 17m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5a6631538fc04b9a85a0e0877edf4a40 : SUCCESS in 1h 45m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b2f8a4bd4bdd42619674b869ee93af22 : SUCCESS in 1h 03m 59s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/41d347df858a4cc0804bf8f7ce765090 : SUCCESS in 1h 03m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2ac162b4bd0d49718197328e6aabf6fb : SUCCESS in 1h 17m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/61dc67230a0248b9b8f642bca9873814 : SUCCESS in 20m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ba0045b8c3b147f4a25d1e564d951001 : SUCCESS in 19m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8c3346f4e03d4b6a87c2919bb6d3b09d : SUCCESS in 1h 02m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/861654c3ea134d0490c119b54b5779ba : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0271ef420d34b3eb160022f608edbe2 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7663ce30186a4037b88ee0f665346b5e : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29afaa5b52d1448faade65c347ebb3fb : SUCCESS in 6m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd934bba47ca4939b2f96c65558ec54a : SUCCESS in 20m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a53805ed16ef4abc8e83dee1f3237310 : SUCCESS in 22m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/42f74ed7dfd44dbc952a51c024a10480 : SUCCESS in 1h 03m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5635228679a7473dbb68328f9b33e92a : SUCCESS in 49m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5ef3897a0ba54a21845d694dd9f34612 : SUCCESS in 1h 37m 21s","accounts_in_message":[],"_revision_number":1},{"id":"617baa9128af8d08938d7f118d1d2d0d432c0476","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 12:41:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/126b5290bec24076a79ae0ecd389e79b : SUCCESS in 2h 18m 28s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b99dee011b79411d8ca357a559d78aab : SUCCESS in 2h 13m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0572da153d69487389abca49b1e52acd : SUCCESS in 1h 50m 54s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ebc27b9ad271426aba9fcef88b6cd438 : SUCCESS in 2h 26m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/369fe4d60ad747589b5a8a89a4d7d58a : SUCCESS in 1h 47m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1de53539e58f9e70d17c7fa117ba41b0c243e811","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-01-05 14:52:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86093ac6c7e7fe2e9fc02ec370e2235f6103f1ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 14:53:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7201fe23bf4c27662adedbcc08520b7d5cec95ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 16:15:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d978464e521c46e0b0f45724459f413f : SUCCESS in 31m 12s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5dbfcf6205524096a64dc627b9b3bfb6 : SUCCESS in 33m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/514a075591164629a5f2f3c316741fcf : SUCCESS in 1h 10m 40s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/55d49c6c08344a05b21bf8fe5ae9a6dd : SUCCESS in 1h 17m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/127da94e37c34c43855d7045a01af591 : SUCCESS in 1h 07m 30s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b7e2f229d0bc4d21824b64fc39c03b51 : SUCCESS in 1h 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f70796385e64699b844887df28cdb38 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78912cff97ec42fba837b67d4ca733a5 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea8c63f026bc4b6ebfb267fae5d0ac04 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaef527cc11e4712b190556b3ad74d5c : SUCCESS in 6m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/44d478ca875c498aa800ee52085e1d29 : SUCCESS in 26m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f5372d3e5a834238b3cf31e3b3eaed87 : SUCCESS in 49m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/024f924412f8419db2a99749c5b3c670 : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5c93146231c861e053885a5cf15b3b562baad208","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 16:15:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"08c968f88de5806094841d1f58b982026346b23d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 16:17:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b793a23034c48499ac27d28652959e7 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"464d1517534fddefc70367abd19cb0acc8c6d694","revisions":{"464d1517534fddefc70367abd19cb0acc8c6d694":{"kind":"REWORK","_number":1,"created":"2022-01-05 10:12:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/823455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/823455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/823455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/823455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/823455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/823455/1"}}},"commit":{"parents":[{"commit":"d034341d50a1ef69f759bfc9e18398e0f27cf024","subject":"Ensure nvme-cli is present in nova-compute images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d034341d50a1ef69f759bfc9e18398e0f27cf024"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-22 07:45:58.000000000","tz":180},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-05 10:12:01.000000000","tz":0},"subject":"Fix variable name","message":"Fix variable name\n\nTrivialFix\n\nThe customization format of the variable name should contain the\n\u003cimage_name\u003e [1]. This change fixes the name of the \u0027remote_apt_keys\u0027\nvariable in base Dockerfile.\n\n1. https://docs.openstack.org/kolla/latest/admin/image-building.html#dockerfile-customisation\n\nChange-Id: I423db39e77f337a73d3d88e3dbd56469be45110b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit cb99f46929298f8e1d939cfca2a1d4831cd1efca)\n(cherry picked from commit c8d3709436c6bc0f9d6bc8c038f3f3fb64dd6cb1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/464d1517534fddefc70367abd19cb0acc8c6d694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/464d1517534fddefc70367abd19cb0acc8c6d694"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
