)]}'
{"id":"openstack%2Fkolla~823557","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I080ad3bbc0e53cec2ac11bfd485c97d88388ef79","project":"openstack/kolla","branch":"stable/victoria","topic":"ci-stop-testing-non-stream-centos","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-01-14 11:39:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I080ad3bbc0e53cec2ac11bfd485c97d88388ef79","subject":"[CI] Stop testing non-stream CentOS","status":"MERGED","created":"2022-01-05 16:17:28.000000000","updated":"2022-01-14 11:48:16.000000000","submitted":"2022-01-14 11:39:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823557-ci-stop-testing-non-stream-centos","meta_rev_id":"e7990738e1839a6252a69d89880bc7b9cc5fd5b3","_number":823557,"virtual_id_number":823557,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-14 11:39:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-01-14 11:03:41.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-01-06 09:00:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-01-14 11:03:41.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 16:18:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-05 16:18:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 09:00:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-14 11:03:41.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"dd2c4ae83a43aa6cf617968062039de214b9b7a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-05 16:17:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18c2fc5825ad05c4ef0dec7b9b7cda1fa60bddf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 16:18:03.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/victoria.  The error was:\n\n  Nodeset kolla-centos8 does not match existing definition in branch stable/ussuri\n\nThe error appears in the following nodeset stanza:\n\n  nodeset:\n      name: kolla-centos8\n      nodes:\n        - name: primary\n          label: centos-8-stream\n\n  in \"openstack/kolla/.zuul.d/base.yaml@stable/victoria\", line 54, column 3","accounts_in_message":[],"_revision_number":1},{"id":"5b042b6b5393dc3676ddb7891f37c8a42b042b1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 16:18:36.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/victoria.  The error was:\n\n  Nodeset kolla-centos8 does not match existing definition in branch stable/ussuri\n\nThe error appears in the following nodeset stanza:\n\n  nodeset:\n      name: kolla-centos8\n      nodes:\n        - name: primary\n          label: centos-8-stream\n\n  in \"openstack/kolla/.zuul.d/base.yaml@stable/victoria\", line 54, column 3","accounts_in_message":[],"_revision_number":1},{"id":"8e55bd3821457836082d5c6df9526f60502a386c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-05 16:21:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"449f926174579969ec9ef7d8b1c77403d755ed54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 16:56:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d9631c30c63e4fb0a3039d06ed739798 : SUCCESS in 28m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/901d54ef4bb8461c9ae16bd27364578e : SUCCESS in 31m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a6fa170fb3f4745b3c98ea1bde44e3e : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fd4d19e1b294689a1cb17de5805921a : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94547afdc92d451da6d3a06dc4ae3e5d : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7738e5d691b43dcbbbba59a4d46712a : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":2},{"id":"9799d8990f09dc71214ce8175621d623cc4df5f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 17:12:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/584e5c16b27e43d7a57924bbf3d0b74b : SUCCESS in 45m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f53c4305e956b8c772d120d2f2370b1919a80f56","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-06 09:00:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eb1c0997e39664d0c823676941ce15993697d131","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-14 11:03:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c9edfa68828f363fabfbc54bb5c856566e91081e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:04:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"22eb225cc19fd64951c2977c072280103d1d7c33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:39:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/613935d52d184973811a25bdf8384314 : SUCCESS in 30m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61dcef7fac7f4c2bb9fd75147e77c207 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51c7a7deb3b94aa5be9fb39dcbb4fae7 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4ac4b101c3742fea13d77084074d6ce : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb35848d34f54627a138e92c664ec804 : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":2},{"id":"5fe1e07598d2ae74ec27ad6c836ea5d04caeb7be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:39:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e7990738e1839a6252a69d89880bc7b9cc5fd5b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:48:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b264516c5c38456482260cdb18803f8c : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bfc4c8e3c5a73bbd0130c0a4328914ee80072b2d","revisions":{"a79c520e81e7ec56431b8c9bcef80fe41596a1e6":{"kind":"REWORK","_number":1,"created":"2022-01-05 16:17:28.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/823557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/823557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/823557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/823557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/823557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/823557/1"}}},"commit":{"parents":[{"commit":"d034341d50a1ef69f759bfc9e18398e0f27cf024","subject":"Ensure nvme-cli is present in nova-compute images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d034341d50a1ef69f759bfc9e18398e0f27cf024"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-05 16:16:33.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-05 16:16:33.000000000","tz":0},"subject":"[CI] Stop testing non-stream CentOS","message":"[CI] Stop testing non-stream CentOS\n\nPer [1].\nIt was agreed on during a Kolla IRC meeting. [2]\n\nThis continues building non-stream images but using stream on\nthe host.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-December/026408.html\n[2] https://meetings.opendev.org/meetings/kolla/2021/kolla.2021-12-22-15.01.log.html#l-53\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/823555\nChange-Id: I080ad3bbc0e53cec2ac11bfd485c97d88388ef79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a79c520e81e7ec56431b8c9bcef80fe41596a1e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a79c520e81e7ec56431b8c9bcef80fe41596a1e6"}]},"branch":"refs/heads/stable/victoria"},"bfc4c8e3c5a73bbd0130c0a4328914ee80072b2d":{"kind":"REWORK","_number":2,"created":"2022-01-05 16:21:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/823557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/823557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/823557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/823557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/823557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/823557/2"}}},"commit":{"parents":[{"commit":"d034341d50a1ef69f759bfc9e18398e0f27cf024","subject":"Ensure nvme-cli is present in nova-compute images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d034341d50a1ef69f759bfc9e18398e0f27cf024"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-05 16:16:33.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-05 16:21:19.000000000","tz":0},"subject":"[CI] Stop testing non-stream CentOS","message":"[CI] Stop testing non-stream CentOS\n\nPer [1].\nIt was agreed on during a Kolla IRC meeting. [2]\n\nThis continues building non-stream images but using stream on\nthe host.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-December/026408.html\n[2] https://meetings.opendev.org/meetings/kolla/2021/kolla.2021-12-22-15.01.log.html#l-53\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/823555\nChange-Id: I080ad3bbc0e53cec2ac11bfd485c97d88388ef79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfc4c8e3c5a73bbd0130c0a4328914ee80072b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfc4c8e3c5a73bbd0130c0a4328914ee80072b2d"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
