)]}'
{"id":"openstack%2Fkolla~823855","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Ia1ece3b74bd9bc1b0b8cc123abb23048bc50894d","project":"openstack/kolla","branch":"stable/ussuri","topic":"ci-stop-testing-non-stream-centos","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-01-14 11:39:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia1ece3b74bd9bc1b0b8cc123abb23048bc50894d","subject":"[CI] Stop testing non-stream CentOS","status":"MERGED","created":"2022-01-07 18:28:23.000000000","updated":"2022-01-14 11:47:20.000000000","submitted":"2022-01-14 11:39:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823855-ci-stop-testing-non-stream-centos","meta_rev_id":"9a66c4d0de2c2ab021ea71c6d82c8acda0edbc32","_number":823855,"virtual_id_number":823855,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-14 11:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-01-14 11:03:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-01-10 09:21:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-01-14 11:03:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 18:58:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 09:21:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-14 11:03:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"74526f52fb4b092a82034ff0a427493dc0d4b68a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-07 18:28:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f94afe4e8118937b00e06f4bfdb5029a0e2f2a51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-07 18:29:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cae0e4a5b500a8de6746f3b82704b63d07eb3de9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 18:58:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5b88498263674ad58ea1ddf04bae1735 : SUCCESS in 23m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/574b9727bc6447bd86a3d9bac65d683d : SUCCESS in 27m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19a51158362f4d55b1ec6f98afc3846c : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c016c14aa2944ceaccc0d22648a3afd : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50b8ab514aea4d8a93d822741227c483 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf74cd3971e544308639c6c42621d2f2 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":2},{"id":"fb75046ac4faba0f01a3d503c98aacb5193aa1e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 20:05:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/11f8cdaec4944e13968f3bb5b77bf972 : SUCCESS in 1h 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0970454b8ef4d155bc3ac98553bada21930d4ab3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-10 09:21:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cd67ed75829f112ec740ddcf2cf77b2903c65cf5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-14 11:03:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"659d9f700a553f3e97e96295952db09c29a7883d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:04:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"64bf1b3bc4f2a895df96de1bde0160c6ee73d307","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:39:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d93ddc7a43fb4802b3a2ea0c5fccf940 : SUCCESS in 30m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745d02848ff446e78d7f362e96a6897a : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e421aff939ab4d01ad2cfd7e12d68063 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6cfbc2f93de42a4a164257f16c972f6 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f43ee16cff5547e8b46532ec59d28764 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"99c469ed66d87146394283781af1bdb97c10f5a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:39:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9a66c4d0de2c2ab021ea71c6d82c8acda0edbc32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:47:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf84a0f6a53430fbccc75c19c07a977 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15226bf473798aa28db99978ab3a4fbbbd8be1e3","revisions":{"d7584c9433bca9dd3ff6846dc2d6e4946caac68f":{"kind":"REWORK","_number":1,"created":"2022-01-07 18:28:23.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/55/823855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/823855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/823855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/823855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/823855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/823855/1"}}},"commit":{"parents":[{"commit":"f4030389ab5df8ea59c0618ba3e167e34662050c","subject":"Merge \"Add Swift lock path in Swift containers\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4030389ab5df8ea59c0618ba3e167e34662050c"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-07 18:27:43.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-07 18:27:43.000000000","tz":0},"subject":"[CI] Stop testing non-stream CentOS","message":"[CI] Stop testing non-stream CentOS\n\nPer [1].\nIt was agreed on during a Kolla IRC meeting. [2]\n\nThis continues building non-stream images but using Ubuntu on\nthe host (as that is the most reliable we have left here).\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-December/026408.html\n[2] https://meetings.opendev.org/meetings/kolla/2021/kolla.2021-12-22-15.01.log.html#l-53\n\nChange-Id: Ia1ece3b74bd9bc1b0b8cc123abb23048bc50894d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7584c9433bca9dd3ff6846dc2d6e4946caac68f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7584c9433bca9dd3ff6846dc2d6e4946caac68f"}]},"branch":"refs/heads/stable/ussuri"},"15226bf473798aa28db99978ab3a4fbbbd8be1e3":{"kind":"REWORK","_number":2,"created":"2022-01-07 18:29:32.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/55/823855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/823855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/823855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/823855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/823855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/823855/2"}}},"commit":{"parents":[{"commit":"f4030389ab5df8ea59c0618ba3e167e34662050c","subject":"Merge \"Add Swift lock path in Swift containers\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4030389ab5df8ea59c0618ba3e167e34662050c"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-07 18:27:43.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-07 18:29:23.000000000","tz":0},"subject":"[CI] Stop testing non-stream CentOS","message":"[CI] Stop testing non-stream CentOS\n\nPer [1].\nIt was agreed on during a Kolla IRC meeting. [2]\n\nThis continues building non-stream images but using Ubuntu on\nthe host (as that is the most reliable we have left here).\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-December/026408.html\n[2] https://meetings.opendev.org/meetings/kolla/2021/kolla.2021-12-22-15.01.log.html#l-53\n\nChange-Id: Ia1ece3b74bd9bc1b0b8cc123abb23048bc50894d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15226bf473798aa28db99978ab3a4fbbbd8be1e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15226bf473798aa28db99978ab3a4fbbbd8be1e3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
