)]}'
{"id":"openstack%2Fkolla~824370","triplet_id":"openstack%2Fkolla~master~I042da2234d0a06f1b7ddcac9eb6db8b86f13d205","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-01-18 17:00:08.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-01-18 17:00:08.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-01-17 11:52:51.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I042da2234d0a06f1b7ddcac9eb6db8b86f13d205","subject":"magnum: fix issue with keystone auth default policy","status":"MERGED","created":"2022-01-12 10:18:00.000000000","updated":"2022-01-18 17:03:10.000000000","submitted":"2022-01-18 17:00:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824370","meta_rev_id":"0052a3075aaa6782012a4d4e7f9da025fcdf824e","_number":824370,"virtual_id_number":824370,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-18 17:00:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"all":[{"value":1,"date":"2022-01-17 11:34:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-01-13 09:17:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-01-17 11:52:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-15 06:32:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-01-17 11:52:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2022-01-17 11:34:19.000000000","_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 10:58:04.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-01-12 11:31:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-12 14:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-13 09:17:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-15 06:32:19.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-01-15 07:00:01.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-01-15 07:00:01.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-01-17 11:34:10.000000000","updated_by":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"reviewer":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"state":"REVIEWER"}],"messages":[{"id":"6a146f4cb83ac003630a3e19bd1b986fd082658e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-12 10:18:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc5c3b3fd2f19a8a60b4eefbe6ed5e4c3624b898","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-12 10:58:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"38bf529f5a28a35b7b411a4df1c8aa191abdb2fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 11:31:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/41e5270bca88401c8d2bc1986dac808a : FAILURE in 1h 01m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8bd6d04a172a45b3b8bdd0d69a0e2477 : FAILURE in 21m 34s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dbf804ab96644d4087c157a9c7d50069 : FAILURE in 26m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3860dd7326e43398aa8e717e08e6296 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81160afafa1344c8ae7ad3d79dd98814 : SUCCESS in 9m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/115d009a17e548c089d8a62029c889f4 : FAILURE in 59m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c13781587311b344aa5e58447ac688455ee30a31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-12 12:01:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1d8fc0cd25d9e21f78c3af93bbe6afc97c9199c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 13:41:50.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/cfcfa7ef2e86418ba3b312cffca9f45b : SUCCESS in 1h 25m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/387b7dcc843247e1b063dfabee6eebdb : FAILURE in 19m 13s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d51fbe63957476d8dc0c1249b23c33e : FAILURE in 27m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2a714c1f3764db488b29a5b5bde87cb : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9cc63b68523e4a3f864ff52e12c467e5 : SUCCESS in 8m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d98048447034f3b979b7cdcb630d7cc : SUCCESS in 1h 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"88b1a7a934d769e34bc9b54ee2578043314686ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 14:03:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/aef4e8b5ee7649dbac530914bbf24656 : SUCCESS in 31m 22s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/31a84f6b031549eb8567e102d4a73b3c : SUCCESS in 43m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a900eca8d6d04614bc7cea4f58615618 : SUCCESS in 1h 11m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d3f81d91c1bd4da4a24cae96cf73dea4 : SUCCESS in 1h 04m 46s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0e4507ae12d047758e4475f9c859df46 : SUCCESS in 1h 23m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f7f8b71cbaf7469697006fe9da43e37a : SUCCESS in 39m 55s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/71429b8049434a08810eec5601f8c797 : SUCCESS in 21m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/122eabe612f54126b2881e41c44c41b9 : SUCCESS in 56m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6cc14dbd7a66444eaf46dac68c5c8a62 : SUCCESS in 1h 08m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bab66b92899646dea5d0e5f58d500dc0 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bc64ded2457472598780207f2675367 : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0c54c3b14124ae584c931cab0df46cf : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95871506fbe74cd99c815bff0a1cad3b : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a25cb519b664b4ea7c01f9007fbc835 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5183cd5ef28742e4b9f347733189b57a : SUCCESS in 22m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c62c548bcba74f28b9b14248c95a3710 : SUCCESS in 34m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8135ae53a6f946f9809c0547eaf8cf6f : SUCCESS in 55m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1224a919197b46879ada5312606912a3 : SUCCESS in 58m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0d31f280922148a89cbf0430a2eaefb0 : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c27ab260a1600738b9d0f5dfd6f4b280f980be8f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-13 09:17:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c5a8b398ea27e07e279dff24c02550902ccd623f","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-01-15 06:32:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e261e7c66354305d6bf3e0ff0d98b098cbc241c9","author":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"date":"2022-01-17 11:34:10.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"6c61743283665ed90ce456d0c1227f90d1496bbf","author":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"date":"2022-01-17 11:34:19.000000000","message":"Patch Set 2: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5c8880efdfe56aaa5809fb038500afc64330cfeb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-17 11:52:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"876bb1276a152c283fb3da3d1f1ff0727acca3cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 11:53:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"00f777d1d6644d7f627ac8feeb319f53b88e341d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 12:57:03.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9703dfa2a11b48b99faa7735f6d8b4da : SUCCESS in 48m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c1d1cf76f176423fa099a50704c1b09b : RETRY_LIMIT in 3m 52s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6391c09792c64389b7e07c7606349144 : RETRY_LIMIT in 4m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1821addace3b4c3cb0238975a6925fa3 : SUCCESS in 33m 34s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/593624e2c58743ff8a02a61c29cac9f3 : SUCCESS in 1h 01m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d270124dbb4740e8b8ce0a83438edc5f : SUCCESS in 59m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64adf6bcf42b4b47bd9e503edc40a60c : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/967bb7794cca43a2b94da0c6e04e9219 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e88dc7b5a6641cbbb53e5ebf9a2fbc6 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20795f7ad8f54c59a26429131ebb1424 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4932847e067c46c8ba3f32c50169c900 : SUCCESS in 3m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a56ff3f86b0b4e5dae6a12e78cbc3d12 : SUCCESS in 37m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e84d86128b604a84a3631e4be2b1a713 : SUCCESS in 45m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/041c843becb4455e9df8a2f39e2ebbce : SUCCESS in 1h 02m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4af357333172c19863b267f42a920475f3d10e51","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-18 09:27:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7764221f59d2a071b72aaf14d0c24d7cc35ba6ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 11:05:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ac67e8f061dc4714ba6960b08a15a4f9 : SUCCESS in 30m 03s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7f9aead3b8d24ef8ad40f844d11242ca : SUCCESS in 44m 30s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9b66b20da4a64b3db056719e890521e6 : SUCCESS in 1h 23m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/306f3ac8157441b3b1c6a66453fc860c : SUCCESS in 1h 09m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/847b4c6b2c4f46e298665e0fafad04b9 : FAILURE in 25m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/16dd58df989c4d4596e4be773e8ef3fe : SUCCESS in 30m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f0ce534d788844d0a921bbe5dfd3903f : SUCCESS in 19m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fa2088a3940d440eae1e5ddce5c71436 : SUCCESS in 1h 02m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/126dda5a1ca64a8cbde18ab40c095d12 : SUCCESS in 1h 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc9e880abeb84be0a642b1433bfb7cee : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d95909b92d64cd5b5790b6683205b18 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ee4ff492f734ea6973fd81e2ad9d801 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aac0944dff24b7a8496c1e7af682d9d : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/806122d842184b9ea03061b2eb477cf2 : SUCCESS in 5m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/04d617c66f0a4bafae0fed01d26b0071 : SUCCESS in 24m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d5d482ed99b45bc80aafd3ed88a7b2a : SUCCESS in 49m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2717504f44a4625887320c11e983ba0 : SUCCESS in 51m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b85c0964f5834369a91760efb9f9959e : SUCCESS in 51m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b8dfb2a2e08f4c898d24e943f9431f68 : SUCCESS in 1h 07m 55s","accounts_in_message":[],"_revision_number":2},{"id":"dda5c735995f1729006140c361796c3501e456c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 12:23:55.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b3dcbadd2e64df49f943f8ff30614dd : SUCCESS in 2h 23m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c60d435b91949ccb9c0e3cdee424c9a : FAILURE in 37m 21s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/22f5d4eb19d44e2c8828e293c8dd2bf3 : FAILURE in 44m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a94d3044a2984e4c8f12e5012cb733ef : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/775e85298a934682a52b329bf90de14b : SUCCESS in 9m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a41f492dada648959169eee6adeff082 : SUCCESS in 59m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3277d922b2425aec63ced6278e27217bae844461","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-18 13:16:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0be85c85c17166d14538e4c52f2c670df014b898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 14:50:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f8655d51d22848dd8c9c778d034ce5e4 : SUCCESS in 31m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/196e59e72e3540f99b9a7be0bbe22479 : SUCCESS in 41m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/99b2510004d7449f8a829730e534c2ec : SUCCESS in 1h 18m 27s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c1852763fb7c484db830c57383ea2fb5 : SUCCESS in 1h 11m 53s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/29255a564a8d40739b3b48bd5d7902cd : SUCCESS in 1h 29m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b9dc6d1bbdc94be3be46d8ce32a998dc : SUCCESS in 38m 41s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/21db32d98fe74d70b7c8648654fa52bf : SUCCESS in 24m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/31f41bf6655843a9a879f199cca71e98 : SUCCESS in 58m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ccf6c5bb483143daadd823b8cbaa1468 : SUCCESS in 1h 21m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b7f83821054f51b9f88310422a4dc7 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8c73ee84b004d4ab7c0da57ba810166 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ece274acded041b2ba61934d54196579 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae0796791154967ae438a76826818df : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd6d2882b32b47eea99cf885157f5e6a : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5bf5b2c41144424e8d569b70b868c3e6 : SUCCESS in 22m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5b00e01a7a8b4b14ab59d2f5c9bea454 : SUCCESS in 30m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/815be78f5eb343e2881bdb0e7838e391 : SUCCESS in 55m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fef9617c89354be9a83cc5117bbdbccb : SUCCESS in 58m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bae429575c72475ea333be721af68ada : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":2},{"id":"00dfb781d79152fd2fd8a7f1cac9f521a4fba96b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 14:59:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"41c7c512f9c00f94b4caac46bb4d44f397e5acc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 15:22:39.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ba207516b064e738a82117648cb5ad0 : SUCCESS in 2h 03m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9b4f7d1efa34efeb2ce2d93622d1ab3 : FAILURE in 26m 01s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/616295d289114d31a0529fcd08283904 : FAILURE in 26m 51s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3aae5ff528f489fac6bf90bfde5e7b8 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c12b9080db64207b1324d18cf290b72 : SUCCESS in 7m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f3d7a5357374cb598d121ec97c67f7c : SUCCESS in 52m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c12897c47908db632d48e5b98f4254fe18306205","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 17:00:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8d454b0548a648718a230c0fad704510 : SUCCESS in 53m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/392134aa0d3640f79e6c7900a4c73cdf : SUCCESS in 1h 34m 46s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/71aaa279cbb14d5b8076cab2f5219d64 : SUCCESS in 1h 23m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/352aaf4e7c854c96835efb43295c4172 : SUCCESS in 37m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a139fc304ec7486f91ca93b5d228f009 : SUCCESS in 1h 00m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4ec83fe2d8c64b33af85c597b8d183f1 : SUCCESS in 1h 17m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3ece1c5cd88488291ab9f81c4e3738c : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4854e69500284120ab2bd2ed2424778f : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b40e5c7d0dc4c1b83cbe2e7fcdc9e0a : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d558af463d49c9915867d07cf9ecab : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f851e2050af8431a9847943d79836109 : SUCCESS in 5m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ba1aed2a7de46a292defb6b6bec9f9e : SUCCESS in 30m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/12c7b6a38a1f43afbdc6751890f86257 : SUCCESS in 51m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/568873d2ad2645de8d6b39fc889e6171 : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f80530b4b4d61e9c68b696dd94ae198cf359f9ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 17:00:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0052a3075aaa6782012a4d4e7f9da025fcdf824e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 17:03:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897dd18bcd9a48cfa788287bc2391194 : SUCCESS in 1m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dd5e4dd54ee4162b639b7c3ea5137f2 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a536f47a8e077d37ccff3b3201512c2b793b0ab","revisions":{"e0346602c4167cea17946ce454c8fd6da72413d7":{"kind":"REWORK","_number":1,"created":"2022-01-12 10:18:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/824370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/824370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/824370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/824370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/824370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/824370/1"}}},"commit":{"parents":[{"commit":"788e340f37f95a7406aa619fbff797c97348134c","subject":"Merge \"[CI] Make changes to fluentd trigger the EFK job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/788e340f37f95a7406aa619fbff797c97348134c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-12 10:17:06.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-12 10:17:56.000000000","tz":60},"subject":"magnum: fix issue with keystone auth default policy","message":"magnum: fix issue with keystone auth default policy\n\nCloses-Bug: #1957159\nChange-Id: I042da2234d0a06f1b7ddcac9eb6db8b86f13d205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0346602c4167cea17946ce454c8fd6da72413d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0346602c4167cea17946ce454c8fd6da72413d7"}]},"branch":"refs/heads/master"},"0a536f47a8e077d37ccff3b3201512c2b793b0ab":{"kind":"REWORK","_number":2,"created":"2022-01-12 12:01:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/70/824370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/824370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/824370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/824370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/824370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/824370/2"}}},"commit":{"parents":[{"commit":"788e340f37f95a7406aa619fbff797c97348134c","subject":"Merge \"[CI] Make changes to fluentd trigger the EFK job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/788e340f37f95a7406aa619fbff797c97348134c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-12 10:17:06.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-12 12:01:52.000000000","tz":0},"subject":"magnum: fix issue with keystone auth default policy","message":"magnum: fix issue with keystone auth default policy\n\nCloses-Bug: #1957159\nChange-Id: I042da2234d0a06f1b7ddcac9eb6db8b86f13d205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a536f47a8e077d37ccff3b3201512c2b793b0ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a536f47a8e077d37ccff3b3201512c2b793b0ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
