)]}'
{"id":"openstack%2Fkolla~824651","triplet_id":"openstack%2Fkolla~master~Ib3ecd73f9e39e320acb2c5f0962b8af9b1a817e9","project":"openstack/kolla","branch":"master","topic":"apache-storm","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-01-21 16:22:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3ecd73f9e39e320acb2c5f0962b8af9b1a817e9","subject":"Mitigate two Log4j vulnerabilities in Apache Storm","status":"MERGED","created":"2022-01-13 21:35:47.000000000","updated":"2022-01-21 16:24:47.000000000","submitted":"2022-01-21 16:22:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824651-apache-storm","meta_rev_id":"57514db0592fd2023f3da24f076824b8cf38df93","_number":824651,"virtual_id_number":824651,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-21 16:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-01-21 14:40:25.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-01-21 14:46:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-01-21 14:46:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-13 22:56:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 09:36:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-01-20 13:06:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-01-21 14:40:25.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-01-21 14:46:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"57f7f574b06f616444515afad7b57437821c37f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-13 21:35:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b437e88bbf937528588e1211e9e14f00321af83d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 22:56:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0ee2fd3e74ca46ce8070707f3ce4058b : SUCCESS in 27m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8f1a83035867402ab98158c6d8aac21d : SUCCESS in 38m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/02edd79a7c0b4696874a06dd2953e830 : RETRY_LIMIT in 4m 06s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4dcce15e10c5474880f31f8a1347ba59 : RETRY_LIMIT in 3m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ec585a2a6bec4d48960907f7d821c2da : RETRY_LIMIT in 5m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b8f6b4c1987640eabfb388b464adfc15 : SUCCESS in 30m 36s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/272cf8edc3e4434e89b45b198eb8f1a6 : SUCCESS in 19m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/03318b027c4447bf81e6e4250f0f2f9a : SUCCESS in 1h 03m 08s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ec144b371f354ea58222b8ebd6b6cccc : SUCCESS in 1h 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42b7fabba49e43df969b92fcbdd7bdfd : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acd23d69be724dfe8f81ec408dda0c7f : SUCCESS in 8m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24791ef596c948c98ae0298ce7b9755e : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72c79a9b664641b9ab9bf7816980c0ff : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/781845d67d1b4f239ef9b0925898c1c9 : SUCCESS in 6m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c110852751db40398de16f3922da2d48 : SUCCESS in 18m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4bea5cf0c5d64f4e830d22e5d88adf56 : SUCCESS in 32m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/891484db5e5244ae8437be3dc93ce3e9 : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/640c976eee9244ce8483fc3f39ecbba1 : SUCCESS in 54m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1939408204814774a7ad9b639cbb37f9 : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9893eef5098480e6785a70239e1d61f58dafe2da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 22:57:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/645af5a638714d638d57ba7991e19761 : SUCCESS in 1h 18m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1ad1fd65325d46b791de08a47a041f66 : FAILURE in 25m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c363d85ba1a44645b79d51b76387ae32 : FAILURE in 31m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1ed9638deca4717b860099d6ea86458 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3152c5f40d94a6ca11814b8c38f9014 : SUCCESS in 4m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d760d6deef09428c8c8cd982d34e97e8 : SUCCESS in 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d184e6dd564466be003577a77b01aa843b653e9f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-18 09:36:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76c04bd2c1eac5db7bcc7a7ef0c41ae5bb732ae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 11:07:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/516824d7ff034eca96886169f1a473cc : SUCCESS in 31m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c862bcc53fba4a1b85104d7330740ed6 : SUCCESS in 53m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/84e177630bb84efb884c3cd93f6f878a : SUCCESS in 1h 14m 30s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/45774c9500bd45a9aaa35ed0db8a2506 : SUCCESS in 1h 13m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7056f925efa34886bba6591412c5bc8b : FAILURE in 27m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/34233337f57e42828ced9479aface365 : SUCCESS in 23m 52s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ff0578fbf02945abab4c92c841c1a73a : SUCCESS in 19m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/183239fd83c9467980908627ca5aa696 : SUCCESS in 53m 20s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fc877ffa9fca4066b4c8837f21e4204a : SUCCESS in 1h 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cce935116804addb263f4aadd1ab4e3 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/255f207c0678451b88df461f0a2c1955 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/114b7a972c9b4767b14ff028820fcf6e : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045c72c800e8446fb049fd1a40909738 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52c15b33a4a4466785a919dfd505a13a : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/65ba45c48a1045c3985c827f6056a7d2 : SUCCESS in 21m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6be1f0cea96746309e75d7a9ef934c74 : SUCCESS in 32m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b664d2479684ede9352a25791f37f8e : SUCCESS in 59m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2a187700a684ff7b9147243357ce70b : SUCCESS in 52m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d6d9a487be6d480f84dda685dfeefaf2 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5651a228dca9984a9c3ccc7618697c36d1d29ec8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 12:17:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c16715ea6c1466aaec6702f792e2446 : SUCCESS in 2h 30m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0bac932d01484ca7b15edabc008a19e5 : FAILURE in 59m 02s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f0901dc4b2f4098bcb3a3fdb4b6a4a4 : FAILURE in 56m 39s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98719607fff04e2a8625eb02d00b8f2d : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47cce8abc7bd480aba3bbb5584fce1b6 : SUCCESS in 6m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/75927b21927a477ab6b61dd0b6cab2c2 : SUCCESS in 58m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7b1fe7f6af51d5e3c3eef9081f4c0c543c0da5f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-20 13:06:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44de225f54a62e3c169def013da3361632f7f8c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 14:44:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8c163405fe95439ea4139993569c84ed : SUCCESS in 36m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/556cb4cbf0cc4eb19d2a6ae2c3a320db : SUCCESS in 37m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a1bb97c8ded3424aa87272d318b56367 : SUCCESS in 1h 07m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8fe87a7ceba54de8b61f4dbdf4a63127 : SUCCESS in 1h 18m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ba252a6fd2344b48918f84732fd9535e : SUCCESS in 1h 33m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/10b8004ecf274bdaa3bac018e0d206c6 : SUCCESS in 25m 15s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b3d391c9ddfe4d34a572b74880707d38 : SUCCESS in 22m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a05cf301a84940818fb8cb3b57127fae : SUCCESS in 51m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c92bf17880fa4aa1918789fb49d09370 : SUCCESS in 1h 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ef681cbe48d4133a7e3615173540f0c : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e98ac9d93caa48d3a974e48a6e89c2bc : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68f143fe420444c784f6cfa756e40546 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cffb2d2c406545fe8e68413981fce843 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34b435b8febd4622b2b28860794ae13b : SUCCESS in 3m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ebbf60aac7d44dfaa1b03006147cc5dc : SUCCESS in 22m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3108e3771d984b8f8d98ae65c1bce2ae : SUCCESS in 31m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9765f194ff042e5ae35843634731ab6 : SUCCESS in 58m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/33354d0f6f874829b1a9b2e3a16d6a6c : SUCCESS in 48m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ab375808634d436aa19c7e456f220da7 : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":1},{"id":"981335b4aebaf65c5cd54a4fb35440a8d4bdbf4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 15:08:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ead71b7a96240baac9a0016c3873d03 : SUCCESS in 1h 59m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e23d9026ef5844baa858eef759f3c4f4 : FAILURE in 14m 48s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/90c4199b4b28449f8416f4f17f959508 : FAILURE in 30m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e9df1c144bd4d8886ff73e11c5d9649 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42de168458f646a69587f82d24b51e8b : SUCCESS in 5m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8af9d9b5d34a4521b0f6bfe721026ea6 : SUCCESS in 46m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c95a635bd6ff76cee86b97f622562a073ea5bba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-20 15:10:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85370a400dd089af55e13b282625f1ea4b36601b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-21 14:40:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a5f6a3bc4c01c98ca4d0d41a5b588d163843a16","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-21 14:46:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a062bdf8128ce21c1c74f03479bbc490cd6da886","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 14:46:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bca531dcbfc28a7cdefbc8058512eff567b609c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 16:22:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d09ecf0536fa467ebbe5bee1074191f4 : SUCCESS in 45m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/33c1a17b5f6946b1a41c587d87771535 : SUCCESS in 1h 23m 45s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7d8f223a91924d328ec5d6c5db163110 : SUCCESS in 1h 15m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7ea5f07a7a574ff59ce51b0c204f7610 : SUCCESS in 32m 23s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/544bda194ca442cd9fef72ac5dab830e : SUCCESS in 57m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fffc668080da4b428c7d29ec114fee15 : SUCCESS in 1h 28m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8d8c3dbe80432fb821896ded7c3559 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/126fc6605bbc4a598032b6d20c7bd173 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7935489425ef4bb8a70c378c09e33c5c : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e00481d9fa5404589a179950dada08e : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5494531920af452da350eb05faa9f6b0 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/330da9ecffb34df68cfaaaf89aebe38c : SUCCESS in 34m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cea986527e31481ea51b53131ad512a7 : SUCCESS in 49m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d59dccb0cae942f997838d63e33f04e4 : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1991d683cbee307001ae81c37cfd49de8e574aac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 16:22:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"57514db0592fd2023f3da24f076824b8cf38df93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 16:24:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a4a83a28e1c4708b3a7e40f364bc76b : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c072ce61dcb4fbdae39c72de4e44fb4 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"448e4f56aa6218a66b0074329d322c0b6089b2e6","revisions":{"448e4f56aa6218a66b0074329d322c0b6089b2e6":{"kind":"REWORK","_number":1,"created":"2022-01-13 21:35:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/51/824651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/824651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/824651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/824651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/824651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/824651/1"}}},"commit":{"parents":[{"commit":"788e340f37f95a7406aa619fbff797c97348134c","subject":"Merge \"[CI] Make changes to fluentd trigger the EFK job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/788e340f37f95a7406aa619fbff797c97348134c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-01-13 21:35:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-01-13 21:35:27.000000000","tz":60},"subject":"Mitigate two Log4j vulnerabilities in Apache Storm","message":"Mitigate two Log4j vulnerabilities in Apache Storm\n\nThe Log4j version was bumped on GitHub [1] but it is still pending\ninclusion in a release of Apache Storm.\n\nApply the alternative mitigation recommended by Log4j [2] of removing\nthe JndiLookup class from the classpath.\n\n[1] https://github.com/apache/storm/pull/3427\n[2] https://logging.apache.org/log4j/2.x/security.html\n\nChange-Id: Ib3ecd73f9e39e320acb2c5f0962b8af9b1a817e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/448e4f56aa6218a66b0074329d322c0b6089b2e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/448e4f56aa6218a66b0074329d322c0b6089b2e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
