)]}'
{"id":"openstack%2Fkolla~825786","triplet_id":"openstack%2Fkolla~master~I88842c17d8e6ddbe89b984c7b5e0367ed356fa50","project":"openstack/kolla","branch":"master","topic":"kolla-podman","attention_set":{},"removed_from_attention_set":{"35104":{"account":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"last_update":"2023-05-08 12:53:10.000000000","reason":"Change was marked work in progress"},"34113":{"account":{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"},"last_update":"2023-08-29 09:11:34.000000000","reason":"Change was submitted"},"29636":{"account":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"last_update":"2023-08-29 09:11:34.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-08-29 07:41:16.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-29 09:11:34.000000000","reason":"Change was submitted"},"34501":{"account":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"last_update":"2023-08-29 09:11:34.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-08-29 09:11:34.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-08-29 09:11:34.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-08-29 09:11:34.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-08-29 09:11:34.000000000","reason":"Change was submitted"},"33708":{"account":{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true},"last_update":"2023-08-29 09:11:34.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-08-29 09:11:34.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-29 07:59:17.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2023-03-25 20:21:46.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},"34911":{"account":{"_account_id":34911,"name":"Martin Hiner","email":"m.hiner@partner.samsung.com","username":"m.hiner"},"last_update":"2023-05-08 12:53:10.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I88842c17d8e6ddbe89b984c7b5e0367ed356fa50","subject":"Transition to DockerClient","status":"MERGED","created":"2022-01-21 13:05:29.000000000","updated":"2023-08-29 09:12:41.000000000","submitted":"2023-08-29 09:11:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":74,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825786-kolla-podman","meta_rev_id":"dc785fe5b5ed341987c314d51a7f06af29993523","_number":825786,"virtual_id_number":825786,"owner":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-29 09:11:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-08-29 07:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"value":2,"date":"2023-08-29 07:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-29 07:59:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"value":1,"date":"2023-08-29 07:44:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-08-29 07:41:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true},{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"}],"CC":[{"_account_id":34911,"name":"Martin Hiner","email":"m.hiner@partner.samsung.com","username":"m.hiner"},{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-21 13:26:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-21 13:26:54.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-01-21 13:26:54.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-21 13:26:54.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-01-21 13:26:54.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"},"state":"REVIEWER"},{"updated":"2022-01-21 13:26:54.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true},"state":"REVIEWER"},{"updated":"2022-01-21 13:26:54.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"state":"REVIEWER"},{"updated":"2022-01-21 14:30:11.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-02-08 07:15:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-12-30 09:27:30.000000000","updated_by":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"reviewer":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"state":"CC"},{"updated":"2023-01-02 13:51:49.000000000","updated_by":{"_account_id":34911,"name":"Martin Hiner","email":"m.hiner@partner.samsung.com","username":"m.hiner"},"reviewer":{"_account_id":34911,"name":"Martin Hiner","email":"m.hiner@partner.samsung.com","username":"m.hiner"},"state":"CC"},{"updated":"2023-03-25 20:21:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"},{"updated":"2023-04-12 15:02:03.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-04-19 13:35:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-19 14:28:18.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-04-19 15:46:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-05-10 13:17:39.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-08-29 07:41:16.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"70862b8a97393058c00c39bbfb1121da39102376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-21 13:05:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c32f145e1ac9689568e00b02aef8869b6b514ee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 13:26:01.000000000","message":"Patch Set 1: Verified-1\n\n(50 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/62c3b02662744bc39282c851152ed5d8 : FAILURE in 5m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b04a0a84b17f4afbb923ba25a7ac561d : FAILURE in 5m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/67536baeb2514e4f832721879850ba9b : FAILURE in 13m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b6b4eae0716f4af89d421fd3a9c2c120 : FAILURE in 12m 00s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/45d4d3a6a0d44670a9c2119a75e422bc : FAILURE in 11m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e5f4d1b3d35f491ebd6aaf434bb15a0e : FAILURE in 3m 03s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/222098315a42419ba9b8f4179746439d : FAILURE in 3m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e2553f3130774b1e97d0e76a5e675bf9 : FAILURE in 15m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4c278a40157f47b98d5f10c0395d082c : FAILURE in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bde2b44a51f4cad97227780ba6d35aa : FAILURE in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4c0768adb124d16ab7138524ae5eb05 : FAILURE in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f08fb4a383dd4d8b91b0fd0e303f3d70 : FAILURE in 4m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2bd52791db794c05af7f6ac61a8dc072 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9507946032b4c8680436e3cceb0f29d : SUCCESS in 5m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2997b647bc0e4b08a1577bd465c207e3 : FAILURE in 4m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c011aee0305a4a61b84f0f0f71bf18c9 : FAILURE in 4m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b3e633135c444438b2d7540fee53c25 : FAILURE in 9m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e86cc36f469641b4a69142c8f64f80d8 : FAILURE in 9m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f1d5b8779aa946d0a15ba566d4833ac8 : FAILURE in 9m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1d0e3e807140e507f107569a87c8b3c3e9b22444","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 13:38:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/66c9a9052c44435a9dcbf10f136157d4 : FAILURE in 10m 16s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6fbe15f8df4a4a4a990e37e41ddeb057 : FAILURE in 8m 10s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/99cc19f87bdd434e92fe812d7bcbf58d : FAILURE in 19m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77d0d8e3f4eb434ca3e0f240f2c68fd5 : FAILURE in 5m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fbecb2746e7492691c4e02ea8fba5ab : FAILURE in 8m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc00767e78a54ab482e969024f3138ef : FAILURE in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d681283591cae6ceecbc68aa60c7ecbf2c31daa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-21 14:11:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa8633acb4af105851efb73100a018955c65888f","tag":"autogenerated:gerrit:setPsDescription","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-21 14:12:16.000000000","message":"Description of patch set 2 set to \"Fixing pep8\"","accounts_in_message":[],"_revision_number":2},{"id":"4f2b7eb1f2b930936c904bf4f2f2984a32252174","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-21 14:30:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21e00e2c9ccee533a55d8e9cfb31625012ea8eff","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-21 14:37:27.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1b187fd4dec7515725a2a8caed73579dae372d43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 14:40:13.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/079c1d93cc3642e897dffa14ff45a8c9 : FAILURE in 5m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a33c4585f47a4a1da0aa4dcf96417536 : FAILURE in 5m 29s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1176d6f043194439b513dd4ec0db89a0 : FAILURE in 25m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2de125c3e963416eb8cb1afdf792c3d9 : FAILURE in 3m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3898aef5ab0b46c09da5256e13a03a59 : FAILURE in 6m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4d75246383f49908c9cd5ea7f1732fc : FAILURE in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cfa8ae2434db6a57cdc3a1ff6bd67e6d682ab7e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 14:44:07.000000000","message":"Patch Set 2: Verified-1\n\n(16 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3c16294900e24df8b6571faffa72cf4d : FAILURE in 5m 08s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ccd8c591c87b40738a45bdebef3939ee : FAILURE in 6m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/53ca128c625548b9be84009ce3293ad4 : FAILURE in 10m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f42ea544f79246e095b3562534c42927 : FAILURE in 11m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d6c20cad1b9a442398a7c5a3db95a55d : FAILURE in 13m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2a8bcd9ec592460ea23eb71e71261522 : FAILURE in 4m 15s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8d2ef443f5b24a2a8941b6ebb6866ac7 : FAILURE in 3m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0f072968164441b4baf52efe6acf43b9 : FAILURE in 12m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/eeda507b932e4c30ab64bd1bdd879c31 : FAILURE in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e16777a6c0b41cd8f3c99efb3b9d8e9 : FAILURE in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b7ea1e7d4ca446c92e91235d4621ce4 : FAILURE in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5204603e783745adac8ecc8b5464f9b1 : FAILURE in 3m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fcf44ec215814d80bb9fdc4cd819726c : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9b77a5841b4195a39f2e8068710118 : SUCCESS in 4m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1cf1f5ac946b4deaa8f812c74383f63d : FAILURE in 3m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9201b36644384fcfb3e159b75b0e1b6b : FAILURE in 3m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/72267233823a4d3a933e656e9e719f64 : FAILURE in 8m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/67aec99f94d54380a2be452bdf7b0d0c : FAILURE in 8m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/82b4b9c628ce48288e528bf43c260dcd : FAILURE in 8m 52s","accounts_in_message":[],"_revision_number":2},{"id":"8447c74ef441b880dc59d3567e88609cffd02793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-21 15:21:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"54712abf7be129fc4f1295fb891ef1659872a0c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-21 15:38:41.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"cb2b6fca1ca2ba89711b0fd74bfe16384be7e0dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-21 15:51:55.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"ab746d1480db943286612fa6735c90f7026909be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 17:15:08.000000000","message":"Patch Set 5: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4578df21d8fa4ce6aea74d4b5828fcc7 : SUCCESS in 34m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8e167c3679be40b4bfad76cbab790a7b : SUCCESS in 46m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9c92276e8eb54a8ab8478470dcf1f797 : SUCCESS in 1h 08m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9532fca2652a4dfaa9397bc6e31d46cb : SUCCESS in 1h 07m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/63503ffe7c8b46079c3f0233c6c3858f : SUCCESS in 1h 21m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2eb5d439047a4d05a456bfdf4ed6d36b : SUCCESS in 27m 06s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4f6ceeccb9f549fbb24da8108059fa94 : SUCCESS in 17m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e6bf3af7e1414bdb888df554ec9f31a5 : SUCCESS in 49m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f03a62074b36408b9fb8447bb0eb863f : SUCCESS in 1h 04m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb650f6ee33d4da3be4ce3a4db9cefd1 : FAILURE in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2df23de67a6e4fa5afe86426dbfd5d92 : FAILURE in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caff4203cc8d44238c43dc90ab3c8716 : FAILURE in 3m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b7fd03ccc07447bb84fe75ff0872d5e : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2189cedcb345ada79ea795b6380832 : SUCCESS in 6m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf062a20c6c448cca86b7b6f91720c2a : SUCCESS in 19m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2ed455b593a4ada831f848e33753621 : SUCCESS in 29m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/696efbd791dd47829ccce96969a4d4be : SUCCESS in 49m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/402081f17ec84fb9aac692e32b260360 : SUCCESS in 51m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0c8453805cf54a78b8fbad05cde237cc : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":5},{"id":"a63d201e548158ff57bdb0ee6686c8a1066f4c15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 17:17:51.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f56ed239cbd14401bf73c3ac284ab4b0 : SUCCESS in 47m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/572e74cd112e4b1eb0c89d376c8c0396 : FAILURE in 25m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2736752e9c2f4ebaad2a6c032a27fb99 : FAILURE in 31m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/065260f9eec7456289384adb17c5838a : FAILURE in 5m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/965976062e324b55897a30933f8b3560 : FAILURE in 5m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/47efe1168ebc4f8aabb794956b509798 : SUCCESS in 1h 23m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"17c519590e7a563ec82d4dfa0e5b595e28c578cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-25 13:08:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5be7be2599c344689851b5f63ffac897a70ca2ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-25 13:33:49.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"f21a980771c9f28e7debf130aa17239be3f9ddd5","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-25 13:45:38.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"369696568539693439625af0d1cbbcc03618f972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-25 14:01:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a1a473cc33b0077dfcc364228cbe3c9e196c8fcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-25 14:03:15.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"0ae23733f0f047cc5ef17af62e36290508988d35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 15:49:46.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ad512559c1864fcfbd0dfbe6cfe22ee9 : SUCCESS in 24m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ee304d42e5f145fead76d049faf75e73 : SUCCESS in 43m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ded97cdf0bba47ae929693b533c026bb : SUCCESS in 1h 20m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/25a4de82232f4a3f937e1ae4a0cac8dc : SUCCESS in 1h 08m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/74b8a5920c5540ad8d27099880b76476 : SUCCESS in 1h 40m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/48cc3715d2ad4a5197a01ae0076ea751 : SUCCESS in 27m 34s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7d601bcc9c264a6eb404d3770b4d6ca6 : SUCCESS in 17m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/923169fe4100497aa916e2727d82c9e9 : SUCCESS in 53m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8725b14c847d44718c6065d926c86fa1 : SUCCESS in 1h 09m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89df4e79ea414f93a1c718cef7858d84 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12cb110345094be6ad7d38f573ad0a01 : FAILURE in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a6bc36b10ed479f9d457d6e50136184 : FAILURE in 3m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/05a47d5b34494ee3812c338e8a8b13c8 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18395c2e4bd04336b5064e0c5f25e673 : SUCCESS in 6m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/15388bcb0cb740269d81b91b808a8136 : SUCCESS in 40m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b58f075fd7a24aaa9277574ecbd23960 : SUCCESS in 50m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2a1c8cd4961416bae435caffe6d4507 : SUCCESS in 57m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/717b0244b33d4273aba40af338f2362f : SUCCESS in 53m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/af278c29e79a4f6c8c2f122e804e0314 : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":9},{"id":"2510e20f11c238ff8d242b3a3d91c639b3c4374b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:02:42.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/af6533a93ee04bbd9db56ad5e1aea202 : SUCCESS in 1h 56m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/74216878f7f24f1db995ff21d07e8f3c : FAILURE in 12m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/68936551661a49caa165a347bea5fedd : FAILURE in 33m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c27c3dd73c7d440aa8228a698d97393a : FAILURE in 5m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05a2a097816e499ab3fcab2f3a887c30 : FAILURE in 4m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/679bca2c2b804ea1b65455037274d8e7 : SUCCESS in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"dd963895e0ed4ee56368b1b86668248e6af92c8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-25 18:23:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f6ab73da773624a90f8b51795eb540b29235e4a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 19:26:48.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9562ccea89a443528660f1ff7127cd72 : SUCCESS in 54m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/576f53ee545844aa965740e467eb8339 : FAILURE in 30m 55s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d7a48dccc3c6491aabe1e060ed982589 : FAILURE in 29m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d648bc8ee384ed4aa6fa95ba8d6be61 : FAILURE in 5m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/94f2ce3ebb1848028e98a6d576640cfe : FAILURE in 5m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/993cbdbaebcd4865bf6be9e2368aab53 : SUCCESS in 49m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4721c64f82a667dc5a15b06d75c19be020ff01f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 19:34:52.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ca6416986d1a447f81a626fadbe8c953 : SUCCESS in 26m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4821874dcc2e44c7a9067aef91383ad8 : SUCCESS in 43m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8ac3323cc1f34a0e86bd1000402821a1 : FAILURE in 20m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d48c1335f2f54c6186212fdde158e383 : FAILURE in 19m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9091dc59808f4d51a829d5f0bc58fbde : FAILURE in 25m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8ba72f27f2614fd2a9424faea5f84bce : SUCCESS in 33m 05s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/197a67213d2c4b5e829638ae3c92a69f : SUCCESS in 17m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/375dbee647b8427c83b3ddf49719c378 : SUCCESS in 58m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cdec110b60e54803b538fe5226b39a7c : SUCCESS in 1h 06m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82197d6e563e45669798a17153a137ca : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ce049b11fad4ac7982b5364040a14b4 : FAILURE in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/917227a40ea04a36896ba4be727a805a : FAILURE in 5m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/425b43d54fab43b1856dcb3001366800 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd3bc183e8bc49f2a96b8471902f99bc : SUCCESS in 6m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ffff8f1e93924803966463f944a467cd : SUCCESS in 18m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cca7ae2a366a43439419a9a4fa48d53b : SUCCESS in 33m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8b4eed1769a413fba7460fa09be742d : FAILURE in 16m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9a23676f7f8649c8b3f1a214471fb111 : FAILURE in 12m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/839214cebf534baba91db5c266c20c8f : FAILURE in 13m 31s","accounts_in_message":[],"_revision_number":10},{"id":"99f28da74bb15d121dc40e9494a6822d8597fda1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-31 17:34:42.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"dba61d64e87d43694fd2f7eabce09e58bc668b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 19:21:13.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/327d0184d8da43adb15fde8eaa00f59c : FAILURE in 12m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e5fbfc107c384542937720058856bd4d : FAILURE in 9m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/329991f470794df69941b27797e4fe78 : FAILURE in 18m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/badcd9433b48439793a6fbec50713ad6 : FAILURE in 17m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2bd6d12098674dde9c4f1eee01d13add : FAILURE in 17m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/78abdca6c92d4dfb9c9103838273a196 : SUCCESS in 39m 15s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a85a8742173745acb5d188abf0a34e6d : SUCCESS in 17m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a3b65e10bceb4703898aaf8406d426eb : SUCCESS in 56m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3f89498ad3864015b4fc90af38fb3739 : SUCCESS in 1h 07m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b55ffab0908348d7a46521f398e8eb02 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93f8539302694eaea03bb9886834c288 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d31bb93523764ed8a88e97b5e86833b4 : SUCCESS in 5m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1026bcff96ba45f7bd016f271632f2d0 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/765d4ce50f9440b594688b7e41de406d : SUCCESS in 7m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c4103b8b90e435d8bd90ae598795ef1 : SUCCESS in 24m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac570f246c574ca1973a45cace37f533 : SUCCESS in 57m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ba6aa4486f64bd69e31f49bb4c2c18b : SUCCESS in 59m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3c7e982f1124b488d6cf3fb0a6de88c : SUCCESS in 54m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4d71a3934d0845f59ecd46f8e0bcfc88 : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":11},{"id":"f20af951e050585179bfd0ca66c9f6ac1312bcc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 19:41:44.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/77dcb5ed04794de2b574f2d31a9ce65d : FAILURE in 19m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/187c0eec27d64e77a28590f6a987b630 : FAILURE in 17m 26s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b83877e7f264d92820d7dfbb5b2b37f : FAILURE in 22m 55s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1990b3034ec047a8974910b420f8f418 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f3149ff5d9446079af47cd44ec83c76 : SUCCESS in 8m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4869990ff9ff4990a233262f390aa1e7 : SUCCESS in 45m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d936a6021efb20f1a76561cae4476dbeb3b81c3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-02-01 10:06:51.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"0153086064e032e06f74b8a6aa6810aa90f7f29c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 10:23:30.000000000","message":"Patch Set 12: Verified-1\n\n(50 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c084d08bb5674ba8952b0d1296f1bc7a : FAILURE in 5m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c9f93ed53ad248ceb95ff9d01ceef9e5 : FAILURE in 5m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/cac4bb7d824b4ccba3a72d08c939a6c4 : FAILURE in 11m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f640517059c64eeda9d22a1bdb1ac90d : FAILURE in 10m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a8626b58005040099b8ea5e869591cbe : FAILURE in 11m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/60498474011c40fab9057eb3f74a7d9d : FAILURE in 3m 50s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c54d751bc3074013bce0e56114d42ae1 : FAILURE in 4m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bba990556f9b4e75a2acdb5148535470 : FAILURE in 13m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5cadbddb83b146faa52ca7e0702a946f : FAILURE in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca5874bafcd42d89f96a6f0c68f3847 : FAILURE in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34dac7484a51450288bb02ae604dcedb : FAILURE in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b69c32f9df684c58bf52bc7ad86c5166 : FAILURE in 4m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/672156220d1c4da197e132edfd2a842b : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce41bd3dc18492aacbc191b86f06384 : SUCCESS in 5m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/04aa5b5b5b6b4f88a9235a15d00fdb6e : FAILURE in 4m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c84f58c333a4f3688e9e26d0e57920f : FAILURE in 3m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/24c80235390b4f96912d92fcdf230386 : FAILURE in 11m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ec4169d796c8417aa7a4e3e06b1ca822 : FAILURE in 10m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c8e531f93f8c4ed3a9b9714eaafaa6e1 : FAILURE in 12m 24s","accounts_in_message":[],"_revision_number":12},{"id":"b4f21273fb39cd98eb310770e41dac293412e32c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 10:35:53.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c0fa824ec224b3d97509cb55cca7a13 : FAILURE in 9m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6427f841a2b54a7b9bdd82555cc7877d : FAILURE in 4m 01s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5706bff51cc9426a98678c5ae85b7098 : FAILURE in 20m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b706456361e942f2a7bae01ac88ff57e : FAILURE in 5m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2559f3dd1a2941e7ad365807a9462e09 : FAILURE in 5m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ab3c97ab0354f74be8eaa3edaee1b58 : FAILURE in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9b154ff901077fba91fd457f4f780d22e5172ba4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-01 12:42:59.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"367c4c62d9adb0f3bf672849e4808eaf80985db0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 14:16:22.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9a875045cc9340d493f4895a2c827899 : FAILURE in 27m 21s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1430d5109e924a56b16ebe914f9ab79c : FAILURE in 45m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/089b8885a92c4c61a27f11b34ad9ec85 : SUCCESS in 1h 17m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/01d434af2abd4273af6f20457f3d8725 : FAILURE in 24m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/543f99b84923407f928e684923a56948 : SUCCESS in 1h 30m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/70513ba9051f49a1b6a0e43bf8a49dd9 : SUCCESS in 34m 02s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/537615ffde6045489e6e699c5d0ef410 : SUCCESS in 20m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eb0b7cabb0c14c309e5f7821c6d6cd0c : SUCCESS in 58m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b07de79d610244d4ba340eed42ff9e7a : SUCCESS in 1h 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/089b62ddc94a4d6a9674645064ecede0 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e6158446c82406b98a9be97a5958ad9 : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47c3ec1f245742a386cacb045d759bad : SUCCESS in 4m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/31e58f19f2b24a82bec4e3d55f21711e : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8f19db36554ed5a32077b4653cf08f : SUCCESS in 8m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eaa9dfac4c7945dc91f0d95ecb2252eb : SUCCESS in 37m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a71e560d790249049369c7fdc165ea39 : SUCCESS in 36m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b9024328dc9459480ad19e526253420 : SUCCESS in 1h 04m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1744587da88d4621bb88ff6f0f7e7900 : SUCCESS in 54m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/51d01f952afb4cfbb3b5de37475fc4b5 : SUCCESS in 1h 06m 40s","accounts_in_message":[],"_revision_number":13},{"id":"c4850617fe2ad5aefb2487d9e548be6794f338e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 14:52:47.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6e026a4670414ef4b652530e41d17bc3 : FAILURE in 1h 54m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f7552292681a4b6cb856f6378557421c : FAILURE in 18m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/79f1bba992fd409fa599fc3ecb7e46b2 : FAILURE in 39m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe2d9a25c99c4fcf825199969356ad8b : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9627855404464dc99b02c3237a96068e : SUCCESS in 7m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/846d8904e886414ab506c449865526e3 : SUCCESS in 41m 43s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fad889a4fdb30d2fc3d3da8e0a747f84e957e59c","author":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"date":"2022-02-01 14:55:04.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"afd7aa874686729014124c402516e532a02996b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 16:56:27.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/38391681996d4bf3826710b9975a7d79 : FAILURE in 18m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2533dee80f6547cea950087dc04179c9 : FAILURE in 1h 10m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4ac52e1016a14181989080000af6ef47 : SUCCESS in 1h 17m 22s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0c35e5511da145bcba6fe97ac1d5e4cd : FAILURE in 24m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8bcc962367e441f4a4ac50cfdc1f3478 : SUCCESS in 1h 52m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8306f8cc45e34d89b8ed8e382b97778d : SUCCESS in 24m 06s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5be8bb284389455da2b6412b65dc6edb : SUCCESS in 21m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/29d811999e3547aa8a00a1d8c5d5aeca : SUCCESS in 58m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/35461353f0224e50b196c2f4a0805df2 : SUCCESS in 1h 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6497cae42bbf466896f667a6d661c03a : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38d48e492eaf47d6849ad460ec21826c : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ae0451b1b3c47919e72686e19170a2b : SUCCESS in 3m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/424979e2f5894426aaf3ab2db983249c : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31650fd694be44548acb0cde821f8d1a : SUCCESS in 7m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dc2403263ff84904becc1ff48259378a : SUCCESS in 20m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b21220d6ef04a8587ecde466aedc1ac : SUCCESS in 33m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e48b9ce9023442f4a5acd135279b978a : SUCCESS in 53m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4df45b3c248147faba57ae49c36dfbd8 : SUCCESS in 49m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/39dbdc2830f140b9a8c75616e987fde4 : SUCCESS in 59m 19s","accounts_in_message":[],"_revision_number":13},{"id":"ee364442db4fb36926d1902d80b69a8e70b3ecc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 17:46:12.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d69045013e9a498bb3fef9ddece26258 : FAILURE in 50m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/db24c040ba274032a2732b276b7edcf3 : FAILURE in 1h 01m 28s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/061e64a764434c158eb7f9792e20d075 : FAILURE in 41m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d99f5e6159f14f17a58be2e04ac168ff : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b233d69d666748828ce8739ebc1830df : SUCCESS in 8m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/85c71ee21a6e404d8d8d351bdc3b3c06 : SUCCESS in 2h 41m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"1e76dbdbf33d31740d817b69e60968290c1e6dcd","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-03 10:50:06.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e6bc98b5f72775305dca5916a15294fd474a05fb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-03 12:16:44.000000000","message":"Patch Set 13: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"164dd62e1a0e94bb6230a7d6a81adbb12811aa05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 12:30:19.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e68b313b7d0447deadcda56617bba09e : FAILURE in 16m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c972046fdaf246568f5561626c2bb02e : FAILURE in 41m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b0955f8c72944149983c3391898998a0 : SUCCESS in 1h 06m 27s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0e75fa435eaf41ad902f4587525422c8 : FAILURE in 21m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3806dd83ad57492a902871289d24d6a0 : SUCCESS in 1h 25m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2239b72d802043ea87f1251d72d3d92f : SUCCESS in 34m 37s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d2bb032736f04b1086426956727c289a : SUCCESS in 18m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6bc7b4a32e08462d84b7d249bc2d20b3 : SUCCESS in 1h 02m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/feac58d128664412a1c533e64f03dae8 : SUCCESS in 1h 05m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c426899385b4a37a209df03cdaacdda : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af2d0f651e8540089e5c06d97d1353cb : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af5313ad7c4c4cf7a624d68111788ff4 : SUCCESS in 3m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2e795058620f4400b4bc988783224673 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcf07f0e5c3c470594b4d22627dfab42 : SUCCESS in 5m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ed8fe40a3e9b4690bcbe0c32887d3887 : SUCCESS in 18m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbcdf5351e9247eea9c21b4599a6f975 : SUCCESS in 40m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/44b0f6c1fadc43508feb6a2a500d296c : SUCCESS in 56m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8f0fcfc800ac42158e81f64c39317904 : SUCCESS in 50m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c8712cb2e53d4c7c96e66b8fd95aad36 : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":13},{"id":"496aad9900b0c252b420202f2e25549df01035f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 12:45:24.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e86ec79141834de4a32235e1beb77b6b : FAILURE in 1h 51m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7018e5edb590476696051c058ff88122 : FAILURE in 21m 11s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/104fe4637de24fc58cf2f23974a84ee6 : FAILURE in 28m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab6bc42468b14306b92d4b84ea203519 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d9d4e10934e430a977b50b3c442ade2 : SUCCESS in 5m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/51d5aeb5e7ef4211b8bbc3b1e72cd060 : SUCCESS in 44m 27s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d39d3b719a9c2956a4191d1ad617fdc5e0181812","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-04 08:22:37.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"06b09c27ebbab4803443fe16de33d8aa94d110c9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 08:45:30.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"aac00f83ff7089b05f6d26cfd9e83d20e852b7b2","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-04 09:16:18.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"962b989206bc413459bc0388de3e405142785fa6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 09:22:40.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c2280b6c200dbff0a1602044f4f67b476fdf75af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-04 09:29:31.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"86cfd686241f40256dac934cf3abd4a62b8fd042","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-04 09:30:18.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"0f09c5f33104465dc0ddbd64958459c76f408804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:01:26.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/73dd69d6457542ce92da493cd81578c4 : SUCCESS in 37m 45s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/aa64222ce102490ca07f2a8920c7408f : SUCCESS in 48m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8426fcf8cd1049b1bd0175c9d8f4df86 : SUCCESS in 1h 11m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/82a8613fab7a40c2926390cfc19fded4 : SUCCESS in 1h 08m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c9d832473f51420980b84faee801abd2 : SUCCESS in 1h 25m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/aefcb96de65143ba8496dd9dff33a643 : SUCCESS in 34m 57s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/182e83332c0a402e8e2137c7a1ca7827 : SUCCESS in 21m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bbcd9d712c2a443ca67802a94824db32 : SUCCESS in 56m 05s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/38394e683bf5424fa9bbebfc64b241e0 : SUCCESS in 1h 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b2f34a13634ba7a7465cb117d4ccbb : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2581ecf01a043148e48a10476135e7c : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/600b394c52aa440e99a9071e7ce92192 : SUCCESS in 5m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a3509b130f64e97979819b1b72fb41a : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c00ebd98ce94fba95a74b94c3a7e1dd : SUCCESS in 7m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f77499205f6943a7a07f6443d13172b5 : SUCCESS in 22m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f3ffcda05b346109ad0f965613809c0 : SUCCESS in 37m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff97f12cb89147efaff67ef404ff34d5 : SUCCESS in 53m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/518a07d6f1d64a9288b6dccd73869fa9 : SUCCESS in 48m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3782b241ce58439a89ef677ce977c765 : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":14},{"id":"181f3ba2bcaea3874c3d6f9ac1bcc8b8ee33539d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 16:22:57.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"969601a5150490a50ca84796b1a73e0734e3fef5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 19:22:05.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/cbd66b9c61904d1da51cf508f6c2b556 : FAILURE in 2h 41m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4056f22159d9453d80ca875610c6aa86 : FAILURE in 58m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/de9a2249fbb94454bac7ac9ac9cfbc5c : FAILURE in 37m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cf14eaa007e4c5fb7da4885af5c9d35 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/524ff4fd2c9c4baf98003f9be665e5be : SUCCESS in 8m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f0a392e7393a465a8f0210d46af074a6 : SUCCESS in 2h 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"59daf1e01b33e48f2b922a8289674d4e8f0c4d11","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 07:15:51.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7598ed7354e5300f3ec4500c92b5863eeccb86f3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-08 08:14:53.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"b95b04f9d1c305ee6ac28b0227b4e21e65a4dd5e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 08:43:54.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"b185bf38aae6a485ac14cad37ddb3b0a71a03e04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-10 14:13:43.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"9abe7d543f56cb255638e2def00524042a335d1a","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-10 14:15:52.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2313ac8ff0f23eed397be16196746665a52a671f","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-10 14:32:23.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"1eccd40b70fc5ef25c62d7d88b1da21ecb8ceade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 15:39:38.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/00fac82c01454e4fa9ceafb830f75786 : SUCCESS in 28m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c9cf53eb183047969b15c078151e1b0a : SUCCESS in 40m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1b76d2c909cd4237b9d18b6ac714e60e : SUCCESS in 1h 08m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/35208d526bb84dda8de0ca280860b789 : SUCCESS in 1h 12m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c713a9eb043b4a87a61b8137120ebbb9 : SUCCESS in 1h 22m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/45eda0f5f8f844cc8c55c332a377fdd4 : SUCCESS in 28m 24s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5966df503a5f4b4b80ef27f161eb3221 : SUCCESS in 19m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cec56ac3166047b9962a1f78170f119c : SUCCESS in 1h 06m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6e7758b87fce4170ad35423352e1e771 : SUCCESS in 1h 09m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b336e74c567d4e5eb28b4bcd43a3ef62 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e47d8cce338b4d3cacb54509ab1052a7 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/237ffa63da11483a86c688fc58e21b60 : SUCCESS in 4m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1705ee40beac4158ab8156c31929bdef : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c312db917b70471880007e885c06586b : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb5e54a883dd47d8a3cfe29deaa6efbb : SUCCESS in 19m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9fcfe61f2a7646f8804e69e2bbda791b : SUCCESS in 27m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/18429d2d7d2447489092fab6cd69927d : SUCCESS in 54m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c9519a01a6034cc1911517723b8a6ce6 : SUCCESS in 52m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/17277f8010764b1faf7f1b008817a15d : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":15},{"id":"c1719998fd3962aa6c66ff30940ad2b8586b9017","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 22:10:29.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0a86ad39cd34a1db0e954261b1363d6 : TIMED_OUT in 6h 02m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/053609b6abd7419fadd4f952fe084037 : SUCCESS in 2h 04m 42s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/53de080d58164f21a23c5801231f6a4e : SUCCESS in 2h 53m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a19d99697754ec69a1c92b6554174a9 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a91a1aa10fa74e46b6f2105170450901 : SUCCESS in 7m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4eb62c32355a4176bb38bfa51494974d : TIMED_OUT in 6h 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a83a75391dfdc2af1848dd981c46ac72e1c988d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-02-11 14:29:28.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"563318a630ca6cb349a953da409fcaca9eabc547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-11 15:01:27.000000000","message":"Patch Set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"7505c32086ae80ad1995782dda7b047a1e99e8a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 16:43:57.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/657b1009dadb423db28b8d6ec6c0f916 : SUCCESS in 31m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2d6d674b9cbe4ca48ad03a230e249454 : SUCCESS in 35m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c162dbb669374a719803e09d768b4cb4 : SUCCESS in 1h 11m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4a829f07cc584c09bb291cc5c1f1970c : SUCCESS in 1h 07m 13s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d8cf008edb404716bfe9c03594c9949c : SUCCESS in 1h 36m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b82be3a4fc624ba59f72a81fab222c25 : SUCCESS in 36m 53s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fe63b31b94ac45ef80335e2d23f5808a : SUCCESS in 24m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/187cf132c7e246539be4b9de154bb1af : SUCCESS in 57m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/624ec7f8fa31499b862c1d9934933049 : SUCCESS in 1h 04m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0720a3bd98d44f39bb0999b5d228c192 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a774fb85bd034cd384a99a3ed993163e : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9095037a629b4cacad05ba8206fa5368 : SUCCESS in 3m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4dab8a6edb114e1f86360389b52deee0 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0582c9c22ee24634a085e991a373b20f : SUCCESS in 5m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e51f415cde54c6f9267cd50ddea70d2 : SUCCESS in 18m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f6cf71c7395247f1933761a48425bbe8 : SUCCESS in 38m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/51d165f4e6684ecdb1e6ab4bd9772d78 : SUCCESS in 47m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/30f2f670679b49bc903d64326507bad1 : SUCCESS in 48m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e025dda1f7b8427ebd258dca59ff6801 : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":17},{"id":"5edcc7b5d90c1ae52eda72a28b379db562e43c2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 17:23:56.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8a272567f8ba497fbd8ce43553963877 : SUCCESS in 2h 11m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c8eaa989ef2d48e0aeef14a9bc001b5d : SUCCESS in 1h 39m 32s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0bf766fca312412abb5b53f4a4d7fbbd : SUCCESS in 2h 06m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0fcffd997b74d7f953d9a28b2485e93 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c3a73008dce43c0b5953ba5541bddfa : SUCCESS in 7m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/220de9e29f9f4666b489d70a4dc02880 : SUCCESS in 1h 44m 38s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"2b27b0fbbf33820cc813ee4da5ba08ee9fee6ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-14 13:13:25.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"877160b2606deb113392897be7cc7046e25280d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 14:35:43.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9b57198dd42d4363bb32a2db647bbaa8 : SUCCESS in 32m 01s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/27e5cfed97004dbd94739f56938ad1a8 : SUCCESS in 42m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fef2f309c4f84c5199f32514621eb054 : SUCCESS in 1h 06m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/62b8d232da634b15937166e7e8776ab3 : SUCCESS in 1h 14m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/918a5541a704467481f75afcb8116a01 : SUCCESS in 1h 18m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/98c9fee3e7c54ddca6e53b465f219aab : SUCCESS in 37m 37s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/272906b701ec4f739e116c09afdb43e3 : SUCCESS in 18m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ae60dfef835b43a89989857bf2fa3ed5 : SUCCESS in 56m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/594ea92fa18243949da1a0f1ea53c4e7 : SUCCESS in 1h 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99978952d9624866b79d7aea0dd89156 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22478fd26c55455880611864ac1ac7ac : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a625be5fec6c49a1989ac03b52d19365 : SUCCESS in 4m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b332ea67763545f4bcd95961b1f96399 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb2aff840b2349e18ca4af191500e3ae : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f5a99c81b9db49efbd65368ff1a9fe01 : SUCCESS in 18m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b18f10ef256448b4af75862d982601f1 : SUCCESS in 34m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6594cf8c5dfb4a688f02cbb236c5c1cf : SUCCESS in 53m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/98e07a317f4b4ab586f789f9793e3b93 : SUCCESS in 49m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c933d4d4871c433c942b8587ea75fefa : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":18},{"id":"3f3838101d2122bc8b0ab3f1d78e4ea7176eeeb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 15:47:55.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8084ad7b83f94b22a3de7cceda0a7a85 : SUCCESS in 2h 19m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f44cf3f4eab40ecb9d7c3676f2c5806 : SUCCESS in 37m 06s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef450e91975d40a88a27d7d0715f26ec : SUCCESS in 2h 28m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0c6a22d3c574fea889ed8463bfdaff8 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f3572f9eacd471a982e3bc3850b53f0 : SUCCESS in 7m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c73baaeca1c45d1ac0594639636efaa : SUCCESS in 1h 30m 31s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"31dd65fd361d5b398602b32dea230439cd029763","author":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"date":"2022-09-01 11:36:52.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"314d6cb5f46d2d310eb560902c841f7f7b7407ea","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"date":"2022-09-01 15:02:46.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_29636\u003e\n","accounts_in_message":[{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"}],"_revision_number":18},{"id":"3dc03d392dceee73feb7b62dce88bab54a3bf551","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-29 07:53:44.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"15f91dbf65e1b368f9d41a24792cf48092012884","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-12-30 09:27:30.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"06144061a6eeb2e9786a42c57ac1cd497a5017f0","author":{"_account_id":34911,"name":"Martin Hiner","email":"m.hiner@partner.samsung.com","username":"m.hiner"},"date":"2023-01-02 13:51:49.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"dc5839f96e2f4f23c453f2ef0663eb75a0c1b8de","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-27 09:44:30.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"4122d8e0e3154b6f989bcd97ad46f5b401ad614c","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2023-02-27 13:07:33.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"e386540d15ae34c768ee6e32d6ebee3da218b58b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2023-02-28 13:07:12.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"1cf3b0701a030b16817ee8291d0ef426a522cabb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-28 13:17:45.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"d07fea791a425e79fe601c27f21d866f60c46820","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 15:06:02.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ec9f2bc33c9459a9e105ca83eb81d21\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ede21b1e1b1642b4b9961a84d29683e3 : SUCCESS in 28m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0f6f9f7585a84e8a8506aef40d3a90e9 : SUCCESS in 29m 05s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d07008ef081742c6b2205090f30700b0 : SUCCESS in 1h 22m 18s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb94cda7b5244ff598c2da949cd3dfbf : FAILURE in 3m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4749367a47ab4d4c8104506d801edf65 : FAILURE in 5m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/36e7a275d8314221aaf4a32d579f41bc : FAILURE in 1h 39m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3354c800c5624c4e8bd1ed126f490d69 : SUCCESS in 1h 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"77df95e3efdc6413aa8287d606de058833b8e9b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 16:02:33.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90f6882f58e34855a56717d7caaddcb0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f37304ebfa6e4a2abe001e32959e4eef : SUCCESS in 23m 50s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6f2db4fefecc435d8f7674a1a5df7a6b : FAILURE in 19m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e6d3247d8ec444dc804edcf073e41be3 : SUCCESS in 18m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0dbd0ce7109d4e23a8c049e402b58f13 : SUCCESS in 1h 09m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c0652f10d53f4029bc38c100fae6de57 : SUCCESS in 1h 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e7396fb82c44a6bfcc1cf7159c3b1c : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b5e3f18378b4a38aedf6cdc25064a96 : FAILURE in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3346685ed3cf4eec9a021acc0e23add6 : FAILURE in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7fd7d01c75c4987a2f8d51fd0c5aa9b : FAILURE in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e8db666a324e5d9e2a90c728d00379 : SUCCESS in 5m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c4836dbf7abf44aead7e18d7177f5e1a : SUCCESS in 23m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/93c2b2d936644db1b6b79a66a24f79b9 : FAILURE in 12m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/61ee4a2fe034403bb986404122aad4e8 : FAILURE in 13m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/402a529aa0e742c691c819eedbdffbc9 : SUCCESS in 16m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e923ddee36b24e8ca40a189099c4bc21 : FAILURE in 17m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4748a7aa943849fca9b48c94c7e6c395 : FAILURE in 18m 08s","accounts_in_message":[],"_revision_number":19},{"id":"7537b8a5f68179ff033c7b706b591cba3da6aed9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2023-03-03 10:15:11.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"be861d8f86e49757c9b58752e09da757b3f4b742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 11:33:32.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d20cc45071f64789a4659eb2726ccc74\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3cb58f82e55b4ac99b384bb125828a05 : SUCCESS in 38m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/75591a2d19174f5484e8dc53df80de2b : FAILURE in 22m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c6415ead6d3e4483b6005a1d059adf20 : SUCCESS in 16m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b69075d2d32f4165815dfe372982c8da : SUCCESS in 1h 10m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f53ef1bcc1ef49d89bbaf69cd2afbb9d : SUCCESS in 1h 08m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4015bf8f21484f1b8d4ed2585ad5f744 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2ff038ab7bc4fbbbeda906947924f34 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fefbba902fd427e95feda03c495f63e : SUCCESS in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/783fcbe552de49efa61e768bd4a5853d : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e39d625c934f3286a6633873fdcabe : SUCCESS in 5m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/00c5a6f153b34fed9415daeeff092fc0 : SUCCESS in 17m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a6a93a604b5d40b78eafa1fd3b0f64d7 : FAILURE in 16m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0892cb6cc6754371aac5b4f7f573c5ca : FAILURE in 20m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/92f70ef8bd404fd8bd3f6e93aff122f8 : SUCCESS in 14m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/73af03c5ef3d4fd8b193e1a81a03e44b : FAILURE in 15m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3ed185ffaf794aec9132d01fa65e05ae : FAILURE in 16m 33s","accounts_in_message":[],"_revision_number":20},{"id":"5a3cc9b53c4f4b51e611f931219838f22a632dde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 13:20:25.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc88d27b31c44d37abded5017710ca6e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4573d2c18b81443d960520b4d8bd83ef : SUCCESS in 25m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/94136148abb742c0830dde7f37cf6ecc : SUCCESS in 23m 26s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/06a5577bec90484bb5677e35c9bd01ee : TIMED_OUT in 3h 02m 44s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a2ee491683047d4a08ead726a1c50b0 : SUCCESS in 2m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f3293555efad464fb06c4bc70d81b423 : SUCCESS in 3m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9385ecde3e464999b0198301e902c5aa : FAILURE in 27m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1ef4a9f7ce65428a85b307be28699b46 : FAILURE in 1h 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"4f07a51b6b6c37b452569ef1b456d34a11fa51aa","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2023-03-03 13:20:40.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"32d2f1aed358b34b640de134306e83ce19ba9509","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34911,"name":"Martin Hiner","email":"m.hiner@partner.samsung.com","username":"m.hiner"},"date":"2023-03-08 09:45:51.000000000","message":"Topic podman removed","accounts_in_message":[],"_revision_number":20},{"id":"504e1a701e16ba641f10a8aa65ff8a831f0f0eed","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34911,"name":"Martin Hiner","email":"m.hiner@partner.samsung.com","username":"m.hiner"},"date":"2023-03-08 09:46:00.000000000","message":"Topic set to kolla-podman","accounts_in_message":[],"_revision_number":20},{"id":"879387bf3f3cf1b7f7aa150dfea5e4eb40061519","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-15 10:35:32.000000000","message":"Patch Set 20: Review-Priority+1","accounts_in_message":[],"_revision_number":20},{"id":"fe71c740da021692491a81f6d8e28cb25017a76c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-15 10:36:05.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"81a6db398014ef7f7e9ac4d20079c169a5979177","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-15 10:42:06.000000000","message":"Patch Set 21: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"29cd6e9d9708cbbe1727aed2714729d639f02890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 12:17:27.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/203ec4bd7c694f039741e6393a9cbae9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a23271e83d9e4f4bb0a2e9400efabc08 : FAILURE in 22m 18s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d0b4171fc2b64f90932ce199b91cba41 : FAILURE in 17m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b9ad2ea368ea494ea59c6b55cc8d3a57 : SUCCESS in 15m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4aec6e4552da41459036835bb6099b5a : SUCCESS in 1h 10m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/02b6c051e0b54696925e98684309ab93 : FAILURE in 1h 15m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd414c58076e4c069f63cf3d85105da2 : FAILURE in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/993016d8d71742d09d296c711fa31d06 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86dcc287333049b981ab439dc41ad08a : SUCCESS in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a9f4854c20242c789ec989fcb9de42f : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e386a24b4ae4020be26b2cb05446a5f : SUCCESS in 5m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4e899b72edd94ae69f166a839b5f4bcf : FAILURE in 24m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d4ff8ac91fb24ab18f7de12c439a3f0b : FAILURE in 12m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/af01312541a44e59b14140516e33337e : FAILURE in 16m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f3b684609aa547a0a4e59c3c00be08ba : SUCCESS in 18m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b995367fa589413ebd043452b1cd085a : FAILURE in 11m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/790c38d0558d4816a256025da9df008a : FAILURE in 11m 55s","accounts_in_message":[],"_revision_number":21},{"id":"23bc015a626556cd4919223747c95117bd2153f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 13:41:35.000000000","message":"Patch Set 21:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9f7bac993e540c68412fa0d20ebaca3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/93803338dc674d8c9c4dd2038be5ebdb : SUCCESS in 2h 10m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/535351439355458fae2959ecd17446f8 : SUCCESS in 2h 08m 52s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/11dd7fd871a1457c8ce43f108c1671cb : TIMED_OUT in 3h 02m 48s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/206cee41cdcf4ef58f2b69e62c85629e : SUCCESS in 2m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9499f82318924a9ea79f3f3f37193194 : SUCCESS in 5m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8853327c698f437db543f01ec6dedd7b : FAILURE in 16m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ad222b19c0994ef18acd93e35cd478e0 : FAILURE in 2h 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"b5a4919275f98f345d7277981876b7e7569a670a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-20 07:07:57.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"aec78c01839f0e7fe490cfc58b6c5e8ed7ff1264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 08:34:10.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/357b5386f92b45af9053ef29e98ae62c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f8adb350a4694e5f8cea8211da5b6f43 : SUCCESS in 22m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a22415d578a549d39174e02273d766de : FAILURE in 13m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f80ee3711c2c4e99a067dca3bda97935 : SUCCESS in 24m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bc60ab4362394dd4b31d5b19a5f56db4 : SUCCESS in 58m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/404d98a7288844e4bb0b1d11f065bb9d : SUCCESS in 1h 19m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d439374a9e8f4a3087870d595d0e4e02 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84ba73868a5345d7ae6f83cf7efb2005 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9945c2b5377e47e7ad707f3b599ebbdd : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d00ae4c9e8e5493a9d5beff6b843f6da : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66ebd7ddf30442249e3de2e4c6bde430 : SUCCESS in 4m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/887357ab5a794dd9b8ec3000e9d561c8 : SUCCESS in 23m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/173d170bcc204a5b8daab416f5e33531 : FAILURE in 17m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/45a1b1d400064aacb0e39fcaa57e8ed4 : FAILURE in 17m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6d7f8bc770554f9889f06c4fe1f3dad0 : SUCCESS in 19m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/688dc442b86c44859ad3887f890b09ec : FAILURE in 13m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4783dfd762a04970accc192f589746b1 : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":22},{"id":"df5216e57cb6cd2e52fe547e7276aaf383b8b83e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 08:35:57.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c08115730b584ea6b13832eb46352aa2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e123d21bc45b460a839b61d82d24c99f : SUCCESS in 25m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f509f392c965482c953d410474823e53 : SUCCESS in 25m 36s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/836af97a7a7341ea8b82b98dd552d35c : SUCCESS in 1h 24m 31s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29f01099f26c4bd2ab42ebf78b023489 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc4c2da39ede4d3b8db2a91c54a412f7 : SUCCESS in 3m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d40609bd10c040778771ac04619e6a5e : FAILURE in 29m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a585d53951fc41dfbe5e29781f4fa933 : FAILURE in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"592f5c27d8ce41ee46925f4d4b76c8a88b92dadf","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-03-25 20:21:46.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_30491\u003e.","accounts_in_message":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"_revision_number":22},{"id":"ea95249e5f1297ea7f33f16a78378e00f30b6bc6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-04-12 13:43:40.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"2ee145b9e2fc1fd9db8488569dd6bf5cd9c1f084","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-04-12 15:02:03.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"1ab8ed79307df73104e083aa91d715e9ae2e30a7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-19 13:17:30.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"54098237a2e0b2e7cdb4d27f9ad03b85f0839568","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-04-19 14:28:18.000000000","message":"Patch Set 22: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"a3b88af84ac29a5c86724800aef0e007d244e2d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 14:31:53.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f0a68b79dd045febdf56ffa01320a21\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ee3074e24c8d470482acb4912da1bdf2 : SUCCESS in 22m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/097c50bc4a0248fdad602c70ff08c642 : SUCCESS in 19m 33s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7b6ed0b239c4483485b0a4b5ed04237b : SUCCESS in 1h 12m 44s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42c5a31f507445f1b6a5849c0a976038 : FAILURE in 2m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45b3b7d7f77d4684a6dbf50861d896d1 : FAILURE in 2m 53s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3fab692724064b578f5ad5638cd9347f : SUCCESS in 1h 07m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9357dce211a14dda91e7020e51c8e819 : SUCCESS in 20m 30s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"7d52d701b5b8a9df2ea6180590fc3d84d4d64ebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 14:39:02.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3054caf0dd80404087a937f7331c6094\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/08b6af2597464e9099def70e0505c70b : SUCCESS in 20m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e6cb6c62158649b796a1adfe943682a6 : FAILURE in 13m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/884f56f6d6314a9d93a621502c8e1682 : SUCCESS in 18m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/78136a4f3561458aa95de122bab81259 : SUCCESS in 1h 04m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/222bb912aeb945948bd66db439a7563e : SUCCESS in 1h 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9e012bdf3734b3db53eb2b0047a2a1b : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5fd910e655f440d8cc7b928d63201f4 : FAILURE in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e58169d909ac49a9be1b3c8c0ecb4ddb : FAILURE in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e184ddf7042d4c518bb58dd40f5af863 : FAILURE in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d66b60399c424f9320d01e37dbfb69 : SUCCESS in 3m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ffe03f53da11417881ac86f8a69eeecb : SUCCESS in 17m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/49bf95d639354db18b57d4d6e47e419d : FAILURE in 14m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/188bb12604ae45508a553fdffde45aee : FAILURE in 15m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/77c9900e027f419099fa9063ca872242 : SUCCESS in 20m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/31678df0377941f080fc41fb624ca79b : FAILURE in 14m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7b129fb956d24076ace2d221b6180a44 : FAILURE in 16m 43s","accounts_in_message":[],"_revision_number":22},{"id":"525d8fadffe9be84bc77e7a57b68afb824f7b2de","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-04-19 15:33:05.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"167c8af82973d3994ba9e7d5165b3fc40b5a2bbb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-19 15:46:00.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"45667c90f63aae1e3a9f1a4488ce90cbca47172a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 11:55:46.000000000","message":"Patch Set 23: Patch Set 22 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"a7c4950b8aeab7fecb75ee9fb8fc26c68709c753","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 11:57:52.000000000","message":"Patch Set 24: Published edit on patch set 23.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":24},{"id":"29775f3964ff34a0d6cba4e247f4baecf42b961d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 12:09:34.000000000","message":"Patch Set 24: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"9e136247ef8dd5aa67565bec5a85bd918ea7dcf6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 12:10:12.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"ff7f86c1fdf59da05028db95afa8419eaf4952e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 13:06:29.000000000","message":"Patch Set 24:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5772e72a63a74ba19ebab16369c4ea91\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a46d28023d8146e8a025bce997766bc9 : SUCCESS in 27m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/50d9ce44802e4410a0d71bc469eb6ff9 : SUCCESS in 17m 37s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dee8fd8d3bd54d38b5e3952bcee12936 : FAILURE in 1h 07m 35s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f930c5f385d4c9facca226885ba9f3f : FAILURE in 2m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/837cf529902643c8af88d60938414906 : FAILURE in 2m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2e4d373514544d0d8e21bae379c4d3fd : SUCCESS in 23m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/03098b3e28dc4008abbaeff162b8af54 : SUCCESS in 16m 29s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"58a66504c7a2fa0d15016e0e8ac18c51503d3adb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 13:30:18.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ea595e186a94a47afc469370c2c63c7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fe3482d7c6d845c8a07a84aeeca30d1d : SUCCESS in 28m 18s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/562b460ece884eca8a3eb575c13b0863 : SUCCESS in 1h 31m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b7324a345bd3432ebebf4354c0af5c1f : SUCCESS in 19m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4b353e9ee8ad41e487a5a8c38ce7bcdd : SUCCESS in 1h 08m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b88e4aa1751d48d380b044e8474c7f6c : SUCCESS in 1h 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7cc48b574804fd285aab1840e488c71 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71fe921b58ec45688c4f3b05fef17a84 : FAILURE in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71dc546ea2054757b98a298f997b7d0c : FAILURE in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0a6d80fa3d04c60b8471014de532863 : FAILURE in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4864d7c32b840e59f99918e6d3e9db3 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5a7c0c9db1064ad3afb4edf582a1ba40 : SUCCESS in 20m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cf539170762e4efe84b428a913bd8686 : SUCCESS in 1h 12m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/22010e1b1acf4b85959b06686f876f61 : SUCCESS in 1h 24m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c6d9ac7c3e0a4f0dbb1cfa684f382eb6 : SUCCESS in 18m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/56f540c07d91457ca09503eb33e374ab : SUCCESS in 51m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3417066a5adf429eae18d7b7c252a5ad : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":24},{"id":"ee788068f888cc0beb566f0e7f5f6bff3b4bce87","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-08 12:53:10.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nThe following files contain Git conflicts:\n* kolla/engine_adapter/engine.py\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"b90b6cb99de3b23d5ccf8b3cc3f53da27985207f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-08 12:54:02.000000000","message":"Patch Set 26: Published edit on patch set 25.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":26},{"id":"dd7f806fb8d94ad8b6b72a5f747c3281e6f90850","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-08 12:55:22.000000000","message":"Patch Set 27: Published edit on patch set 26.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":27},{"id":"42f0f0840deb002e361a139a1f16e799020c77ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 14:34:37.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/274f351777a64ef38764535451027a0d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/85913dc001144e6ca48737fef444baa1 : SUCCESS in 31m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a9ad701e312c47a9953ea428e069c72d : SUCCESS in 1h 28m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7e7ca1120c184ac088eea1e0a9fab6a3 : SUCCESS in 19m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2df3efd6728548fba27550eb5012e2f9 : SUCCESS in 1h 17m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/69683b6f4de8459a88eaaa3434714476 : SUCCESS in 1h 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f38076fdfadc4d5fb100839cae0b581e : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/421ae8f838514761a68c3f7535fcc7b0 : FAILURE in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b23126163b546cc8359ceb2be2eba80 : FAILURE in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7ec2e3a58b34c0ebf2c5181893ec2b4 : FAILURE in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d340c400f2a14a54a677a51aa454fe4e : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6aa230037d7f4c1eaa37cd7c2bc2b601 : SUCCESS in 16m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/799b501dfb234ace9ec0132754f51952 : SUCCESS in 1h 21m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/06e92a659a504e949a5e04b26625a754 : SUCCESS in 1h 17m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f7999de6588c4aad911207b1c4cc2f77 : SUCCESS in 12m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a60ad26a9f124baebdd4295ba69d15c3 : SUCCESS in 57m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a57800c07f334b05a771565e01d8ef85 : SUCCESS in 1h 33m 33s","accounts_in_message":[],"_revision_number":27},{"id":"d7960891b8f1ddb7cc2ce6500716a5bad02233a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 14:37:17.000000000","message":"Patch Set 27:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1edfb4c7da44470d9278334ce641112b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bb49f912c3c74533b6ef13a9d0450cea : SUCCESS in 30m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d8521aac40b040038d5abc4a1b9965c1 : SUCCESS in 23m 05s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/989720329a404151ab437b78f2005942 : FAILURE in 1h 41m 28s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33be199ca1524b6594b116b79975b48f : FAILURE in 3m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/616b35f913d24f879f2129a1509f1536 : FAILURE in 3m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8f9ef079a64d4c8ea171cb90636c8efb : SUCCESS in 25m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d069f45aa11b4c9d9b6c39537dbade14 : SUCCESS in 24m 42s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"6ba58cc13c18d915eb4c5ecc03b8cb0fadee5f50","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 13:12:42.000000000","message":"Uploaded patch set 28.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"166e3ce9ff7ee5545ceb3e4b99634fd040cda337","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 13:22:38.000000000","message":"Patch Set 29: Published edit on patch set 28.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":29},{"id":"60c7c26404f55ace6f855257790b7b6ab6bf51cb","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 13:23:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":29},{"id":"5bc51a09b2fe611e67feaa8a6850c18a87e3665f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-17 13:27:07.000000000","message":"Patch Set 29: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":29},{"id":"9fb65d181296fb4b9a0d6b73945e71aec66a785c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-17 13:32:15.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"96e5433fc47f858a07c2a774bd64358564d97838","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 13:54:47.000000000","message":"Patch Set 29: Review-Priority+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":29},{"id":"1adf69854b49740bc7070f42f64ef3c632599eb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 14:45:07.000000000","message":"Patch Set 29: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d4e688f30d648dfb3ad28d0f1648821\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/523407fd17ea43cbbccdbd7885f7c193 : SUCCESS in 19m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2186a97f4b604c85b2f354b101647a1c : SUCCESS in 1h 14m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/03e2f65b82114dffb776389b2f60e988 : SUCCESS in 18m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2937792f619f4b8fa28fc438baccb83c : SUCCESS in 1h 19m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5191cd5e20a74b019530c5dba066cf74 : SUCCESS in 1h 05m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876ef240885d4675b36793cac45363e2 : FAILURE in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48f6d5628b3e4b42a2f1ccc0253d6fb7 : SUCCESS in 2m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/013a6a4b0443448189fcfd4e0e8e83ab : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d174bd58ee446b8b31b44acb9b16b36 : SUCCESS in 2m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/caef512c899c49339f5f11d0923f9de5 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5d3a544c4634f32bb23610ab32c42d3 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f202520148c64c93b5a0ee1b66fc163e : SUCCESS in 3m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/64c1175eef674e059748d34f17d50f13 : SUCCESS in 17m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1503d516d16542cb86b89642f59b9eb2 : SUCCESS in 1h 09m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/636915eb1e944f518ec4182122b48272 : SUCCESS in 1h 15m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a2d36e53f9604f62b4c509c6cb7799cd : SUCCESS in 22m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/880e9985150b4876aeba5a9a0d8ee90c : SUCCESS in 49m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4ab8ad8c02b84ade99b92e6e8d8e8422 : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":29},{"id":"3298de87f1d2561b22af09f11a723f5ae10f301f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 15:09:51.000000000","message":"Patch Set 30: Published edit on patch set 29.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"0c70ce086c4a6e118354898863b7655cd9c89c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 16:40:09.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e1ef92f3d7c492f9880e743e5861a68\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bc4fc89327474a56bbda572512c212dc : SUCCESS in 20m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b642a3d8186148edb5c8e3b6f16b256e : SUCCESS in 1h 22m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bef57227b392406fba86a9fbbf0b7578 : SUCCESS in 20m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e244bfce315e4b80a02eb5d8f78a5d1b : SUCCESS in 1h 12m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5c1df117c15d44b3933eb3362b029f2a : SUCCESS in 1h 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad8ffc0e9d0433b9049e412449cc182 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91dd7ea9117b467a8cf6a59e8a2bd1b6 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0dfe1d7213c45c9a57bfcca8e703b7f : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/882555c0af5241f5a65339aaaa6b2bee : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0481557d0317407990e6d839e1f1a5c1 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e3223a88534bb2bd918676cd034bd8 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e63f6a5956d94b71be537afa142e4541 : SUCCESS in 3m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2d6a44f19cba496cb307e48a9eb2fa4f : SUCCESS in 18m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b78f330db3ef4fb79ac7c27d69c60103 : SUCCESS in 1h 02m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0e11705056c44b0082e943793d0b0170 : SUCCESS in 1h 24m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bd6c5525be9f49319f7b57adf04d2701 : SUCCESS in 19m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f7fc610b504b4c8594ebea7026961c89 : SUCCESS in 57m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/83c0659530524b8d9fed573c0c909f74 : SUCCESS in 1h 10m 13s","accounts_in_message":[],"_revision_number":30},{"id":"4b16e1f7a51d5083df91b3ae2de2fa7b5940ebc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 16:47:17.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99db5aeb3a064fe9aa26c5aee71f78b0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/69406433b40e419d8fa3dbe169cd1bad : SUCCESS in 1h 34m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a4a5b1eddb4445b0b83c73805f20760b : SUCCESS in 1h 34m 23s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/42caaabc6e0e4998a2ac18fbda84164c : SUCCESS in 1h 31m 27s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f155cd6d85884dbb93d560a7295fd567 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/781c5e5798a34898b36531eda3ba9f1a : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/134dc007894d4374bb734ac9f4a20a62 : SUCCESS in 5m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b842a7999d0840d5882bbaeb76ec528b : SUCCESS in 27m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/215ec386ab1f41ec8d8cf1b0b38577f5 : SUCCESS in 1h 32m 23s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"37737a0b83357d3a9095af0f8f9e3f4dd2b09e3a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-18 05:16:25.000000000","message":"Patch Set 30: -Review-Priority\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"c7e8354515585898caa4457136b4050129a376fa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-23 05:23:37.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"497bbfd00f394d0fe557021518914db4dd861904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-23 13:15:01.000000000","message":"Patch Set 31: Published edit on patch set 30.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"e009acf699952a27b1054d188781a8e76aa24f0c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-23 13:15:08.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"279565857cd24b144f0b6d74b92654b6227f806f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 14:34:19.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cded54130cd24b83a7c3382b58b0106a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6f033eba30f14137a27d32c600c37d6a : SUCCESS in 18m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/74f2513045ec43ffab6ff7aaf8515a30 : SUCCESS in 52m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2071f59546124f6da8b3df170fb07c4c : SUCCESS in 15m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/994ba157053e4f4c92659f89e2fc9925 : SUCCESS in 59m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a67c937203f94c64ab30dfc1975e2a47 : SUCCESS in 1h 04m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b488b4bdd904e8ca106f300d7e17a64 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4adc5fa11de43b1be242b043f9b354f : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a93ba618f7aa4f5d82a64786ca997841 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ea5c7cef3cc445f8256a57aa5cb0a99 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0323220ec73e482184751f4474cdc26d : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db86fc9be28049fb9426a9a8844638d3 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f86c803536243728570470ee77a4b2b : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b79d9bd871574ea59b8e54e86f3a15d6 : SUCCESS in 3m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ca3f3c88cae64d8fb53b1837ae5de2aa : SUCCESS in 16m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3f546688598045b0bf9346ac75ff6348 : SUCCESS in 1h 05m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c8b53b7aed5948c3b5c2781ab1feeaec : SUCCESS in 1h 18m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d1b2d3bbfd3f43e0829dd39beb09653f : SUCCESS in 17m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4b8932c5727546d18d3ecf0cdcc7785c : SUCCESS in 59m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a4c82fd81354414983c409670b1cfeb9 : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":31},{"id":"3d9a9002012c67264b3ce9683c7a319c08eecb5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 16:01:19.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/188f6bc4cb994b9ca847bdc5fecfe39b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8958ba7455664cf0999e7c63dbb76d9c : SUCCESS in 1h 38m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1b14117ed712486aabcf0bf2977b1299 : SUCCESS in 16m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ecdbd87aab64b6e97a97bb68b8dc182 : SUCCESS in 2h 44m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4e9d929002a4fb9ad8c7a57f4d7c804 : SUCCESS in 2m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d70b6aa890774d5f83db3b6740809632 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d6af1bfbe8954de1848da2a0a049b306 : SUCCESS in 2m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c9ceef5100ca4675bdf2e35c96e1ba7f : SUCCESS in 5m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ad8ef19eaade4d57886cc282698774a4 : SUCCESS in 18m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/31e6547665f74983836d561c225127e7 : SUCCESS in 1h 40m 13s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"2a5481302dcf46548244a5abd193b4bf9aea0b02","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2023-08-24 08:49:46.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"469377f6e3ee385df17cc4a10247c045dc9dfa07","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-24 15:19:19.000000000","message":"Patch Set 31: Code-Review+1","accounts_in_message":[],"_revision_number":31},{"id":"da949181dc45ebad94d00cf8300b6e78bfaf2fa7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-24 20:03:09.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"19658196e31a5e4f36839d4579bea913921ddff2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-28 08:58:00.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"ca30b5f9e630cce1b8f468d4f646cb7bf2dc5066","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-08-28 09:49:20.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"5a62a918d405ab225cab8dae57692b4ecce043b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-28 09:55:53.000000000","message":"Patch Set 32: Published edit on patch set 31.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"66fb6b19408bdbf2d43dca7e224af42dbfbc372d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-28 10:03:03.000000000","message":"Patch Set 32:\n\n(5 comments)","accounts_in_message":[],"_revision_number":32},{"id":"81787b40a9939732fe5bb9c087b459d4f1e19370","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-28 10:08:29.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"9dfd72917870f1961594cff81c41abd214227484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 11:30:51.000000000","message":"Patch Set 32: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9856d446ebb949b8b146d305d5395f27\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ae1bdb00c1f542baa62234d03f286a78 : SUCCESS in 24m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/36e0738b4ac743ee9f35314b96200fb2 : SUCCESS in 1h 01m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8b399e0d80b54f9eb6b952443b7f5f34 : SUCCESS in 22m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bca44c52fad54b138fad5b2ddb49008e : SUCCESS in 1h 02m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b6fa3da6a6c24c2d890a1441bb0519c1 : SUCCESS in 1h 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/235580133e8c4d73922bb14868fbbdac : FAILURE in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c896ff6c2b64784af930fefb359c0ed : FAILURE in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44ddc5b863b94994b19ba6b9c5c15a4b : FAILURE in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fab6ea0315545e4912d4f538b5384dc : FAILURE in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a77085fc30dc49349dd2fd4a243427e8 : FAILURE in 3m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/581a227b82894c92a9a588b1e159a75d : FAILURE in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd20758b2134d52b7106046c8ac44f0 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33a142eb1f344cbab57b18851b89abe5 : SUCCESS in 3m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3d60b03ad86d486186249e608f9d73f0 : SUCCESS in 17m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1108c31133f74ca7bd05beb4571cfb71 : SUCCESS in 1h 06m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/25c8b54a33864e7ba073977b39337f59 : SUCCESS in 1h 16m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a27e978e8dc34c3ab0b90980eb5b933d : SUCCESS in 21m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/38aefc0511b44c53852ab17bda0fe95f : SUCCESS in 59m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/00928a1f4d2a4bc68cdf8dc6e1364a40 : SUCCESS in 1h 07m 09s","accounts_in_message":[],"_revision_number":32},{"id":"e313c559c058b89c6b4a777cc44fad40c11fe8a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-28 12:23:00.000000000","message":"Patch Set 33: Published edit on patch set 32.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"31d53d403492ae5de5274e99662779e8487fc784","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-28 13:08:52.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"0272cc9ec05ba1bd3d4796b7760b5dff74b2adc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-28 13:09:39.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"14ca95667e452a54bdfc7e1c15a273521ac12279","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-28 13:12:21.000000000","message":"Patch Set 35:\n\n(4 comments)","accounts_in_message":[],"_revision_number":35},{"id":"9a35f66e0ba7b8930e25579ce5a4d380c5ed1e5c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-28 13:12:30.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"07ee949cafe0b524c52dfed5c2550c57454518d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-28 13:12:56.000000000","message":"Patch Set 36: Patch Set 35 was rebased","accounts_in_message":[],"_revision_number":36},{"id":"63e93066c9c01b1dfc364415a2324bd7615639d5","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-28 13:28:17.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"e4ea43dfe35b5ab2291ee9ae98881146d5198d23","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-28 13:29:43.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"78687fc57e5f4748e6ccfa6550c0f94ef7dbbd70","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-28 13:35:34.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"3e3d1fbf47cc80899cb54f303f789d08c48b5689","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-28 13:54:49.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"d28e84ec4ef1a7fb8e4ec0c522866ce981ec6cae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-28 13:56:30.000000000","message":"Patch Set 37: Published edit on patch set 36.","accounts_in_message":[],"_revision_number":37},{"id":"eae7fbda57ad9055b78c78458be524be79cb80c9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-28 14:02:16.000000000","message":"Patch Set 37:\n\n(2 comments)","accounts_in_message":[],"_revision_number":37},{"id":"d93b3fc6c12d1bc419fcad9546ff0ce28c256fdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 15:28:53.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/321af14f66964fd88c6877dc65756e89\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0e84ef4c46924c5caacef27e49bdc26f : SUCCESS in 21m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8d1a54d72e7c459b9af3d54e16b62c18 : SUCCESS in 59m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/af07e32da1484e199d87fac52b223d1a : SUCCESS in 19m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/eb45a5485c924a6ea882e7fc22498ca0 : SUCCESS in 52m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/12dbe52b840a47a4bfb8b71e5e6d1d78 : SUCCESS in 1h 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc3d3008228d4d689d70536c6d52806b : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7146e7a459fa4616a360db9231c9d396 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1139ef0552a9415da4121f1a4a7f7ce7 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccd555ff0b68404c93a448f425fe08bd : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/519fa25259da4277afab2541a87ec386 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c57effe2191b42938691b2f6519a985f : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e4d6a281b2f4322b0383139fc46d693 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0c988bed084484a95b3d1c858e8f91d : SUCCESS in 3m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7cfdcfc169194b379afaaaa5a6e4c9fd : SUCCESS in 18m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d39672c0ea7f4b39af6dc8ad3449f31a : SUCCESS in 1h 12m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9bbae9c3f0f74362bf3043507f596342 : SUCCESS in 1h 25m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6fc7f00ceb91480f8500c44d5c5deb50 : SUCCESS in 23m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/30760a82a4f241e8a4142d39cdc813f7 : SUCCESS in 1h 02m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/79a2ddaa607049debd1a7339a85e1cc9 : SUCCESS in 1h 08m 40s","accounts_in_message":[],"_revision_number":37},{"id":"5bb168a0c51b311d1057629d2a31bb3b8fbf3e17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:20:36.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b63420de57b4192ad8ba1411ce476fc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/864aa31f43a54fccb3bbdb38cc046671 : SUCCESS in 1h 23m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8b78766a0d454b419782e8ede2edac71 : SUCCESS in 1h 19m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0ac1dbdfd22b4945913c805c6f9e1cd7 : SUCCESS in 2h 22m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3151b5244ecd4ea4b868a69c2b4109f5 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f7a85ba5fc84ff19b4180ceb93f0c8d : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ded64cd0fe0b430b9c42e303ad75f7f7 : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bbbbcaae664a4a8baf9e7192dc844e5b : SUCCESS in 2m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/536e0ddc23564485ad2d7c50b6e69b24 : SUCCESS in 21m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c00487d46b2e455ca2c59adc14152fe3 : SUCCESS in 17m 34s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"b97c7ac2c9900d79ea589998600f15f427daa6f0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-29 07:41:16.000000000","message":"Patch Set 37: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":37},{"id":"992b65ba17611117cdbea15237aee4aa750e462c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-29 07:44:44.000000000","message":"Patch Set 37: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":37},{"id":"2e8a53165314a66064a96b7880d15983350e2c65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 07:45:05.000000000","message":"Patch Set 37: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":37},{"id":"501b1b14ee80153e1206e9fd8d78baa0e6d7bea5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-29 07:59:17.000000000","message":"Patch Set 37: Code-Review+1","accounts_in_message":[],"_revision_number":37},{"id":"2b2e369866b4e8ac2dc84c035b7828803ea902d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 09:11:34.000000000","message":"Patch Set 37: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/781b6ffce3db4b8382f3fb0eb0779e5d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3ee0e080b48849b1b014feebee2a02e5 : SUCCESS in 16m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/736d502461d0490a909c9aeed9ec97b7 : SUCCESS in 54m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6c59e0266ed240cdb46e6b46aad6cdd9 : SUCCESS in 1h 06m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93fa734cc3e4d9ea0580197dfd10c2b : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ea51dba52624174b4b85b18eaf7394a : SUCCESS in 2m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba2c73b5ca1b4b948c22082b249bae0b : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa03bc28439f409eacb4a14900a1d569 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fef20348664ffcab41f37230e222cf : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/332b7fe8a477401d896a9428f5d030f2 : SUCCESS in 3m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5ac60935b1874055adb1f8b9f9ba9b70 : SUCCESS in 18m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/57adc27bd62e4f62a4bfa8c426c3bd91 : SUCCESS in 1h 07m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ef894aacc7124fe291cc805688b80ee1 : SUCCESS in 1h 22m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f3a9856a8c034172a8ccc00bd66249be : SUCCESS in 17m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3d41cda6ed3d4c629000a5e44df12f19 : SUCCESS in 58m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/245d38bf38a34f3aa458db5a8058b33a : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":37},{"id":"95b0862248b3aee069d9362436049de451273b73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 09:11:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":37},{"id":"dc785fe5b5ed341987c314d51a7f06af29993523","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 09:12:41.000000000","message":"Patch Set 37:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcd74601aa3a4b1298db775d8d31922e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/756653bd1b9a41f498bc43bf68b77581 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92c437913fbf46deaf6e1b09fb1f7c4e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":37}],"current_revision_number":37,"current_revision":"c9af85c5844172836d34c69dc7ea1b0a606795f6","revisions":{"8a26497db2c93e16176e109119afce5346192046":{"kind":"REWORK","_number":1,"created":"2022-01-21 13:05:29.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/1"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-21 13:04:59.000000000","tz":60},"subject":"Refactoring of the image module. Adding engine adapter module. Transitioning from docker.APIClient to docker.DockerClient. Preparing for adding podman as container engine for building.","message":"Refactoring of the image module.\nAdding engine adapter module.\nTransitioning from docker.APIClient to docker.DockerClient.\nPreparing for adding podman as container engine for building.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a26497db2c93e16176e109119afce5346192046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a26497db2c93e16176e109119afce5346192046"}]},"branch":"refs/heads/master"},"34d57bcb537ed027c3e91b06230d61359fefeeea":{"kind":"REWORK","_number":2,"created":"2022-01-21 14:11:32.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/2"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-21 14:10:59.000000000","tz":60},"subject":"Refactoring of the image module. Adding engine adapter module. Transitioning from docker.APIClient to docker.DockerClient. Preparing for adding podman as container engine for building.","message":"Refactoring of the image module.\nAdding engine adapter module.\nTransitioning from docker.APIClient to docker.DockerClient.\nPreparing for adding podman as container engine for building.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34d57bcb537ed027c3e91b06230d61359fefeeea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34d57bcb537ed027c3e91b06230d61359fefeeea"}]},"branch":"refs/heads/master","description":"Fixing pep8"},"22d24496c803e6b39a7eadf6c1139ed1b1b664aa":{"kind":"REWORK","_number":3,"created":"2022-01-21 15:21:51.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/3"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-21 15:10:39.000000000","tz":60},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file build.py into multiple smaller ones\nto divide responsibilities and make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other modules.\n\nTransition from docker.APIClient to docker.DockerClient for connection with\ndocker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22d24496c803e6b39a7eadf6c1139ed1b1b664aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22d24496c803e6b39a7eadf6c1139ed1b1b664aa"}]},"branch":"refs/heads/master"},"53d4147a6c14f0d990d72cf5ccef60bd37e0c2b9":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-01-21 15:38:41.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/4"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-21 15:38:41.000000000","tz":0},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file kolla/image/build.py into multiple\nsmaller ones to divide responsibilities and make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other modules.\n\nTransition from docker.APIClient to docker.DockerClient for connection with\ndocker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53d4147a6c14f0d990d72cf5ccef60bd37e0c2b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53d4147a6c14f0d990d72cf5ccef60bd37e0c2b9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7a15fded1cfa236eb7a6a83227e0accb207339d6":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-01-21 15:51:55.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/5"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-21 15:51:55.000000000","tz":0},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a15fded1cfa236eb7a6a83227e0accb207339d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a15fded1cfa236eb7a6a83227e0accb207339d6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"711a8789abc21ba9ec8498eb584107625f765669":{"kind":"REWORK","_number":6,"created":"2022-01-25 13:08:53.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/6"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-25 13:08:05.000000000","tz":60},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file build.py into multiple smaller ones\nto divide responsibilities and make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other modules.\n\nTransition from docker.APIClient to docker.DockerClient for connection with\ndocker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/711a8789abc21ba9ec8498eb584107625f765669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/711a8789abc21ba9ec8498eb584107625f765669"}]},"branch":"refs/heads/master"},"027da1b936a565cf5b29ead6b562b131eda32d41":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-01-25 13:33:49.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/7"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-25 13:33:49.000000000","tz":0},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/027da1b936a565cf5b29ead6b562b131eda32d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/027da1b936a565cf5b29ead6b562b131eda32d41"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cd54ea4f043540be8641a39cd7013504df65f6da":{"kind":"REWORK","_number":8,"created":"2022-01-25 14:01:09.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/8"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-25 14:00:49.000000000","tz":60},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file build.py into multiple smaller ones\nto divide responsibilities and make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other modules.\n\nTransition from docker.APIClient to docker.DockerClient for connection with\ndocker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd54ea4f043540be8641a39cd7013504df65f6da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd54ea4f043540be8641a39cd7013504df65f6da"}]},"branch":"refs/heads/master"},"8441086e732ab897081195b72775d456e9b209e6":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2022-01-25 14:03:15.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/9"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-25 14:03:15.000000000","tz":0},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8441086e732ab897081195b72775d456e9b209e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8441086e732ab897081195b72775d456e9b209e6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0ca31db4c628bb49c436d08fcf2bd1f781af9a95":{"kind":"REWORK","_number":10,"created":"2022-01-25 18:23:28.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/10"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-25 18:22:07.000000000","tz":60},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ca31db4c628bb49c436d08fcf2bd1f781af9a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ca31db4c628bb49c436d08fcf2bd1f781af9a95"}]},"branch":"refs/heads/master"},"cd8103130879c86501c336167b0982068abd1a2b":{"kind":"REWORK","_number":11,"created":"2022-01-31 17:34:42.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/11"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-31 17:34:27.000000000","tz":60},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd8103130879c86501c336167b0982068abd1a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd8103130879c86501c336167b0982068abd1a2b"}]},"branch":"refs/heads/master"},"e5322a3ce1f4c93042d5126eec02a814ff70f70b":{"kind":"REWORK","_number":12,"created":"2022-02-01 10:06:51.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/86/825786/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/12"}}},"commit":{"parents":[{"commit":"d84cd121cad5ed30fc87f97d5332f357b7f81e3a","subject":"Merge \"Ensure set_configs sets execute bit on directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d84cd121cad5ed30fc87f97d5332f357b7f81e3a"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-02-01 10:06:15.000000000","tz":60},"subject":"Refactoring of the image module. Adding engine adapter module. Transitioning from docker.APIClient to docker.DockerClient. Preparing for adding podman as container engine for building.","message":"Refactoring of the image module.\nAdding engine adapter module.\nTransitioning from docker.APIClient to docker.DockerClient.\nPreparing for adding podman as container engine for building.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5322a3ce1f4c93042d5126eec02a814ff70f70b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5322a3ce1f4c93042d5126eec02a814ff70f70b"}]},"branch":"refs/heads/master"},"e6547bf46f17fe19ba2d078475870e58529a4420":{"kind":"REWORK","_number":13,"created":"2022-02-01 12:42:59.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/13"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-01 12:42:45.000000000","tz":60},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6547bf46f17fe19ba2d078475870e58529a4420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6547bf46f17fe19ba2d078475870e58529a4420"}]},"branch":"refs/heads/master"},"32af1b4316522c581dda3a90ac29590cc2ca3890":{"kind":"REWORK","_number":14,"created":"2022-02-04 09:29:31.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/14"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-04 09:29:16.000000000","tz":60},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32af1b4316522c581dda3a90ac29590cc2ca3890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32af1b4316522c581dda3a90ac29590cc2ca3890"}]},"branch":"refs/heads/master"},"fd589ab95a0a92b07e4098666b594b42d8337ad0":{"kind":"REWORK","_number":15,"created":"2022-02-10 14:13:43.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/15"}}},"commit":{"parents":[{"commit":"336c921374c7c7376130c1497a7a0c22087cfb00","subject":"Refactor image module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/336c921374c7c7376130c1497a7a0c22087cfb00"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-10 14:11:38.000000000","tz":60},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd589ab95a0a92b07e4098666b594b42d8337ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd589ab95a0a92b07e4098666b594b42d8337ad0"}]},"branch":"refs/heads/master"},"ad6309c188e23cda1e2dfec4c28ddd60eae5b432":{"kind":"REWORK","_number":16,"created":"2022-02-11 14:29:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/86/825786/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/16"}}},"commit":{"parents":[{"commit":"333fb35c79d2912a10b70001f5d8b5724e9b8651","subject":"Merge \"rabbitmq: update to 3.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333fb35c79d2912a10b70001f5d8b5724e9b8651"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-02-11 14:28:55.000000000","tz":60},"subject":"Refactor image module and transition to DockerClient","message":"Refactor image module and transition to DockerClient\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad6309c188e23cda1e2dfec4c28ddd60eae5b432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad6309c188e23cda1e2dfec4c28ddd60eae5b432"}]},"branch":"refs/heads/master"},"5f69bf9b44c241a29654e32ce45b432021f23379":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2022-02-11 15:01:27.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/17"}}},"commit":{"parents":[{"commit":"333fb35c79d2912a10b70001f5d8b5724e9b8651","subject":"Merge \"rabbitmq: update to 3.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333fb35c79d2912a10b70001f5d8b5724e9b8651"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-11 15:01:27.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0 in requirements\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f69bf9b44c241a29654e32ce45b432021f23379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f69bf9b44c241a29654e32ce45b432021f23379"}]},"branch":"refs/heads/master","description":"Edit commit message"},"31ee7c64d3a3a1208e524c0fdffeddf6131d346c":{"kind":"REWORK","_number":18,"created":"2022-02-14 13:13:25.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/86/825786/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/18"}}},"commit":{"parents":[{"commit":"336c921374c7c7376130c1497a7a0c22087cfb00","subject":"Refactor image module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/336c921374c7c7376130c1497a7a0c22087cfb00"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-14 13:12:56.000000000","tz":60},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31ee7c64d3a3a1208e524c0fdffeddf6131d346c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31ee7c64d3a3a1208e524c0fdffeddf6131d346c"}]},"branch":"refs/heads/master"},"15a454d218e051ea7dffe2d758475c5a90193db3":{"kind":"REWORK","_number":19,"created":"2023-02-28 13:07:12.000000000","uploader":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"ref":"refs/changes/86/825786/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/19"}}},"commit":{"parents":[{"commit":"a2854da1b3e12987f7fa7590fca3bb74ac19c53e","subject":"Add a way to add other container engines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2854da1b3e12987f7fa7590fca3bb74ac19c53e"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","date":"2023-02-28 13:06:46.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15a454d218e051ea7dffe2d758475c5a90193db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15a454d218e051ea7dffe2d758475c5a90193db3"}]},"branch":"refs/heads/master"},"aa92bcf511f371c26d5df751785eec6169d2350e":{"kind":"REWORK","_number":20,"created":"2023-03-03 10:15:11.000000000","uploader":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"ref":"refs/changes/86/825786/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/20"}}},"commit":{"parents":[{"commit":"a2854da1b3e12987f7fa7590fca3bb74ac19c53e","subject":"Add a way to add other container engines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2854da1b3e12987f7fa7590fca3bb74ac19c53e"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","date":"2023-03-03 10:14:31.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa92bcf511f371c26d5df751785eec6169d2350e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa92bcf511f371c26d5df751785eec6169d2350e"}]},"branch":"refs/heads/master"},"adb06216ef0684b609ad1704ac994eca506e3ade":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2023-03-15 10:36:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/21"}}},"commit":{"parents":[{"commit":"12d431e3994023ce735a2352554a99a5cc9431c9","subject":"https: remove dhparams.pem before running gencerts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12d431e3994023ce735a2352554a99a5cc9431c9"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-15 10:36:05.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adb06216ef0684b609ad1704ac994eca506e3ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adb06216ef0684b609ad1704ac994eca506e3ade"}]},"branch":"refs/heads/master","description":"Rebase"},"aa3f1c86302ea75340f16eb1a2ac17bd2e87858f":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2023-03-20 07:07:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/22"}}},"commit":{"parents":[{"commit":"3d008b7f5ec2a54d004e8e9370f303ef9dc7858b","subject":"Test for unsafe files in tarfile.extractall","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d008b7f5ec2a54d004e8e9370f303ef9dc7858b"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-20 07:07:57.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3f1c86302ea75340f16eb1a2ac17bd2e87858f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa3f1c86302ea75340f16eb1a2ac17bd2e87858f"}]},"branch":"refs/heads/master","description":"Rebase"},"69b8375c476d586a016d0841a5225827bf96afc9":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2023-05-04 11:55:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/23"}}},"commit":{"parents":[{"commit":"115b46f76f1b7cf10ce496a16c0f01ccb5cdd385","subject":"Merge \"CI: Ensure quay repo is public on publish\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/115b46f76f1b7cf10ce496a16c0f01ccb5cdd385"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 11:55:46.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69b8375c476d586a016d0841a5225827bf96afc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69b8375c476d586a016d0841a5225827bf96afc9"}]},"branch":"refs/heads/master","description":"Rebase"},"ad6705c839bdd1c58b450fa9013fa776bb924817":{"kind":"REWORK","_number":24,"created":"2023-05-04 11:57:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/24"}}},"commit":{"parents":[{"commit":"115b46f76f1b7cf10ce496a16c0f01ccb5cdd385","subject":"Merge \"CI: Ensure quay repo is public on publish\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/115b46f76f1b7cf10ce496a16c0f01ccb5cdd385"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-04 11:57:52.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad6705c839bdd1c58b450fa9013fa776bb924817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad6705c839bdd1c58b450fa9013fa776bb924817"}]},"branch":"refs/heads/master"},"88bf42fc02d4e1a8b17780f560a7280a71e2aaac":{"kind":"REWORK","_number":25,"created":"2023-05-08 12:53:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/25"}}},"commit":{"parents":[{"commit":"050e26ed89b307a2037f118e314d0a44249a0f89","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-08 12:53:10.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88bf42fc02d4e1a8b17780f560a7280a71e2aaac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88bf42fc02d4e1a8b17780f560a7280a71e2aaac"}]},"branch":"refs/heads/master","description":"Rebase"},"6248931dc5100173cf13b164d9e1e54c63c0febd":{"kind":"REWORK","_number":26,"created":"2023-05-08 12:54:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/26"}}},"commit":{"parents":[{"commit":"050e26ed89b307a2037f118e314d0a44249a0f89","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-08 12:53:55.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6248931dc5100173cf13b164d9e1e54c63c0febd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6248931dc5100173cf13b164d9e1e54c63c0febd"}]},"branch":"refs/heads/master"},"a335789ee6b0b625552e7235a2ba01055adca6c1":{"kind":"REWORK","_number":27,"created":"2023-05-08 12:55:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/27"}}},"commit":{"parents":[{"commit":"050e26ed89b307a2037f118e314d0a44249a0f89","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 14:59:26.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-08 12:55:22.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nTransition from docker.APIClient to docker.DockerClient\nfor connection with docker daemon.\n\nUpdate minimal docker version to 3.0.0\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a335789ee6b0b625552e7235a2ba01055adca6c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a335789ee6b0b625552e7235a2ba01055adca6c1"}]},"branch":"refs/heads/master"},"6343774ecd55a91167a6ad180e63c27d396719f6":{"kind":"REWORK","_number":28,"created":"2023-05-17 13:12:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/28"}}},"commit":{"parents":[{"commit":"050e26ed89b307a2037f118e314d0a44249a0f89","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:05:35.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:05:35.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nChange to use docker.DockerClient instead of the low level\ndocker.APIClient.\n\nUpdate minimal docker version to 3.0.0 so that we can\nremove \u003c3.0.0 handling next cycle.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6343774ecd55a91167a6ad180e63c27d396719f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6343774ecd55a91167a6ad180e63c27d396719f6"}]},"branch":"refs/heads/master"},"ca62b9f5b723850f4f039a194877121637b28b1f":{"kind":"REWORK","_number":29,"created":"2023-05-17 13:22:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/29"}}},"commit":{"parents":[{"commit":"050e26ed89b307a2037f118e314d0a44249a0f89","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:05:35.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:22:36.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nChange to use docker.DockerClient instead of the low level\ndocker.APIClient.\n\nUpdate minimal docker version to 3.0.0 so that we can\nremove \u003c3.0.0 handling next cycle.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca62b9f5b723850f4f039a194877121637b28b1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca62b9f5b723850f4f039a194877121637b28b1f"}]},"branch":"refs/heads/master"},"10efe2a469e1941ed96ea08cfeef011af708cc6d":{"kind":"REWORK","_number":30,"created":"2023-05-17 15:09:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/30"}}},"commit":{"parents":[{"commit":"050e26ed89b307a2037f118e314d0a44249a0f89","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:05:35.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 15:09:51.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nChange to use docker.DockerClient instead of the low level\ndocker.APIClient.\n\nUpdate minimal docker version to 3.0.0 so that we can\nremove \u003c3.0.0 handling next cycle.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10efe2a469e1941ed96ea08cfeef011af708cc6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10efe2a469e1941ed96ea08cfeef011af708cc6d"}]},"branch":"refs/heads/master"},"20edebc636b46bb31ed63aeb998cd0a6c910892c":{"kind":"REWORK","_number":31,"created":"2023-08-23 13:15:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/31"}}},"commit":{"parents":[{"commit":"050e26ed89b307a2037f118e314d0a44249a0f89","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:05:35.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-23 13:15:01.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nChange to use docker.DockerClient instead of the low level\ndocker.APIClient.\n\nUpdate minimal docker version to 3.0.0 so that we can\nremove \u003c3.0.0 handling next cycle.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20edebc636b46bb31ed63aeb998cd0a6c910892c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20edebc636b46bb31ed63aeb998cd0a6c910892c"}]},"branch":"refs/heads/master"},"576890cf731fa1a804286e75f9db040e94b56435":{"kind":"REWORK","_number":32,"created":"2023-08-28 09:55:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/32"}}},"commit":{"parents":[{"commit":"050e26ed89b307a2037f118e314d0a44249a0f89","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:05:35.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-28 09:55:53.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nChange to use docker.DockerClient instead of the low level\ndocker.APIClient.\n\nUpdate minimal docker version to 3.0.0 so that we can\nremove \u003c3.0.0 handling next cycle.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/576890cf731fa1a804286e75f9db040e94b56435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/576890cf731fa1a804286e75f9db040e94b56435"}]},"branch":"refs/heads/master"},"9d6b8d21ea169929b684e47a609516d738e502df":{"kind":"REWORK","_number":33,"created":"2023-08-28 12:23:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/33"}}},"commit":{"parents":[{"commit":"050e26ed89b307a2037f118e314d0a44249a0f89","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:05:35.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-28 12:23:00.000000000","tz":0},"subject":"Transition to DockerClient and docker 3.0.0+","message":"Transition to DockerClient and docker 3.0.0+\n\nChange to use docker.DockerClient instead of the low level\ndocker.APIClient.\n\nUpdate minimal docker version to 3.0.0 so that we can\nremove \u003c3.0.0 handling next cycle.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d6b8d21ea169929b684e47a609516d738e502df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d6b8d21ea169929b684e47a609516d738e502df"}]},"branch":"refs/heads/master"},"6d06dd1556254bfcb1f9ffed311bed2f1441a60c":{"kind":"REWORK","_number":34,"created":"2023-08-28 13:08:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/34"}}},"commit":{"parents":[{"commit":"050e26ed89b307a2037f118e314d0a44249a0f89","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:05:35.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-28 13:08:52.000000000","tz":0},"subject":"Transition to DockerClient","message":"Transition to DockerClient\n\nChange to use docker.DockerClient instead of the low level\ndocker.APIClient.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d06dd1556254bfcb1f9ffed311bed2f1441a60c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d06dd1556254bfcb1f9ffed311bed2f1441a60c"}]},"branch":"refs/heads/master"},"e446b06ab9704e627919cb974a5e7e8172f89d27":{"kind":"REWORK","_number":35,"created":"2023-08-28 13:09:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/35"}}},"commit":{"parents":[{"commit":"050e26ed89b307a2037f118e314d0a44249a0f89","subject":"Fix docker version check after docker-py\u003e\u003d6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/050e26ed89b307a2037f118e314d0a44249a0f89"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:05:35.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-28 13:09:39.000000000","tz":0},"subject":"Transition to DockerClient","message":"Transition to DockerClient\n\nChange to use docker.DockerClient instead of the low level\ndocker.APIClient.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e446b06ab9704e627919cb974a5e7e8172f89d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e446b06ab9704e627919cb974a5e7e8172f89d27"}]},"branch":"refs/heads/master"},"3b5cc22c6d275e48be68cbf23d8635c6659beacd":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2023-08-28 13:12:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/36"}}},"commit":{"parents":[{"commit":"4c31f7a3f2002d77dd715dfbb5c2eb74192149d4","subject":"Merge \"debian: add mariadb bookworm repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c31f7a3f2002d77dd715dfbb5c2eb74192149d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:05:35.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-28 13:12:56.000000000","tz":0},"subject":"Transition to DockerClient","message":"Transition to DockerClient\n\nChange to use docker.DockerClient instead of the low level\ndocker.APIClient.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b5cc22c6d275e48be68cbf23d8635c6659beacd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b5cc22c6d275e48be68cbf23d8635c6659beacd"}]},"branch":"refs/heads/master","description":"Rebase"},"c9af85c5844172836d34c69dc7ea1b0a606795f6":{"kind":"REWORK","_number":37,"created":"2023-08-28 13:56:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/825786/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/825786/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/825786/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/825786/37"}}},"commit":{"parents":[{"commit":"4c31f7a3f2002d77dd715dfbb5c2eb74192149d4","subject":"Merge \"debian: add mariadb bookworm repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c31f7a3f2002d77dd715dfbb5c2eb74192149d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-17 13:05:35.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-28 13:56:30.000000000","tz":0},"subject":"Transition to DockerClient","message":"Transition to DockerClient\n\nChange to use docker.DockerClient instead of the low level\ndocker.APIClient.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I88842c17d8e6ddbe89b984c7b5e0367ed356fa50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9af85c5844172836d34c69dc7ea1b0a606795f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9af85c5844172836d34c69dc7ea1b0a606795f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
