)]}'
{"id":"openstack%2Fkolla~825791","triplet_id":"openstack%2Fkolla~master~I068c906df97745e397408d8c3ef6af47ee037638","project":"openstack/kolla","branch":"master","topic":"kolla-podman","attention_set":{},"removed_from_attention_set":{"35104":{"account":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"last_update":"2023-09-22 09:43:07.000000000","reason":"Change was submitted"},"34113":{"account":{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"},"last_update":"2023-09-22 09:43:07.000000000","reason":"Change was submitted"},"29636":{"account":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"last_update":"2023-09-22 09:43:07.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-21 10:46:49.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-22 09:43:07.000000000","reason":"Change was submitted"},"34501":{"account":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"last_update":"2023-09-22 09:43:07.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-19 16:05:20.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-09-21 15:27:03.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-09-22 09:43:07.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-09-22 09:43:07.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2023-03-25 20:21:50.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},"33708":{"account":{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true},"last_update":"2023-09-22 09:43:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I068c906df97745e397408d8c3ef6af47ee037638","subject":"Add podman for image building","status":"MERGED","created":"2022-01-21 13:17:37.000000000","updated":"2023-09-22 09:44:07.000000000","submitted":"2023-09-22 09:43:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":60,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825791-kolla-podman","meta_rev_id":"883a796b652d81251225c91a205faf45879b6d15","_number":825791,"virtual_id_number":825791,"owner":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 09:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-09-21 10:46:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-09-21 09:01:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-09-21 15:27:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-09-21 10:40:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-09-22 08:01:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":1,"date":"2023-09-21 10:46:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-09-21 09:01:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-09-21 10:40:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"},{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true},{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"}],"CC":[{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-21 13:30:09.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-01-21 13:30:09.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-21 13:30:09.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-01-21 13:30:09.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"},"state":"REVIEWER"},{"updated":"2022-01-21 13:30:09.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true},"state":"REVIEWER"},{"updated":"2022-01-21 13:30:09.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"state":"REVIEWER"},{"updated":"2022-01-21 13:37:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-21 13:44:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-25 20:21:50.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"},{"updated":"2023-05-04 12:10:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-30 13:25:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-09-12 06:11:12.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-09-12 08:33:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-09-14 12:00:06.000000000","updated_by":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"reviewer":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"state":"CC"}],"messages":[{"id":"cdbb431492b444121aef37ea48ad2db3039aed22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-01-21 13:17:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d4cb22f4cb2892ef6f039b82b5d489bec570655","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 13:37:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/56557961718844639d16ced912fab0b0 : FAILURE in 7m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/39d05fdea5794f4caf5cf2f0d7f331a3 : FAILURE in 6m 51s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4fa7044c743b45bb9976716309da63df : FAILURE in 17m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec213de59e0b46fbbcf9370f2a3d1022 : FAILURE in 6m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f328eac559e49cfbd57a66ab86969ec : FAILURE in 6m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/59dfcba2369a43c9948e6361bb802834 : FAILURE in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"099eefa909fdaf90d7029e85ef7efa3c56188dd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 13:44:30.000000000","message":"Patch Set 1: Verified-1\n\n(55 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f516429068d848c68d1656ebf1c327d2 : FAILURE in 5m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9854ad6c98a64fcbbc448316c932886f : FAILURE in 4m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/05a68beb7d53431ea405b1e9f889624d : FAILURE in 11m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6860ba60f21d477aa8e593eea7658ecb : FAILURE in 12m 42s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/662db31949cc4cada5894fbc24796742 : FAILURE in 12m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/89d2e0866b9044b7ad3b328cd673c4c6 : FAILURE in 3m 03s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7c8789d693d24df8b9d7202a35f9eef4 : FAILURE in 3m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9cfcedd2ec0e40339c9bc8750f88e617 : FAILURE in 21m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/245d3a8eca94460bacd538143a1e774f : FAILURE in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e738fad5c3b64928b3b04575fc89a389 : FAILURE in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5916e5a6e5454dac918a011a57c38092 : FAILURE in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d07de0012cfb44e0946a821c828ee024 : FAILURE in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f3946049ae40eca30e96255986ebb6 : SUCCESS in 5m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/32d8a0b657ac4723a923d6543ef320fb : FAILURE in 3m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/acb02bb2d61441869fdc1181b43c27dc : FAILURE in 3m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b143f3926f7648e08b1736aaa26555ed : FAILURE in 9m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e1a58443e9844331a1eefad075b37816 : FAILURE in 9m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/27fc3198bb3b4ec7a30834c0ab9de9b6 : FAILURE in 10m 15s\n\nWarning:\n  Comments left for invalid file kolla/image/kolla_worker.py\n  Comments left for invalid file kolla/image/utils.py","accounts_in_message":[],"_revision_number":1},{"id":"8dd3674c4c6eb38e0a0739c8842bb5fe76909668","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-02-01 10:06:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"eea3b7295bea74cb5e29979a274c2993a75055c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 10:26:08.000000000","message":"Patch Set 2: Verified-1\n\n(55 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9a5aa60d81e24e9da3d8fba95c7feb0f : FAILURE in 5m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dd30056f98e640cf9234b957433a6f20 : FAILURE in 5m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c974b55207294728b8837ebf16c20287 : FAILURE in 12m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a32d79401b7842aebbba588495765323 : FAILURE in 14m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e62bf405f4a347538969aa9f31512127 : FAILURE in 12m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f4881cd2cfc749aca3621482047d7742 : FAILURE in 4m 25s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6995b283bb214f088d8fef512c2ec2bb : FAILURE in 3m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/85289ab8c5c04379bc7c5772508ab9ff : FAILURE in 13m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7797f3c3849346a286d8b1b9fbb8da2a : FAILURE in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3ee884a20f44e91859ec9cce286e871 : FAILURE in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/508703342c5144659b28e03d3969a1b1 : FAILURE in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71824cdaa71142a6aa199d34e09d9a7a : FAILURE in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65e794a01544e43859f77481c03dea4 : SUCCESS in 6m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c34ef0e6d0594759a2db4b5bbb0903e2 : FAILURE in 4m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/66654e8bbdf64664ad8e510bee94e6a5 : FAILURE in 3m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c09b7c9221a349f49ead15ae45f940e0 : FAILURE in 9m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/82a261c109314fabbedcca515d2df3ff : FAILURE in 10m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a3234a0c3cc941408df9a1fa3c3ddfb5 : FAILURE in 9m 00s\n\nWarning:\n  Comments left for invalid file kolla/image/kolla_worker.py\n  Comments left for invalid file kolla/image/utils.py","accounts_in_message":[],"_revision_number":2},{"id":"c9823ca56123b4fcdc9bdae61f6c7c0436700f73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 10:27:04.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c8f619ee67f243729402ccab8133d34f : FAILURE in 8m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/600bb3fb282c47d98b24423f19361de7 : FAILURE in 5m 41s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c41ef61dbe854b0b9192d784e2e1ab9e : FAILURE in 17m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fcd43ae04e24ad9954e192fdf64315d : FAILURE in 5m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b167ef575fb64ed98ef77db53876dc0c : FAILURE in 5m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b2fa1bca6ff44778b472596309f963ad : FAILURE in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae9fa4b55feb5853c992326e2a710b3b4d3670b2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-02-02 14:58:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93c0ddc4c266bc3c45aa8a67bae9220157e469d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 15:23:20.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e5f87e9a02d40db99dd20bc6d9f25f9 : FAILURE in 10m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/68d4d1c3e5d94c3c995a890f465e0ede : FAILURE in 6m 43s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8d2a702aabd43f293d20a2bf1d0c44d : FAILURE in 21m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ac9f5f63d0e48b1bf6d86f1a40e8eaa : FAILURE in 6m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8437897dcfca42d1a01d4de05baa9c78 : FAILURE in 6m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ac5cf4a8574474fac6d1fc5bf3f7c20 : FAILURE in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4bdf63e30c4c6834fccce3c4b1ad10324ea73e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 15:28:07.000000000","message":"Patch Set 2:\n\n(55 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e492fd40552c4a88b85ee174e42ee629 : FAILURE in 5m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/50006195f2d94596a10988381079e4b1 : FAILURE in 7m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7d4ee2b31d9448c99267ee6ac05919f8 : FAILURE in 16m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fb391f5d514041d296c5ec3a78d4cf9a : FAILURE in 16m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1a8e32238a4a43c08653c8cc83d55366 : FAILURE in 12m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/39d3a6379fde4865a0c2db95a2f7ee6f : FAILURE in 4m 41s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/78952a7120b748a6abc6c59cbec9ea81 : FAILURE in 4m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/77a88b7a97d94d838e1478142cb1f7f4 : FAILURE in 21m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b06b51464fd34fafa455fb338179f75e : FAILURE in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4111af060914c20bacb83adde86ec8d : FAILURE in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5452e82a567e4a919ae716b30df21197 : FAILURE in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ba3cbc603554f71b5d6e1f525a4ea23 : FAILURE in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e96857905510441c81842abb9df3f713 : SUCCESS in 7m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c0ee8fb03dff4508a4bd868d83f7ab7a : FAILURE in 4m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3faf7617e034adba0709bd2c829b35d : FAILURE in 4m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1243e0005ab14ca483379a029a35276a : FAILURE in 10m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/55fe772e076d49b1b20b1736b5fbfb94 : FAILURE in 11m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3847a7db61ae4dba88ddf24c0233332a : FAILURE in 10m 00s\n\nWarning:\n  Comments left for invalid file kolla/image/kolla_worker.py\n  Comments left for invalid file kolla/image/utils.py","accounts_in_message":[],"_revision_number":2},{"id":"0cc2898f4e482bb88e9c3880608f041456d5e25f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-04 09:43:07.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"afff6f182e7c08a6d0341b9501cc1fbd6c40ee1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 10:08:08.000000000","message":"Patch Set 3: Verified-1\n\n(55 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1261e3efe7a3427dbc590d9cd6c2d0ed : FAILURE in 7m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d501e4321d124ce0975cb322f7c179b4 : FAILURE in 6m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/821d3289f2fb4e4cabdae4f9970986be : FAILURE in 16m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/82c70a71f58e407096543de9d6954578 : FAILURE in 14m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5f12651e286c4fd7a4f25d5b374a7271 : FAILURE in 13m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/25591000f89748a6af7f8ea5122817cb : FAILURE in 5m 03s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/45514644d47e40659adcd4b666ebed2e : FAILURE in 4m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/331c6377566d4ed28ac32e61d2516bbb : FAILURE in 10m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0f2ee7c6bd4546808f043e7449c101b9 : FAILURE in 13m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9567f11d8164b718a4ee8eea33fc369 : FAILURE in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad45191cbdd74072903e2ca0afd76718 : FAILURE in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d763461e02445b58fff0f69855ac64c : FAILURE in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25bc62a9718849eba30bc65d71f8fd22 : SUCCESS in 6m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/05f066a84c7647c2a49441e1447c8f87 : FAILURE in 4m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6a986ae30b52487f8267e4f8e2360478 : FAILURE in 4m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca5cfed27a1c42ab8bbe661dacc83630 : FAILURE in 12m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bdf235224feb4bf0a645d175a47bc7d6 : FAILURE in 11m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8f08d0e294694d45997378c56d5f13ff : FAILURE in 10m 54s\n\nWarning:\n  Comments left for invalid file kolla/image/kolla_worker.py\n  Comments left for invalid file kolla/image/utils.py","accounts_in_message":[],"_revision_number":3},{"id":"df080bd1d582cf5dc1abc4fdd99cf4d3aabca04a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 18:35:34.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/547f712ca0cb4b4f8bca5904eab71c4c : FAILURE in 9m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/29fc9d550c684df79f0b8a054a431a93 : FAILURE in 6m 14s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fb59e90894154d30bbf0edb295611bc8 : FAILURE in 17m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32184ffbe2ee47698da9fbce03bdb973 : FAILURE in 5m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cfa182e86baa46a3939e7a609249a6cc : FAILURE in 6m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a58cca360f8f4067bfb862248f981101 : FAILURE in 6m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f38ebf1e3c2f8f15c20a38d6e5a8b090609995d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-07 09:15:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0337901a5cc45e33d295a4e190958796ac7b7931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 10:43:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f933fc42e9d54d5e90f81f568893599a : SUCCESS in 39m 49s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f09426fa3a474dde99abd94ffc230365 : SUCCESS in 31m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d094aa6793bb4fe7b338ff5797a6cf36 : SUCCESS in 1h 03m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c46159971cbc4de783767643aef378a5 : SUCCESS in 1h 22m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2d457d7c83cf4985b9411e1e554b7600 : SUCCESS in 1h 10m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/18d5e5976d5346e5b9f1cc19cc8df764 : SUCCESS in 27m 48s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/557ee0ab50dd49749b0c4419563b0e39 : SUCCESS in 16m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/adcebfb51a804841804a1fb441a47e7c : SUCCESS in 56m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ae70e009eac14858b4378cdf2bc3758b : SUCCESS in 55m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cd47719fe004abcb9c8868ed51b826b : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/434dc812d7004194a338edfda7804453 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9160e2d545e5495396d6dd9e5d617cd9 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7178a0ed54444742a1d0f852ad6a2b69 : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d90e9755d0f4ee5bb74f1362f543f9f : SUCCESS in 17m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a46af98f1b24b3eb72e7572b98c25d9 : SUCCESS in 30m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7928ad50fcb49d0b5a4efe2ab26e9cc : SUCCESS in 53m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c8428c9e0fb94e4383dcdf1d0a54d9f0 : SUCCESS in 41m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9f6ea7732e9b43c5bd2f5da43e9a2ef6 : SUCCESS in 1h 06m 31s","accounts_in_message":[],"_revision_number":4},{"id":"1bdd8327b5568d4a0ecd204316356e38db38863f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 11:53:15.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/951f2581ceb64a52b154e958270c892a : FAILURE in 1h 59m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c484d13ce344efebc5e810b3d51c075 : FAILURE in 33m 23s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ca21330162e48eaaf9b77daf63758ad : FAILURE in 30m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7687b9182a5e4138a68a9cf9a49615a2 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2910fc3dbf014b058dc2aace525b4619 : SUCCESS in 6m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad56b17d1f934766870aaae81223d8ad : SUCCESS in 1h 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2d010573ab2ab5a8777ebd6ab70be1a76b492d0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-11 14:21:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ba9226d8fd5152d63d05e7a584cc385f88ad119d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 14:22:13.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"19f6fb05029a861423aabe90492831b6a3284635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 14:22:31.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"ccda13e036dc7eee0acded70718375bc1a8a9df6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-02-11 14:29:28.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"0d305cfb227903fc2f36899e0b7373d1e5182605","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-11 14:46:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"57ce43dc732229ccf80b28a045453a1d1fc513c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 14:46:45.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"d10f8b3282198d1bd6e7516029bad5077172c70e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 14:46:49.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"bed56fb2daa8342f08e0c0b568030e9951cdc8ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-11 14:48:39.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"5ba576f476dc7c8d763f14fc7146dae6fa88ec9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 15:11:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d14dcd9ac1824b47a65679028acce3bb : FAILURE in 5m 11s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ae05e5fdde9a48acae115f7e5962c29e : FAILURE in 5m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e4590e2147734324a33c4c7c3bdbc579 : FAILURE in 12m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/54808f7b849e409890c792ed3e157839 : FAILURE in 12m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8abe017d2a624f2fae3db6525de5acfa : FAILURE in 12m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8ec46edb05894537bc92a56d407eba9b : FAILURE in 3m 29s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9f87057fd3c047208dc037a4f235fa5e : FAILURE in 3m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d942ed68966f498f974cc4df8c0773f9 : FAILURE in 14m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2701c0f293cc4372ab293c4bf6c924eb : FAILURE in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d50bedc98fd5495fa92c9ac64da875c1 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f354b2a79694695b9d3c5dcd741368a : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/447c3b38d47e49dbbb1e51e6b5be9caf : SUCCESS in 4m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ce68771f77684b27a1a8ae39b311f867 : FAILURE in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7898da3cc677425c804e4b8777fd686c : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16f1a7f0f8a54218a7d9ee35d6032836 : SUCCESS in 3m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f357140c17c74294ab83d990b8a0db43 : FAILURE in 3m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b02ff68491b640f0a3fb545b8582cc62 : FAILURE in 4m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1afe04cd96304387aaed1b85adaca057 : FAILURE in 10m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/55bf39700be34ea9857707db798ff090 : FAILURE in 12m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/815bd1e8e272406e8c363d82e9e49092 : FAILURE in 9m 39s","accounts_in_message":[],"_revision_number":8},{"id":"44d3622c1e6d35c15dbe47fa59b13573bc87fddd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 15:21:29.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2086f2285fb04480bcb0270654f24dcf : FAILURE in 10m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1440f80df96e43e892b00d18ccb1aa1d : FAILURE in 6m 07s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/800885ad4e8a4cc2af1ac51c2967bc8a : FAILURE in 17m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c966edf311d24eadbabf08e7bcdbaf7a : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f88591fd024f46adbcd0444dd0a8b8cd : SUCCESS in 8m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bbb2b4fe13364ab58b9f7a303fc52d6a : FAILURE in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a1caf30a399ada4e8210d057470788dbf91ddd94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-14 12:11:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"929fd345dc06d7e3c961bd45b044bb2b50f587ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 12:30:39.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/482740ad730441b88d0b0ebee029feba : FAILURE in 4m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/89c68024c02f4857a777e6109325b773 : FAILURE in 5m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/acea49995d6e432aa10adb80701df6fc : FAILURE in 13m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8f5a4dbc767d4277927df2272e861b71 : FAILURE in 13m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cfece1a5d1ae40a989b4801925ed0306 : FAILURE in 13m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bb88816cd8b9490892e1526a29fbf50c : FAILURE in 3m 16s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ca62978332c94ea28a77642c792bed91 : FAILURE in 3m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/61598e66594f486fac873816f12cf4c0 : FAILURE in 11m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/989355dab6f947b4aeead26458e12cc9 : FAILURE in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aad3f29e11494cec9fa1a3e383fd0e26 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b7734d20530495bacab6e62ff94623d : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e051eef73924166bac591e0f6d67b16 : SUCCESS in 3m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/618b43a016df4892908b2527310ee6ef : FAILURE in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299a6244f0a741308f84dffc04271790 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/335ca86a5f6c45e9a7ce31fa2c9a34d5 : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/100b32fcb5fa4b458e3c8be887354c33 : FAILURE in 2m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7983c2130c347e7bfce4bf53a6c9164 : FAILURE in 3m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c44e5832dce46c2a14f3c53f5fdcb46 : FAILURE in 9m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a416e9372d60401fb6a032d69b7f713b : FAILURE in 10m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4fe69eb033a14f05abf5da4d3236344c : FAILURE in 9m 09s","accounts_in_message":[],"_revision_number":9},{"id":"33f1d27891edf6dd6b5a369117915f69803f0d32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 12:32:30.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/61aacd4b91cb42b7aef2a36a2fbd9991 : FAILURE in 13m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a926d452025b43f3a92b287d2dba6289 : FAILURE in 5m 36s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a870f1b8edf4036bb9c6c00e5ff59b3 : FAILURE in 16m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/daf75e16851945448e09c822d49a468f : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63e2357cfe414323a3a581577b1fe8f9 : SUCCESS in 10m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/00594d9ec465452eacfa7c5f90f46276 : FAILURE in 7m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ddb28584bcab3d2c0304a1002588f3d5fa9ce572","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-14 13:04:17.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"cd11ebc7694cbc4772a46f8485a3271c8c04c9d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 13:32:36.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/449f10112fd34f23ae9dfa74e22aeda1 : FAILURE in 5m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/836724b17d7c436fb4499c78cef7cae8 : FAILURE in 5m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/44e6d894ff7b4cd787bcc256d603d676 : FAILURE in 12m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a54c2b499bc04bdfb7128893440974bf : FAILURE in 14m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f60e157c68f442578c106d9bdfe2f0d9 : FAILURE in 11m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4a12ab94f40e491da94019a59f8f1ab4 : FAILURE in 3m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6cbfd7e2e55244e0abced54c358ad7a5 : FAILURE in 3m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/54fe1447e0e04ae4b54d93ad367f1d7b : FAILURE in 15m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cdcfb96b8a864333b15e375afb5e0a91 : FAILURE in 22m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62c4c5a43b3475b839fac6150bf2cb7 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b9b1f698c7e4572849749b9fab11df4 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf6395e827cc4eafad1c7d47195cb994 : SUCCESS in 4m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/063509e63b604921aa70f642d6a0ff69 : FAILURE in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acac9a4d240745d6b96ba6dc0bd430bf : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/034fa11813554dfdbf1b688fda01593d : SUCCESS in 4m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f62c6f3272a45838ec08e5b3a2f6f67 : FAILURE in 3m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f55251c6c74a45fab4c2b18ba71d97e2 : FAILURE in 3m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36ff70129ded4cc3b6ead4245b318b72 : FAILURE in 10m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/caa1779355094c978c5ae90d0257d5d6 : FAILURE in 9m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b029f595aa734be5b764efbd396ab59d : FAILURE in 11m 03s","accounts_in_message":[],"_revision_number":10},{"id":"6375090c3288fdbe32927aff34b0a09075ca6506","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 13:33:47.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b4a7d712409c47938fc459ea99abee72 : FAILURE in 10m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7db8be624e014010ab5fa9aeba189f0e : FAILURE in 6m 15s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b2b04c862b1042bb92747eb38d365125 : FAILURE in 26m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b8c4c3f7c6e4f80975abcb5a02465a5 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3fc987b1a93a4d48b6231dc099dfa120 : SUCCESS in 5m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/76807ab0f0d94ecab1bd093a3b0adba9 : FAILURE in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5ec70e1200a11b68d2937ea161a446e793cb4824","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-14 13:50:24.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7d4fd8ebe6a90d815dabacb67fd46dbe0d5a270c","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-03-25 20:21:50.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_30491\u003e.","accounts_in_message":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"_revision_number":10},{"id":"18e44e25e52f1e28f93d6450d841716ba5e7b598","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-04 12:10:45.000000000","message":"Patch Set 10: Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"d9809fa290ed0f3363f69f8cc0e9eece5ae88f87","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-18 05:16:30.000000000","message":"Patch Set 10: -Review-Priority","accounts_in_message":[],"_revision_number":10},{"id":"ae2fe5045568cdf96ea834ae98608b91bb6fcf34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 00:21:59.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"ed597e1f5a2ef504f95fc4e7b6233567eff12953","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 00:28:45.000000000","message":"Uploaded patch set 12: New patch set was added with same tree, parent tree, and commit message as Patch Set 11.","accounts_in_message":[],"_revision_number":12},{"id":"0ccc8573cba01d82b7f07455596f484b067d6715","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 00:31:51.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"161dcd535b7180f4f6193ee7dc856a83b558c2bd","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 00:39:40.000000000","message":"Patch Set 13: Review-Priority+1","accounts_in_message":[],"_revision_number":13},{"id":"4832bf6270c2dea57f1d00346df1852a5c2eefca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 01:38:50.000000000","message":"Uploaded patch set 14: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":14},{"id":"3b35bbfb895364162304cb408a8ac3bcac07ebae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 02:54:34.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dbcef99af354460e8b6f72fd3bd1e437\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fe93cea7a676413fa1ce4af27fdad7cd : SUCCESS in 20m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/062d4e2034114b618c30993ba773db29 : SUCCESS in 1h 09m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c20932a1a687492bb4d8198b733fa762 : SUCCESS in 14m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/09d2e771f1754b78b24a661524da17e3 : SUCCESS in 1h 06m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1fa18c835a5c49339ef32e970da77f70 : FAILURE in 13m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f188e4ce3a4607bedae42f06cf45cd : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03e5c32e40b84bc5abc792ce364265b2 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8be28baaf1234039a2779fad6e08cb9e : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da1e0c31a1c74bedacf9aa50edd90f45 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f0e446955d641309823f55f71a9dfc9 : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d51d9091838c4beba6518f6c4a54d111 : SUCCESS in 2m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a91157a221a244f8a3fe86a7219c27b2 : SUCCESS in 1m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c133e7fd93924ae1b7ac18a9597c1fb4 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/487e15eed8af41fb880e530e70d09a75 : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/515fa9bd1d7547de9a613670dc883333 : SUCCESS in 19m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cb414f0cdf774750a7eff3735b398431 : SUCCESS in 1h 07m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cccb12a5fd264c4fa6c445e9c811775a : FAILURE in 17m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/88b78d44fefe4c338f78a7e95c4ff993 : SUCCESS in 16m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f09b17bf76294ff792bb158e3b281bb0 : SUCCESS in 59m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6793e4042c7b46a9a415a5e243f246ee : FAILURE in 14m 59s","accounts_in_message":[],"_revision_number":14},{"id":"3fa6447d4253fbdc2426f8cb7881dbe83bc8ecfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 04:29:35.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23cb35d131f3469895f5dd0501b37410\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/faf62d8f9463425cbb3f0683cdcd50b9 : SUCCESS in 1h 05m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/203fda74486048af827901adfe7e7500 : SUCCESS in 1h 29m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a5e514fc678940129cb5c77136c17811 : SUCCESS in 2h 48m 55s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/752c2db8a3ae4364afb6edde6b6ff04e : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef4affc7707643e68ac2a87df5084026 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2ff0abebd5704386ace8ab1ceecfa165 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1dadda329b004affa82ac4c287a89519 : SUCCESS in 4m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ad62d01ecc6648f58a7092034da357a7 : SUCCESS in 1h 05m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f6d63cd586db49dfa1ac39f1c62b4cc6 : SUCCESS in 1h 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"62cb365d03986b91ad95c9cb5e94b878c6249f02","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-12 06:19:30.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"fb2bc3284253155c203e7cf0e76c2c517048e3d9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 08:09:12.000000000","message":"Patch Set 14: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"72029be5ebc674ca94ae64977221f5d7a5270f51","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-12 08:33:56.000000000","message":"Patch Set 14: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":14},{"id":"232cc761bdaa5b87fdca71b82f62758655556b9f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 09:38:26.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a6f7c88f0d615670bec1e388ecc3cd19774c693f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 10:06:19.000000000","message":"Uploaded patch set 15.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"5f41d091e1c311dc6ecb24bf1b847272b8fe6459","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 10:08:08.000000000","message":"Patch Set 14: Review-Priority+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":14},{"id":"b1727fe7b439cf15c79b716cb39f5448ea198e41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 11:15:22.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8820e4ae06574e65b52dedb468b5872c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/141bdcda5aa84cceb91a540664eca18c : SUCCESS in 21m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/527adecbd04b444a8baa886e24e09341 : SUCCESS in 1h 04m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2fe9ef5e13124861b09f5e9af91ab901 : SUCCESS in 17m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dbcf10eab2a84944b6f16e4dba76b6a0 : SUCCESS in 54m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8aff621049b7425a8a77397ab7a6905c : FAILURE in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d653ae8041a34efb8212f1830e24130c : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb047f8a281d47ffbd5dab8d29be6476 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5b230bd22934a4e8ba1b7925f6818f5 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb80aed559f6473c8f68938664cfc89a : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f37906edcf6a4f2480f0c03414a5d50f : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e8edcef696d4953945195f8d88a9fb5 : SUCCESS in 2m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/54b30f613fa44efca85f72d71957b855 : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382f4d65eeee43a18d294e5868cdf7ba : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9d89bc505d94d1782ffa266dcc4e8cc : SUCCESS in 3m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fda6ccf9bf2342d3a92e5b0d42b10cc7 : SUCCESS in 18m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dd940c1fbba04a72b11bd4715e98fa4c : SUCCESS in 1h 07m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ecdee0d4c84c4ba7b26de5cbf31de3c7 : FAILURE in 21m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c0fa6511666f4a78984b8dee54c5b2f5 : SUCCESS in 24m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4e567b68083e4c6d8946392b14b80dca : SUCCESS in 1h 02m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f92e6cf0b3cb42a7af4beb3bb083c806 : FAILURE in 19m 23s","accounts_in_message":[],"_revision_number":15},{"id":"b56f7e768490d67555ab671aab92b9bbcf1cf830","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 12:17:44.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f7d07c7aa27428da7daac39df1ba31a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d1ca5064dd2e4a78bcbe82e2e13f5ba7 : SUCCESS in 1h 12m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/389fed984cae4bb185038455eaaf57fb : SUCCESS in 1h 11m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/40c64ca1756d40d5ab3e8583b9bb1217 : SUCCESS in 2h 09m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c92ee9fe09b94cb3acc1128d867f77be : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/043e16e6caa54c0db3798e00ebebf35f : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aefde82aa13549109eabf91581a34e62 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b85aaced86c64eff813ef10ef4484cf3 : SUCCESS in 4m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d450d167dcfc4367bb59c9e7a4c7a5a1 : SUCCESS in 1h 16m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/94faa050a3b445aebbb39b4652283c72 : SUCCESS in 1h 11m 58s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"155b8dfe8baab39fc1053afce90209b586de4310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 17:49:46.000000000","message":"Uploaded patch set 16.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"cb8e4d73bb30b1969f4b8df8a7b7fca808e36e3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 17:56:48.000000000","message":"Uploaded patch set 17.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":17},{"id":"b076af6057f7c314a0f18f81c41392b862af4c0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 18:52:00.000000000","message":"Uploaded patch set 18.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":18},{"id":"adeb088fd4451318225f4dff40b5eb1408de7c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 19:43:08.000000000","message":"Uploaded patch set 19.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":19},{"id":"f9fc9d89887634e627d515dbbb22bf3db88bf934","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-12 20:27:50.000000000","message":"Uploaded patch set 20.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":20},{"id":"7a17df5ebc265a193d8db11d8b853538828f49f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 21:46:47.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df8ec5e9a2b1460da704222a5cde6e00\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/62b6d0f45e554d62b053ed28164c0cc5 : SUCCESS in 18m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0b82ab51e0d643258347fd1eef1ef6bf : SUCCESS in 1h 12m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c255272d30624f8294dce4631a0aaed7 : SUCCESS in 18m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a1a06bff3ba543c79346913b14cc5295 : SUCCESS in 1h 06m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8252bfac2f9d4ce3904ae78ce59b6e48 : FAILURE in 18m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b167626589f41bfaf459b61362d84b9 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3afb5176dc3a44b18fe94c9924ef5ece : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/838d4f0310624764a13dddea45d1eda9 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/345059e08a2448d1991fa6e820954d68 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5022f1cb39e4958b3abc17bbdfb0176 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5587fc1206db43488fc770acc117cb8e : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631d535fa93e44c3a5d80be266703d4f : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65ffb86d738742478c078ed5b1d7fcd3 : SUCCESS in 3m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5143732645404490a4592b1904862519 : SUCCESS in 19m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b1582af126ab45919e0954c402f0b72c : SUCCESS in 1h 09m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e117c207abc246abbc052df2ed2aa5f4 : FAILURE in 18m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8e0f63eb9e3a4cc39517d9ad431d9215 : SUCCESS in 20m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9a5ab9e8bc6a4569a61e84a0440d9697 : SUCCESS in 50m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ebfa9fdc0d2340eaa48d07838aec29ba : FAILURE in 17m 36s","accounts_in_message":[],"_revision_number":20},{"id":"3c22a8272d2cde9fa0b2729aac964a7aafa608c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 01:13:54.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afcc3940c1ad419ab654c96165fd895f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1f128199529b451faee592b4a3450b66 : SUCCESS in 1h 29m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/455f94495e54406dad406a02b1e53a34 : SUCCESS in 1h 35m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6beb0fc7f44e416483e35e187771abc1 : SUCCESS in 2h 34m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccfccf1cd9784714972871acc44b8499 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0404659bfa04507a29ce4be5189a16e : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f8f2f62d7a4f45bea39e2e84c7f975a0 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3d7c8e612eb14e5cbc472f8778a87242 : SUCCESS in 4m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/176ce29b0b8c40a6ac7ff929d9938b8e : SUCCESS in 1h 13m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4d4cd4f017f14094b6ebfcc7be29c164 : SUCCESS in 1h 30m 13s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"77d130f16f326177d8f0d11ea6c56b8b7e221d5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-14 10:33:07.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"7528bf96e403de3ea88c444fed11a5e32bc59778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 11:49:27.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de82c435d6984616aac33ff86e360fec\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/721c53490f14483892680cabf7f3daa1 : SUCCESS in 19m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/df2f4be53f1c4d589cd62b2e4faf9570 : SUCCESS in 1h 08m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c9c319dd43494d0b8d5221e629ac335c : SUCCESS in 20m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d87af8746c21437ca98919d15ec1f916 : SUCCESS in 1h 09m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5e65e3d067b6426299255745f307fa1d : FAILURE in 27m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/908e9936e47f432e81642f437097c3f6 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d676f7856aa4a8991cc6f1a432ef899 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9701b72ff2a745fbbbc2be63ef18074c : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ccf7439c9354dfa8c82630dc8b8d82b : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/688ec83af3614774ac0185bf9759e53b : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e80a2c197afc441099c37d82265f6580 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593144d4785a436ba05f065c3342a3c2 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89554e2ea74748c4a6222f7324434010 : SUCCESS in 3m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/38170535dc924986a8a02cc86ed74126 : SUCCESS in 17m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0708190fc5054b26a3275348c546c4bd : SUCCESS in 58m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f683cab26132459ebc97bcb9119bee2c : FAILURE in 16m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cd59847683f240798f1fd2a6ca830879 : SUCCESS in 21m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a3fbabd7a532473890e4218b573c9772 : SUCCESS in 50m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3fd8314347a646f1a9a51e2c04532802 : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":21},{"id":"bbba68fb618a2e08fb76017c152e3e87b257c46f","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2023-09-14 12:00:06.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"85ab1a5db70f12fc41441fb81160b5b8a7dcc75b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-14 12:45:56.000000000","message":"Uploaded patch set 22: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"aab0e635a12a8a8d46caf1693bb0ec666804c0f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 14:20:17.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa25cfbbb3664c4cbbb26eb8a8116ba2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6b2622ad470143c899dc80801cf37fee : SUCCESS in 17m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e8bcf83f9d80412da6c7be047e9b80fb : SUCCESS in 52m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bed0f7cb9b484d1688134017ae0cc502 : SUCCESS in 21m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bccf10a74faf4e5493af21ed241f84c9 : SUCCESS in 1h 12m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/86724e5efcfa458a836e461e21e8a0b3 : SUCCESS in 1h 29m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed9a36a64f74c4297265a21fdeec7aa : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6892ceb5cb654339863fb9d4a485f8e8 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adef5ff0b64d4563b46750a8fa5d67b0 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22f3a8bc59654dd5837765ffa6206a52 : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e399e01429f24122bbfc65f3fdff7ac1 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f178fcf121f43cbbde5d99f70593270 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5fcfd5cb28642669eb225e58b115204 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7ab9861922043699be82f070741e24d : SUCCESS in 3m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/86b15cd1d7e54bc4bb270022296ad195 : SUCCESS in 18m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2dd9d2f8a6fa402faf70db414f255d1a : SUCCESS in 1h 07m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/354525e9511240f090c1a88c178ffa7d : SUCCESS in 1h 24m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d3ab8025b9a842988f3d69ac626df675 : SUCCESS in 24m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8980ab3d5e8e4446b34cde55e3660286 : SUCCESS in 54m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4bb3ee958ddd46daba0a4071c647bb57 : SUCCESS in 1h 06m 45s","accounts_in_message":[],"_revision_number":22},{"id":"8e5bba6e69ab14637cb15df7b48bd40fc410586a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 16:32:00.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c590b530e04249cd94b4b01a13c8533f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cab03983d15d437cbeef701758eb1b95 : SUCCESS in 1h 16m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7936698d398043b8aac36d5d6228ec24 : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/26b5c36ef9494e4bb8820c65853482a8 : SUCCESS in 2h 09m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e1ca93e2e254fb78ed7145fc0476828 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00de819426da4a72bd95d69c1de721a6 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7438aa624d76481eba8d4443dac6e67f : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4530e96575514b1c85d0e403a06686bb : SUCCESS in 6m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6712c88e6d6c474ebf867aee59e6e5fc : SUCCESS in 1h 11m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5fa0586009e4479985a4f6cc712f1df8 : SUCCESS in 1h 21m 45s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"6a82510fa920707f248c80069a88d552ae6563a8","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-14 17:15:39.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"28acd11608449c75689f0089b09624648b307b4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-14 18:51:00.000000000","message":"Uploaded patch set 23.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"d727fa13b1e32f199f43bfbbf9b6db10edd0a248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-14 18:52:48.000000000","message":"Uploaded patch set 24.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":24},{"id":"60cf5e3f1f76652660b068e0764a4e102063b662","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-14 18:52:52.000000000","message":"Patch Set 23: Review-Priority+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":23},{"id":"5c7c46622fd87f6bbcc934b23db16c8edb93cabb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 21:16:13.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19f7659c94e74dd1b339dbc4004cf61e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6dafadb8803d4de08938b07ec6ba9015 : SUCCESS in 1h 22m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9b012243f7a94387ac85682107bad7ed : SUCCESS in 1h 21m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e4bf80e27cc544c7b94985a3fc44c011 : SUCCESS in 2h 18m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0563d9af2f79476ab3c3b14e06a74921 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bcbae99ce57440498e405e01f286e8d1 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7bd5c55342e346329edd4c7bea1b8b41 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/82113430148e4e9ebf542ba4178af821 : SUCCESS in 5m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f13ffb67f7c04c768ae9d7b9eaacaccb : SUCCESS in 1h 14m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bc4e758b4440449f9463bd04cb7a0666 : SUCCESS in 1h 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"85f722c5430a471f2df5f3898299d57684999ba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 22:01:58.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75392cf73b0442c4a5ebbd4a8c1f09c4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b5bf8166f212471aa3f289b2fb4bb8c1 : SUCCESS in 15m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2061b14c6f794e9a8516c8fe92f0bcaf : SUCCESS in 50m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/db7fcbf3681f402c88b784024e69d63d : SUCCESS in 14m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/72f5ec8ca655472bb27c55a70b6570c9 : SUCCESS in 1h 02m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7befbe5d58fc4dd1927c6a8dd216d014 : SUCCESS in 1h 02m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16893456867749b9954b20491bc70bea : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a4874787eec4152a8f11ddb99e8edb7 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5612b74d8c704c3a990630a629ca6fd2 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c66252e726bc4f1b86352dff3f47801d : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5479e565ae1438397f524778c523c71 : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22a9a368f8b348e98459e4b06023aa21 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90eb3808ef254cb98758dbd270186ba1 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16340b2959834efda11e8f239d845bf2 : SUCCESS in 3m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/63da01bdb69a49979bf146dd4e5b512b : SUCCESS in 18m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/deca09458cd74171bed0ab111994de6a : SUCCESS in 1h 08m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9b0db5ef054c4f77b424c74ab885490c : SUCCESS in 1h 26m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dbb8627120c540418b238979836bb5f3 : TIMED_OUT in 3h 03m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1eac8c59e2c4442eb600807a76d4b647 : SUCCESS in 1h 03m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/03f1316862ea42ea8cfad4f19c0014b8 : SUCCESS in 1h 17m 32s","accounts_in_message":[],"_revision_number":24},{"id":"e7931dc20234e62c602c70c684836695689203aa","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-15 09:03:23.000000000","message":"Patch Set 24: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"046b0ad9ba41e92facae820df22dabb08878b952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 10:20:18.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfae6ab9cb4745ff8ab6913dfd5dfceb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/83818caffc6647aeac83503a2e759819 : SUCCESS in 20m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4188dd37dbf541b1812505705ae0d092 : SUCCESS in 1h 06m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/19978992ca1b417bb9127305e1e3dc5f : SUCCESS in 21m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f4b1ea1e1f5642f0b3a743c0aed54581 : SUCCESS in 1h 04m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d1bc3003d01c44109f132a962a23f7a1 : SUCCESS in 1h 05m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e094cd05f186452aa7bf3bc513ab576f : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd9ef94bdc914b5183d3f0fcae5c7880 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fde4549b4af045e4931ad123280eeccd : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd074a8730c34e0fad84c97bc184e70e : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa628dde5f3d480e946d56ac24af1a1e : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/389d538a707f445d8a9eb63b11b1ae5a : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e4fb8e8dac64acdbbaf17b1bd4d2bc0 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/262d9373d9594c3f9fedb72defa253fa : SUCCESS in 5m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/931b6b733b9b42348f6856f77feded4a : SUCCESS in 17m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ed57707847ce42999761f39d8e84c9c0 : SUCCESS in 59m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/03fd5ca74318442ca3c08828996b0c15 : SUCCESS in 1h 10m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/475c8e7fa3574c89a9cdd647d52c2cb8 : SUCCESS in 18m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/89c2a3102fb44fca8c8e43e5ef99e275 : SUCCESS in 54m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ea17fcc4aa57428c92e278097fb56c5f : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":24},{"id":"3f0d6a2fdd6895629f1c0fe6c9e72513a95f1ee5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 12:32:10.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79238259c64649e7b3839a10dbf50c89\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/43fa8e6336b44facb802b6124ffefe60 : SUCCESS in 1h 09m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/339a039c9bdf40f89ccff0ac9c8e2bfe : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1683c3654bc54dd790e0164e8457f920 : TIMED_OUT in 3h 02m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2740be56d634748b4d03dcf59b6fa4a : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6aa03dce89fd49399ab67f893e13010c : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c1ab7eae32b432f86e3cbc86128ba01 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/78bd3d47076b41b4bc42a6bbc269cf95 : SUCCESS in 4m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b4a5a51a272d432c8352d8415607a0d0 : SUCCESS in 1h 10m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/894a3bc9c926497ca240d4bbfba22bd8 : SUCCESS in 1h 28m 41s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"4652e7c1504c98f862387944422245ec629800de","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-18 07:30:04.000000000","message":"Patch Set 24:\n\n(3 comments)","accounts_in_message":[],"_revision_number":24},{"id":"a58da9a7e63aac8cfa24e7f477c3d34ba2a118cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 07:15:13.000000000","message":"Uploaded patch set 25.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"25edb2a93c8ec7ac5658e8246aed703d1d963ef6","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 07:16:08.000000000","message":"Patch Set 24: Review-Priority+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":24},{"id":"d317d9237a664a435c03a4c8199e8e9decab8cb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 07:17:38.000000000","message":"Uploaded patch set 26.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":26},{"id":"76741928ad2aef46925d8b3e38369381e59f80bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 07:34:52.000000000","message":"Uploaded patch set 27.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":27},{"id":"7e885d0070fd85b4213aa093cea2d416ad6ba428","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 07:42:02.000000000","message":"Uploaded patch set 28.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":28},{"id":"b369adddb4351295fa6dfbe196d142b6207e2440","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-19 07:54:40.000000000","message":"Patch Set 28: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":28},{"id":"ce6d23e2b83ad82c1617702a1b75950cc0a9c939","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 07:57:10.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44321bbb79004767acb4faa2de109fa0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c5aad976333849baacb6cf1c611a7727 : FAILURE in 5m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/353dc5c5f15940b7a43e14ca9da879af : FAILURE in 4m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/24f3fcd4b81c496894e324b529c80168 : FAILURE in 12m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/425667bf70d04a78b5c7bd3657cedbd4 : FAILURE in 3m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66e6707633c94af3bb052ba00a493a3c : FAILURE in 3m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7ad318fc2e42476c8ab64ee5fe6d2377 : FAILURE in 3m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/67706e55180448dca4d3d608fb099c48 : FAILURE in 4m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/52b15c215f714410a47792418896eb26 : FAILURE in 5m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a1ccaa40c23a46578e8118cb0d4eb76a : FAILURE in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"3498ff6259f7f2efc737c07a172fa8a640ca7f45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 08:00:46.000000000","message":"Patch Set 28: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae8e1af866f94260b8df9cf3447f4eb1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a7d7d3015d534b1aa7e79b4ba143e4b8 : FAILURE in 3m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c05063f68b6d414fbfd16ed090626682 : FAILURE in 8m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5553ca4281684ac0bdaa6d51d3a0273f : FAILURE in 2m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1c908b9ad8d24b299235f969cfb10e2e : FAILURE in 9m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/552bbff49d8747b383eb0f7d580e0b99 : FAILURE in 13m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef0e920e1d74e1b92e97a899e57d9f7 : FAILURE in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a916646f5f16460f8f30918ed4e048c2 : FAILURE in 2m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ac495be7ea2413b95934fb3d92c5875 : FAILURE in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7ac6036263c46dbb0b5a10da8c14798 : FAILURE in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c82fd7c6093a48a4906e6442f0e6269a : FAILURE in 2m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20aa0f0b87e341e98421ea547bd0fdf4 : FAILURE in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/348d338f82044a088d531b7825edbbec : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a07fd60eea414c3f9ba2f946e555d1e4 : SUCCESS in 3m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/511d1e1b82c2474ebc6c51398f26c796 : FAILURE in 3m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/be39ef2c8f5946c7a3702d7ec9dd32e0 : FAILURE in 7m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ad3fcc922cd44ef5b67b248cffcb8c38 : FAILURE in 11m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9ccb9895154742e39a73330da508b973 : FAILURE in 2m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4b5dc3d7863a4e9ca4abc4dd12c79dd7 : FAILURE in 6m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/222bfd089f3a469ca73cf015a4f74b96 : FAILURE in 7m 22s","accounts_in_message":[],"_revision_number":28},{"id":"ba690c80cb121a21b2c94d083dee18a5a751cb84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 08:05:20.000000000","message":"Uploaded patch set 29.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"db141024b1a4018a0aac74d929f09f29de7d2ae8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 08:20:43.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/961c22f4ac4a49ffb7df661687fc09d6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/061ece80aa7f49cea49d120ae73cd176 : FAILURE in 6m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3c3c7a59201c4356ada5717ef3ae8f7e : FAILURE in 4m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/559a69cb21b84539860d7df96d0ab055 : FAILURE in 13m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6bfb1b1d3004be5bf4d5de422ce8089 : FAILURE in 3m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c318644e94f401690ad91d7d09c2ce3 : FAILURE in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5236f4545a3b4d4e9a3f59cca2de8d62 : FAILURE in 4m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4c646118e2f540a18edfccf6a9601bb6 : FAILURE in 4m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/39a0f3180e1147fdbe7265e3b19341e5 : FAILURE in 5m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e672169a948046c2873abf924c54e762 : FAILURE in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"72fc7c233d5dbcdf64eb127a2b0aa1e41d463702","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 08:21:53.000000000","message":"Uploaded patch set 30.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":30},{"id":"891c0f3deaec18ef67d4c7e894ded17ee240784f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-19 08:31:44.000000000","message":"Patch Set 30: Code-Review+1","accounts_in_message":[],"_revision_number":30},{"id":"53cce0ebf3df4790b7e6326b14dc345f49ef9797","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-19 09:13:42.000000000","message":"Patch Set 27:\n\n(5 comments)","accounts_in_message":[],"_revision_number":27},{"id":"c048b53b4f3282c799b4b5254ffea9fba88da00f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 10:00:19.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2a6f9a37c6148f79ddc6f7ab57fca38\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5f7df3e8677d40cea7aecb868bfa6606 : SUCCESS in 16m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fb1ebe58a567483995fc4af346105caf : SUCCESS in 51m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/29935e75d37e4cffa628e57c1b39401e : SUCCESS in 22m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e0e80224c98d41b9aa7cdc1736ef7f84 : SUCCESS in 50m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0315277d8ac94a5086f94b602ec8b1bc : SUCCESS in 1h 31m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efac76130df747b994c7e2b0bb7f9afb : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33aaaff2f5cb48218e0eb723ff68415c : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b218bc06a734af79421602bcf590d9e : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b22adbd53144b00b3cb70b7db20bd4c : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4f9ef400a5a4c91aed94ad5f4155b94 : SUCCESS in 2m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f093b52c36ad4628b08fe21b4550deaa : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/178b63f90b2841c5a58cc5f1e1518baf : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9efcda11d64542079a3373af15105214 : SUCCESS in 3m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5425ff57a11442ef9adf561af7b15a9c : SUCCESS in 21m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/04b1be6524a64661800a479355accdc3 : SUCCESS in 1h 05m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/697c2a5823b5407f9576713b58dfd303 : SUCCESS in 1h 08m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a04c16b6daf64848ac858066d1919ee6 : SUCCESS in 15m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d56d3d70ee9840e884507cf4f4dbb9ff : SUCCESS in 55m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c3b88ee71784418e92c9d2c89fa18cbf : SUCCESS in 1h 12m 06s","accounts_in_message":[],"_revision_number":30},{"id":"d99b21735ef45f4bab68bf2fcac2765ce39f8406","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 10:47:58.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff58507fe2ac4cf199f1a26a102afba1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3138e9359ec0495396a188d21d8c591a : SUCCESS in 1h 06m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6d88378d3aa040b8b659509a90196e46 : SUCCESS in 1h 16m 04s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/09978a97e48241a280ef6d4f0c52eb72 : SUCCESS in 2h 24m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fa193c0469145b98c0623c45d217daa : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1e617c4f1994ae2bedc530601889ef0 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ab40f3943b694ee3b81b0fe0c4f0dcc9 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/94fd9aa333a84b839a2a806d6a390926 : SUCCESS in 4m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/39763e6bf187420c8f1a04465ea4ba16 : SUCCESS in 1h 16m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3d402a18860c484789f7728cb05566f8 : SUCCESS in 1h 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"8260f69da723a789b36e569b0e5da46426175cd2","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-19 13:50:12.000000000","message":"Patch Set 30: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"7d4e167932a6ff72d86ff5c303e5e19608e72bb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 14:08:40.000000000","message":"Uploaded patch set 31.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"f3ea387b3c6569e60b221bf1b7a8165be6e35115","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 14:10:18.000000000","message":"Patch Set 30: Review-Priority+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":30},{"id":"d3b4f2292309cb8856f43dbfcc0d9e6d0fa394af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 14:10:27.000000000","message":"Uploaded patch set 32.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":32},{"id":"d69a4b8681ebc1139d097ea22785db47e3e5654b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 14:38:10.000000000","message":"Patch Set 32: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"b0f49f2286045d4591fb8c6364f06b1b2de3f706","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-19 14:43:25.000000000","message":"Patch Set 32: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"91253a1ce2ca873788f7f4a7b907f1b562a82af2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 15:13:05.000000000","message":"Uploaded patch set 33.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":33},{"id":"2c2f371fd07aa008f14590db8fd9c4ec272cf5dc","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 15:13:33.000000000","message":"Patch Set 33: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"405677acc5eb9941b0d9efb7bf95851b5420f349","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 15:14:35.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"a46b01a5cd5419741c523da67808dd527b4ea6f2","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-19 15:58:13.000000000","message":"Patch Set 33: Code-Review+1","accounts_in_message":[],"_revision_number":33},{"id":"505bbed9f5bdcf68a83df903be719ea16b65248f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-09-19 16:05:20.000000000","message":"Patch Set 33: Code-Review+1","accounts_in_message":[],"_revision_number":33},{"id":"5ce9c851363011887b2cd17e6edff67d934c75c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 16:50:56.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f45b3538eda942c48ee30d1ab36f14e1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/febc9c69545a4de584d2c252fa0d5e3b : SUCCESS in 20m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5849075a49d5434eaad19ed8c245797f : SUCCESS in 1h 04m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2de99b711a2f4291a5092d67291dd64e : SUCCESS in 23m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e045c1d72f834933bca31a39505e1f33 : SUCCESS in 58m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/aaa93e4889f744159cb7cb335058efd2 : SUCCESS in 1h 25m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/544599b4563543579b7cc456eaf5bd38 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cd9eadf1c5e4513a4ae8141524ae856 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e11486574bb84692871d65b3765a8ae6 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/950a9d366a2f48d3bd6adee7af3cae28 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/115991aa035d45c3bfd568346bef0f83 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f57953ba8e94ba5b81824887b503d46 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f09dd4d4f6f4bb9b89c31cc94b3c747 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55474a7d457f4458b4e774076be98f3a : SUCCESS in 4m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5bc236480dd74b86a96a90f706ccef63 : SUCCESS in 16m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b772811c24d444f58697ba19bff96386 : SUCCESS in 58m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c970af313a5b4a0698f28fb606b0f82b : SUCCESS in 1h 32m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a76371a7704e4bbeb09235d5a2a53ca3 : SUCCESS in 25m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ada9dc88c5c74f41b02e67ad241ba941 : SUCCESS in 49m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ab5c5feac11842579b1263c0258d32ce : SUCCESS in 1h 17m 22s","accounts_in_message":[],"_revision_number":33},{"id":"8d1527456083471aff903fe858c1564e5e070830","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 17:43:10.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2ab73f6b37240f08b92917058a8b093\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/81b42ca1b49743f983b90f8bca2f98f4 : SUCCESS in 1h 16m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/32d3b3bb4f584464ba47776d6cfc09e7 : SUCCESS in 1h 29m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0507e8b066aa472283482b1ab8f92287 : SUCCESS in 2h 27m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d47ebbeeb2342eaa0b6ee704951f74b : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb2fc0e2138848f6921692fd1a9e70cd : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d151f1782ef4f329f9b45b89d2c4a49 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/42ab09763a464c398de3824439700052 : SUCCESS in 5m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/db6de4864253484b9a4712a6d036fb62 : SUCCESS in 1h 22m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5f5de1003920422d8e0acd8b1846ca51 : SUCCESS in 1h 17m 59s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"359e1b4ccd48bf3a6d5de2be7a5d752917a28b34","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-19 19:28:29.000000000","message":"Patch Set 33: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"376b1da417a0d0001ecd54149035c1060be22384","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-19 23:14:45.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"cb9e76c85aa9305bc2f39314d53ea91c0daa22d0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-20 13:20:38.000000000","message":"Patch Set 33: Review-Priority+1","accounts_in_message":[],"_revision_number":33},{"id":"cbcf4478960a8c869cd9b8c3b446c2f939fc6c6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-21 09:01:39.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":34},{"id":"7db14c9a73908b4adede679e1e915d9051fb4363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 10:34:41.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c265dd751f244f0e9bb1f05205e21317\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1f7bacef48eb48bc88dd2709665daf6e : SUCCESS in 20m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b0dab64e77474a31b9fb5a71955741d4 : SUCCESS in 1h 08m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4ffcec829d74437da99a0a91bd86b58d : SUCCESS in 20m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6a71da1c1a974779ab03436711f10924 : SUCCESS in 58m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ee39ee7093864365b19f1909c6f5e042 : SUCCESS in 1h 32m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be32478dc7d4b5fb8ff606397c89ce6 : SUCCESS in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8a2872067f84416971439bff8a28e7f : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/675fbcbc3cfe41a0a1de2fae34a36850 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbd40b26f26f42e6b79c2884905867d5 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/096e68eb11924998a3f9d61a099e764b : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4455047403b46f2999f1db07c13f4dc : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d46197169c94878a497c7889aeec400 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a067d589d1d4733b1e91fdb5c030436 : SUCCESS in 3m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/869a5ee74da64b458afd076f255d3f5c : SUCCESS in 18m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b220b51d2fc4486ca6840a4a35c78f99 : SUCCESS in 1h 04m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/674b0427404944bfb31484da550fcec7 : SUCCESS in 1h 05m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9f3da96fbc894a628379d80e044ffb2d : SUCCESS in 20m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4fe3759da87a45abb6e9d193497959af : SUCCESS in 53m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/56f6827f02e144e8a3ff2a3ded4456a6 : SUCCESS in 1h 18m 39s","accounts_in_message":[],"_revision_number":34},{"id":"e2dfb44d56af34f5713c0058b156af95df15af5f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-21 10:40:37.000000000","message":"Patch Set 34: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"403c427d24d5992441cb2c12f3dd34e64916f026","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-21 10:46:49.000000000","message":"Patch Set 34: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":34},{"id":"6868904eef2abe07ff29a6056ea2ab7bbdd80ba6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 11:37:30.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa0463e85924d79b720334e6ecc428a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4bf59101037f4a24923e552f7a09f270 : SUCCESS in 1h 45m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4d4bbd678cdc46ce83ca6bbed0b19ed0 : SUCCESS in 1h 56m 15s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f810b82e066749c7a5d23d534cd330b6 : SUCCESS in 2h 33m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/544ba613c12b42b19549a8180de5a395 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/720089d9443b44959e7a1115c49656dd : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1a9ef31bc856461b9985faf79db63a15 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cb80bfd4ae9641d89b96fa4af763ea44 : SUCCESS in 5m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/81fcb26bbc7b4c3fbcc194d7cf496a66 : SUCCESS in 1h 32m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1eb468cb03224dddb9dd83663d1033b3 : SUCCESS in 2h 01m 48s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"168be35160ffa61d1bca3f8954f167229d88758e","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-21 11:53:36.000000000","message":"Patch Set 34: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"04d5abc778b1db2a531b77ea51e1c051c903740a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-09-21 11:55:42.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"09d030a3b1d39607ea1999dc049b287da76971f7","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-09-21 15:27:03.000000000","message":"Patch Set 34: Code-Review+1","accounts_in_message":[],"_revision_number":34},{"id":"d8c894b295b42cd852da9f58aeca16b34a8749dc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-22 08:01:52.000000000","message":"Patch Set 34: Workflow+1","accounts_in_message":[],"_revision_number":34},{"id":"dcaa1558cfb593f260eec76755f90a6d87cc0fea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 08:02:14.000000000","message":"Patch Set 34: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":34},{"id":"aecc5c19435c67668d8b1844bb7a826fcc2cdaff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 09:43:06.000000000","message":"Patch Set 34: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042216ba35db485f82fe03115207529a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0f2560cddec14a718a1f06ec86702c6b : SUCCESS in 20m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/96bc9c55aea84aaba47ac9eaf3f49f2a : SUCCESS in 53m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/451a1e36fca04c8b904dab002468d298 : SUCCESS in 1h 17m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcd076f857d04f5e861e3ad13b461eb7 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1a9917246804ecc87f506fe9250e8bf : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b03c8ca85224461188abf30a6bd4985f : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85973158b6124fa5b1885668bff10762 : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ac55ebb7d6d4322bf59ea22af68be13 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3be261a90e3c4bddbf97514c9eccc542 : SUCCESS in 3m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bcfc838c652b4e1ca7c0466ceb4a30fc : SUCCESS in 21m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e130e5670da84cd28f677fae4de53712 : SUCCESS in 1h 15m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/591df8fa9689476a851d16ca64a145e6 : SUCCESS in 1h 36m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f48cb1ee60eb40b3b249f02f34518ca4 : SUCCESS in 32m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dec49735048340bca692214baa1b72c2 : SUCCESS in 1h 01m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/edfcee2ad8014bc3a6f5551a37810aeb : SUCCESS in 1h 27m 54s","accounts_in_message":[],"_revision_number":34},{"id":"2e89f0f07615590b32fbbb2ca95c9734339f2bf9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 09:43:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":34},{"id":"883a796b652d81251225c91a205faf45879b6d15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 09:44:07.000000000","message":"Patch Set 34:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee34a2c8e47f457b807050846e417d94\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aac7185395c4984881773842b9896e8 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ab57f23bb71482c80d62fe34493d417 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":34}],"current_revision_number":34,"current_revision":"b76bd4c6efa22af43d7895a7b5e95c75806d3893","revisions":{"b893725d38fbe608b8e53287aca83758422c8695":{"kind":"REWORK","_number":1,"created":"2022-01-21 13:17:37.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/91/825791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/1"}}},"commit":{"parents":[{"commit":"8a26497db2c93e16176e109119afce5346192046","subject":"Refactoring of the image module. Adding engine adapter module. Transitioning from docker.APIClient to docker.DockerClient. Preparing for adding podman as container engine for building.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a26497db2c93e16176e109119afce5346192046"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 16:36:59.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-21 13:12:46.000000000","tz":60},"subject":"Adding podman for image building (--engine podman)","message":"Adding podman for image building (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b893725d38fbe608b8e53287aca83758422c8695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b893725d38fbe608b8e53287aca83758422c8695"}]},"branch":"refs/heads/master"},"7f74a7abdd6b16c1dca14c368d5a035ec5ae0fd5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-01 10:06:51.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/2"}}},"commit":{"parents":[{"commit":"e5322a3ce1f4c93042d5126eec02a814ff70f70b","subject":"Refactoring of the image module. Adding engine adapter module. Transitioning from docker.APIClient to docker.DockerClient. Preparing for adding podman as container engine for building.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5322a3ce1f4c93042d5126eec02a814ff70f70b"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 16:36:59.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-02-01 10:06:15.000000000","tz":60},"subject":"Adding podman for image building (--engine podman)","message":"Adding podman for image building (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f74a7abdd6b16c1dca14c368d5a035ec5ae0fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f74a7abdd6b16c1dca14c368d5a035ec5ae0fd5"}]},"branch":"refs/heads/master"},"9a36d4cbddac5e7ad3a137da8b16215548154878":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-04 09:43:07.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/91/825791/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/3"}}},"commit":{"parents":[{"commit":"e5322a3ce1f4c93042d5126eec02a814ff70f70b","subject":"Refactoring of the image module. Adding engine adapter module. Transitioning from docker.APIClient to docker.DockerClient. Preparing for adding podman as container engine for building.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5322a3ce1f4c93042d5126eec02a814ff70f70b"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 16:36:59.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-04 09:43:07.000000000","tz":0},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a36d4cbddac5e7ad3a137da8b16215548154878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a36d4cbddac5e7ad3a137da8b16215548154878"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2b02a4795fa10889e4883d55ea31b4276c567779":{"kind":"REWORK","_number":4,"created":"2022-02-07 09:15:39.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/91/825791/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/4"}}},"commit":{"parents":[{"commit":"32af1b4316522c581dda3a90ac29590cc2ca3890","subject":"Refactor image module and transition to DockerClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32af1b4316522c581dda3a90ac29590cc2ca3890"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 16:36:59.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-07 09:14:48.000000000","tz":60},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b02a4795fa10889e4883d55ea31b4276c567779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b02a4795fa10889e4883d55ea31b4276c567779"}]},"branch":"refs/heads/master"},"b639826ffbeff1c5d145120eb4d92fd3290acd3e":{"kind":"REWORK","_number":5,"created":"2022-02-11 14:21:45.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/91/825791/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/5"}}},"commit":{"parents":[{"commit":"32af1b4316522c581dda3a90ac29590cc2ca3890","subject":"Refactor image module and transition to DockerClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32af1b4316522c581dda3a90ac29590cc2ca3890"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 16:36:59.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-11 14:18:45.000000000","tz":0},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b639826ffbeff1c5d145120eb4d92fd3290acd3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b639826ffbeff1c5d145120eb4d92fd3290acd3e"}]},"branch":"refs/heads/master"},"b4f690a805766557f37665289c0dfef01a8e2237":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-02-11 14:29:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/6"}}},"commit":{"parents":[{"commit":"ad6309c188e23cda1e2dfec4c28ddd60eae5b432","subject":"Refactor image module and transition to DockerClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad6309c188e23cda1e2dfec4c28ddd60eae5b432"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 16:36:59.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-02-11 14:28:55.000000000","tz":60},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4f690a805766557f37665289c0dfef01a8e2237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4f690a805766557f37665289c0dfef01a8e2237"}]},"branch":"refs/heads/master"},"c218c458a9a0a644f595e1b3c48ae0bd69fe92ca":{"kind":"REWORK","_number":7,"created":"2022-02-11 14:46:09.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/91/825791/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/7"}}},"commit":{"parents":[{"commit":"32af1b4316522c581dda3a90ac29590cc2ca3890","subject":"Refactor image module and transition to DockerClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32af1b4316522c581dda3a90ac29590cc2ca3890"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 16:36:59.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-11 14:45:40.000000000","tz":0},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c218c458a9a0a644f595e1b3c48ae0bd69fe92ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c218c458a9a0a644f595e1b3c48ae0bd69fe92ca"}]},"branch":"refs/heads/master"},"7a4a40bf826db02eb7887091bbc27eb1ce64a811":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-02-11 14:48:39.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/91/825791/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/8"}}},"commit":{"parents":[{"commit":"ad6309c188e23cda1e2dfec4c28ddd60eae5b432","subject":"Refactor image module and transition to DockerClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad6309c188e23cda1e2dfec4c28ddd60eae5b432"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 16:36:59.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-11 14:48:39.000000000","tz":0},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a4a40bf826db02eb7887091bbc27eb1ce64a811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a4a40bf826db02eb7887091bbc27eb1ce64a811"}]},"branch":"refs/heads/master","description":"Rebase"},"48064e45186667056b0ba7a86eef5e6d4de9f89c":{"kind":"REWORK","_number":9,"created":"2022-02-14 12:11:01.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/91/825791/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/9"}}},"commit":{"parents":[{"commit":"ad6309c188e23cda1e2dfec4c28ddd60eae5b432","subject":"Refactor image module and transition to DockerClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad6309c188e23cda1e2dfec4c28ddd60eae5b432"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 16:36:59.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-14 12:10:27.000000000","tz":0},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48064e45186667056b0ba7a86eef5e6d4de9f89c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48064e45186667056b0ba7a86eef5e6d4de9f89c"}]},"branch":"refs/heads/master"},"6a9710730e13240366d2b550cc43f37079f6de02":{"kind":"REWORK","_number":10,"created":"2022-02-14 13:04:17.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/91/825791/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/10"}}},"commit":{"parents":[{"commit":"ad6309c188e23cda1e2dfec4c28ddd60eae5b432","subject":"Refactor image module and transition to DockerClient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad6309c188e23cda1e2dfec4c28ddd60eae5b432"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 16:36:59.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-14 12:36:40.000000000","tz":0},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a9710730e13240366d2b550cc43f37079f6de02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a9710730e13240366d2b550cc43f37079f6de02"}]},"branch":"refs/heads/master"},"5a3093f0dc8ac6b0cf39f0807f35720320b00fa1":{"kind":"REWORK","_number":11,"created":"2023-09-12 00:21:59.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/11"}}},"commit":{"parents":[{"commit":"e183e9a21e615476262c4380f7c8e6ed17a80a3b","subject":"Merge \"build: distutils.StrictVersion() is deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e183e9a21e615476262c4380f7c8e6ed17a80a3b"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-01-19 16:36:59.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:21:49.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a3093f0dc8ac6b0cf39f0807f35720320b00fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a3093f0dc8ac6b0cf39f0807f35720320b00fa1"}]},"branch":"refs/heads/master"},"992637c0c05b79b75bfa6d084a804f8894ca2b53":{"kind":"NO_CHANGE","_number":12,"created":"2023-09-12 00:28:45.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/12"}}},"commit":{"parents":[{"commit":"e183e9a21e615476262c4380f7c8e6ed17a80a3b","subject":"Merge \"build: distutils.StrictVersion() is deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e183e9a21e615476262c4380f7c8e6ed17a80a3b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:43.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/992637c0c05b79b75bfa6d084a804f8894ca2b53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/992637c0c05b79b75bfa6d084a804f8894ca2b53"}]},"branch":"refs/heads/master"},"deef02843a9f10d830d9a97c575b80e31d47f706":{"kind":"REWORK","_number":13,"created":"2023-09-12 00:31:51.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/13"}}},"commit":{"parents":[{"commit":"e183e9a21e615476262c4380f7c8e6ed17a80a3b","subject":"Merge \"build: distutils.StrictVersion() is deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e183e9a21e615476262c4380f7c8e6ed17a80a3b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:31:41.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/deef02843a9f10d830d9a97c575b80e31d47f706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/deef02843a9f10d830d9a97c575b80e31d47f706"}]},"branch":"refs/heads/master"},"733289b8830e793abd1daa75005f9e6aad3353b6":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2023-09-12 01:38:50.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/14"}}},"commit":{"parents":[{"commit":"e183e9a21e615476262c4380f7c8e6ed17a80a3b","subject":"Merge \"build: distutils.StrictVersion() is deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e183e9a21e615476262c4380f7c8e6ed17a80a3b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 01:37:41.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\nDepends-On: Ie44bbdea1dd5af2219f9359c1b0f428a84049bdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/733289b8830e793abd1daa75005f9e6aad3353b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/733289b8830e793abd1daa75005f9e6aad3353b6"}]},"branch":"refs/heads/master"},"f3409d30e1a555cd18c633f7deb1d56978366cac":{"kind":"REWORK","_number":15,"created":"2023-09-12 10:06:19.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/15"}}},"commit":{"parents":[{"commit":"e183e9a21e615476262c4380f7c8e6ed17a80a3b","subject":"Merge \"build: distutils.StrictVersion() is deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e183e9a21e615476262c4380f7c8e6ed17a80a3b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 10:02:21.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\nDepends-On: Ie44bbdea1dd5af2219f9359c1b0f428a84049bdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3409d30e1a555cd18c633f7deb1d56978366cac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3409d30e1a555cd18c633f7deb1d56978366cac"}]},"branch":"refs/heads/master"},"b823dd938cd662f9e57d9bae42a0369c7b30cc64":{"kind":"REWORK","_number":16,"created":"2023-09-12 17:49:46.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/16"}}},"commit":{"parents":[{"commit":"e183e9a21e615476262c4380f7c8e6ed17a80a3b","subject":"Merge \"build: distutils.StrictVersion() is deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e183e9a21e615476262c4380f7c8e6ed17a80a3b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 17:49:25.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\nDepends-On: Ie44bbdea1dd5af2219f9359c1b0f428a84049bdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b823dd938cd662f9e57d9bae42a0369c7b30cc64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b823dd938cd662f9e57d9bae42a0369c7b30cc64"}]},"branch":"refs/heads/master"},"8c7932bef4789537d9508baab95e4ec806ac0bf6":{"kind":"REWORK","_number":17,"created":"2023-09-12 17:56:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/17"}}},"commit":{"parents":[{"commit":"e183e9a21e615476262c4380f7c8e6ed17a80a3b","subject":"Merge \"build: distutils.StrictVersion() is deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e183e9a21e615476262c4380f7c8e6ed17a80a3b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 17:56:39.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\nDepends-On: Ie44bbdea1dd5af2219f9359c1b0f428a84049bdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c7932bef4789537d9508baab95e4ec806ac0bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c7932bef4789537d9508baab95e4ec806ac0bf6"}]},"branch":"refs/heads/master"},"33d4ebf2369af8445227523d79dbe65003b83a14":{"kind":"REWORK","_number":18,"created":"2023-09-12 18:52:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/18"}}},"commit":{"parents":[{"commit":"e183e9a21e615476262c4380f7c8e6ed17a80a3b","subject":"Merge \"build: distutils.StrictVersion() is deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e183e9a21e615476262c4380f7c8e6ed17a80a3b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 18:51:52.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\nDepends-On: Ie44bbdea1dd5af2219f9359c1b0f428a84049bdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33d4ebf2369af8445227523d79dbe65003b83a14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33d4ebf2369af8445227523d79dbe65003b83a14"}]},"branch":"refs/heads/master"},"d4f4f9d210fae4e1626bdbb6aac4fa6d9f61e456":{"kind":"REWORK","_number":19,"created":"2023-09-12 19:43:08.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/19"}}},"commit":{"parents":[{"commit":"e183e9a21e615476262c4380f7c8e6ed17a80a3b","subject":"Merge \"build: distutils.StrictVersion() is deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e183e9a21e615476262c4380f7c8e6ed17a80a3b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 19:43:00.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\nDepends-On: Ie44bbdea1dd5af2219f9359c1b0f428a84049bdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4f4f9d210fae4e1626bdbb6aac4fa6d9f61e456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4f4f9d210fae4e1626bdbb6aac4fa6d9f61e456"}]},"branch":"refs/heads/master"},"71f6e17676184f2cc187955fce39d84a94dc43c6":{"kind":"REWORK","_number":20,"created":"2023-09-12 20:27:50.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/20"}}},"commit":{"parents":[{"commit":"e183e9a21e615476262c4380f7c8e6ed17a80a3b","subject":"Merge \"build: distutils.StrictVersion() is deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e183e9a21e615476262c4380f7c8e6ed17a80a3b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 20:27:10.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\nDepends-On: Ie44bbdea1dd5af2219f9359c1b0f428a84049bdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71f6e17676184f2cc187955fce39d84a94dc43c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71f6e17676184f2cc187955fce39d84a94dc43c6"}]},"branch":"refs/heads/master"},"222fd6f291c3f58e904ee9a25a7817779d37bd66":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2023-09-14 10:33:07.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/21"}}},"commit":{"parents":[{"commit":"299f1d6bdd94018532ec27fad80830e8a8f1d4eb","subject":"Fix stderr logging when image build fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/299f1d6bdd94018532ec27fad80830e8a8f1d4eb"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-14 10:32:56.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\nDepends-On: Ie44bbdea1dd5af2219f9359c1b0f428a84049bdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/222fd6f291c3f58e904ee9a25a7817779d37bd66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/222fd6f291c3f58e904ee9a25a7817779d37bd66"}]},"branch":"refs/heads/master"},"f4866c0235b286583729b3bed4e1123271aacea5":{"kind":"NO_CODE_CHANGE","_number":22,"created":"2023-09-14 12:45:56.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/22"}}},"commit":{"parents":[{"commit":"299f1d6bdd94018532ec27fad80830e8a8f1d4eb","subject":"Fix stderr logging when image build fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/299f1d6bdd94018532ec27fad80830e8a8f1d4eb"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-14 12:44:26.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4866c0235b286583729b3bed4e1123271aacea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4866c0235b286583729b3bed4e1123271aacea5"}]},"branch":"refs/heads/master"},"06bd7b8316b1d4464c64db77bb48110e51b4cb14":{"kind":"REWORK","_number":23,"created":"2023-09-14 18:51:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/23"}}},"commit":{"parents":[{"commit":"299f1d6bdd94018532ec27fad80830e8a8f1d4eb","subject":"Fix stderr logging when image build fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/299f1d6bdd94018532ec27fad80830e8a8f1d4eb"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-14 18:50:50.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06bd7b8316b1d4464c64db77bb48110e51b4cb14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06bd7b8316b1d4464c64db77bb48110e51b4cb14"}]},"branch":"refs/heads/master"},"38cbbda033ce7660b81ce073d80602cd179d9435":{"kind":"REWORK","_number":24,"created":"2023-09-14 18:52:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/24"}}},"commit":{"parents":[{"commit":"299f1d6bdd94018532ec27fad80830e8a8f1d4eb","subject":"Fix stderr logging when image build fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/299f1d6bdd94018532ec27fad80830e8a8f1d4eb"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-14 18:52:00.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38cbbda033ce7660b81ce073d80602cd179d9435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38cbbda033ce7660b81ce073d80602cd179d9435"}]},"branch":"refs/heads/master"},"cb307d3a16a111e04624765f072c05f3c87428f9":{"kind":"REWORK","_number":25,"created":"2023-09-19 07:15:13.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/25"}}},"commit":{"parents":[{"commit":"299f1d6bdd94018532ec27fad80830e8a8f1d4eb","subject":"Fix stderr logging when image build fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/299f1d6bdd94018532ec27fad80830e8a8f1d4eb"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-19 07:15:00.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb307d3a16a111e04624765f072c05f3c87428f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb307d3a16a111e04624765f072c05f3c87428f9"}]},"branch":"refs/heads/master"},"27927c15d2c09eb47369869f832023ac539b91f9":{"kind":"REWORK","_number":26,"created":"2023-09-19 07:17:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/26"}}},"commit":{"parents":[{"commit":"299f1d6bdd94018532ec27fad80830e8a8f1d4eb","subject":"Fix stderr logging when image build fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/299f1d6bdd94018532ec27fad80830e8a8f1d4eb"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-19 07:17:29.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27927c15d2c09eb47369869f832023ac539b91f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27927c15d2c09eb47369869f832023ac539b91f9"}]},"branch":"refs/heads/master"},"c749c55be57d6eca77e4c5ff8738908ed46e22c3":{"kind":"REWORK","_number":27,"created":"2023-09-19 07:34:52.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/27"}}},"commit":{"parents":[{"commit":"3d703b60e6fb1062313b57293696f1c34151c27f","subject":"Merge \"toolbox: Add sleep script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d703b60e6fb1062313b57293696f1c34151c27f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-19 07:34:42.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c749c55be57d6eca77e4c5ff8738908ed46e22c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c749c55be57d6eca77e4c5ff8738908ed46e22c3"}]},"branch":"refs/heads/master"},"65ad0db955466aa7a6855fd9f682289b597d2bd7":{"kind":"REWORK","_number":28,"created":"2023-09-19 07:42:02.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/28"}}},"commit":{"parents":[{"commit":"3d703b60e6fb1062313b57293696f1c34151c27f","subject":"Merge \"toolbox: Add sleep script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d703b60e6fb1062313b57293696f1c34151c27f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-19 07:41:54.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65ad0db955466aa7a6855fd9f682289b597d2bd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65ad0db955466aa7a6855fd9f682289b597d2bd7"}]},"branch":"refs/heads/master"},"66cc3a60958c1ae1100b0f1e11654eaac723749f":{"kind":"REWORK","_number":29,"created":"2023-09-19 08:05:20.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/29"}}},"commit":{"parents":[{"commit":"3d703b60e6fb1062313b57293696f1c34151c27f","subject":"Merge \"toolbox: Add sleep script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d703b60e6fb1062313b57293696f1c34151c27f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-19 08:03:44.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nCo-Authored-By: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66cc3a60958c1ae1100b0f1e11654eaac723749f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66cc3a60958c1ae1100b0f1e11654eaac723749f"}]},"branch":"refs/heads/master"},"2cbc152f70f02961f973d03fa0de84cd69372809":{"kind":"REWORK","_number":30,"created":"2023-09-19 08:21:53.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/30"}}},"commit":{"parents":[{"commit":"3d703b60e6fb1062313b57293696f1c34151c27f","subject":"Merge \"toolbox: Add sleep script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d703b60e6fb1062313b57293696f1c34151c27f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-19 08:21:43.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nCo-Authored-By: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cbc152f70f02961f973d03fa0de84cd69372809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cbc152f70f02961f973d03fa0de84cd69372809"}]},"branch":"refs/heads/master"},"0e881d91275f12d09ae9247c86fd6ad3e6faa7e5":{"kind":"REWORK","_number":31,"created":"2023-09-19 14:08:40.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/31"}}},"commit":{"parents":[{"commit":"3d703b60e6fb1062313b57293696f1c34151c27f","subject":"Merge \"toolbox: Add sleep script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d703b60e6fb1062313b57293696f1c34151c27f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-19 14:08:23.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nCo-Authored-By: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e881d91275f12d09ae9247c86fd6ad3e6faa7e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e881d91275f12d09ae9247c86fd6ad3e6faa7e5"}]},"branch":"refs/heads/master"},"e8f1f414a289b8b39c80a02b8fa00060f0ddce5f":{"kind":"REWORK","_number":32,"created":"2023-09-19 14:10:27.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/32"}}},"commit":{"parents":[{"commit":"3d703b60e6fb1062313b57293696f1c34151c27f","subject":"Merge \"toolbox: Add sleep script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d703b60e6fb1062313b57293696f1c34151c27f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-19 14:09:53.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nCo-Authored-By: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8f1f414a289b8b39c80a02b8fa00060f0ddce5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8f1f414a289b8b39c80a02b8fa00060f0ddce5f"}]},"branch":"refs/heads/master"},"b33153b1c5053bf82f4879a41acd9719c9ff72b7":{"kind":"REWORK","_number":33,"created":"2023-09-19 15:13:05.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/33"}}},"commit":{"parents":[{"commit":"3d703b60e6fb1062313b57293696f1c34151c27f","subject":"Merge \"toolbox: Add sleep script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d703b60e6fb1062313b57293696f1c34151c27f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-19 15:12:42.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nCo-Authored-By: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b33153b1c5053bf82f4879a41acd9719c9ff72b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b33153b1c5053bf82f4879a41acd9719c9ff72b7"}]},"branch":"refs/heads/master"},"b76bd4c6efa22af43d7895a7b5e95c75806d3893":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2023-09-21 09:01:39.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/91/825791/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/825791/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/825791/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/825791/34"}}},"commit":{"parents":[{"commit":"4871ca4d6b83f6d28d790722eae52de5e56becee","subject":"Lower limit for open files when installing fluent gem","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4871ca4d6b83f6d28d790722eae52de5e56becee"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-12 00:28:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-09-21 09:01:21.000000000","tz":120},"subject":"Add podman for image building","message":"Add podman for image building\n\nAdd podman as an option to choose for container engine\nfor kolla-build (--engine podman)\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nCo-Authored-By: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nChange-Id: I068c906df97745e397408d8c3ef6af47ee037638\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b76bd4c6efa22af43d7895a7b5e95c75806d3893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b76bd4c6efa22af43d7895a7b5e95c75806d3893"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
