)]}'
{"id":"openstack%2Fkolla~825948","triplet_id":"openstack%2Fkolla~stable%2Fxena~Ib3ecd73f9e39e320acb2c5f0962b8af9b1a817e9","project":"openstack/kolla","branch":"stable/xena","topic":"apache-storm-stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-25 10:54:01.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-01-25 10:54:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3ecd73f9e39e320acb2c5f0962b8af9b1a817e9","subject":"Mitigate two Log4j vulnerabilities in Apache Storm","status":"MERGED","created":"2022-01-24 09:49:37.000000000","updated":"2022-01-25 10:55:50.000000000","submitted":"2022-01-25 10:54:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825948-apache-storm-stable/xena","cherry_pick_of_change":824651,"cherry_pick_of_patch_set":1,"meta_rev_id":"d73d601a3b1cdb919f32af7841e3c319eb9799d4","_number":825948,"virtual_id_number":825948,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 10:54:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":2,"date":"2022-01-25 09:11:07.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":2,"date":"2022-01-25 08:58:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":1,"date":"2022-01-25 09:11:07.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"date":"2022-01-25 08:58:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 09:49:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-01-24 11:11:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 09:11:07.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"}],"messages":[{"id":"6fc93025c8aa94bed11f710895609463a2361e7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-24 09:49:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6e359b040f0ec97287a0574668393ec197f14fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 11:11:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c3b81d4e82cf4c8b8a032bcd4236f04e : SUCCESS in 32m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3c3d0024c9d14ff69b9fe3aeffcf1d20 : SUCCESS in 40m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a4bde2d943af40d380c6dee401a82c45 : SUCCESS in 1h 12m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/870af1cb3fe74f549e71e6421beedab1 : SUCCESS in 1h 12m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/33cfbf8c93fa4bbe93ee74846096e2e9 : SUCCESS in 1h 17m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/19e18552f9f84cdb91486f3c1d43eb53 : SUCCESS in 27m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/92ae17ebda4e4ad78acd262c2f3e7225 : SUCCESS in 24m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/59e13200b49847aba10d3c1c861c80c4 : SUCCESS in 1h 01m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98dfdf9fffb14dcfbefb1cb1843d2a11 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4f4efb31a1244e69116c5033e20e314 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d2369e7e20048ca8d4b475231a430c0 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39f48a2212054f0b867740e6ab86261a : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc56e8bd5e840c5a62dd4089e54d2fc : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d916a098c3a49ddb56119acdd75af69 : SUCCESS in 4m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/168282cd071d45479d63549c8a1ff79b : SUCCESS in 21m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/853952290dfd40d39dfbdf6d44c921d7 : SUCCESS in 35m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d52372df9fb840309c8b6cde8f823aae : SUCCESS in 54m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e14bc30444147cbbc9434c518876f9d : SUCCESS in 50m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/611db6e601c44a44969fe7d743549f81 : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7d11aace930fe9274c769cb54ec3c10ed114b722","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 13:17:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8d3db6b2cb0146f1a973ebcfd1334321 : TIMED_OUT in 3h 14m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/236259cdc92a40bf844aa855afac0cfa : FAILURE in 49m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/79a187af4bac4538bd95fd177fd95b7b : FAILURE in 54m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1512f6dd72ff4c58ba2a86fdc9636da7 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c138f49cd4340bb8e43b28199a5a49f : SUCCESS in 5m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/333ecc9682804f13b4972e587d75d0a6 : TIMED_OUT in 3h 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c36d167564d6c302971f2fd9223e4dc4dacc2f6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-25 08:56:55.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"54724461a33e83c577d736d60b766e023d87894c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-25 08:58:55.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33eb2babc823d322079c4a522561c0831ee751a5","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2022-01-25 09:11:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80dc1a878e85fcf21ac3e1c428d9aa14c56f774f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 09:11:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a6f4c70d9106e2a9907011c0f893629c048ad25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 10:54:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5e631abc016d4b198b9d84c84aeadaa7 : SUCCESS in 39m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7c5503c3119c4d838569496799442655 : SUCCESS in 1h 17m 30s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0fed5f6a059940d58b89f5b3e66ab745 : SUCCESS in 1h 37m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c9a6e76740f411491d16f9e73e36228 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54b049fe667b4b43816c4ff8824b2d71 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46855bfbc602400984b6faf414f2c02b : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0871d2e1580841289f6174095fe2f97f : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bab7a2a6037140909037d43a4e1f1423 : SUCCESS in 4m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3df2ec53659d45a4997ba5a1497eb6d7 : SUCCESS in 42m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0888abab18064d05bd248397336f7281 : SUCCESS in 58m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cf4250c077694b70b74d07f3730d58ec : SUCCESS in 1h 09m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3a7da642c8be340fe8621015dc9ea8452d462251","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 10:54:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d73d601a3b1cdb919f32af7841e3c319eb9799d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 10:55:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8f1ff97f2d84ce9996655229aa01c67 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abebf1cde6894d8e9796c961f20dfdeb : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdd9506de38e5917a0406cf447238d42dc17e8ca","revisions":{"fdd9506de38e5917a0406cf447238d42dc17e8ca":{"kind":"REWORK","_number":1,"created":"2022-01-24 09:49:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/48/825948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/825948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/825948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/825948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/825948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/825948/1"}}},"commit":{"parents":[{"commit":"c8d3709436c6bc0f9d6bc8c038f3f3fb64dd6cb1","subject":"Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8d3709436c6bc0f9d6bc8c038f3f3fb64dd6cb1"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-01-13 21:35:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-24 09:49:37.000000000","tz":0},"subject":"Mitigate two Log4j vulnerabilities in Apache Storm","message":"Mitigate two Log4j vulnerabilities in Apache Storm\n\nThe Log4j version was bumped on GitHub [1] but it is still pending\ninclusion in a release of Apache Storm.\n\nApply the alternative mitigation recommended by Log4j [2] of removing\nthe JndiLookup class from the classpath.\n\n[1] https://github.com/apache/storm/pull/3427\n[2] https://logging.apache.org/log4j/2.x/security.html\n\nChange-Id: Ib3ecd73f9e39e320acb2c5f0962b8af9b1a817e9\n(cherry picked from commit 448e4f56aa6218a66b0074329d322c0b6089b2e6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fdd9506de38e5917a0406cf447238d42dc17e8ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fdd9506de38e5917a0406cf447238d42dc17e8ca"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
