)]}'
{"id":"openstack%2Fkolla~826052","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Ic8ebe4089c84f923c24697d632eb294a262059a8","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-01-25 16:20:59.000000000","reason":"Change was submitted"},"34505":{"account":{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},"last_update":"2022-01-25 16:20:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8ebe4089c84f923c24697d632eb294a262059a8","subject":"openstack-base: drop anyjson","status":"MERGED","created":"2022-01-24 11:00:14.000000000","updated":"2022-01-25 16:25:33.000000000","submitted":"2022-01-25 16:20:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826052","cherry_pick_of_change":814779,"cherry_pick_of_patch_set":1,"meta_rev_id":"24a8634adce6b3ee776a602ee9d0b6dcba41fd4e","_number":826052,"virtual_id_number":826052,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 16:20:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-01-25 14:55:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-25 09:27:45.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-01-25 14:55:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-01-25 08:54:58.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 11:00:14.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},"state":"REVIEWER"},{"updated":"2022-01-24 12:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 08:54:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-25 09:27:45.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2022-01-25 14:55:31.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"890c4d474521ce34e895c64e890ea1b84fdeab37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-24 11:00:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bd59a4398d156203e20045421fb1a3f0d0e70448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 12:09:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fb72d69e9f04406faa79f18bc327b449 : SUCCESS in 23m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/31521811ae304ae09d91f98ef7c6933a : SUCCESS in 43m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1959231a2ae74b1ca47996fbdea2024b : SUCCESS in 54m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e5711c9843f84f2ab9e0c82e5e8f42ff : SUCCESS in 58m 00s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/047fa01531c04728b1c27d1e074405c2 : SUCCESS in 1h 06m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cd46d15b4b0840aa876f0a037b5a99a9 : SUCCESS in 34m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7ee1299083a342d1803acb3f308f9315 : SUCCESS in 15m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/173892d40ce2494eb4da4221f8e62843 : SUCCESS in 57m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f3ced03a82241199ce529eacd64076b : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b895688db6c434ab2853fd4c6c56320 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14753997430e4342bccecdc86bdbcf59 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ab9dbe5fd864336a9056caad39505fa : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e62e1853052c43d8ba51fcd1144b3001 : SUCCESS in 5m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ad2fa70a4104449da8d3f7b5c1f1c2be : SUCCESS in 23m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ce81f1f75c24d64b187c8c84be00271 : SUCCESS in 48m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a890268c783f49fab6c567f9d69a5bf8 : SUCCESS in 51m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3a66c376309540d8879f1b690637927b : SUCCESS in 44m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/50e8e8d81c2e47c7b19b1464bf424648 : SUCCESS in 51m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e966d62db34fb0f80cb863619bc2127916286229","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:12:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/680f5681fdfd43e38e738e44e816bdd9 : SUCCESS in 2h 45m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ce9a729475804db999fd8049ae8f9fb1 : SUCCESS in 2h 02m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/96be223384b8421c9b197a6d9fec41e5 : TIMED_OUT in 3h 02m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/25602786947b4d70873ae060257af684 : SUCCESS in 2h 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89f091a208508c5c6d532c1e3e823afa6a1e9816","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-25 08:54:58.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"7ea598e9323168a8515e15f34081f05a70ada54d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-01-25 09:27:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7ebcedc4553bfafa454e9295fd0fd4f226360773","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-25 14:55:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"958aed60e1bebc0bbce02713a2eb84a6c20ede0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 14:56:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e7b8cf6d0df6fdbc1de51c2ca761c21a1d5b950","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:20:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b9a28eea9b4f4f71a9e52f3ed2d7eba1 : SUCCESS in 42m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0ccfa2f1e7594965b1b18bb75516e30d : SUCCESS in 1h 04m 04s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b037a986781443e39a9d5b19352c76c4 : SUCCESS in 1h 08m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00dbc4f22e7544919606035cbf5a3862 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31f0b8d07e6a45e18f510a8c782c5914 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c05f6ec97f6641548679329f16b9a19a : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c95d01637ced4be28b7208d88e2a1ca8 : SUCCESS in 5m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ccbb3a5096f7435790e615df681fd9bd : SUCCESS in 28m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2775a31653d64ee8aceee21d2f4121cf : SUCCESS in 42m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a81d1f93785e4dcd906a0c632fb8b683 : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":1},{"id":"42621dbb6718ebdde3e28b6fecab2a2bdd65e7c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:20:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"24a8634adce6b3ee776a602ee9d0b6dcba41fd4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:25:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f27a9f44d04148e5a3ca0836997bede4 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ba4fb275da333d7f71a17c8755d25b851e7f07b","revisions":{"5ba4fb275da333d7f71a17c8755d25b851e7f07b":{"kind":"REWORK","_number":1,"created":"2022-01-24 11:00:14.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/52/826052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/826052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/826052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/826052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/826052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/826052/1"}}},"commit":{"parents":[{"commit":"59adcfd80ad19e67d16cae3f23117d3ff4a1a1bc","subject":"magnum: fix issue with keystone auth default policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59adcfd80ad19e67d16cae3f23117d3ff4a1a1bc"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:28:16.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"hrw@review.opendev.org","date":"2022-01-24 11:00:14.000000000","tz":0},"subject":"openstack-base: drop anyjson","message":"openstack-base: drop anyjson\n\nIt is required only in Tacker and fails to be installed.\n\nAnd even Tacker does not use it...\n\nChange-Id: Ic8ebe4089c84f923c24697d632eb294a262059a8\n(cherry picked from commit ec06d44d2c095199a9e14aef2f036e957082537e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ba4fb275da333d7f71a17c8755d25b851e7f07b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ba4fb275da333d7f71a17c8755d25b851e7f07b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
