)]}'
{"id":"openstack%2Fkolla~826053","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Ic8ebe4089c84f923c24697d632eb294a262059a8","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-01-25 16:21:02.000000000","reason":"Change was submitted"},"34505":{"account":{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},"last_update":"2022-01-25 16:21:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8ebe4089c84f923c24697d632eb294a262059a8","subject":"openstack-base: drop anyjson","status":"MERGED","created":"2022-01-24 11:00:28.000000000","updated":"2022-01-25 16:25:34.000000000","submitted":"2022-01-25 16:21:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826053","cherry_pick_of_change":826052,"cherry_pick_of_patch_set":1,"meta_rev_id":"3375f563b0215b4d4efd2727b79e39975d954964","_number":826053,"virtual_id_number":826053,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 16:21:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-01-25 14:55:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-25 09:28:29.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-01-25 14:55:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-01-25 08:55:11.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2022-01-25 09:28:29.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 11:00:28.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},"state":"REVIEWER"},{"updated":"2022-01-24 13:46:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-24 14:06:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 08:55:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-25 09:28:12.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2022-01-25 14:55:38.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"c08d7a88b51355dae54cfa13e68dee9b16353edf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-24 11:00:28.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"82518bc126632d0a66c9796a494c4bcb3c273053","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 13:46:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e3f50c0523d34c62b3941871efcfeed6 : SUCCESS in 2h 39m 39s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/29ad5d0aeea94c8ab4cf9ee4d88d795f : SUCCESS in 2h 43m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb426e0ae0a44089a70bae8e5368c8e9 : SUCCESS in 1h 41m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b73f3943c00d4f198a2ff2e80b93fe62 : SUCCESS in 2h 29m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c77c1cd4b2e34639b29c5101de731baf : SUCCESS in 2h 03m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c5021cc68dcfbff31650af6546d218d56a495f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:06:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4cada13c703a452fbcfb76972b872298 : SUCCESS in 25m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a5e99c2d57864458814da7a90ef77f1f : SUCCESS in 43m 30s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/06daec379f6d453e8cf2d13485ad72c2 : TIMED_OUT in 3h 01m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f02b49c358e244a2be3f9670a3c8aa93 : SUCCESS in 44m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a37b7b760e0f4aef9c9146e2dc941dbb : SUCCESS in 1h 10m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b934d7517cde49848f2839e2cb5b4ee6 : SUCCESS in 1h 12m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f4c72fc7f38c4f1ba099ae1e8ad460fc : SUCCESS in 1h 18m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e7f34e4e90134dd1b252d8fde5a8af0f : SUCCESS in 19m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a0682aaf427d4a01bcfdc17df72de75c : SUCCESS in 26m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/889ea519acd34e4e8aa41f0023606f05 : SUCCESS in 42m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c7f59a056d488b8327dc94f480cdc8 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b27d5694e60843c49642f92eafbc4931 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36d68f4a7f7244c9a429ca50d96f350a : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293458a02fec4ca38027d9514b1be347 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c0d5c1fbc2b420da2e79bf0e53174de : SUCCESS in 22m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a75336bb09649fcb9eb46c53c972d8b : SUCCESS in 20m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8bbe888446b048fe9efc2f059c34c1a5 : SUCCESS in 52m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/752464c9b2eb4e22b1946e3453d2e60a : SUCCESS in 46m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2fed635e1ebb41fea678e60724570284 : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b12d39fab6d9ca943c711e7f91a4d02d7ae42c6f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-25 08:55:11.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"14ad452239637a69094269369d2a4a8c90162da1","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-01-25 09:28:12.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"453af30491160cebdb50765ce8e1fa7e8a6ca7f7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-01-25 09:28:29.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"457c96203e7e70d6939b66edf3c162b5e55cc52f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-25 14:55:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46c6a908a23f220413f76076ae0b2d7b77fb6b73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 14:56:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"548710567039b1fdc0bafb52e56992be0de26d27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:21:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0167f16cc2614acca4e7e04fa7aa45da : SUCCESS in 29m 29s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/072ab33633d546cd8b0049ae0b50abb1 : SUCCESS in 31m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ee8e151dbec14e2d8bcae10b1132866e : SUCCESS in 1h 04m 04s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c0bfed825eae4d9b815adbdda8266fb1 : SUCCESS in 1h 06m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2dd0e291b4941bd8e1d8c390a0c2a70 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29b8ab723abf462dbc271553a30a0310 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d80a6a3b495d42748d6b4fff4ae25870 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fcc2f691479455f9ae81c2460bbd9b3 : SUCCESS in 6m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2f5370cd6e074f8cb662a76edab37560 : SUCCESS in 23m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5bef21b5a52c464fb18f4f0a0c25735f : SUCCESS in 1h 09m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ff36ed5dfc1a4d2c85ab0e45bdf4990c : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":1},{"id":"244368255476a6d7666e5866859726202dc88694","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:21:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3375f563b0215b4d4efd2727b79e39975d954964","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:25:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b393457ad214531b81b07f87ef91625 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56f4942309d18ab241db8ebf0fe4a926caba3fbe","revisions":{"56f4942309d18ab241db8ebf0fe4a926caba3fbe":{"kind":"REWORK","_number":1,"created":"2022-01-24 11:00:28.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/53/826053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/826053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/826053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/826053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/826053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/826053/1"}}},"commit":{"parents":[{"commit":"ca6951744f273f20bcf05916f090ee84abb163ea","subject":"magnum: fix issue with keystone auth default policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca6951744f273f20bcf05916f090ee84abb163ea"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:28:16.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"hrw@review.opendev.org","date":"2022-01-24 11:00:28.000000000","tz":0},"subject":"openstack-base: drop anyjson","message":"openstack-base: drop anyjson\n\nIt is required only in Tacker and fails to be installed.\n\nAnd even Tacker does not use it...\n\nChange-Id: Ic8ebe4089c84f923c24697d632eb294a262059a8\n(cherry picked from commit ec06d44d2c095199a9e14aef2f036e957082537e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56f4942309d18ab241db8ebf0fe4a926caba3fbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56f4942309d18ab241db8ebf0fe4a926caba3fbe"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
