)]}'
{"id":"openstack%2Fkolla~826054","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Ic8ebe4089c84f923c24697d632eb294a262059a8","project":"openstack/kolla","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-01-25 16:21:04.000000000","reason":"Change was submitted"},"34505":{"account":{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},"last_update":"2022-01-25 16:21:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8ebe4089c84f923c24697d632eb294a262059a8","subject":"openstack-base: drop anyjson","status":"MERGED","created":"2022-01-24 11:00:38.000000000","updated":"2022-01-25 16:25:35.000000000","submitted":"2022-01-25 16:21:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826054","cherry_pick_of_change":826053,"cherry_pick_of_patch_set":1,"meta_rev_id":"530b6239d55a4671d3c2299b4b6f14c5674ea180","_number":826054,"virtual_id_number":826054,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 16:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"recommended":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"all":[{"value":1,"date":"2022-01-25 07:25:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-01-25 14:55:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-25 09:28:42.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-01-25 14:55:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":1,"date":"2022-01-25 08:55:25.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2022-01-25 09:28:42.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 11:00:38.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":34505,"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org"},"state":"REVIEWER"},{"updated":"2022-01-24 11:58:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 07:25:37.000000000","updated_by":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"reviewer":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"state":"REVIEWER"},{"updated":"2022-01-25 08:55:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-25 09:28:08.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2022-01-25 14:55:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"e350a5aac64d4c6d5800dab29d7d233f8b2b4943","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-24 11:00:38.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"45edf7fb596382ff6083c1d928a67a226e012f43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 11:58:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4e7e8751ea2c4377beb53c618dcba9ea : SUCCESS in 19m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d5d2ea4dbe0d480abb77bf0c0a9a0a9f : SUCCESS in 24m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/21f776f06fc44a0eb146bfefd7bd6dd9 : SUCCESS in 21m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e73ed2811eba442ea4a49ed8fb1dbbba : SUCCESS in 16m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/580c0a01e8104fcdb6739f4a1c9788b0 : SUCCESS in 48m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/006d00b4cb6c4eb0b28ca89f35a2c075 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/624024b037fa4afc8ae4ff0ac7b8d429 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e091140e5844c689e839f571b7a8e30 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbdfa6a66185498287b330e16920bc06 : SUCCESS in 5m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1a3f6c037df64f8daa0dd41fc196d99f : SUCCESS in 22m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e72d0424286497b9ef6eb12e53bfc14 : SUCCESS in 24m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c3172f1432e4548b5e4dcafc1dcfdf5 : SUCCESS in 41m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b1c43807fdd04314ac9acb50b7ac144a : SUCCESS in 43m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/065855b76cbf4f2aac7933975f40b9cd : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a144abdc4150cc54572562092ba4d37672585360","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:13:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5fc70b61ab124b54af44edaa253585e2 : SUCCESS in 2h 42m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d784bbe78ff04bbab98ad2e0cba8d596 : SUCCESS in 2h 34m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4262a1b679cc46808d2fe05d4fcbf670 : SUCCESS in 2h 50m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e85262feefda48bf84f0f77830e39978 : SUCCESS in 1h 58m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d0c4aa11803c96ee4639450ae9323940e9daeff","author":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"date":"2022-01-25 07:25:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dcf37cc0033c73e9e17ff74ee9a309fdae4565d3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-25 08:55:25.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"65f254bccaec44b42e6497f3f76f6a7100dffbd9","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-01-25 09:28:08.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"e406e80a202273208d274e7c551751aae8c382da","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-01-25 09:28:42.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ada536876b3f2bf47cef51b9c42f748a7be53d33","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-25 14:55:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c7d68c5bf7d2dc885b04f93b96319f6969f77bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 14:56:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93d05c1e950244da7d024e0397121f5776c9402d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:21:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ac22a90f790f4170bc84d72ebe043285 : SUCCESS in 25m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/302600df2c69435a988c07b1d157e374 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/189c2f797c3e495ab498e07a9ae8cbef : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b3edbf7d29347de9a9d340e20e4d203 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213cc1037eff4cd08b118942a3824e16 : SUCCESS in 8m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e50f0e547e3a438a9637d4e50845af9f : SUCCESS in 36m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5cb6acc0cfe34e388f8916a4c5689dbb : SUCCESS in 57m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/934496748b57480c9ac369b96909d04a : SUCCESS in 1h 01m 09s","accounts_in_message":[],"_revision_number":1},{"id":"dd329e7f596511990b38833feca6cb4779e6ab39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:21:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"530b6239d55a4671d3c2299b4b6f14c5674ea180","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:25:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e03a5db3c3499d8f4131d813c59a04 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"490d1bdbd1fbdabeb5e82bd5edd100cd22492066","revisions":{"490d1bdbd1fbdabeb5e82bd5edd100cd22492066":{"kind":"REWORK","_number":1,"created":"2022-01-24 11:00:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/826054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/826054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/826054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/826054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/826054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/826054/1"}}},"commit":{"parents":[{"commit":"15226bf473798aa28db99978ab3a4fbbbd8be1e3","subject":"[CI] Stop testing non-stream CentOS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15226bf473798aa28db99978ab3a4fbbbd8be1e3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-10-20 13:28:16.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"hrw@review.opendev.org","date":"2022-01-24 11:00:38.000000000","tz":0},"subject":"openstack-base: drop anyjson","message":"openstack-base: drop anyjson\n\nIt is required only in Tacker and fails to be installed.\n\nAnd even Tacker does not use it...\n\nChange-Id: Ic8ebe4089c84f923c24697d632eb294a262059a8\n(cherry picked from commit ec06d44d2c095199a9e14aef2f036e957082537e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/490d1bdbd1fbdabeb5e82bd5edd100cd22492066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/490d1bdbd1fbdabeb5e82bd5edd100cd22492066"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
