)]}'
{"id":"openstack%2Fkolla~826273","triplet_id":"openstack%2Fkolla~master~I0d69009b8b736b59b122ad29a9a5f6a22b041513","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-01-25 16:20:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0d69009b8b736b59b122ad29a9a5f6a22b041513","subject":"Upgrade pip and friends in source images (part 2)","status":"MERGED","created":"2022-01-25 12:09:51.000000000","updated":"2022-01-25 16:26:09.000000000","submitted":"2022-01-25 16:20:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826273","meta_rev_id":"3d95cdabc7e820ffac73e991d7da7edcc3ab6fc2","_number":826273,"virtual_id_number":826273,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 16:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-01-25 15:02:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-01-25 14:06:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-01-25 14:49:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2022-01-25 15:02:02.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-25 13:00:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 14:06:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-01-25 14:49:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"065a548943f37e8d8ff59c4c2ecf0c673b506ac5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-25 12:09:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f4767ba1f7babbfb9621690cb3a3e1628978230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 13:00:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/88cfed706d5140d3a4342e0603682f44 : FAILURE in 29m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/be6b775f72b04a07b00012e7db99f6ff : FAILURE in 31m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/932ab6d31a3f45bfa64d677d053c9efe : FAILURE in 27m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4080da87afe643ca9c52a3c389c88d32 : FAILURE in 27m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/16408785ccbd48468321065a77839da5 : FAILURE in 26m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c82fd5020cfe49fbad175ce9770f5c9e : FAILURE in 32m 53s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/789306e5aaeb4ee89a6877cfbee4673a : FAILURE in 15m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/466d55e9726c45d7809fb425dd581200 : FAILURE in 22m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6bf94cd9aa9849c29ac552130b4c1eee : FAILURE in 25m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c341221591d54a4e9f169def4b858bbe : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e588dab4b454925bab219ce2acc556a : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47e8d0fda7de4ad6abede563b1a695ac : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36a05ed63ec4fda9502aa538d4526ff : SUCCESS in 6m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ee0538416ed41f58ce078ab20d728c3 : FAILURE in 28m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/549a16c0a7ce406995bb3b5154d6bfb3 : FAILURE in 47m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c186958e32e64f499363ea6b6616b4b7 : FAILURE in 18m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a76d91279c5b49cbb31e9dbb57efa7f6 : FAILURE in 20m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5d7774d950ba448fb814073ff98884d8 : FAILURE in 19m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c23c21e75308ca0314618cd9800faaf0522e2cd2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-25 13:06:43.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"131d61d34f0961d08ceb99aa31e3e5ae0c3e02c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 13:17:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4dbb2e1135ae463fa80499bafe9d1fec : FAILURE in 1h 00m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ae39b3ab71b74754bbdc0653721c741c : FAILURE in 28m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5f21d049518b4a3a9c385721b7d352ec : FAILURE in 30m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a76492ce4888429d9cbc4205e25f3ad8 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1673c4028c346bb841ec03da848ce42 : SUCCESS in 4m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/839ab7e4b8a64a7b9e3371e6225912a3 : FAILURE in 44m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"294dc2e1f9f02db3dd8fcf0d46069f0c112e48b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-25 13:21:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9d43a229fe5aca1cd48dad4b45cc6e0dff560296","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-25 14:06:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"73188c50b013bdc98e20f1e5376502ca2952e50d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 14:41:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6d210772c45845339c3c56fc45b8cc3c : SUCCESS in 30m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/be8d67d38e394883a59a93058717e6ae : SUCCESS in 48m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/61a7f5cffb3943528b186716447936d5 : SUCCESS in 1h 04m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6af8fb6d6e1944f0bd0796759462eee3 : SUCCESS in 1h 08m 37s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7b2b69f4b1e84cbcb6deb27859bb9f8f : SUCCESS in 1h 13m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bbedd6f2fd1c44dfa7f3aaabf0291ec0 : SUCCESS in 28m 11s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/eca31a957ced460a8994d9fd19240073 : SUCCESS in 20m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/26ebd0dbc33c4386a66c54d21852a137 : SUCCESS in 56m 20s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c2fce041b65c4fe182ba1086dbef335c : SUCCESS in 1h 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce78b659af03463e93b8f0a9121049a5 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8190d3a059e450396eade21c3f601ae : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cc1389e0eca4ad7966830d2895f8242 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d803e3dab8f34476908a06d7f45c59cc : SUCCESS in 5m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d330bc395dbc450dbe8eee2fdeb2b7fc : SUCCESS in 18m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e6841f059504956bc53a2ed74349724 : SUCCESS in 23m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ab9751384c60400480d3895d0a056f80 : SUCCESS in 48m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/568febde707b49bdb953b97b05a8fff0 : SUCCESS in 48m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f6487538fde841feaede9afb06deeaed : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":2},{"id":"db9b452408d8fce158f5bf46f5263e1bc57f05d2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-25 14:49:14.000000000","message":"Patch Set 2: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"38b5b212fee3bdd03b06c231ddca59525173a015","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 14:49:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2bb94cd56d9c064f34e26d58daa189ab84d3c5e5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-25 14:51:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d609b94aa05adf273f0ff1a5d5b6ca00ae430cd","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-25 15:01:57.000000000","message":"Removed Backport-Candidate+2 by Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"8cd157f32b542dc76bb6d5973ba359fbba38bd95","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-25 15:02:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8f35cece2311934ca12fd54cf87729f993c4ceae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 15:13:42.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2aa120b90bc64161b067527c1765b059 : SUCCESS in 1h 49m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/721db26d28b24dc19f0081c27311a0be : FAILURE in 58m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b453f5540370416b9c55faa3d0b0811f : FAILURE in 33m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a49a5473e1d841c6a504cee31040ffaf : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b4992d7bab94d6fa3cf34311a99c68d : SUCCESS in 6m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/add0b1e2d92a4e57a39ce446efa3d28f : SUCCESS in 1h 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2d78bdede651d892d4927b4a6530686e2a75aaa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:20:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c48b81ad878c46dfbbdcfa6972bfdea2 : SUCCESS in 33m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/365161352c414dfd9a662de3f0994a5b : SUCCESS in 1h 01m 07s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b1780ba673174161b62ee42ea9ebdce6 : SUCCESS in 1h 26m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/39d95dd9f48448858dc3e473ebf71b1f : SUCCESS in 29m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eba1c73b9c3342bdaa26fec98e194a43 : SUCCESS in 1h 16m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f2cde5ddb7454796a26acfc572c8e1fc : SUCCESS in 59m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862d0562f9104ce98059663760d99e4f : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3889fb9cdd9a46238d6cd776d254a30b : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4518d429eeff4b53b950be2d36278bc9 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3359ee00e4484181a72c0294c10529 : SUCCESS in 7m 47s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b76df3f931c4ec383a744ee99eb2d21 : SUCCESS in 35m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/53d18e5b732948e9a0917ba04a5a9669 : SUCCESS in 52m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2478dd49e49b4e80a40f895ef5547c57 : SUCCESS in 1h 11m 34s","accounts_in_message":[],"_revision_number":2},{"id":"c2137cfe7ed39cbb195b98d01c412cf0d055e03d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:20:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3d95cdabc7e820ffac73e991d7da7edcc3ab6fc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:26:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5300a558ecdc4c7c8360bc29b9faaefc : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d13c41b776b092a5fd4934962032d354c1822539","revisions":{"d5720023333360343dbea6557a6c111c779fef7e":{"kind":"REWORK","_number":1,"created":"2022-01-25 12:09:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/73/826273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/826273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/826273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/826273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/826273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/826273/1"}}},"commit":{"parents":[{"commit":"d7dde0265383e2498cfec9420b1a35d1f754b47f","subject":"Merge \"Mitigate two Log4j vulnerabilities in Apache Storm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7dde0265383e2498cfec9420b1a35d1f754b47f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-01-25 12:08:45.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-01-25 12:08:45.000000000","tz":60},"subject":"Upgrade pip and friends in source images (part 2)","message":"Upgrade pip and friends in source images (part 2)\n\nNew pip has nice features for detection of conflicting\nrequirements.\n\nWhen installing from PyPI, as we do in source images, running\nthe latest pip+setuptools+wheel is recommended.\n\nThis change covers entries missed in I4ae3a82cc796a60450c2a35beba32972964bc5d0\n\nChange-Id: I0d69009b8b736b59b122ad29a9a5f6a22b041513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5720023333360343dbea6557a6c111c779fef7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5720023333360343dbea6557a6c111c779fef7e"}]},"branch":"refs/heads/master"},"d13c41b776b092a5fd4934962032d354c1822539":{"kind":"REWORK","_number":2,"created":"2022-01-25 13:21:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/73/826273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/826273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/826273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/826273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/826273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/826273/2"}}},"commit":{"parents":[{"commit":"d7dde0265383e2498cfec9420b1a35d1f754b47f","subject":"Merge \"Mitigate two Log4j vulnerabilities in Apache Storm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7dde0265383e2498cfec9420b1a35d1f754b47f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-01-25 12:08:45.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-01-25 13:21:35.000000000","tz":0},"subject":"Upgrade pip and friends in source images (part 2)","message":"Upgrade pip and friends in source images (part 2)\n\nNew pip has nice features for detection of conflicting\nrequirements.\n\nWhen installing from PyPI, as we do in source images, running\nthe latest pip+setuptools+wheel is recommended.\n\nThis change covers entries missed in I4ae3a82cc796a60450c2a35beba32972964bc5d0\n\nChange-Id: I0d69009b8b736b59b122ad29a9a5f6a22b041513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d13c41b776b092a5fd4934962032d354c1822539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d13c41b776b092a5fd4934962032d354c1822539"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
