)]}'
{"id":"openstack%2Fkolla~826488","triplet_id":"openstack%2Fkolla~master~I5e620cd07326b3d69b07f7c4667176918d7c1b24","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-02-04 11:56:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5e620cd07326b3d69b07f7c4667176918d7c1b24","subject":"Switch to use Yoga binary packages","status":"MERGED","created":"2022-01-26 15:26:21.000000000","updated":"2022-02-04 13:12:13.000000000","submitted":"2022-02-04 11:56:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826488","meta_rev_id":"5a599a89b27cbb77bfa6051029af93dca4d93d66","_number":826488,"virtual_id_number":826488,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 11:56:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-02-04 13:12:13.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2022-02-04 09:15:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-04 10:03:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-26 16:26:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-26 16:52:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-26 19:52:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-04 09:15:44.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-02-04 10:03:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2dc3ef0db3e7e1fbf3550413b7d751b4c2737058","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-26 15:26:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dad40950d0769f890c270cebbb87b82cb304c3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 16:26:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/10acf67ecbce4b46959a0547b7351c61 : SUCCESS in 52m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/483f5daf69c44e0faf92205ffefcf081 : FAILURE in 14m 23s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5bb1a7597734fd4b4de2066aad4ee3b : FAILURE in 34m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45b52cb719ce4b86a17538d10481dad3 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/126ccdb5665f4dd982fc9ac32684b7b5 : SUCCESS in 6m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1371dd3f70924a2c9cd28092f6d1cbf7 : FAILURE in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"028981809b1169e3445c8af30597a27b9ccc24cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 16:52:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/42e347343e5f45c7a8a73fa539b7afb1 : SUCCESS in 29m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5dde33708f7e427eb3d1b02edb368f90 : SUCCESS in 48m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1abcc2b26ef54eaa80279e8d50d7ea36 : SUCCESS in 1h 05m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/97cf7d93ab454f14be21e0bd0db9bdfd : SUCCESS in 1h 17m 37s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8e2192077b6a43549ba6edcdafe69454 : SUCCESS in 1h 19m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/621f08abe4f7459c8f20ddad2c33c8a5 : SUCCESS in 21m 40s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2e01bee6a994454fb6c3bdb5b7bf2509 : SUCCESS in 19m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c02a10e7792241b1a954412618e3e540 : SUCCESS in 58m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/623da75113e3453989613196d958b91b : SUCCESS in 1h 09m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/173e7a7624ff4c958a673b00467ba869 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/943e782679984df2944763633b91875b : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0e7ddebebb54831bc87270bf697a2d5 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e2a916f32d43eb986287c3cc907c7a : SUCCESS in 7m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ea776fe82724303b32d755d9ca34c80 : FAILURE in 6m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2a1189efdce43a594fd2e62fd7d29ff : FAILURE in 6m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4800f5d7b16049109382fb31695c68f5 : FAILURE in 11m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/480dccb07adf4cbf8a6bfb6da0adecf5 : FAILURE in 11m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/775f5fea752b4e84bdf63f9716a06e49 : FAILURE in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"56b70fcba6fdc600d75a092a408fddc0a0e9ecab","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-26 17:18:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d2daf833ace6ceb651401eae7e099e491da6424","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-26 19:52:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fbb4dd2e10f52166002791ccd759c4f66aa4487b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-26 19:53:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"040e712fac9bb305ed2f70aa1759ebd47e35d5bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-27 20:46:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"551e8e2c50429c7007128529e3060f62767eeff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 22:45:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/249fe1b815014f9684336e0224214bfc : SUCCESS in 37m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/06450c1436ab4237970832997ae4c9dc : SUCCESS in 45m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2841cb3efc994f11b5d76c77f72e7511 : SUCCESS in 1h 47m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/624ae1a69f1d4f79a914e65d96c7dedf : SUCCESS in 1h 07m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e599db200652493ab4700d2481d20594 : SUCCESS in 1h 24m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/089cabf5226b41ebafa6449dc6d7bb4e : SUCCESS in 35m 32s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9a9c9a8a17a345a99461d127a7882b53 : SUCCESS in 22m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/290cef7f6e344baeb8eabaffeaa3baca : SUCCESS in 50m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/964f018f74c442b0ae46383b81712d8c : SUCCESS in 50m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9a67f5e580541429843b81ac62e551b : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0352ce6ee2c473884700dacf8a0766c : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e91d33410d9a458cb30459a02b253f9c : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ba56d1f7504fada911497ec3ebddf1 : SUCCESS in 7m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2736789edf804389b2e1ac7a3ac0241d : FAILURE in 23m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d52546efd9a4cd294cf713e9a179d76 : SUCCESS in 28m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/17bdfbcc092e4231b8289fdc17e99316 : SUCCESS in 54m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/80333af42d6744df8e2677722390e62d : FAILURE in 19m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3efe5b2e8d7f4c44b52a4bf9e9e98e4b : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":2},{"id":"1dcaa4645103f4c72c572175c48ee7c3ca353ffe","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-27 22:57:46.000000000","message":"Patch Set 2: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dd08d2a6e8494064b69f410fe9da7ec6d72055a8","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-27 22:58:04.000000000","message":"Removed Verified+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":2},{"id":"dd9e33499e5cf6cb1885f2878651e43a20141f89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 23:24:12.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e77ce7adcc9e4b5b895b42e6c7626c19 : SUCCESS in 2h 27m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/860ca5b7ced1423eaa34a6eb515456fc : FAILURE in 58m 48s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7279483c2ada4de28c72d9163c6a1154 : FAILURE in 32m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7f3105989844bea8bc1c56f8308d0b4 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a3d552b09474d8490624e4f625efa48 : SUCCESS in 7m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c21929a3e1624d80be28c9ebe8ebaebe : SUCCESS in 51m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e3af9d9b5a9045106b2e7aa1302c29c43b69df5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-28 14:04:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f495a6504eb3194c1f298b3224231561c8db11d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-01-29 08:25:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"007cdea1dbbe40c2ef50b166a89244d5beab813c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 09:45:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3f71db9495194945b565e3d20bdc63d9 : SUCCESS in 24m 28s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9f8e6aa3ae3a4fa7a475e5f2d93ed644 : SUCCESS in 40m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/47c02ea8cd1f467e9362a306fe7c692c : SUCCESS in 1h 19m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/63ab0678ec024e5a87e6ee8fd4303c20 : SUCCESS in 1h 04m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d240b7266b6b430c9412fcdc9ae57015 : SUCCESS in 1h 06m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/35f4d266e1634c5c8bba27ed770f7d3f : SUCCESS in 28m 19s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/46cd6e74f00c43fbadc6295684e1a974 : SUCCESS in 16m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/64ff5975500041b187add97750654560 : SUCCESS in 55m 05s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c5fa26480fff45c09d8ce21afa476a69 : SUCCESS in 49m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23e4dc23f28e4da9b5ee743d4535cce1 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dccf7675fdc64245856592a7824d967d : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf9a318c7f5d43f19b8950c9949ed4a7 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee82414843c4d808be46ca624a372d0 : SUCCESS in 5m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7959b71f6be44e7da49385556a6e9d1a : FAILURE in 17m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ee27e5aacece4223bf9272522604d892 : SUCCESS in 23m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/58d268033d10405583b502ebdafd9042 : SUCCESS in 35m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/68e92898d27f4a75b51f03c5d441a514 : FAILURE in 19m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6de7eeaf50974be894ce0f462761ff41 : SUCCESS in 48m 23s","accounts_in_message":[],"_revision_number":2},{"id":"35d7264ed9985f5eb89655704e5a40e8a80cfb15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 10:07:33.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/86264d3435ae4322a11fa19fcb27f4c3 : SUCCESS in 1h 40m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b77757d6e2b4519bc4cf3ec89789901 : FAILURE in 31m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/38ce91706f9b4ddbbe5625e33e0568c8 : FAILURE in 28m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9636289f4a704567807f6c40b74d1131 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c2ae00c5979d4c0d8d16b2e2f02ed3a3 : SUCCESS in 3m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/be43d3052241407b89ace04f58566141 : SUCCESS in 1h 08m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6a32c52cdc678a3deaa83d6eeed556021bd89d95","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-03 19:48:18.000000000","message":"Removed Workflow-1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz+kolla@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"90ec57427e0559c8c9e72b66ceaedbdf0a6b5524","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-03 19:48:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ecec2ce8a429df8c6406f3c1051d85196af8d243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-03 19:48:54.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"095fb8ecbf0153d15f8a98f144bb8bf7e24a5a89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 21:16:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f81bf819e0a64d00ada26cdd2e47b5e4 : FAILURE in 14m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/79195ba1d83a4fe792e5174236156538 : FAILURE in 33m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/843eb7adcf9e41ecbaa37b1747c72331 : SUCCESS in 1h 16m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c8f56a74e981446daac25da619edf99d : FAILURE in 20m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ab1e3a24f34248ad94c1ac8e34c65fbd : SUCCESS in 1h 14m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f639c7a55b384fa4b476cabbba1002dd : SUCCESS in 30m 59s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/441ff148e352439c8209acb47b52acdd : SUCCESS in 26m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/989337120c79498ca67fe167358422c8 : SUCCESS in 56m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/aa3e025d7b204de9910de794ffc27e64 : SUCCESS in 1h 05m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74765b37f3f44de4a862aaceb247755d : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27cd00a0ccce4831b5be1ce280b28136 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ed131ffe3e4430cb800568162c741c9 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d11862a90f742b6b8b5c38e897a6f45 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8266e636c40f4d6f8a78a37c37eb3fba : FAILURE in 18m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2fc86f45b1b47579b57f81b41f8083f : SUCCESS in 28m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c24e66854ccd400e952cb7264fd77eb7 : SUCCESS in 49m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/95339b3ede29401cb719e35fd1c10cad : FAILURE in 18m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/daf2749cef2a47968fa53179739ee035 : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":3},{"id":"febacbd039319cddec4b3920d92e64bcdc2e0e11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 00:38:09.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b129fdfb254f47bfbd4240cf258de9dc : FAILURE in 2h 16m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/93af7b6ecedb4f7297e79767fa597ca5 : FAILURE in 39m 04s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0cb9c3cacabb493cb8e860b288be480c : FAILURE in 34m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb967e74edae457b9643d2a95df50243 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db2a12fb596a4312939ece9343a4213c : SUCCESS in 7m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/18ffa6fdac7c4938992e75f4ce1b97a2 : SUCCESS in 2h 38m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"39a28a8fec18b9ddac02b28595d0637f69548bac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 08:50:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"73a1fccf8cd35369df3ce8e92f593cfcf5f76909","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-02-04 09:15:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0756fd87be13addc23856c2686eaade9a2a26636","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-04 10:03:42.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0eea96fcfc291403bdd99e65bd10fdb6a6065191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 10:19:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ba3a19fda91641beaecd036cabccba88 : SUCCESS in 24m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e7c9accfadff4e04b3535a30d66bdf15 : SUCCESS in 38m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2a1ef96c0e23461fa72bf439195df42d : SUCCESS in 1h 15m 05s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/833c3c463157417ca0a35694dbc1772b : SUCCESS in 1h 17m 53s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3441d46ca6da4cd085c16245c3d8ccf3 : SUCCESS in 1h 27m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bec0a61461af48c3af35735abb472025 : SUCCESS in 40m 59s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dfeadae4423f44878cf0923a972c70c3 : SUCCESS in 15m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3ae22148ef594ecf8366780597cb3eb4 : SUCCESS in 1h 08m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/274fcce572b840fdb6e3ae25998530c1 : SUCCESS in 1h 08m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/093fedbc8f5c4a59b1b18e5b6895147d : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7197608b4f7f4a7aa29844c3d3c839f8 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad00938ad958456ca32e972a2659ab58 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7200c4b419a041b08f2591af8a378151 : SUCCESS in 6m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ae4ebc381e240ccb3272235da755480 : SUCCESS in 21m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc35064012504d89800de6c6933f1b49 : SUCCESS in 32m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3162c2921b074bd787390e32a8a65d5f : SUCCESS in 50m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/330d33fc52c640beb2b49873051077ec : SUCCESS in 51m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/37d9b0b13e3546bebc49dee825ac346c : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":4},{"id":"cc2d6df0fea2252c84d609722963aa09549c4eef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 10:20:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"48035ee97a98f22bb9390f81bdfada21dae0ca47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:56:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a7984c251fa241f09d43bb8745686cae : SUCCESS in 43m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/aa584dcfd06c4e84a162761c5dec1f7b : SUCCESS in 1h 15m 28s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f243506709f047b299ec0f682ec6b74d : SUCCESS in 1h 15m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c7d921c013684ee0a1beb2901b6e287e : SUCCESS in 35m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e9d3c993dc144aebaabe86c45804b410 : SUCCESS in 1h 00m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/81a63f9c3fa04e519aabdafd6badba4c : SUCCESS in 1h 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b08453d1124a6eb347b65088832674 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fc035f0052f436b897874a75f1ecbba : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2090cf7a77a74ae7ae87f19a5cf3d6c1 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/686503161632405c91efe734358cd80b : SUCCESS in 5m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/484d47e9d8e74034bf36dbeb8a48a55d : SUCCESS in 26m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/57275ef912b743f499a9462070c57ba5 : SUCCESS in 43m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f1a1e6060459488f88032734bc5fd496 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":4},{"id":"34949b037264eea5b680642e85024cf1d5a3579a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:56:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8724d2179d9b00f18a4e7cdfeb28db1f0a28898b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 12:00:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950998f172c548038031eabb61cfd44e : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":4},{"id":"550894134cb407d59eaca1d758da013975939a5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:12:13.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fc14c7772c0746a9b813545770922882 : FAILURE in 2h 24m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/048a8b1b9e984405abcbfcd4e5ed18d3 : FAILURE in 52m 05s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/301d3e95b6e44b64895f681083517edb : FAILURE in 40m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4d26795bc53411dab50e735639cb5ae : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5118883c4f524601a9268d796b2ae26d : SUCCESS in 9m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d6832fc60afb4facb84aefe6ec919d81 : SUCCESS in 2h 08m 43s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"57f56e70623a5cb9bec3a97a1f05b087c99f1ccd","revisions":{"821f25954de0b1243d1d065a9e8bfc088a2a24e8":{"kind":"REWORK","_number":1,"created":"2022-01-26 15:26:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/826488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/826488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/826488/1"}}},"commit":{"parents":[{"commit":"e8651961929a00fb8684c56983c2f9c46822553d","subject":"Merge \"Clean up fluentd image recipe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8651961929a00fb8684c56983c2f9c46822553d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-01-26 15:25:17.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-01-26 15:25:17.000000000","tz":60},"subject":"Switch to use Yoga binary packages","message":"Switch to use Yoga binary packages\n\nAt R-8 week we are moving to current release binary packages for CentOS\nand Ubuntu (Debian OpenStack team do not publish before RC).\n\nChange-Id: I5e620cd07326b3d69b07f7c4667176918d7c1b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/821f25954de0b1243d1d065a9e8bfc088a2a24e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/821f25954de0b1243d1d065a9e8bfc088a2a24e8"}]},"branch":"refs/heads/master"},"0626fb320b7512e7e75662c55de22f227d0269ef":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-27 20:46:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/826488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/826488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/826488/2"}}},"commit":{"parents":[{"commit":"5b2d5881d8182c9532da72c353e10834833777a7","subject":"Merge \"Unpin td-agent and cap elasticsearch gem\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b2d5881d8182c9532da72c353e10834833777a7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-01-26 15:25:17.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-01-27 20:46:40.000000000","tz":0},"subject":"Switch to use Yoga binary packages","message":"Switch to use Yoga binary packages\n\nAt R-8 week we are moving to current release binary packages for CentOS\nand Ubuntu (Debian OpenStack team do not publish before RC).\n\nChange-Id: I5e620cd07326b3d69b07f7c4667176918d7c1b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0626fb320b7512e7e75662c55de22f227d0269ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0626fb320b7512e7e75662c55de22f227d0269ef"}]},"branch":"refs/heads/master","description":"Rebase"},"617e3809741a370d38a1db2e754c27bb2ce886b5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-03 19:48:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/826488/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/826488/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/826488/3"}}},"commit":{"parents":[{"commit":"d84cd121cad5ed30fc87f97d5332f357b7f81e3a","subject":"Merge \"Ensure set_configs sets execute bit on directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d84cd121cad5ed30fc87f97d5332f357b7f81e3a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-01-26 15:25:17.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-03 19:48:54.000000000","tz":0},"subject":"Switch to use Yoga binary packages","message":"Switch to use Yoga binary packages\n\nAt R-8 week we are moving to current release binary packages for CentOS\nand Ubuntu (Debian OpenStack team do not publish before RC).\n\nChange-Id: I5e620cd07326b3d69b07f7c4667176918d7c1b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/617e3809741a370d38a1db2e754c27bb2ce886b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/617e3809741a370d38a1db2e754c27bb2ce886b5"}]},"branch":"refs/heads/master","description":"Rebase"},"57f56e70623a5cb9bec3a97a1f05b087c99f1ccd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-04 08:50:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/826488/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/826488/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/826488/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/826488/4"}}},"commit":{"parents":[{"commit":"405fca90f2c8d3763ac56ce8b925611d136bba6c","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/405fca90f2c8d3763ac56ce8b925611d136bba6c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-01-26 15:25:17.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 08:50:12.000000000","tz":0},"subject":"Switch to use Yoga binary packages","message":"Switch to use Yoga binary packages\n\nAt R-8 week we are moving to current release binary packages for CentOS\nand Ubuntu (Debian OpenStack team do not publish before RC).\n\nChange-Id: I5e620cd07326b3d69b07f7c4667176918d7c1b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f56e70623a5cb9bec3a97a1f05b087c99f1ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f56e70623a5cb9bec3a97a1f05b087c99f1ccd"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
