)]}'
{"id":"openstack%2Fkolla~826610","triplet_id":"openstack%2Fkolla~stable%2Fxena~I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591","project":"openstack/kolla","branch":"stable/xena","topic":"bug/1879265-stable/xena","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-01-27 13:08:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591","subject":"Use distro provided GRUB efi","status":"MERGED","created":"2022-01-27 08:38:43.000000000","updated":"2022-01-27 13:10:00.000000000","submitted":"2022-01-27 13:08:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826610-bug/1879265-stable/xena","cherry_pick_of_change":724630,"cherry_pick_of_patch_set":8,"meta_rev_id":"be83bb77009a192fa16935d179be0eb617d2e698","_number":826610,"virtual_id_number":826610,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 13:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-01-27 09:10:16.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-01-27 11:31:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-01-27 11:31:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 08:38:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"state":"CC"},{"updated":"2022-01-27 10:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-27 11:31:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"246690380c8f2f649fa8a937e2695ce4f6bed47b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 08:38:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4804f4557708640ca868eea4ee2dc4cf4840316f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:10:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"38fa3356cf4b27be2032a0f97d70e3b33c4bf35f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 10:09:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7846f9507dae41149af0a4c34c23fdfa : SUCCESS in 31m 21s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ff052293dee34eb58794f4e776d264f2 : SUCCESS in 42m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8f823b81ae98420c8efcc6735e4d0482 : SUCCESS in 1h 15m 00s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/82b79f7a05384f50b2a5f4392a18133b : SUCCESS in 1h 23m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ea156874aec148449412bf3662d20a94 : SUCCESS in 1h 28m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/528e4634a5e3496cbf0b1ba9df6e6c93 : SUCCESS in 32m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b3460f89722c45d6ab2410c9f2e5dd23 : SUCCESS in 23m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c509dd9bbc3b4f11ac62448dcd78e4b2 : SUCCESS in 49m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cea16aa2e434de5ae5b9b429d36d3f2 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/958bbc3c654344f3b87645862e025d09 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24e8aae2d55f451d879469fa3cdb8fe9 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d475ea492f614f62887753429bea5835 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f43c433784f45d0a8f5df03c0ae1667 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcec3a5911d64d549568d75d808ae1bf : SUCCESS in 5m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0eede94f569344749d5a16041bdc786f : SUCCESS in 19m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/03bb4e4ec0074192b48bffe443de4753 : SUCCESS in 34m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/601eeba92bcf4ae49839ffc75669a81f : SUCCESS in 1h 02m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ea81df3ddaa94d7f9de1de87f06af93b : SUCCESS in 42m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/798eb75a012d4b8d9adfcfb3f9da69b1 : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2abbb9d584271258977fe7f680ea61b726c07d37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 11:00:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c86725d7c6a4748bf033d6cc7a77fde : SUCCESS in 1h 19m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf336291ae1449b4bfcd7b71da74a6e5 : FAILURE in 14m 59s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c4acf2b9f3a3490dbb48ad2123186a14 : FAILURE in 20m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dd0ed737e3f4467a148822d24fbc355 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0795b09aa4ee43d1babb602da44b621c : SUCCESS in 5m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e7f0ae3fa67045b6bc85518f339f629e : SUCCESS in 2h 19m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bce11641c42bba580c50d50f7b3ccfdc250d65bb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-27 11:31:17.000000000","message":"Patch Set 1: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d24d81a97434d177553cf75ed119323f15cdd9a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 11:36:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37734a1a92cb644191fd1c637071ac1216312754","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:08:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/573871bbf44f4d209a6a13d898897d2d : SUCCESS in 49m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/34a51ea39ea64af6b27d28bc5954a9e1 : SUCCESS in 1h 12m 11s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/aad0864173154e3f8f950bf459bfcf6b : SUCCESS in 1h 25m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb16f00ee63437e8efe9c75da029f81 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea8b64db3d35403d917935d2a13f8047 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0322eb6245d4a3e8782b3e4a25c5949 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52189c35df724c07b62d7c7e967f50b5 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a028835438254d5699f432ad8d8d0223 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce7d733de4dc453caeafdafad7568391 : SUCCESS in 31m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c2a9d9c61fc8417298e3f62182808061 : SUCCESS in 58m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1eb5293c41504087b413d7868bd83a58 : SUCCESS in 1h 06m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d22cd7c85cadbccd1f16f4b1f0b41cc993f00653","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:08:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"be83bb77009a192fa16935d179be0eb617d2e698","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:10:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b599f09d08fb48c79fb6f204082c1653 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a14f41d723804bb58e9a93d20b9f0467 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2171f0a7babe1e25381a430a079198e079202b67","revisions":{"2171f0a7babe1e25381a430a079198e079202b67":{"kind":"REWORK","_number":1,"created":"2022-01-27 08:38:43.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/10/826610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/826610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/826610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/826610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/826610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/826610/1"}}},"commit":{"parents":[{"commit":"ff75a2e1f960e9420a5232aebcb6b92fb834af18","subject":"Merge \"Mitigate two Log4j vulnerabilities in Apache Storm\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff75a2e1f960e9420a5232aebcb6b92fb834af18"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-15 07:12:56.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-27 08:38:43.000000000","tz":0},"subject":"Use distro provided GRUB efi","message":"Use distro provided GRUB efi\n\nUse distro provided GRUB efi instead of creating it like ironic GRUB\nsetup doc[1]. This avoids below ubuntu ironic-python-agent images PXE\nbooting failure.\n----------\nerror: invalid magic number.\nerror: you need to load the kernel first.\n\nPress any key to continue...\n----------\n\nThis also fixes x86_64 uefi pxe booting issue by setting up GRUB efi\nfor x86_64.\n\nBesides, GRUB setup only needs to do once at bootstrap stage.\n\n[1]:\nhttps://docs.openstack.org/ironic/train/install/configure-pxe.html#uefi-pxe-GRUB-setup\n\nCloses-Bug: #1879265\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n(cherry picked from commit b6e057d694d90d50ea278411d0edf20afe427e13)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2171f0a7babe1e25381a430a079198e079202b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2171f0a7babe1e25381a430a079198e079202b67"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
