)]}'
{"id":"openstack%2Fkolla~826611","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1879265-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-01-27 13:08:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591","subject":"Use distro provided GRUB efi","status":"MERGED","created":"2022-01-27 08:39:00.000000000","updated":"2022-01-27 13:10:01.000000000","submitted":"2022-01-27 13:08:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826611-bug/1879265-stable/xena-stable/wallaby","cherry_pick_of_change":826610,"cherry_pick_of_patch_set":1,"meta_rev_id":"d645652c209d1fe549ffce7c3c764ba420e67356","_number":826611,"virtual_id_number":826611,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 13:08:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-01-27 09:10:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-01-27 11:31:31.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-01-27 11:31:31.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 08:39:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22997,"name":"xinliang","email":"xinliang.liu@linaro.org","username":"xinliang.liu"},"state":"CC"},{"updated":"2022-01-27 08:56:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-27 11:31:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"bb1f8277641c2d66260a2bc111a1e4b72872bcac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 08:39:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.\n\nThe following files contain Git conflicts:\n* docker/ironic/ironic-pxe/Dockerfile.j2\n* docker/ironic/ironic-pxe/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"58cff326f172744a6330fbec76852540b18535bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 08:56:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1766afac850a4f2c85cf172856b04e57 : FAILURE in 5m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c67210ff67e048878e09115226871204 : FAILURE in 6m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ffe69c4a96804616bc34cac4626c7667 : FAILURE in 12m 05s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e4e1bd4f31d54ab7a83cde56b0ad8ac8 : FAILURE in 10m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/34c79f7df79c4766aa9c894d45ba1e27 : FAILURE in 12m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3a3687db234247fb81598f9585f46974 : FAILURE in 5m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f7f7bd1b68bf43bfa879b83ab00fd0c7 : FAILURE in 4m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fe05a9cca3ad4914bb454d75fecfa382 : FAILURE in 12m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b6870eef1349be8b80c08aab866994 : FAILURE in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3116626562414cd18e0da094a62d3eb5 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebd5861668d44b1e8ada44e461c76f2f : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26dd6b9bf1bc49cd828134a09a4598ec : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6669e9f9772149838eef0944579ebd51 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99399d20221c46278e14ffcaca589d55 : SUCCESS in 5m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c4a4add21a0404da93f7093c88249ab : FAILURE in 6m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9141266fc3a449f8948b4dffebe399de : FAILURE in 5m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/69afc9d0c07b430e8052251a57b84dd4 : FAILURE in 10m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/070c0cfc07784a3da25e1dbab58496e8 : FAILURE in 10m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/972cb7211f6445118325368786f6c827 : FAILURE in 8m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7273a9d7d4a73f392fda88227d4e3a09298209e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 08:58:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1ed998841d484d89bebfe14e5aa40670 : FAILURE in 11m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2fb6efb38004033961bf57ad5c4bae0 : FAILURE in 8m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9c63f01d60a42d18f6212ba7109734f : FAILURE in 17m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e6b5bb7364a940df9d6a2e951d7207e2 : FAILURE in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5e714b041d7aaa4e9beeae71230d12c4282801c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:06:14.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e9d4b6ab46c6b5e10d6987a512565fc0c4096ae5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:09:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b0bc3c114b0d402496eb42757e4792057ef935c9","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:09:51.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"0f8511c30981c586fca73dd599ea1218ad10e3ea","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-27 09:10:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"feae7ffab1c57bc6b070a68826b71649bf1b245e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 10:29:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d0959028cec24cdc8ddd27f601a6be6e : SUCCESS in 38m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b089911e164742b59b1324518be9b702 : SUCCESS in 47m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/20172bdaa7bc4a8785bdf1113d437a85 : SUCCESS in 1h 08m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/53373cb2e8974e47a514791f9097acce : SUCCESS in 1h 09m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/326814e0b2404abc9f4df667dcd64b2c : SUCCESS in 1h 18m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d439751b85ce453ea6569eaef5be594e : SUCCESS in 34m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b6ed1ac469614a169e95a145eb513e94 : SUCCESS in 22m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/27c00e9f12ed4b33957cfc353863f2ce : SUCCESS in 1h 00m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f730775493f046e2b83095b78c5fae16 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81f7ebeeba2947dc941fb68a39244263 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab97d8be500a42f981fc74be5ed8aaa6 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45f0cb18d83b484290548144ff4103d2 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1766d3cd19a4bab9bfcff73dbe04397 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c680de20cff4af093bbfb2b2c741654 : SUCCESS in 4m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9a532c397efe477e809d8205e1704d69 : SUCCESS in 18m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba4c4373060c46c1abd1b84f5902bfca : SUCCESS in 39m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba577a4301cc4da599c6c49416aff360 : SUCCESS in 48m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6969bbf428d645b7a25d8e59b7c7dc67 : SUCCESS in 53m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/563d3319321146718670fb5911c71ec9 : SUCCESS in 55m 36s","accounts_in_message":[],"_revision_number":3},{"id":"248dcb74cde308940952bba04012914ef9f30b1d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-27 11:31:31.000000000","message":"Patch Set 3: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cb8d226eb1f4c160cb0c4479045f0e8659cc3fab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 11:36:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"88a21eeebe8a6c769dbef99e5f94a287f16afcda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 12:14:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/79cb2926fc4d470593f58f92913a09f9 : SUCCESS in 1h 16m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b53d90fcf71045b3b127b1efa2d7e987 : SUCCESS in 1h 03m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eaa088571ada4bdca95e286a6b953608 : FAILURE in 2h 57m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/400874321cb44f7cae755f6d0419b09a : SUCCESS in 1h 09m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"778156953ebf4881e91ab51670946bc1d257d904","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:08:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ca73fa114f72428889fc491366b3c9d9 : SUCCESS in 54m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/be7cb084f45941389e8076a87e1a1fbf : SUCCESS in 1h 09m 50s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e5fa5806036f447ebfdae0b9bc0858a4 : SUCCESS in 1h 08m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a17213b2294a1185fdccb0dc5a23a9 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7b8346c865f4da0a2dd0bf986af001d : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea7a0d4d25a5454a871da2796a45e873 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c12acb7eca8c46a299d15bf5d278c761 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1265f239f6340b8b8a950c21de24933 : SUCCESS in 4m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6ede57d731a442483c2d5560ac78131 : SUCCESS in 39m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/662371a9ae9642fdaf29f576ea28dd71 : SUCCESS in 1h 16m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/31b742c85cdd407aa6bfd704f815269b : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":3},{"id":"0c38463775a5d4688e70bd80f90b61bf78450c5b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:08:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d645652c209d1fe549ffce7c3c764ba420e67356","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:10:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957cf276a35b49eda163ada79dab5278 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26e11541f4464d0bbd5d9fba8e8e51d7 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"67e4f50bf7cd7166b27dcd38611215327b722cb8","revisions":{"a53c93c1306862ae28b1d533150c24e99e86f83c":{"kind":"REWORK","_number":1,"created":"2022-01-27 08:39:00.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/11/826611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/826611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/826611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/826611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/826611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/826611/1"}}},"commit":{"parents":[{"commit":"6d8b6c585667cb2d8a1221c17c6120a1e3823791","subject":"Merge \"openstack-base: drop anyjson\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d8b6c585667cb2d8a1221c17c6120a1e3823791"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-15 07:12:56.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-27 08:39:00.000000000","tz":0},"subject":"Use distro provided GRUB efi","message":"Use distro provided GRUB efi\n\nUse distro provided GRUB efi instead of creating it like ironic GRUB\nsetup doc[1]. This avoids below ubuntu ironic-python-agent images PXE\nbooting failure.\n----------\nerror: invalid magic number.\nerror: you need to load the kernel first.\n\nPress any key to continue...\n----------\n\nThis also fixes x86_64 uefi pxe booting issue by setting up GRUB efi\nfor x86_64.\n\nBesides, GRUB setup only needs to do once at bootstrap stage.\n\n[1]:\nhttps://docs.openstack.org/ironic/train/install/configure-pxe.html#uefi-pxe-GRUB-setup\n\nCloses-Bug: #1879265\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n(cherry picked from commit b6e057d694d90d50ea278411d0edf20afe427e13)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a53c93c1306862ae28b1d533150c24e99e86f83c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a53c93c1306862ae28b1d533150c24e99e86f83c"}]},"branch":"refs/heads/stable/wallaby"},"fabe59fc6bbfab57b2002fc5d6797211d8934aca":{"kind":"REWORK","_number":2,"created":"2022-01-27 09:06:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/11/826611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/826611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/826611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/826611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/826611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/826611/2"}}},"commit":{"parents":[{"commit":"6d8b6c585667cb2d8a1221c17c6120a1e3823791","subject":"Merge \"openstack-base: drop anyjson\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d8b6c585667cb2d8a1221c17c6120a1e3823791"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-15 07:12:56.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-27 09:06:14.000000000","tz":0},"subject":"Use distro provided GRUB efi","message":"Use distro provided GRUB efi\n\nUse distro provided GRUB efi instead of creating it like ironic GRUB\nsetup doc[1]. This avoids below ubuntu ironic-python-agent images PXE\nbooting failure.\n----------\nerror: invalid magic number.\nerror: you need to load the kernel first.\n\nPress any key to continue...\n----------\n\nThis also fixes x86_64 uefi pxe booting issue by setting up GRUB efi\nfor x86_64.\n\nBesides, GRUB setup only needs to do once at bootstrap stage.\n\n[1]:\nhttps://docs.openstack.org/ironic/train/install/configure-pxe.html#uefi-pxe-GRUB-setup\n\nCloses-Bug: #1879265\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n(cherry picked from commit b6e057d694d90d50ea278411d0edf20afe427e13)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fabe59fc6bbfab57b2002fc5d6797211d8934aca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fabe59fc6bbfab57b2002fc5d6797211d8934aca"}]},"branch":"refs/heads/stable/wallaby"},"67e4f50bf7cd7166b27dcd38611215327b722cb8":{"kind":"REWORK","_number":3,"created":"2022-01-27 09:09:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/11/826611/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/826611/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/826611/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/826611/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/826611/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/826611/3"}}},"commit":{"parents":[{"commit":"6d8b6c585667cb2d8a1221c17c6120a1e3823791","subject":"Merge \"openstack-base: drop anyjson\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d8b6c585667cb2d8a1221c17c6120a1e3823791"}]}],"author":{"name":"Xinliang Liu","email":"xinliang.liu@linaro.org","date":"2020-05-15 07:12:56.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-27 09:09:11.000000000","tz":0},"subject":"Use distro provided GRUB efi","message":"Use distro provided GRUB efi\n\nUse distro provided GRUB efi instead of creating it like ironic GRUB\nsetup doc[1]. This avoids below ubuntu ironic-python-agent images PXE\nbooting failure.\n----------\nerror: invalid magic number.\nerror: you need to load the kernel first.\n\nPress any key to continue...\n----------\n\nThis also fixes x86_64 uefi pxe booting issue by setting up GRUB efi\nfor x86_64.\n\nBesides, GRUB setup only needs to do once at bootstrap stage.\n\n[1]:\nhttps://docs.openstack.org/ironic/train/install/configure-pxe.html#uefi-pxe-GRUB-setup\n\nCloses-Bug: #1879265\nChange-Id: I8be5bdf5f1a62751aefe6bd0959e8f558fcfe591\n(cherry picked from commit b6e057d694d90d50ea278411d0edf20afe427e13)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67e4f50bf7cd7166b27dcd38611215327b722cb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67e4f50bf7cd7166b27dcd38611215327b722cb8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
