)]}'
{"id":"openstack%2Fkolla~826948","triplet_id":"openstack%2Fkolla~master~I68a562f15bf4fdda82b86c446e14fc8b5594413a","project":"openstack/kolla","branch":"master","topic":"support-skyline","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-01-30 23:34:41.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-02-01 07:11:14.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-01 07:11:14.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-02-01 07:11:14.000000000","reason":"Change was submitted"},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2023-02-01 07:11:14.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-12-01 12:28:05.000000000","reason":"Reviewer/Cc was removed"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-02-01 07:11:14.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-10-18 13:08:39.000000000","reason":"Reviewer/Cc was removed"},"33196":{"account":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"last_update":"2023-02-01 07:11:14.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-02-01 05:57:23.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}},"hashtags":[],"change_id":"I68a562f15bf4fdda82b86c446e14fc8b5594413a","subject":"skyline: Add new dashboard","status":"MERGED","created":"2022-01-29 02:18:05.000000000","updated":"2023-02-01 07:12:17.000000000","submitted":"2023-02-01 07:11:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":141,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826948-support-skyline","meta_rev_id":"7dfe8156ebce0a12418dd1064d1378fc7dd86277","_number":826948,"virtual_id_number":826948,"owner":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-01 07:11:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":2,"date":"2023-01-31 01:18:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-31 02:22:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2023-02-01 05:57:23.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-02-01 05:57:23.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"}],"CC":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-29 02:25:08.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2022-01-29 04:25:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-29 06:18:55.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-01-29 08:06:56.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-01-29 08:06:56.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"},{"updated":"2022-01-29 08:06:56.000000000","updated_by":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-08 04:06:00.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"CC"},{"updated":"2022-02-08 04:06:00.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-08 09:20:06.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"},{"updated":"2022-02-19 08:47:11.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"CC"},{"updated":"2022-03-07 10:42:35.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-05-30 02:18:40.000000000","updated_by":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"reviewer":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"state":"REVIEWER"},{"updated":"2022-05-30 08:20:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-09-30 10:31:14.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-10-13 07:38:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-10-18 13:08:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"},{"updated":"2022-11-29 11:21:28.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2022-12-01 12:28:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REMOVED"},{"updated":"2023-01-31 02:22:53.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"a3f9b5718e7221a221240b19888bace9b55e139d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-01-29 02:18:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccd1c4276a797c1684a1aef3ec7ec13e90f656df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 04:25:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/33954fc4bc4245d6a6c87fa1cd174c7a : SUCCESS in 28m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cb288a9e492a4054bdec3ba090517b49 : SUCCESS in 35m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/48885263f49e4351aed87ab7c7f74b42 : SUCCESS in 1h 13m 09s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/69626adba454423bb26ce3df546874ff : SUCCESS in 1h 10m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/939c34975930471b8143fc6c33c37dc4 : SUCCESS in 1h 33m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b0c05a5931a4480aaf2e961460b714ec : SUCCESS in 30m 44s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4284c7cb77474b98b8edc4013e0674c1 : SUCCESS in 18m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f9d2bc47b591419e910e09039bab87e2 : SUCCESS in 58m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/85b86191b6504a1484ba3bb13b1e42a3 : SUCCESS in 1h 18m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8ba9a36e3bf406f92cca1796e17ec37 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8d603aecd774eb9882fe6df810b2d50 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c6d2731b7f041229806bbd1738758e2 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be55879189a410fadbc157164a2fc84 : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae21bb9712e042da864c3475fbfba9df : SUCCESS in 19m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c672c05a552a464da6ed8d4d62c6bcd7 : SUCCESS in 30m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e791d0b9098b4c57af97a57ad08de9c9 : SUCCESS in 49m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c8eff9099ddd4bae87b3f02db35bd8b2 : SUCCESS in 54m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1128e02cb77b489594437db5c11f20cf : SUCCESS in 1h 06m 34s","accounts_in_message":[],"_revision_number":1},{"id":"dfff0316e1e335a11b4720021201bdaef38e63ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 04:54:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a98fd625efa146afbf8df5ea73636c75 : SUCCESS in 2h 05m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e09e37e47c1345a0be78b098557e0868 : FAILURE in 18m 09s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b775024f714a4900bb92c3229bf18c5c : FAILURE in 31m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d5af4592be841af94d45d93235a798c : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9b73370dfb94c44beea735edfe9e7da : SUCCESS in 7m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/57eaa7767c96443a94f30d903810d66b : SUCCESS in 1h 53m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a6a4defc1e1326da51615d92acf285464077820","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-01-29 06:16:56.000000000","message":"Topic master-ts removed","accounts_in_message":[],"_revision_number":1},{"id":"78f39a9177e8f84c51186817741556cd6d1bf01f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-01-29 06:17:00.000000000","message":"Topic set to skyline","accounts_in_message":[],"_revision_number":1},{"id":"19414a259b12b26ba879cd01be470de7de26b78a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-08 06:59:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2d02a8e6f2d19cd6f7a51583673b6ba7e568af7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 08:14:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3ccc11f838c84c29862ac70e923f53ba : SUCCESS in 29m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/69f5f6e9070445e2aa53955c592d8220 : SUCCESS in 33m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ad32bde588984af88501c068fda77845 : SUCCESS in 1h 13m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/65d1ebce5bca436293fc52e741e23095 : SUCCESS in 1h 12m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/307639262e50464e8a80bc27826b9f1e : SUCCESS in 1h 06m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f749e1d3076848f89e7d76730fe594a1 : SUCCESS in 25m 41s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2dfec2fc8e334f26b58a4492e2bacd7b : SUCCESS in 17m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d098452878504d6bb25d3a1a31578007 : SUCCESS in 55m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/93c8a5c7a2304e83939d828dda216a82 : SUCCESS in 56m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82450f371b64b1f933cbeb63318522f : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8436398bc1b94195887542bc14887fc9 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c706b497b4e9445a8e1b5a4aedabffcd : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e5a9dfb36844c41a69a5c44808dd372 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f62dc18f4c3a49a79313019f96163359 : SUCCESS in 15m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf99d1155c204d9f89d1406fc4fdb6da : SUCCESS in 30m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d75fcf54a7c0433cad0ec1a3b560c660 : SUCCESS in 48m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c44b330d25ac401fa7438e51dc07d87e : SUCCESS in 51m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc7fa84b5e794e6fbe551546f250b38a : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":2},{"id":"1d6825c5c17d1b7c316e212482c176f097718afe","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-08 09:20:06.000000000","message":"Removed reviewer Radosław Piliszek.","accounts_in_message":[],"_revision_number":2},{"id":"93b586a02692ce7f19d4eb7e49ec8ebc04b52c36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 10:04:24.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab3addb0a1f44f60a865bf85c587bb06 : FAILURE in 3h 01m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e31a8599fdd4c20b861f1222deb3d68 : FAILURE in 37m 31s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf21ffdb9ae944328b510524c9f51be5 : FAILURE in 41m 21s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67fda2b1af9847fbbf6eca49c169b91b : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b9f7f3a5a244f4e80252549fb449fd9 : SUCCESS in 5m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/60ab79b534544a509691ec80b5248d05 : SUCCESS in 1h 24m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2dab90ae4692a3303d47d747bf0d55dd24786868","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-08 14:55:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8d1656b9903827c8b9c836606c5baeedc44aa2e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 16:29:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4eddb8cd1b7a4ae78f3f05f6967814d4 : SUCCESS in 33m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b088c5eed1354c24ad9d2f715ac434f8 : SUCCESS in 40m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a9120ab9ee8a4547a0e43abdfdd43222 : SUCCESS in 1h 19m 59s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8c25854839b24b20b0d51d72b02942bd : SUCCESS in 1h 17m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/21ab896831c24a54a40f2faf3ec62ee3 : SUCCESS in 1h 22m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6ff0902ec69d42ef92c506e1bab08db0 : SUCCESS in 27m 33s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4ed5e644ab694591b3b5ce549b2f1e92 : SUCCESS in 17m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7cafff213044470490ef9f374ed7afd9 : SUCCESS in 51m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f2b7438034de4449b2b0576fbd5423b2 : SUCCESS in 56m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e936de9ac31453e9615327957072303 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85d8256da0b145ba8ea88ab813e4ae88 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de37b690e24c40f49b9d2e943c4447f7 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3888fb0385d42dc85ae0e6bef9f09af : SUCCESS in 7m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e8f9a98acc14aa9af55376d87a6e22f : SUCCESS in 20m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/430af7abc84342c1bd1710b2ff27858a : SUCCESS in 38m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4240dc5a99fb4be6b12d2a3eb1c116d1 : SUCCESS in 49m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c5633bddbf349328fb3b26c8e0ae00f : SUCCESS in 43m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bfeb9e30a7834d6b8a26a59353abbdc2 : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":3},{"id":"80670136bdc7586461e4d91d8ce90271886736f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 18:24:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/580ed71e286945e092c319275596c75d : SUCCESS in 2h 16m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/30dd727fcbe24b178ccf4bb757570147 : SUCCESS in 1h 37m 25s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b207ce7601a472087d56e618af572c2 : SUCCESS in 2h 10m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53566378793b4f938e3e1a306b024abd : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa4506f7e6b14d20a8a8c0671218af6d : SUCCESS in 6m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f56e5da7a76f484a9b7daaf3a9cd34c3 : SUCCESS in 2h 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"726a1b9c117f469fb1339398d4c29076bb293310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-09 07:29:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2ea4bf8a3a3baa45d0c321fa42d31b3f6e8b3d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 08:49:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/048aecac993143e9b42bab150cc19e30 : SUCCESS in 29m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9d1dabefc55a422ebde3abb79a9185c7 : SUCCESS in 34m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b1128a4a7363488894cd292c41e067d3 : SUCCESS in 54m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2e25b418984041ac8af730acc2fe5511 : SUCCESS in 1h 00m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5f357bd791754cb1898fc33eee407eae : SUCCESS in 1h 09m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/66f66fd4c8c2471cba0a7a8ca167748f : SUCCESS in 24m 38s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b44d47f4b57f4c10ba86fc7fdd25de82 : SUCCESS in 16m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f75e6edfbb5840b78b72fea50100c43c : SUCCESS in 1h 02m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1774c4c06e8542579f1ffb94f5eb4f32 : SUCCESS in 58m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba48981648244e10b8cac5f85b6a9d46 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/666410fbc0d5440188b2167619320f93 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa305ec9bc5a47c8830ee1a0e6c2501a : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79038f640f1438dad1d21a0d963b754 : SUCCESS in 6m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/af2cdf77cfab42aa84c1a14ab9317d7f : SUCCESS in 17m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8743d17eb074f40a6586571fbdd3e93 : SUCCESS in 23m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e9a1932e8e88489a8dbebe165be1762f : SUCCESS in 48m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e6abd8e29ae4215afc2d5e6f929fd42 : SUCCESS in 47m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4f4358431deb4983b570c6b1d54213a0 : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":4},{"id":"f89ad9e021f1ae2a21795cc2e031ada0d539e1e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 10:46:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c361ee8b8a204fc7b483eb246c11e501 : SUCCESS in 2h 13m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/31c026d7352a49fba53746625a3c4645 : SUCCESS in 2h 15m 27s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/70c322ea5569497ba9857e862b44065b : SUCCESS in 2h 55m 21s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92d8cc2deda14597ba055c94003dd891 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/241624a412244cebaab588977b992ba9 : SUCCESS in 7m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed936b4fd5d34d03b0a1d890b0a31918 : SUCCESS in 2h 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fb7bf202c264edea9c5ab7437ba056268ef25a80","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2022-02-10 00:46:48.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"917104726b029cdc3fb112d5e1e6ac7014f62de8","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2022-02-10 00:48:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c8ea36fa56ccaf9d39759d5a4b389789eadd5834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-14 05:39:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"09308caa3e3cbe5b10b48cd92b8159d33507d74e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 06:57:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0261f9bdc1ac4ed1a6374b7a314eadf1 : SUCCESS in 27m 08s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/00c59e4e2b934b8e892124397bfa5f5b : SUCCESS in 35m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8abc5b66a92b494188f2be34acdce869 : SUCCESS in 59m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/37f43059a7a14a9d81ed77705501a3e4 : SUCCESS in 1h 08m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ef43eec8557a4e4793e565b5294a4220 : SUCCESS in 1h 16m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0ea6ade9fe9242caaa4bdf778f8998fa : SUCCESS in 31m 13s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/92739dff50794b4a9ab8eda2be20f36c : SUCCESS in 20m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c662b422328e46d6a3dce12fd711875b : SUCCESS in 53m 00s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/14c811fd722149bcbdb6a83c7d61a91b : SUCCESS in 1h 09m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0ecb8112a3c421ab0bbf5339a328925 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f50b987483b14156b7e918b29ba9c4f5 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c35c486c9564a348fa0ddb16a50be16 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7acd3c0756485dbdf339506ea56efe : SUCCESS in 6m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fd8f6ef1b72f4259bb6b5a30bbd8d2fd : SUCCESS in 19m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/27406365492d4c70807a4ade6c2f6154 : SUCCESS in 31m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/11d6a5a60e01415aa33f2332d49f6814 : SUCCESS in 56m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6d228812b203416389520461992449fa : SUCCESS in 52m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b6f915cad40a4b28b6fe867661568462 : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":5},{"id":"2816e74c471fed01fb4e0d063e2ff9dbfddd7213","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 07:59:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc836e4ca38a4b63a40ca3921ed3a9cc : SUCCESS in 1h 57m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f0b10c09632e4e3e8ce2d4f2188296bd : SUCCESS in 1h 21m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d31e9d4b8dbb418bb93b3af673f6b857 : SUCCESS in 2h 07m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cef387eabfd4ac9b00141bfb25db15c : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c027a2f06ac417a977ec9289e5fe77c : SUCCESS in 5m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c754c83d1704801b9ab2fb3691b6b41 : SUCCESS in 2h 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2279e696267521f8d3c7d827d721a420cb499ae4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 05:24:33.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fb60a303090be16bfededaeb4177d0b2bf19f6bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-15 08:12:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a7c7538cc18dba7e136a9cc503a1b73318c0c5fd","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-15 08:18:12.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"32e0a52484fe038773de9b63d7693eccf949e61d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-15 08:24:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"494dc2d335d2cf2531721f0914ecba51cfddab13","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-15 08:35:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f32f847f9efcbfd89634d4649731f426cad10218","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-15 08:35:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ee2c599c67077592ad2de8a1cd97176d3efb413c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 09:51:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b26df2a096c54446b80af63626fe0dee : SUCCESS in 29m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7b0cf9252c8948f6b4c7c55ac8626275 : FAILURE in 52m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2d7bded2b81a48b4b9d1ffc0abf3a6f0 : SUCCESS in 1h 04m 59s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6217b217feaa47d29bf809b936cd6ef4 : SUCCESS in 58m 37s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3b43ca0cb31f47e48148b5bd9c9273fe : SUCCESS in 1h 13m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/32fb1b543a41473e843b856f87a489d1 : SUCCESS in 26m 05s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b9bb466f4bc544b7a06e64968c98cf4e : SUCCESS in 15m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1b5c0d5e44fe4deba2543a37e003edff : SUCCESS in 1h 00m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/77534edf866645b68c1d84f35c6956b3 : SUCCESS in 59m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2304f06c45b404dad868b03b6c14792 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d1f922785c9498e8afca83d618110ba : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b784ecffed8f494bba9d30abe4835b72 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92b56782fa94f6783825f98b6b21b0f : SUCCESS in 7m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f58b846cc5cb4301a599eb5ba257c0ae : SUCCESS in 26m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/59db481fcca34898aa02714f2745fee6 : SUCCESS in 28m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bee00dc3ec9e4890b419cda6fc95eb9d : SUCCESS in 49m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/009fb576b7e44e24ab211bf3abb4702e : SUCCESS in 54m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ad4a4cabb9d045ccae7fd6ac092f15c8 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":8},{"id":"72e2bc0f3db69de75cfbbc7183f86e2836a3f86d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 10:42:37.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/591e89d5b71c4770936117b204de3ffc : FAILURE in 46m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b840df249144cedb83044c26a2235dd : SUCCESS in 40m 50s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/30d2d8f0494b4b01aad04454cdc2b05c : SUCCESS in 1h 22m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2448661bef664227b60f78ed5ca304fd : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f263df8638804a38970552376f674468 : SUCCESS in 4m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/de76a05151494c5583ce75c00dfe894c : SUCCESS in 2h 04m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6b559b74a5bab109935e646c884dad020f14c524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-15 13:45:24.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4b66c8f71c187b4008af24bda4d2bcc6f6e3c9a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 15:26:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/213a995eb7a84b0087942bb29bd03cc5 : SUCCESS in 48m 36s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/caa924b8161f4f04b883429b36d8db3e : SUCCESS in 53m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bedfef57a0c24ac6aa17dab5a2b0330d : SUCCESS in 1h 17m 52s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6db7c5d0f8e04d4a970f9f03c81db020 : SUCCESS in 1h 22m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/467e87208fea4ba1807f47a016d6a9e4 : SUCCESS in 1h 32m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7d56ee0a5e4e47fdb740982c8446462e : SUCCESS in 33m 30s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ce121345403f4fe6bc67a0e8e53c22a3 : SUCCESS in 17m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/94931c7bc200473ba8f91661b13b2518 : SUCCESS in 1h 00m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/42925c69a8014512840349c5768b0dc7 : SUCCESS in 1h 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/572ac582609e4821a8dc604c0be323da : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dcddcf8a2394efa844b7af19aa52c59 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4ed1fc935ff414193f67ddb6924923c : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d246a90d323c4ecd90de247d2986bc66 : SUCCESS in 8m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1080faa17e794c5db245499d15e9a3dd : SUCCESS in 16m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/78bbd08e10644d29a2235701e66cd3e1 : SUCCESS in 27m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd6c4f595e424332863dd98e1ecd9e8f : SUCCESS in 52m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d734ea6559b7459bb320ca2abfe6204c : SUCCESS in 46m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b83ba8a326da450daf510aacf4cc2469 : SUCCESS in 1h 26m 13s","accounts_in_message":[],"_revision_number":9},{"id":"7b07511b7b4e218531b2759d198e541b35ca73e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 17:37:43.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5f24842fc35a44d29cfdb07bc95d72d2 : SUCCESS in 2h 46m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e62124d49ffc4071828666b2c7352920 : FAILURE in 2h 09m 17s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/24755381ffdf4b07b599bf27fabf3272 : FAILURE in 58m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8925367a69f4165b95c384f78f1209b : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2bc462d31ef04b70923ba14f5943dc0c : SUCCESS in 7m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0c03d95048e0433d87bc69e9a139ef6b : SUCCESS in 40m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1744eaff0f37ae785d9c684432c07505650fdef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-16 02:12:00.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"94f0cbeee71087d7b61756dea0cfe75acf310a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 03:51:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8064ae3c311d4d609d223988d03f6803 : SUCCESS in 31m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/50e9ed317f7b4b6b82c292403c51747f : SUCCESS in 58m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/98c13bd5285b4be6888a1eea49801fdc : SUCCESS in 1h 18m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9fb3e59dd3484f35845e29fe31e095fa : SUCCESS in 1h 13m 25s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3c733d5ff46e4a5792020402a9726c16 : SUCCESS in 1h 21m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/60cfcba4f2ae45c3ad792ce8d9ba8997 : SUCCESS in 36m 21s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fb56f93644fa4a40a555e4caa3a509cd : SUCCESS in 20m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/761859fa6bad475aa1ccb1ce810425d8 : SUCCESS in 56m 30s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3e6277e5d2c44b06bb1b175ded2188ac : SUCCESS in 1h 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7c0239b185646d6807abc36468a8528 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16b82b66d8f24ed89fcae39a5f9e9510 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64f6a3cec6ed4fba88c7be502d64171c : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946fe86cbf9f46c7befaf24f2dd49480 : SUCCESS in 7m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/12ce3d69462546529cd5ff5d16288eeb : SUCCESS in 19m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c3972de92844826aa44ac87ed6a0949 : SUCCESS in 36m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b4e863fbf974c2bbc8499d4ad798aee : SUCCESS in 50m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/375baa0eb2a9427aae13126ccbbf4cef : SUCCESS in 50m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e056d5ac632e4e7494b05bdf89442325 : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":10},{"id":"5dda0d9937cc81e3eb2c7cbf53f5353c4f6851ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 05:24:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e3758abec894ad7b6204dfcfe184f56 : SUCCESS in 2h 50m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e6a3221af98a4bdd88c7a03a40ecce2b : SUCCESS in 2h 39m 39s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/51b8a5c960ed48f0a269ff7fbc91ba1c : SUCCESS in 2h 56m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cda627b4f6343efa1dd9e8eea4ec263 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e6358231e2b04399b7525d4984ffc74d : SUCCESS in 7m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/19bce16f4904444da083996de631e88b : SUCCESS in 2h 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"07c1c13a4515c07e6ef900732c54e5a3f4205247","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 06:41:39.000000000","message":"Patch Set 10: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"13609c1d78f91b61713edb55bc29e11925667609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-17 02:08:59.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"af392b9f6ba631ae554534603cdf23a1b1d95c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 05:00:47.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0494d7312cbf466d8969d6253b27b537 : SUCCESS in 31m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/30739e68aa8e442990f664e54c2e166a : SUCCESS in 53m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ba0c52dbb61745a1b53488073cc358ad : SUCCESS in 1h 14m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7031d3001f4145609b7aa97d2e331d14 : SUCCESS in 1h 22m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2a02d55ae31f45f7bf05f97b3f6d258a : SUCCESS in 1h 26m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/85cb7a4bdc814ade8ad234cbd50b48dd : SUCCESS in 29m 15s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/23bd998a7c644f8ba1ff3d4193ff5922 : SUCCESS in 19m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e06487b095e440a2b10f976f97fb36ac : SUCCESS in 58m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/17798308091142ef91df04b0f962653b : TIMED_OUT in 2h 32m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d9a666a2cc4413bf2b7a64734250fe : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/566563baafaa43ca9f29a6a558f73295 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8988d0dfacf4123a4229732e5881b80 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5311ce38655f4ba6a99b370d8d63f2f9 : SUCCESS in 6m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/90b6b6aa5f1e4037959556bf4a8dd389 : SUCCESS in 21m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/575dd63c3853488db6f4cbf1eb1501a0 : SUCCESS in 39m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b839f3659bb4a41b35603d5178ba812 : SUCCESS in 57m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4cbe3ae3d5894d1ea0045ff483cd27d6 : SUCCESS in 58m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e3bf92cb5ae04f9a82c9c86db331660f : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":11},{"id":"b7c8d7637f05712eccd02815d56ecacdc9af00fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 05:13:24.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/20aefe5acf4a4e3c9737310371fcf6fa : SUCCESS in 2h 45m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/56b67c7717cc4d3d832f031967f2c722 : SUCCESS in 1h 50m 38s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/755267a8a5434c1985d56f17d378dcc9 : SUCCESS in 2h 27m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aafa15d20d784a87b5c1b01f6505d0cd : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3b8713d55bb4d4ab6f9cf3b79d3c6cc : SUCCESS in 7m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/da672530696e42308b4184b8ed298b4f : SUCCESS in 1h 54m 38s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4cc94b96a1d54f47a9b10a6938dbb7df9a72a0a2","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-17 05:58:35.000000000","message":"Patch Set 11:\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ae0ee60e3460369a960839c3792044c61b610d3b","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-17 05:59:10.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"157dc263dc3e58d5b9485a84cb4c93ff566609fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-17 06:01:13.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"0e0da1ecf63ae15839bb6cb2497bebcee747f10d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 07:28:21.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/02b5428fac6c49d48306e5157a80bcde : SUCCESS in 28m 14s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/965cc6c0a8f94d5682892f6510246ee5 : SUCCESS in 40m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0aaf146b0eaf4e41bc126f7d7d96fcff : SUCCESS in 1h 12m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d328a0bc89354bd680987cbe29289a44 : SUCCESS in 1h 10m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/87be460145e24965bd998226335935eb : SUCCESS in 1h 21m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9534c659835b43279f830a4f41d2b8e8 : SUCCESS in 29m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/08ab933277a9490091de50086461c742 : SUCCESS in 17m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/25720859974947f585e3ba173aa7dd3b : SUCCESS in 59m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0e74993d9f3544c5878a970efcdb1b26 : SUCCESS in 1h 05m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f288588e7496471480ad943ae079fd82 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/990af4586bd4486c8710d082753d8aab : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f20033be928944649b70be3d07da2ebc : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c01f46eb7fc41e682b3732cad3101af : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/54924c59aa51484ca121e91a59868b9d : SUCCESS in 18m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d94db942519c459b8cb1e12c81921fa1 : SUCCESS in 34m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b1828431d17d4b8a9ac2afb9fd02785f : SUCCESS in 49m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2483b6457d944339406ba1496d8a4c0 : SUCCESS in 54m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bff1a3f51a61448e99ed0a839c5c098f : SUCCESS in 1h 02m 40s","accounts_in_message":[],"_revision_number":12},{"id":"5bb0e285b03824e645e75a4a6a502e1c3f0fceeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 08:47:25.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/65abd0eb5ec34d6a94f00d6814cd5dd1 : SUCCESS in 2h 21m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9835128b87c2471e8a333e82eceed3cf : SUCCESS in 1h 48m 43s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/37861260ce6348ce9e690f8edfb3a300 : SUCCESS in 2h 22m 55s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e59df76229d64afeb147bc860784ea10 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a347edc24cc4458b4f3df4636dc9ddd : SUCCESS in 6m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c364a38e970441f849bf23d0a6bf525 : SUCCESS in 1h 31m 27s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c974bdc05cf8a03157c35a827ac09c29f79c82ff","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-02-25 01:56:26.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"a859e82f9e92d5587b4f731dc39d68716f089e11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-25 02:17:30.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"56e0def7643f8838b610fa0091ea2d4c51d542b3","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-25 02:18:13.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"768ef72c3e5d425bc7f357202a0feda28011f592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 03:56:00.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2242475192954268bf5fdbf877c7591e : SUCCESS in 29m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/63c102ae85924907923c4314649d01ba : SUCCESS in 45m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b9fae3e961224c51a8101aa5d23daf0a : SUCCESS in 1h 08m 54s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/aea94fa6b770409ca00db770b7439b3b : SUCCESS in 1h 09m 10s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/15e67fe043ee4546947706dbe1948687 : SUCCESS in 1h 20m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b1ca4ff97d7b47ecbf7305406e41ac79 : SUCCESS in 36m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2a12f0a4d1b546039ed8fa73341a60c0 : SUCCESS in 20m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f249fe43883548c2a7d79aaadb5d717d : SUCCESS in 46m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8502576766f5462eb7f0048fb37c2150 : SUCCESS in 55m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/951b844beaf74fbcbe49296ef0f40a5d : FAILURE in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eb985132f614715a2a5c1af4d3e8561 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2b743fcd6754f2494bc87cfa1202a17 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c94680b2d64aa6af9a9f2f1a5b9fc6 : SUCCESS in 8m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0af2bb0dd45f44d6b5d31ec40d3bbeaa : SUCCESS in 18m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/87e49b5c5c1543219ff2d6b23b0f89b5 : SUCCESS in 35m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aba2efb5ae564125a8ab2e09984d5388 : SUCCESS in 50m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bbfc888ed57348189e329c4ad5fcf0a9 : SUCCESS in 54m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6e69860e599145c0a0d995d4305b14c8 : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":13},{"id":"0ac62def0877a428a17be852661fc9ad13dee198","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 06:17:21.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd2b358c6fa14f5197614db309f08de7 : TIMED_OUT in 3h 19m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/262dde34326d4322ad388dc3feab14bf : SUCCESS in 3h 00m 17s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/24fa68580884422192ab71ca05d0ad11 : TIMED_OUT in 3h 04m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27ccf46852ad4d7fad2450421f467ef5 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0928aa01a8124d3ea0401f5eba00f21a : SUCCESS in 9m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f193317bb1094d329a7419037570741e : SUCCESS in 2h 24m 41s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"68809ca2cd6378688ce3dd874c36474c07b83051","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-25 07:41:33.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"04f9080c7140fa6ec65d3c6b71e29c71d65bdfb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 09:32:39.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e63b623bc2104cf69f7edbb6ad8d7b07 : SUCCESS in 34m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6c05b777ccd2484694fd7f1fdcf9072f : SUCCESS in 43m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ad91ab0a29354224a0aec8ba6c2985a3 : SUCCESS in 1h 13m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/493c42280a8442598323cc3ec3ada982 : SUCCESS in 1h 15m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/663104d7d90e4d3ab2de61c7797121f9 : SUCCESS in 1h 44m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7fb5d9e8547148a997e07f61b87cee62 : SUCCESS in 31m 55s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cbb3e1c269bb45afaaa4182bfebc83a1 : SUCCESS in 19m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/967f1fdb482e40ae91392dd645522893 : SUCCESS in 1h 00m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0bcd66399cfc46048dc8ade9e514eb72 : SUCCESS in 1h 07m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1883d570bf1f47af9e883409245a9f22 : FAILURE in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79f90c1ef8754632ad52e42a98d4b953 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4266162181e540f79bcb240787c937a0 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dfe84aee18e44d1a0bfccb205888f6c : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f243b6a4b29f4738a22a819731a79d0d : SUCCESS in 5m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db71643844a440d0af3d5a66835691ce : SUCCESS in 18m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5599bbd082b54cf1970eb0a369eb459a : SUCCESS in 32m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/01f3711891934224bda0ad18087ce602 : SUCCESS in 50m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/adb13250491340b5b66722a2505bc9ae : SUCCESS in 48m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ed123bee2dee47cfa063e6388aea2f87 : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":14},{"id":"80e00f59d55467c382578e191f84a9be4f41742c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"date":"2022-02-25 10:18:07.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"7e5513750cd78e996627fed5bbfc06e70724128e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 11:42:27.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6e03a4fe3761478bb59ed59296144369 : SUCCESS in 25m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2ad1d36552fb404d99240935b164f411 : SUCCESS in 38m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f388bb46b7e94d7d932997140ea2771f : SUCCESS in 57m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9482fea76f884e5b94110d7c939ce1d2 : SUCCESS in 1h 03m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cb3f6326cb0b4df3afa61bf76609ed70 : SUCCESS in 1h 06m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1cf23c292e3c4028839c0c6e3e532b65 : SUCCESS in 28m 58s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8107f12bc15b4b1cb4f1de94e7ff449f : SUCCESS in 15m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/18ee03c6041d4091a374c3d48643f354 : SUCCESS in 1h 00m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/49e462d74ec84bcba78bb6edc2542d44 : SUCCESS in 1h 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110491cbdaf64450b911c45a29ccee47 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/791a046add5f48ffb9fbb3265e9a346c : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aadcd1afe2af4afc9787e2fa6483f173 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b70061f133245c1a6269cd7d514a467 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43eab3a572e741d69fc9db80de59d72e : SUCCESS in 3m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8228953a74d49c6b276d552ed8c5399 : SUCCESS in 20m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad3a4aab104542ad82110c77a60c6acc : SUCCESS in 32m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de7efefc3003496ebee081e34bb0e4bf : SUCCESS in 56m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7e32a2824b1743c3a05e7015fe615e4d : SUCCESS in 38m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e220ca60cab144fe80425a2bfff5d72d : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":15},{"id":"7a0a27ad5780c1fae675634f9203d64bb650845c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 13:45:41.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a468905d67854cca94415ded74693f68 : TIMED_OUT in 3h 23m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/02c11f74138541eb8e21f7d8ef03c22b : SUCCESS in 3h 04m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4e64541715bf4ad9bfa1d8b2ca744530 : TIMED_OUT in 3h 03m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61f68144f4fe4c3fbb79c4d519766a2b : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55536508898f4f84a85a18b38f60a077 : SUCCESS in 7m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3114d67afde34a40baa0ada71b270996 : SUCCESS in 2h 47m 43s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"111b0ad876be24ded54ce57e1956cafd636126e0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-07 10:42:35.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"2bfa1e29f96635fc5c4096dbba368b886bd3276f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-24 10:54:03.000000000","message":"Patch Set 15: Review-Priority-1","accounts_in_message":[],"_revision_number":15},{"id":"1507678bc9beeffc8357b4ee02d16d002303dc8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-05-27 06:43:57.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"9d67e8b02cbd0ee111078bc07aed9be61bc5c7dd","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-05-27 06:48:14.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"b84f737cc4e5f282659082098cc9ff0823382709","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-27 06:55:11.000000000","message":"Removed Review-Priority-1 by Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":16},{"id":"95325756e27267e1901c51a2fab0a74d81c24b1a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-27 06:55:15.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b7e7f4dbc0d8c18f1be2bc0b9ab0feef1ff31cd4","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-05-27 07:48:45.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"581326051b02b086cea56d1c325917fbde22b190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 07:55:08.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/c33377ec002f4d03bc231343f176b961 : SUCCESS in 54m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/16933758c9b14396b4e9871d314a0f35 : FAILURE in 31m 19s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/28d4d9b3362d4124b1c8290a95c99694 : FAILURE in 1h 10m 18s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cb36819a12f5434daaeb2108e2402945 : FAILURE in 21m 10s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2c29d569c83a4f8bb5a37f46c291d0ae : FAILURE in 36m 20s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/035df7cdf6a5404e8da3f1e355a3b1c9 : FAILURE in 50m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbbf7585c3b94a4e953be243c86381f0 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aee4f28fc9cf45929d5aa7a4e662ff00 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3292f65217064a2da15d9a8a48e9674a : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20daac0724564d9fb623f0aede4a1f22 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ad665609be34f10928cccd62f0dd8c2 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf85206085774e3aae5214b7b16d76de : SUCCESS in 4m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/eafe9f8cf1d240a8bcb012a1a3b817b0 : FAILURE in 17m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/279a9005336d4dd0933fa2b8a19905af : FAILURE in 31m 54s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fe35ee139f5044f4950b47a563193556 : FAILURE in 33m 26s","accounts_in_message":[],"_revision_number":16},{"id":"af933d26e85b14cee92e8fc5ef7104f2afa38bb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-05-27 08:13:11.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"3ada0fd8ec9796677289e8fcb0986f83144c68ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 09:31:14.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/65b61bb22e9545d6b6236773978e8616 : SUCCESS in 32m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9524c136626845fe88cccfac556a8b16 : FAILURE in 32m 27s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c2edd6bdf5674b78a1120c7ccebbf431 : FAILURE in 1h 11m 54s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/45d6937042d344b8b3dfe4b469728772 : SUCCESS in 17m 41s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f71fdfe73cbe436ea9ed7e825f5e4b7b : FAILURE in 29m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a3b0454106d54c0a86978e1c5b5658e0 : FAILURE in 53m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84caaeddbf924357b1d49108c60991af : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9d8c7cf907b4db099e79304e06d7be7 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37698b562f314c32ab21bab55aebc288 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbee0eebe9b54d6ead9652e5de8604e2 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8654674add6b49d2b5bb9cd945ea5169 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0fa663b18b5425fa3c56037f99ffbc4 : SUCCESS in 3m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d4fd7668caf14ec5bc10f08504d54fdf : FAILURE in 29m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad161b8948c34afbab38677b6dd4e1f9 : FAILURE in 19m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1099536786b542159b3bdbb94d251b94 : FAILURE in 21m 10s","accounts_in_message":[],"_revision_number":17},{"id":"7620cf318bc1805df695aec18387632cc8a7bfb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 09:56:24.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/24d68cec0f314e19b7f9db9433d8d677 : SUCCESS in 1h 41m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/661b4bc9c7ff4ce8ada266f2fa339563 : SUCCESS in 1h 28m 28s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2926bd7d67f74fc089aeafa516cbe2e1 : FAILURE in 39m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a263ba739c2408a8c278c3a3e73f4fd : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/983fde5cc04e45fba3afb24498688e57 : SUCCESS in 5m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6a1b636718f344c49b7f64ff0a36054d : FAILURE in 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0b312a4e41614492e2999020bfda521f4d25c49f","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-05-28 11:17:38.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"8239dc2dc747e3500bd862d88a9a2ffc94d6b31a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 12:01:47.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/61137833381f48c48ee68439cdfb56bf : FAILURE in 28m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c680766dbaa04c2191042348ac584700 : FAILURE in 26m 24s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b4b8299cbd3049fd862b2d4206d9eef8 : FAILURE in 20m 40s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1a92b997cdab46deaccb2f18d8d7bd56 : SUCCESS in 18m 05s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/299da1bca1504fe2b42cd3ac4bddb0b2 : FAILURE in 24m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/623c01f45ac24977a86f793223295fa9 : FAILURE in 43m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d54b5b68118d41729b800c0e8f3b5eb0 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6820cb4e029f4af990e98373b88cf20e : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2458dbf79404566b2a0c032db30eddd : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6604ebc48ef040c7a4f7c6160c0e5f3b : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8252e32c93fb47acb92a57c69f770c24 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd16b14d6d244ae39c378f49c48f8c7b : SUCCESS in 3m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/505a8862f10a4a67b8f767d263de4c16 : FAILURE in 20m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2048007ca5784c5d9564396b1c41c0ca : FAILURE in 14m 51s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a27b8505dcb54c27b6ff4f4cf046322f : FAILURE in 17m 26s","accounts_in_message":[],"_revision_number":17},{"id":"ba213400f9e5abd51d736a9a03296123fd214b37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 12:19:02.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/a68a5fc294bc4d9db2c5dde7d2676f71 : FAILURE in 35m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/721710454ffc47f683f37aad57ef5752 : SUCCESS in 26m 28s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/06d395ad0c544122bc714bb0b089db96 : FAILURE in 46m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17277c5505af4f4c88eae4163f54f24f : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5f48994266141ffa89cd54553e0839b : SUCCESS in 6m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/af6197eda7bf4fe6b6179e3471872a70 : FAILURE in 58m 43s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"22269df799cab2719eff8eea4103ed7a543cfa45","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-05-30 02:18:40.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"dc89edfff196d19e2a9c36a52539b433798bed97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-30 08:12:45.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"06ad81dc910d001dfc8adb636eb1370a09cb0856","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-30 08:20:02.000000000","message":"Patch Set 18: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":18},{"id":"52833e42046e6af38db18c112c1e4520691534d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 09:44:41.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/3ca256ba7300444da19796280d13f423 : SUCCESS in 26m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a6812548aa9a4ab59fb41649c3420caf : SUCCESS in 59m 03s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e1b50e5c520d4f699fada3d0a7770788 : SUCCESS in 1h 22m 09s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bea91be4c07b4ec49efa08deadd80a85 : SUCCESS in 17m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6ac99560f2c947d093601fb58513d3f7 : SUCCESS in 1h 00m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/dad326f3e7e84bcea35239e26595a919 : SUCCESS in 1h 25m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/102482fe79f147ca9805c733daa365b4 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b122f18ddeb413bb1f7a3371b643854 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92f4dc2ff01a480bafdc8ecc03273ff4 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65b331a6da0b40fc91008d8f900fad2b : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a10a25ca52444897d859e9ffb32750 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba3c69fb9d9d48c48061f2c57e7776bd : SUCCESS in 3m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6e6ff14e68eb41c9b2cc75d3e9d88db2 : SUCCESS in 17m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a39b3d4519b8400d91135bc2c772c49b : SUCCESS in 51m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac6f485e5e174188b97cefaa4400c3a7 : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":18},{"id":"c23815d1aa80204c7f892f824869a932ba2dcb13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 10:17:40.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/574fcd3c1e8447ef871102c985684dfc : SUCCESS in 2h 01m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3a5bba4f9ac9420bad74e084d9946458 : SUCCESS in 1h 44m 35s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b129a0e8577645cc9fbe6c71d1fb88f9 : FAILURE in 1h 03m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f96fb00e00e64009945291359592511e : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ad229ef3a814aa496eac4e237181b3c : SUCCESS in 6m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0d51af4520754900ae6b156738dc4408 : FAILURE in 1h 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"cb18d44155afec2335805703a893fa0a7e082771","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-05-31 12:26:33.000000000","message":"Patch Set 18: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"292ee71cf103b31394b4185d206f2338bb6c1395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-08-19 10:12:34.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"e9c6014dee380e23df2ed5a376fed0c9552d35a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 11:27:16.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/373d88b255c942f8b4e22d6516e98235\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4f4f5fb3fba1479d8384af0a8da2c34f : SUCCESS in 24m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c833c30a04b54bd2a5040043f2bbf46d : SUCCESS in 1h 01m 10s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fe761570a24c4366b9d729813bae924c : SUCCESS in 1h 08m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/083b4e7d73144b7483c73c4df6eb00df : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24d3f6e850a44e3eb3822535caf2d50f : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee549b12d35844149733342cf29ae0c6 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/773ec77011934947816ef8507f5554ec : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db4fc1b696e459684870eafadb2ee30 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a8fa7f2e3ba4b9d9497da18b082e3ff : SUCCESS in 6m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b43902f1238c49c786dcf7bdd7c40956 : SUCCESS in 22m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/175fa5f07e8e48d7a89066a90fa4f4c0 : FAILURE in 22m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7a7c533d1b9b4de6a17381ce1c035fbb : FAILURE in 21m 08s","accounts_in_message":[],"_revision_number":19},{"id":"498bc650daab5aadcd5c99abce24ff6753b8e0ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 11:37:47.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05ddba7a9d43498281986495af2ccbed\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1f4b0b0a730a4029b2db43ad5a9b7fee : SUCCESS in 49m 08s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a2bf8ebc93645cea7ad33a4f8498432 : SUCCESS in 1h 20m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/425152f17de848bc9bf617b1a9fee676 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/910b26ba42ee4f94bb0934639111304a : SUCCESS in 6m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4473f5545f84499cae55fac85d0c1eb0 : SUCCESS in 21m 52s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"91d06340fdcf5c895bce74ce129ec6b60e51bee1","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-08-23 09:43:43.000000000","message":"Patch Set 19:\n\n(10 comments)","accounts_in_message":[],"_revision_number":19},{"id":"3de0816c17d6a6a5d313e8389e8cd91e870c68a9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-30 10:31:14.000000000","message":"Patch Set 19: Review-Priority-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"873056f6b81c66b24788f93516cccb3ff7ff5c37","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-01 11:29:06.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e085afd74584fde41d813e8b77583ad96edadf5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-13 06:56:55.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"9456180c4c2bb0a18e7a10bb1f07c810850c99af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-13 06:58:17.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"06044b1a88622fa54a21f51e07e090e96802d2bb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-13 07:38:03.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"7427d98f9b8b2ce088d25ca52c53daa93ede544a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 08:24:45.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9bf781c82344083a8d90f6f21ef1ae0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e901ef4023004bcbae8824085f723c2f : SUCCESS in 22m 38s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/835fbf997dba4bf58416b4c97b9427f2 : SUCCESS in 1h 11m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ab52ed68eb0142e88fba3ea72e202765 : SUCCESS in 20m 20s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ed16606a2ace43c4846f86148b6baa77 : SUCCESS in 59m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7c1e0d43fcd349c5bb29f01338407803 : SUCCESS in 1h 05m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec90ae530be42d8adc63165b12cfc8e : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e23ebb4c1f748c4839df02635865133 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37f1c21f43ea48929523139738db77dc : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52f3372a190d460ead43cca23dc3a4f1 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7499ae556a914fe0a804abecfcd9add2 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef83843270f048e4b2d805c4dab756b8 : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/97335a2d9efe4af6bf34e10baaea73ad : SUCCESS in 20m 59s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8235a10097174d2ca1e5e4f7fd55bf86 : SUCCESS in 1h 00m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d497b8f20de7469c8cb2786424b695f5 : SUCCESS in 18m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ca826ecd74749208ead2060cfa1b1d6 : SUCCESS in 47m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cae444ce8caa4d579ae15c040de9a219 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":21},{"id":"15340ac43c0111b091099918533268ae6bb3e7c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 08:40:32.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebddf559f9f24e1a87a6e7b4c2f57d60\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2016f818f08b4b2a94d0fb3274e56f64 : SUCCESS in 26m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/562af9d3f3524cef86cebcc2ec07f501 : SUCCESS in 1h 31m 29s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/15d58184637d45ee978e29bc419b31bd : SUCCESS in 1h 40m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/771a3c235b794a34bc2bcf8a4ee77900 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad852863d005434bb060cff239cedac4 : SUCCESS in 4m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/526d2f922abb4612bd260ac5b1b8277a : SUCCESS in 27m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4cfeb817f44d4fe6b668968278f44be8 : SUCCESS in 55m 50s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"7487545c48e9d080e902daec2ee8fd1bd97334e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-13 09:36:41.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"38f7143877addabf7b5321d0dc72379c6dd65594","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-13 09:41:14.000000000","message":"Patch Set 22: Code-Review-1 Review-Priority-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"670d20d76e64a8acdb3805598488016c25e08c29","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-13 09:43:34.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"dfb15954774ffa3944f17cea5e64a855081f24d0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-13 09:48:43.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"21d385c3d5815cca32f08fcd8201931841bac89a","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-13 10:13:06.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"a748499a67af26741346b9ebafd1984f6a8a1725","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-13 10:18:35.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"94d7ac48bcd51410bafbd62f79a6733581b7c641","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-13 10:18:44.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"00dc8c5ff2aff5d19ac71eccf0ca8b0b4e96ba78","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-13 10:20:58.000000000","message":"Patch Set 23:\n\n(4 comments)","accounts_in_message":[],"_revision_number":23},{"id":"c37a110488b9d5afcdc833361e8e18bc19c7e266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-13 10:21:00.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"8dd717735882054dd8301135c539e505273a6e5b","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-13 10:23:54.000000000","message":"Removed Review-Priority-1 by \u003cGERRIT_ACCOUNT_24072\u003e\n","accounts_in_message":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"_revision_number":24},{"id":"24ba0ebd93d40c3dd1b46a6d41d8825ad08835a7","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-13 10:25:08.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"8e07aa1c4f3a1b2843375abf7d7e824dcf982abb","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-13 10:25:37.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"c880e8fb9a9f789f672495f1e68cd5bc7bbc8d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 11:36:08.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/393153f92c5844988290b0dc2df2a959\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/aef26cd58f974c21b7e0636fe8a9879a : SUCCESS in 26m 03s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a114d990c193470ebbc5824893904b71 : SUCCESS in 1h 09m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5801cab443424ec28549121ab59e3e95 : SUCCESS in 22m 07s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1ed682c7e0864089a32701827e8170a9 : SUCCESS in 1h 04m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0b7fb71e144f479ab502e3002df12785 : SUCCESS in 1h 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de38ea78efd4c31bf43ad571ea7e0b3 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a0c427068c04c72be1b665c8abe2ed9 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e76ca58ed024ab1aeb1a2b2560dd2f4 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1a5c0e0f3684b1bbe0663d4ba1c482a : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a532ab2c23c94736ac250dc90ef9720a : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb56d43bfedb46b7964720cd36d4b5f7 : SUCCESS in 2m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b022f5c736f641b7af058bb3974f6325 : SUCCESS in 27m 42s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/5542187141c84816b2b92eec5fe09a5d : SUCCESS in 1h 03m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/93ae5c86928b4fb38147424653d70b7f : SUCCESS in 19m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/964249507f20468ebbfae4bb41248030 : SUCCESS in 52m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7c3021c3423841aba0f6dc0b116e3d05 : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":24},{"id":"65546d03bf477ebe1dfa6987a40b8bffb423406d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 11:47:45.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1a6a9340c7f4bf9968df6a55c0ebe14\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/debed5d38c5c4df09d0438ebf6fabb5c : SUCCESS in 31m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b86d86cfac864b629a80a62206f73939 : SUCCESS in 1h 11m 08s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ba6d593fed3843a3bedee683fec54a77 : SUCCESS in 1h 19m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db272075da324a2580a783646732db28 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eac78f5439ec45e98d27800543612976 : SUCCESS in 4m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1dd3df2f09af4514b9a4367da45f34db : SUCCESS in 1h 14m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/814c2af5995049c2a59391a05ce59282 : SUCCESS in 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"d97de182ffca2e4489f095e998d43436857e5b0a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-18 06:14:50.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"8674917b88618bbd5f2d97c29776f97b6b939722","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-18 13:08:39.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_30491\u003e.","accounts_in_message":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"_revision_number":24},{"id":"8011c522e24496041da07c9bf268a2ea3e63ee73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-20 09:45:51.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"0d4291fb3e5067e9a0517fd8240fc55acde1c31c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 10:10:45.000000000","message":"Patch Set 25:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82bd15a35f824dfa84dd30c0938dcdbb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/221911e978bd4a9d9aa7a7d73d026572 : FAILURE in 6m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7bb2f1515b6b41beb75f3078d20581cd : FAILURE in 3m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad6e5edc69d2445d9c1d6d8207fe5b64 : FAILURE in 22m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c93f90bcb3354cfda0e9eb1db79ba480 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d2e4d75d82d4deba97f82c07f5408f2 : SUCCESS in 5m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2eed477ca445482995aa8599ca021b5e : FAILURE in 3m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9add13c3db1e4e2aa083bbbc76c537f3 : FAILURE in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"d611dd66ad845a290a87de2f1bbcdb327307de21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 10:24:04.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3095e11853a848a9a2f65b4d9d0a1d1c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d4a7f16453144e0b99e27a05ef495b2e : FAILURE in 3m 55s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/de267f15bd4745ccb7189e02b1245779 : FAILURE in 10m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c5a3b4f42bbd477dbdfbca5f919bc43a : FAILURE in 3m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e5b91ed75b24429d8c947dd7d89490b8 : FAILURE in 9m 30s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/df8d92942ac04a7485674924ac56128a : FAILURE in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2009d7fc23044d6bcbb14d7860c83ba : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f36ca9540e9412ca9427139a8389da4 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c380726547c149f99c997eb092e00c3b : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/231150b5d2f74dc696778a1fbfc55b83 : SUCCESS in 2m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5a67929df9439fbe6c60075b5b1785 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d59b63deeb341b0ad8a5a14014782b3 : SUCCESS in 3m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/27a828cef0764b0e82185a1222bc3407 : FAILURE in 3m 24s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/c5090cff523b4acfbafe6c749f9d4000 : FAILURE in 8m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a8a369f669f34232b18484916831b56b : FAILURE in 2m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ecaf86f0989749e783280f8772d8ea54 : FAILURE in 8m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c0d283c32e9b48c0a135390d3386fef5 : FAILURE in 9m 08s","accounts_in_message":[],"_revision_number":25},{"id":"68e557d203acb8e8bd4713a83e04f49e0e592667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-20 10:25:46.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"a14aa3a7866432b97fd77b70893ae27cb32e1860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 11:50:51.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b2f5187a7474a72940bd135357e9e7e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/39a38ec3f2e6406e8eaf7f1fe58a01f0 : SUCCESS in 23m 21s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/54306d0cf4374a0ca52822b6e1e61e90 : SUCCESS in 1h 18m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c3f0b7f6345c4e2e92bed0465ec14d67 : SUCCESS in 21m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/40b9d4f7b3e349e8ad3bba3e636c741f : SUCCESS in 54m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/26f67fdb9c604e5f9cf85dd4793eb7c5 : SUCCESS in 1h 13m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53b469e1f5da405fabdbac7e4493718e : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/353fb7773016477b80bb58c536727061 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d71b77a669df42609f0c9b0fcea93f79 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b42097ad30147e99182575e88058152 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a64c9347ec90480bab28e0264ecb6c94 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a85564116a554e229cb157b44afebbcc : SUCCESS in 3m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1c24146c581b482baee211d155c28db9 : SUCCESS in 25m 17s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ee26269e0a6c412797d1f4eeba6bf154 : SUCCESS in 57m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f5da5d9db72142628738b99cc86dc7f4 : SUCCESS in 20m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a12226faf1b8448896f1c599a019e981 : SUCCESS in 37m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/90f346bf64aa4d42bc3cf134169116d9 : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":26},{"id":"06e1014f353a8e7fae0e5dd7b495a0bc2b495aca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 12:01:17.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/177c930249f943899b6fb1b4c78819a2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/54eb9c1de8874ac3bd7ba81fdb4bee58 : FAILURE in 39m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/47adbc2f0be6451690e7ef7dc2337d98 : SUCCESS in 38m 27s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/85e0c2f2f4d143deae73d09b561e0292 : SUCCESS in 1h 27m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed2c82577a9c4f48b4d11f77f1b53dc0 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90b0042cbe324a99926b51e402392d7c : SUCCESS in 3m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1478cbf8ed0140cabcee114a8e7b6f8c : FAILURE in 39m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5c765e2fa13b473584a84808a6943814 : SUCCESS in 37m 20s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"5683924c2907491abe5cc8327b63b9a99149127c","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-21 13:58:30.000000000","message":"Patch Set 26: Code-Review-1","accounts_in_message":[],"_revision_number":26},{"id":"c09f1e324bff14cc3b1df011d730c10636917757","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-24 01:24:59.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"390cd848b8a9bf62495088d8da3422998a77cfba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 02:59:25.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1cd44b714024e2e8f61473b06179339\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a0674e93ddaa4f069d6e221bd47278c6 : SUCCESS in 27m 58s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f6f5e8d8c6d442feb92aabaa5ba65d74 : SUCCESS in 55m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7f0cd57069494557bcacdd0a195cc9d1 : SUCCESS in 22m 05s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/19e5d806e1084c0d873d644218da8f66 : SUCCESS in 1h 03m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/16648b869d3b4181a04b5bdd3fcc926f : SUCCESS in 1h 17m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b77e44eb704e4fbefd31820d0fab8b : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46182bbc564c4735b17ef540950a4cf0 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9447b45c2b044d52b977917970766413 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/def05499b4974fee806147576ce2322d : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b421f24cc25c40ab96427f12f9ef7ec8 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88dfff19c38d413eba2f9a73809ba775 : SUCCESS in 3m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8c900a899d724290912be3d582002154 : SUCCESS in 22m 32s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/964efa7684d642f8a61eb68fac1891c0 : SUCCESS in 1h 04m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/abf08cbb862d4d8dae55c1c830222782 : SUCCESS in 20m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/14d86a255b3e41a0a7d71feea91dc714 : SUCCESS in 49m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/722e38b19a1a4e049f7f23816bd52941 : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":27},{"id":"2f8390dea191e94ab9905a15dc071d706d42d2fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 03:15:43.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/942a4779dd2a45968d6d5c8859b07268\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9a57c68eccb94ddd841eb42a3ae975ae : FAILURE in 36m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e851b2caea054aa0af2320f85f606d01 : SUCCESS in 37m 36s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/021f1a97b37b4cba82ae378edd31adb6 : FAILURE in 1h 17m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2047db899a754ae08ca950f34588bd78 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d5842a8a4424ed6b3bb4bf2d2be86bc : SUCCESS in 6m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1eedb2587dea4771a6b7b7461254f876 : FAILURE in 1h 48m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3b511b6c05a84aa09bc4230d43a99ad3 : SUCCESS in 36m 33s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"eb772282285a39f437b9069e9b731a879e9aca2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-24 05:30:15.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"4d6d0715b0759aafe6893d0d31721897533b0289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-24 06:16:07.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"fb242e98d27f99a77e37e672edece92d8ddf357f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 07:32:56.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1112b7cfa584e3bb2e9ba65f2645217\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/df4f5bcce96b46f5964af7b44f02a890 : SUCCESS in 22m 08s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/cdb52fe3f8f2494eae82c7fa951a30c4 : SUCCESS in 52m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/109b2b81261e4096a3020f6c33820046 : SUCCESS in 23m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/43d2b30896e040b0b99b02d46d756ec0 : SUCCESS in 1h 08m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f92c25e327c44912ac945bc3175667de : SUCCESS in 1h 03m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a5d61c956754fddb0509b26fa1711e0 : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b40e00be95147a89bbee34080ca602a : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5c4740c23ec4e8b8a0d689cfd1a010b : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e39794b94394573ba497807c98f92e0 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b0279f830341f88670244c62aa9458 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/472d18af55f4417583807ede1db77070 : SUCCESS in 2m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/947d1aba02c541cfb8f8721fcfc8b15f : SUCCESS in 26m 09s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/476cee4ffe8c4b3a8fde91a750c0c6ba : SUCCESS in 1h 14m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f369936733a44dd4a8ffdb52f594435c : SUCCESS in 19m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9554710ab64444cd8792a106f4235ab4 : SUCCESS in 50m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e13d014d89f340a9bdda2988839a1e34 : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":29},{"id":"0d101e9a0ff4b41ccc235f20ffed11c6c6f04767","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 12:27:18.000000000","message":"Patch Set 29:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f88b8ff4464246ff8b5e74247234679a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/70e1192ef14542ef8bb4c199bc5d1dee : TIMED_OUT in 6h 00m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c0faf35f43a8411991aec2c068e66840 : FAILURE in 38m 35s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/db6efe0fd73e40fbb6ca33aca1a7066c : SUCCESS in 1h 29m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a14d1a3ee2684e9ca6d84f23bc77e439 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7e87189d91b430d9d3f7fb0482137a2 : SUCCESS in 4m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/475950f62f6943018067f80baca20c73 : FAILURE in 36m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2c5417b6212b412b8a07c11cd7e8bff3 : SUCCESS in 36m 32s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"5d2bb7b1e9a39993f67965fa4c035248c72f9890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-27 15:21:11.000000000","message":"Patch Set 30: Patch Set 29 was rebased","accounts_in_message":[],"_revision_number":30},{"id":"b9bda7c4ec033bf132377ef0883265f529fb6b8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 16:54:44.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/035bdaddf01645268fef85177db6d8ba\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/eaf489e56fa04b858f75fd2a9a60eafa : SUCCESS in 28m 18s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/e4fcda5cad42405c9f814948b1dfaca4 : SUCCESS in 1h 18m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/426edfe2dd9e44999151b07fc3a95887 : SUCCESS in 22m 47s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ddfedb75faa44dda906eebb87e5762a5 : SUCCESS in 1h 06m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/76729334706c47fea87c0aa2182db729 : SUCCESS in 1h 26m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/254816c947404b63be732b8e6dd1dd52 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/581360117052443da30200b245b3c5cf : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bf4bf72709d4d4ea6708348340b90f4 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9506bf074bd45f08a016896c428ddfa : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e00fde53a58248939fb1295dbd09ddd2 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e07ca6ef93f4310a179803039d85108 : SUCCESS in 3m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/be28b488dd024e0397046f3653d3040d : SUCCESS in 20m 43s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/095c25f8b74948c594338e6d95d8c194 : SUCCESS in 1h 13m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/abac97697e544336a21966f18835eaab : SUCCESS in 23m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4cb1f611cc30416bb0033f8d578c569e : SUCCESS in 39m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a6e24544310445191151e7e0d7fa1e4 : SUCCESS in 51m 46s","accounts_in_message":[],"_revision_number":30},{"id":"41666b4aa202bd52b776a342fb40602de0febf56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 17:22:34.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48b4275a84404baf9dd78242de1a5f52\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/22cb72a5d40f49be9b2533e26a188a12 : SUCCESS in 36m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/850c4cb94aa04c6193a08899b8931727 : SUCCESS in 1h 59m 31s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b307f5d323c4d9c9d0a7e6c9f71df15 : FAILURE in 16m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92d8a3a31f8349a78bd1af8ce37f1575 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c555ac6f1f1c493fafd8f472b6f387e0 : SUCCESS in 4m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/352f29b8c49a456d910b91405f0a5046 : SUCCESS in 1h 13m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5e54c7db54424a60b1a025c5bc8d3f4d : SUCCESS in 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"1244b6a56b8b3ad4f6f07de833e9937af0dd5ba9","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-28 07:19:32.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"166b15a50ab6b4c47599c6c0f38f5189df32be22","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-10-28 07:23:56.000000000","message":"Patch Set 30: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"32a530982562a57d7507ff8c28e20274a263c45d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-11-02 13:29:59.000000000","message":"Patch Set 30:\n\n(2 comments)","accounts_in_message":[],"_revision_number":30},{"id":"68d5b3030a9710e7ccf0c83f13e1bf121de4b67a","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-03 06:48:57.000000000","message":"Patch Set 30:\n\n(2 comments)","accounts_in_message":[],"_revision_number":30},{"id":"15f2b5e607e9a32b0771ce7671a065a76631f8e8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-11-03 07:03:33.000000000","message":"Patch Set 30: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"30beb4362ce899eb2a725659145492b4fc958aa5","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-03 10:40:33.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"276d72b4bde1603b3052a1165bec77480ed89c5d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-11-03 10:58:48.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"0ceac5fba98ead089745998e52b9fca036f7db55","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-09 10:39:50.000000000","message":"Patch Set 30:\n\n(2 comments)","accounts_in_message":[],"_revision_number":30},{"id":"c14e156a98eeff3a7bcc8816a02c7e8be949ac60","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-11-09 10:44:08.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"443b9164ba5326bf14518da0a9e426c514ea6b0e","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-10 02:17:24.000000000","message":"Patch Set 30: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":30},{"id":"edef49d30960153575fe78a048240770245ed928","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-11-15 09:59:46.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"14734ef4f6279e831577b90cd49ac6dd108b2fe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-11-18 14:21:05.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"9b0eb4b7cfbb457390c20bda08ccd3332c4a34de","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-11-18 14:30:25.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"0ef0b3c7faf13af45d78705c15b7cb24c40263ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 15:53:25.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c398b1bf23fc4c8893ff1b96c5ca764d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7756daec4f2e4c61a85565e106b1e2a0 : SUCCESS in 26m 14s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/955da224361742f1b8f88a8971f021da : SUCCESS in 1h 14m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/617bdd6f9426416bbb418b58d0b58178 : SUCCESS in 22m 26s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8f13cbf33a8f43ddb172f0d333d65f90 : SUCCESS in 1h 01m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c132d1d76ca64ec2a68db8b1bf9986c9 : SUCCESS in 1h 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70e593e19d3d4581a5fe3edb8fd4d8dd : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dde434c3b7b4df0ab8f4a763c9f0052 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d7c9ebc92b541648b5bed864ea60923 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c2a2810c1424088894335e0c2ef6fde : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6961a7d2f2cd4a98a92032dfc7f09f6c : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cfbb68f91d24ce48458e3ae53bb0f97 : SUCCESS in 3m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bec677c23a5943d0ad0045aa2f886053 : SUCCESS in 23m 08s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/281694232aef42f286930ff07f1d4304 : SUCCESS in 1h 15m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/384ee9db44f54d3283d0ad162e237e97 : SUCCESS in 25m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2b3af38f7d54715871cd7361c608491 : SUCCESS in 50m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e97241f40bfa4c16b5a433f4aaf1b43d : SUCCESS in 1h 09m 43s","accounts_in_message":[],"_revision_number":31},{"id":"cc00ff67428dbde7efcbded3102d7c97b4cceef5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 20:24:58.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b5430f8a2eb4c089f57def8660b5db4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/41f5782cad7d4a168dafca80a2f7650f : TIMED_OUT in 6h 00m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2ca67759ec7042e39f0504e7f71e32fb : SUCCESS in 1h 26m 18s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/65e49d8deab44e5eade9500f8639efde : SUCCESS in 2h 13m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92e8d85f05cc4db6934b137ab95b179a : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/888df23c410c41c28a6d29124d0816cd : SUCCESS in 4m 59s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7ee7a81f381940c5a1213896a68a382e : SUCCESS in 1h 12m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eac8d6b70c2b42f182e8f3cfa7e71e5b : SUCCESS in 1h 06m 07s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"d2f291ee9b3e467f4fb1af67b989692457102887","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-21 01:54:51.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"9a24323f8ed3dc8f840b2629300135ebdee884c9","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-21 02:02:54.000000000","message":"Patch Set 32:\n\n(2 comments)","accounts_in_message":[],"_revision_number":32},{"id":"bc400b835457283a5bdbe2139a68033cee604d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 04:30:42.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09baa4573dbb48f7897ea9c95fcfa3e8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/05eefc71267e4c2487fdaf7b05003c35 : SUCCESS in 26m 57s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b2820ee638e547799c235bc9cfddfb60 : SUCCESS in 1h 29m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/13f85bffd0694a27986b58fe44e78929 : SUCCESS in 22m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/890d4ad5cf36484e91723c28e59857a5 : SUCCESS in 1h 19m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/523244c1168647cc9051a20f388b860c : SUCCESS in 2h 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ec51ab6fc874116ae74a847538966e2 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9977a1642520437b873319ccfc2503bf : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/040fc15166144fedad920978e21086d6 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed744531b8ad41de87e196571646c9fc : SUCCESS in 2m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5e70a1178b44c1c9f126b2c5e36182d : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95998cdeeb01479880dfd2b76ba98794 : SUCCESS in 3m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/42e3790a40d94c67aef43ce89e460c90 : SUCCESS in 34m 40s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/b1cd176413164dfb98944ed47199ad9e : SUCCESS in 1h 22m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a2509c83d29c40b9864bbf0a999c2326 : SUCCESS in 39m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc64519ad40147f7a1041fb114f6c84c : TIMED_OUT in 2h 30m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aaa259b193234f5c81f1664b7f8dc686 : SUCCESS in 2h 08m 15s","accounts_in_message":[],"_revision_number":32},{"id":"e117ef887ecc6a7aab853e951dcaab3fef5631e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 04:49:41.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df242e97ef7a452eaa99898f7654eb37\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d3f4fa0cbd5a46b39df001d136393648 : SUCCESS in 42m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/38c3c0b49e2d46ee80f71f108a80adba : SUCCESS in 2h 18m 50s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a005baf969e4cf6973cfb83bdb863f9 : SUCCESS in 1h 58m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca191bc77eea4d8ab5f822ddeab9d676 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3ff256dc3e444c6becaf45b7899da63 : SUCCESS in 3m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bf88d39b7a4c436fb97563f00885062f : SUCCESS in 45m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e45a251e68f34590b86d64c81cae0af1 : SUCCESS in 43m 03s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"32cc971441a283396ca4f957c93de5d1e01af97c","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-21 05:15:42.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"15241e882d5b7ea921eb94d9ed282f22a8b8615d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 06:41:25.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8f0b44494014308bde5c383d862409c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c4ca356213084d5885f20a16d6701d9c : SUCCESS in 18m 37s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/4842bafc1b744f828f66f082cc524e37 : SUCCESS in 59m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fca97f8f4f7a4b6496ba1b2d2de9dc45 : SUCCESS in 17m 43s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/81b0fedc7b984214b23a7508684c4cf3 : SUCCESS in 1h 03m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ace22aaa542a469b817a98f5c70cf6de : SUCCESS in 1h 20m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df17c0d0bd4246219737a75a5ce56c5d : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c779e0f831e44e4bd8a1f770cf6f51a : SUCCESS in 2m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dec933ce00c144168baa93daacf8c731 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/568a9f57037745d8b2c37d623a4d37dc : SUCCESS in 2m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5590e1c19e274d26b35d7682fc847213 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e94e780fe4f469eae61dbeedb6cc979 : SUCCESS in 2m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2e1d459e33ec47dc80112f3a2f4e8504 : SUCCESS in 25m 08s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f9d5a2b802b2476a9518df2f1a810d3e : SUCCESS in 1h 08m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cc9a891301e64ca383b7e70c2ef42001 : SUCCESS in 19m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/28a8e7ef0ec34ba99287bde8f60326fc : SUCCESS in 51m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b307e60182a8468f92a7bbe48fab67b7 : SUCCESS in 1h 03m 07s","accounts_in_message":[],"_revision_number":32},{"id":"fd5120bd91cc88ad9b14e2d7233db4f6b5538932","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 07:13:20.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87062521e56d4a059b75f00ada2abeb7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/61935e9529e04850b48d41537bf0c471 : SUCCESS in 57m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dd5c333074e64a8ab160db035064cbf9 : SUCCESS in 59m 35s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9fa74fa790004e2da6e5c8b8695bbeaf : SUCCESS in 1h 44m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/558ead46dc7b405a96b62d651b2dcbe5 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7dbe2f3cf8724ac6875cf024a5c1e9ad : SUCCESS in 4m 53s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/aa5a5307e7214f108cf0a6f8c0b68039 : SUCCESS in 59m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b938efce2fae435f89d3bbdbd3ec9aab : SUCCESS in 1h 04m 08s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"4988304572ddd6fc1b5cfbdf5dcc4521f3b7d69e","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-21 07:17:10.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"274eab569fa5f645ffa529e62262e0badd40cd9e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-11-21 07:39:32.000000000","message":"Patch Set 32:\n\n(2 comments)","accounts_in_message":[],"_revision_number":32},{"id":"89b5501181f670d88e7a67f8325a4d75dc8ff14a","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-21 08:00:25.000000000","message":"Patch Set 32:\n\n(2 comments)","accounts_in_message":[],"_revision_number":32},{"id":"7c3bc5072853e616e2e6ed03b3635ef10b667a36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-23 02:54:11.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"5252c8e4379d763ef7863dc6bdd107c4a163ed56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 04:17:57.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2833875ee6b4babb67d0693fef9fa93\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8309c49697024a088c0ae5f9611df34e : SUCCESS in 27m 27s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/0c53f975f095440b9afde1c4a06eaf89 : SUCCESS in 1h 17m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/67a3b742b1d54df7a5b5a23e9d22e08b : SUCCESS in 21m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/520560638c394a5c83953162345322c4 : SUCCESS in 1h 00m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/345764d710ce499d844a68fbf7ab9dca : SUCCESS in 1h 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35db120be1ff43e2b1b733c04df01ec4 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94511e6f657048b98bb55aaf68b30a70 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9729729e7cc1414d95899a4a63cea746 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba145e5c6063461a8a9d4275a90d970b : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f6c6c85cf443dda531dd7d3a068d2f : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36fe3d69e96846a8b8a1b35a54daf1b8 : SUCCESS in 3m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ebb6a70f25c14e6bad1ce1a2c8611d9f : SUCCESS in 25m 51s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a3dae5599e134d578d79965fb6beb9bd : SUCCESS in 59m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7188fdbe92ba497baef2db5921b8629a : SUCCESS in 17m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/19aaa0d163554213aa3de773613cf316 : SUCCESS in 39m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ba4cbcc7497f463298d52ef5188faeba : SUCCESS in 1h 01m 56s","accounts_in_message":[],"_revision_number":33},{"id":"ba76567a3c6e8a1632b44ce188ab08b6e1995530","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-23 05:34:56.000000000","message":"Patch Set 33:\n\n(2 comments)","accounts_in_message":[],"_revision_number":33},{"id":"f130b2cf1a595e3c586f212d43b74ebc5d55c6d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-23 05:37:26.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.","accounts_in_message":[],"_revision_number":34},{"id":"0de35fb237d8e2685d0435cddd3150ebb12d1a82","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-23 06:51:51.000000000","message":"Patch Set 34:\n\n(3 comments)","accounts_in_message":[],"_revision_number":34},{"id":"431e8d8efbafe9cef13afe4d96011385caa9b070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-23 06:59:57.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"62e7412e548d7d66827edb62671e4f6bacb55ff1","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-23 07:00:19.000000000","message":"Patch Set 35:\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"a26698d7fa34033a650f1b3c6b0560d31d44b00c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 08:19:49.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d9c70f2854348c8a8c42b39f600119a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a07e37e893d44883aa53b079b7cf8b0c : SUCCESS in 23m 12s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/c024ea673e6b4ebfb600cbad780ef791 : SUCCESS in 1h 13m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e8d421f08dac4472926318fd289f0411 : SUCCESS in 21m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9d4965d184564b1caa12942a965e4db4 : SUCCESS in 56m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5f156a7e31104ea49b51ae83a40f8762 : SUCCESS in 1h 08m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95110159b8b94a2f8f9a98a8efad5b46 : SUCCESS in 2m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9c8e61016af49ccaad14a69461e3f28 : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7e9246e546344c891aba8998cf10786 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76f0225f5f644082847819278fd56654 : SUCCESS in 2m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3d4b4e298c84915b21736dad660115d : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0c30bb8c14049a089554aeab2bdbfb1 : SUCCESS in 3m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4014f622f7034156b39843086579720f : SUCCESS in 22m 19s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1215533b09604a42bacc7f34494f7587 : SUCCESS in 59m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cb52e295124a4cf2848efe516d8b809d : SUCCESS in 23m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4046b40d28ae4bb6abd71f5c695123c4 : FAILURE in 30m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a7ae103a55554ed98df0a10aac1455eb : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":35},{"id":"96e569c7cc5634f7d190a52e843f155e594547fe","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-11-23 08:34:24.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"d874ae61c9c8c10c44f33458cfbeb64f371d9265","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 08:57:05.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c92ec55f561f4b2ea7f66506b9d8cc86\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/24bfec4c8b7d4841bbbf3b3a48063be9 : SUCCESS in 51m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/968166ce97974df0b6bf10b4d99b0a3c : SUCCESS in 51m 25s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/42c1a3f99f1c4ea29c7c1c1207556f2a : FAILURE in 1h 34m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca9724ad932042a1b93bfe9a993632e6 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d2c4926cdee4bd1ac99385894606c71 : SUCCESS in 6m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7d29976765a049838ba8174a273679d7 : SUCCESS in 43m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fd8845c068ad4daea6e00b6ebe682e0d : SUCCESS in 49m 25s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"d8a254994459f542c3a2ef1972acdaee5484b59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 09:54:13.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/559ad829eed147c9b116dd686632b603\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/087f551ad41b411784d050b4104b1c4f : SUCCESS in 26m 39s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ff00f2946b1449e98c11f89ae0abc8e4 : SUCCESS in 1h 09m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8cde83295ff64d2f90d2923d8dde6852 : SUCCESS in 22m 22s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/05e2fa17440b4f42bdbd8eeca9de8cdc : SUCCESS in 1h 10m 10s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6e769ba278f941ee82bdf1afd4a9fd6a : SUCCESS in 1h 15m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf8e05b7c3a4bb5a84b44dcd9c2c470 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd6c0e73cced4d4ea8490bf53004b664 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/330a29d5bf6f47f99f21eafda73a82a2 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9758d76e5c24b5fb57cb8d4ab9a31dd : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aef3d39f66540f6ad411ea7390b6d10 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac14545ff27a4e8a91d5392d7c635a9f : SUCCESS in 3m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/367a668556b145f6affd7c014781b0ac : SUCCESS in 29m 34s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/83b75fc940f946cb9ffe240e503bb470 : SUCCESS in 1h 17m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5b4aff765a4f4f5caf77a64c4b1f00f1 : SUCCESS in 25m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/118071ed7ea24f0fbed6fdec026688fa : SUCCESS in 50m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f5afa08a19e44065946705447cc6fa06 : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":35},{"id":"49c9bc2e967ddb9adbcf0e6807a99994f6ee612d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-11-29 11:21:28.000000000","message":"Patch Set 35: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":35},{"id":"5911288aff58218864de572e4d884b1ca6e9c0d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-12-01 05:48:07.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"962de1c444a34004e2bb5f6dbe28ccfc00b23f1f","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-12-01 05:48:40.000000000","message":"Patch Set 36:\n\n(4 comments)","accounts_in_message":[],"_revision_number":36},{"id":"3e6622ddce100bc39d81fc48eed9b40302d602af","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-01 06:40:07.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"24b7f66a20d1afb83af09aa7a59db7a1020985af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 07:15:25.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf5b27b0cebe497f966e3423ca08bd61\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/477e3999c4124cffaa2ec235caf00110 : SUCCESS in 25m 43s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6d4f639a0f784888a4f257869e15b22b : SUCCESS in 1h 07m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/793b46ab35af404bbbcd8a97b7675118 : SUCCESS in 19m 53s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/691965b5cf76401eb14a84e79601a87d : SUCCESS in 1h 05m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/169072f7af9b49719c3cb05a192984b2 : SUCCESS in 1h 20m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e886915697c4e1a9db79f5db94d8f04 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a573df65bd1541ada351c918965d77a5 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c81d7d2038c48fab5eb859625e4999e : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a155d3164e5464588ad00a5b4ec7d09 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ea1d22d63e4a168d2216bbd1777cce : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c389bae4a4e4e27b77ac7517dbf34eb : SUCCESS in 3m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4bd291149bad46d9af332481b0d7c8f1 : SUCCESS in 17m 43s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/e46a284d7d6a4d4285b4cc334d63be8b : SUCCESS in 1h 03m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/376e7efc94134769889bcaedc8a27c25 : SUCCESS in 19m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b08ad0be73434f3bb1e7053e7f61ac97 : SUCCESS in 47m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8032ace8dccd4978be2dbec75c878ac4 : SUCCESS in 52m 16s","accounts_in_message":[],"_revision_number":36},{"id":"6909ee5d66a44388c926a493f8686f37766d1e2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 08:24:35.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d416050f4124eeda5a73b67134bd187\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2bd518b4cc384dacbaab984e36dff4d3 : SUCCESS in 1h 55m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/55c531c7fb704b86b363ea8c941ffd3b : SUCCESS in 2h 25m 09s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d6ec990a5f24d4abc3e8a4a0f3ceb23 : SUCCESS in 2h 33m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b2f58d26f5847f992d9a75a4990c17b : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9717e32dd9047389c30b10adde78a18 : SUCCESS in 5m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/89adf3f59f404403b61d6df650a41722 : SUCCESS in 1h 50m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9b18b19b53e04d82a3797d19796afdd2 : SUCCESS in 1h 49m 47s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"92cb458a8dcc771c5861fb8da88fdcfe3b6ee1e0","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-01 12:28:05.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_14826\u003e.","accounts_in_message":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"_revision_number":36},{"id":"2dfec3e407d5d76e3bfbca905d0619fab701e130","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-01 12:38:12.000000000","message":"Patch Set 36: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":36},{"id":"daa2782d10ff5a4014841982009aab10a631a81d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-12-02 01:57:27.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"83764170f6655f33e4738cc6c4b1df57edd942b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 03:17:16.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0f47ef1b866483594ae4b9105e21a07\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/528e16c6b06649e9b35da21bea22291c : SUCCESS in 28m 33s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/559eab75624247ac95503ad08bc1650b : SUCCESS in 1h 09m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/01db3359490849b5873616595e1f530b : SUCCESS in 25m 13s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3b95fa7ec59e45f794835cc1b596b528 : SUCCESS in 1h 00m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ef48e0ac3ff746fcab287afb42dde169 : SUCCESS in 1h 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5e5ccd60a924d97a832827ed8746901 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e264d837483b4e9ab02503895da36f97 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1505bda9cfa04d65a7b0902ca9705561 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4daccaaf1f14130901b448cf067781f : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9ace26724c4021aa83285f4ecd72ec : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60260cecaf45403390b5af78db1e9b68 : SUCCESS in 3m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/03a1289e93ba402785bd7a6299f0170f : SUCCESS in 25m 26s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/60c10cd192364b4da283c15db9492f07 : SUCCESS in 1h 18m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fcb188b51c75427e98e70bf9b1a158f3 : SUCCESS in 22m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fffbeffe71194b0d9f92751bbd0b73f5 : SUCCESS in 49m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/db9f3a8d8356449e9ce7c7316d8a59a3 : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":37},{"id":"5551ce36781c8eb0ff548037901c2fb875bb4a12","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-12-02 03:59:07.000000000","message":"Patch Set 37:\n\n(2 comments)","accounts_in_message":[],"_revision_number":37},{"id":"7eb04251d341cecb5e74c4e3e3475895ae4e8e34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 04:56:03.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ae550d0e4a4dc996e06d7e0cc76fdc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/24917899b01b4fceb98c6e3cde180e61 : SUCCESS in 2h 06m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/58c8c73fd70b4e22abac94634f68de40 : SUCCESS in 2h 55m 27s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/da6e993761bb4047b749fad31087b400 : SUCCESS in 2h 33m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c5ee195a9fa4c96bb1afda4ba7a527c : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43ec70a820104b36ac8d37d57977b997 : SUCCESS in 4m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ba218618b80e4218b4e8cfb6a09051a5 : SUCCESS in 2h 40m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7d938862f0074255bb200a317b6fa997 : SUCCESS in 2h 29m 42s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"0e43dc6f7dcff271b1e2dd26674ab31294bdc484","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-12-06 11:23:38.000000000","message":"Patch Set 37: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"894756bb39600eb421f61fc076e1356f0332cdc2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-07 07:30:59.000000000","message":"Patch Set 37: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"f5a0dcf4b190d1ccb63b2049d6f876648b230a97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-12-08 03:15:58.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"8e6e28939491eb08936b996e1a3a67a59b8f44e0","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-12-08 03:16:29.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"108927e86d2121365ad33832c01a9690f1cc9284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 04:38:30.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8039ae5efc445ed9aa1a86a70b47c7e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/82c6d730218549b593ce6aeaa02459c2 : SUCCESS in 2m 10s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/93ffcd9168574e3da586e2bcd88d5e39 : SUCCESS in 19m 59s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/d3a553f0fd8b4d23b3ba4611f625c0d3 : SUCCESS in 58m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c4f36d86e8164250b2c1e8abae59a974 : SUCCESS in 19m 05s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5367d446330e4896a2367e79280dadfe : SUCCESS in 50m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d507771cd26947d88f1a0756aa4602b8 : SUCCESS in 1h 00m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e85d8f003a14967901fa62ed7c6c0d7 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7829cab30c234c9aa32861fb915b0479 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32ecf21509764463843fb3db8d7687f5 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a5a0a419d2e430f8b320a9a80ba77a5 : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98eb6001ca664a72a393ad948326f653 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f6d8f10fad744c3b7b857793ea37ce1 : SUCCESS in 3m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b41247ee37174b97934136c875eb013b : SUCCESS in 30m 53s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/7386a6eb7e1249f6bef6f7c256f64a7a : SUCCESS in 59m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/14842af0a68d43819949a94bc559c2d6 : SUCCESS in 18m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c6e0f1b065314524aa5c650a06dce482 : SUCCESS in 41m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/259809fe0cda4a17b21f18c0a3d98c4a : SUCCESS in 1h 18m 03s","accounts_in_message":[],"_revision_number":38},{"id":"fd803a21731cadaff7c75a271750307c77503ee7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 05:03:44.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c314ebb424e41d78e1597e346ae2988\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a377e26144d241c0b85086a203b932cb : FAILURE in 7m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/93551abdccf54bb8bfe0ee97e7ff0b53 : SUCCESS in 59m 57s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/393aa8c7f95143e2a83c201e935e3bd5 : SUCCESS in 1h 44m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e9b5a77f86e4ec59055b17f825b59e0 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8cbd5e162a9b4a83b742900813dd6145 : SUCCESS in 5m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ffe504d27cf64de685b78db6f991ec10 : SUCCESS in 58m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7f5baf2651cb4375a05fb5b7e8df7bdd : SUCCESS in 1h 01m 38s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"fd492b7507f1069e2433ceb11b0a3172602ecc76","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-01-04 15:20:15.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"32f7cfe811d9d6a5c9ea49cf500bfb44276cafe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-01-30 01:44:18.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"6b7d3118d9ae92b112953d1aa33425117787ca6b","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-01-30 01:44:47.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"54eef99eff40bebbcbef8e0815cf9caea76c3060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 03:18:56.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10003486da754faa9b30e74c0243b248\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/108b97e666ec4a97965f0f5c015bbbaf : SUCCESS in 2m 42s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c3e94d3151344c0885128638f2b714a4 : FAILURE in 17m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/45498d8320fb49bd87b79dbce5a43301 : SUCCESS in 1h 15m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0c9abfc6eac6456aae05119de126aa39 : FAILURE in 16m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fbbd8b0615124e72b40d60aca50361a2 : SUCCESS in 1h 01m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3a0a42ba40994523bbf44366b997093d : SUCCESS in 1h 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b647270847c34f9399669fc4eb4a4266 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0b050585ffc4daa86d5c6cb281c8ccc : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e46ddd1e334428f813af29ffb05c231 : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e550bae55c1642029d2fc0a0c22fa8cf : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/598eaf34ca6748d59cacd27386616a65 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d67f346fc7642b69a04bc20bf662091 : SUCCESS in 3m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e3a6b08da84d476080273fd2dde14367 : FAILURE in 16m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9f37b68c90c14e90840ee0bd2632e24d : SUCCESS in 1h 19m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/055673f2ba8945599ae56ba880cc106e : FAILURE in 14m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/227c0a9d5b4647a1bc8af7d3026753ca : SUCCESS in 46m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ed66eb4bd2b54457bd6b7dc1873c08f0 : FAILURE in 24m 29s","accounts_in_message":[],"_revision_number":39},{"id":"bb07b7a13f7feb899dccd24a5c24bd14d1c7696b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 03:50:37.000000000","message":"Patch Set 39:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7b231a09c044a57af57bc7cea97329b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9ebf319d80a742668762c0842bb2016c : FAILURE in 1h 09m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a567f3912b8c46c1a598ac08886c88c9 : FAILURE in 13m 05s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/41329f27fa524fd6ac381323a10ffba6 : SUCCESS in 2h 03m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be84d8eadf3847c8ad216f30e5d5dc1a : SUCCESS in 2m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36048d645a194de495e111aafdd179d5 : SUCCESS in 5m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ad023e451c3741d1ab6bb89d342fec62 : FAILURE in 1h 08m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3ce0599c377047479aaca6609be97833 : FAILURE in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"7335656b7f014fb45c59177bfbb3bea3f247a1e6","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-01-30 05:48:40.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"5dea6e7f68da5f75d5cd7fc1f786c34ef0feb8ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 07:06:25.000000000","message":"Patch Set 39:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c799a2287c764bd08102c21bb0fc7ba2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e77d7fe41ae645d79d1f0db25a35a4b4 : SUCCESS in 2m 14s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/05503d25d337419ab83fd06bc5299897 : FAILURE in 20m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f2c116912077471a864417034ec28fdf : SUCCESS in 1h 12m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9e33c929c7bb4a9c91904bdd9e96550e : FAILURE in 18m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b6d9f4810a2d453691cc4dd09610ea96 : SUCCESS in 59m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7f10ea86a458469b9653d1096b77b63e : SUCCESS in 53m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3363f03b5ee7404c8b3ceceea9a7aa14 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3c409bb478940ce915334012219ed7c : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01495d6a7b074062bfeca6de9561a9bd : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b53ff092ffba47d7ad6a40dbf677bf46 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a98058d30568472aa4ed3f40f1f5266f : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb595a542aae4761b545f12cf3a09d1e : SUCCESS in 3m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/85377b70b8014b4b87aca2f7d0201852 : FAILURE in 23m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/530bbec6bd2949909a7a5d441a860dbe : SUCCESS in 58m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c03e4c6f009f499a9e6dfe49c7b592c9 : FAILURE in 18m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9f61e00a7d954d3cad67bc468a2e32ab : SUCCESS in 44m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a231a7a4d0f14f6680c140f0dba0eb43 : FAILURE in 19m 34s","accounts_in_message":[],"_revision_number":39},{"id":"333e06326710de72905a76f04cdec012983152d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 07:20:30.000000000","message":"Patch Set 39:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/373093dd5cdc4a1da9b727b66cfe7650\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d35d09d430634b3596c87acc6b21d272 : FAILURE in 15m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e98b293e1612416784c42b2e1063870c : FAILURE in 15m 36s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4f91e7b466c34166b46ea268d467b914 : SUCCESS in 1h 15m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6b2d7957a7549d0ac17fbca2b4b59ba : SUCCESS in 2m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a496fe6e522741e2bfd67897a7c912d8 : SUCCESS in 2m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/62d56718e63a4db49e3954ee766e8390 : FAILURE in 1h 29m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9ff07c67f5094051a131513dc16646ce : FAILURE in 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"c024b19aae00262b7be61f2ccfbe8bea0882bcb7","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-01-30 23:34:41.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"60d49bcc7e45aa334950fc96fe6dc255a89e9c5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-01-30 23:40:42.000000000","message":"Patch Set 40: Patch Set 39 was rebased","accounts_in_message":[],"_revision_number":40},{"id":"59dcfd6f9f23c5fdcda7d63ff667052c6c03e6a7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-01-31 00:00:33.000000000","message":"Patch Set 40:\n\n(2 comments)","accounts_in_message":[],"_revision_number":40},{"id":"e971353bba06cdd76b36b5e1476821ee7ac87d71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 01:17:46.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/125d35004d2342d7ad0436a6ca7eddf9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1f6802e6f55740a7b8dcbab9f2377bf9 : SUCCESS in 1m 44s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b49dbf6a1d984a079c8e6043decc2f83 : SUCCESS in 22m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/66e4a474b3a0402aaaa0195b34ad55d5 : SUCCESS in 1h 20m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4fddd471a5f648fc98bd13d1ee24f199 : SUCCESS in 20m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8317f92a94364acd956674bfbd8e5010 : SUCCESS in 51m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8938c4557e584c78ae8ee6dd8ddbe134 : SUCCESS in 1h 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e3589b7667497baf3df25a1a1165c9 : SUCCESS in 2m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fe68671e4c8463cb49a2d572a6bcda1 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09ebd9005d0043548bb61a6f1e981e51 : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf374edb9cf848df8a447ab18283fb9e : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70606451941b4458898171481b30303e : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f388464b09754da6905d30f8e0ca004a : SUCCESS in 2m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1e43d31cae7e450bb99950df1bfb9398 : SUCCESS in 22m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/26210f78dde04e32b0cd3fbf7e71e884 : SUCCESS in 54m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/87ec269de31148cea54dfa0d2259ea50 : SUCCESS in 14m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/da56f3985780441a99f9f4b351eef983 : SUCCESS in 41m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e75a84c5442e4d57a7c03d44e5650694 : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":40},{"id":"3c8640cf27bec538cdcb55d3328653076a8c6523","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-01-31 01:17:49.000000000","message":"Patch Set 40:\n\n(2 comments)","accounts_in_message":[],"_revision_number":40},{"id":"a3c53fa0f34e5b139ec0adde6753022cec6e5dde","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-01-31 01:18:35.000000000","message":"Patch Set 40: Code-Review+2","accounts_in_message":[],"_revision_number":40},{"id":"c7e8d7abd753e52e182bf67ecc414c7c7ed02cfa","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-01-31 02:22:53.000000000","message":"Patch Set 40: Code-Review+1","accounts_in_message":[],"_revision_number":40},{"id":"4042e83e496a3b0345e2e3e876f4b5d14ea6cd99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 02:48:28.000000000","message":"Patch Set 40:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddb688cbe40e459894ff612169b88262\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c61bd9538e054bc1bde6ad9d9d7a77b4 : TIMED_OUT in 3h 04m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9c0fa2b755264a31bf51bc540990ce6d : TIMED_OUT in 3h 03m 09s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5d915569fbe94940b903ac6910d2b3de : TIMED_OUT in 3h 02m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5d3c1aaaeb74b0ea14c5b2783204563 : SUCCESS in 2m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f67fd80665fe45ca9c5a8555b649c0f5 : SUCCESS in 4m 05s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d89e6cd2252f4b66a83e39baa93f128b : FAILURE in 3h 02m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/536560cfcf1541cea7cabdcaca8e465d : TIMED_OUT in 3h 03m 07s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"6c1abe2f7096d06c671b5b0b26c93744df3b879c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-31 02:51:33.000000000","message":"Patch Set 40: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"0869db9b165a51d140a162d9f28ffdb4924f6cc2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-01-31 03:04:34.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"1fc04d7a123f8b4b01f73aaddab871639a162795","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-01-31 08:04:16.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"3ffdaf913a95fc8fae30a37dee299a861ceb876e","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-02-01 05:57:23.000000000","message":"Patch Set 40: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"dd61594181be002d18ef8f3d1f0eedd98bb4f540","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 05:57:49.000000000","message":"Patch Set 40: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":40},{"id":"0aa2e9ed98bc6b5085a5e115152067c89e606693","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 07:11:13.000000000","message":"Patch Set 40: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb874641d54e4851b87561e5b15b2788\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c5914d60bbc544889f0deacc79c0dd4e : SUCCESS in 2m 02s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a8d61d7f23854e328918de4fae51915c : SUCCESS in 27m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/255e363357e44de5840c7ea27c2703a0 : SUCCESS in 50m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7e86c63842d64934b6c61d9733adf689 : SUCCESS in 1h 05m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/591aae67b42140c493fad1857679b324 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30446fd5b5894ab78e6adf6c797d13df : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26f72a6df9a04941a0896b00c0499c31 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ea2b0230674d44ba05c1ecaf7623f5 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/517ce81895ac4ca2890282b05b84461b : SUCCESS in 4m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cb9b80d5a55844199818ad6452f564d5 : SUCCESS in 29m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fa6262bec0984f829a9cfde0605bcd96 : SUCCESS in 50m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/558984006faf405698eea314d82590bd : SUCCESS in 36m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c6d8badf583846e791ed6095f95a05cb : SUCCESS in 48m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8fd05bd76f2c4b4c8a000c9a90bd69c7 : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":40},{"id":"5b2c53472d55dc7319a23c6246d0bda808bf8b9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 07:11:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":40},{"id":"7dfe8156ebce0a12418dd1064d1378fc7dd86277","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 07:12:17.000000000","message":"Patch Set 40:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e87e0351b690465a91eff6e7203f11d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0416cc872276424aaf74d612dfdff92a : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7690cfa3ada4680a6ac047b9a87166a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":40}],"current_revision_number":40,"current_revision":"9d4736820f9c3dc589915acd6b0aaedfeaa68912","revisions":{"4c180f1c7d9c8e871dc60d0d125470636ac88db7":{"kind":"REWORK","_number":1,"created":"2022-01-29 02:18:05.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/1"}}},"commit":{"parents":[{"commit":"5b2d5881d8182c9532da72c353e10834833777a7","subject":"Merge \"Unpin td-agent and cap elasticsearch gem\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b2d5881d8182c9532da72c353e10834833777a7"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-01-29 02:17:58.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c180f1c7d9c8e871dc60d0d125470636ac88db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c180f1c7d9c8e871dc60d0d125470636ac88db7"}]},"branch":"refs/heads/master"},"2074471954871278167ba87f4b80b723c5fe0403":{"kind":"REWORK","_number":2,"created":"2022-02-08 06:59:48.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/2"}}},"commit":{"parents":[{"commit":"941d021e091567d2709bbb868768c6dc4a3372e0","subject":"collectd: pcie-errors is x86-64 only now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/941d021e091567d2709bbb868768c6dc4a3372e0"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-08 06:45:51.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2074471954871278167ba87f4b80b723c5fe0403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2074471954871278167ba87f4b80b723c5fe0403"}]},"branch":"refs/heads/master"},"f1507f093a4c8a7e6b4ff9921a8141c2ea55428a":{"kind":"REWORK","_number":3,"created":"2022-02-08 14:55:37.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/3"}}},"commit":{"parents":[{"commit":"758b53016ac0328a584c843ef1972fa2e8572571","subject":"Merge \"Revert \"nova: drop pypowervm dependency\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/758b53016ac0328a584c843ef1972fa2e8572571"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-08 14:55:13.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1507f093a4c8a7e6b4ff9921a8141c2ea55428a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1507f093a4c8a7e6b4ff9921a8141c2ea55428a"}]},"branch":"refs/heads/master"},"e577d049cc7ea166e29ed63333d3a0bd71e05606":{"kind":"REWORK","_number":4,"created":"2022-02-09 07:29:32.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/4"}}},"commit":{"parents":[{"commit":"758b53016ac0328a584c843ef1972fa2e8572571","subject":"Merge \"Revert \"nova: drop pypowervm dependency\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/758b53016ac0328a584c843ef1972fa2e8572571"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-09 07:29:11.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e577d049cc7ea166e29ed63333d3a0bd71e05606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e577d049cc7ea166e29ed63333d3a0bd71e05606"}]},"branch":"refs/heads/master"},"7307d794c7fbb5e0e0b46e57852ed0345d579a8e":{"kind":"REWORK","_number":5,"created":"2022-02-14 05:39:10.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/5"}}},"commit":{"parents":[{"commit":"333fb35c79d2912a10b70001f5d8b5724e9b8651","subject":"Merge \"rabbitmq: update to 3.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333fb35c79d2912a10b70001f5d8b5724e9b8651"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-14 05:38:49.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7307d794c7fbb5e0e0b46e57852ed0345d579a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7307d794c7fbb5e0e0b46e57852ed0345d579a8e"}]},"branch":"refs/heads/master"},"ffcc9b4f0fb403799b5ae2befccf0768fe4df715":{"kind":"REWORK","_number":6,"created":"2022-02-15 08:12:43.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/6"}}},"commit":{"parents":[{"commit":"333fb35c79d2912a10b70001f5d8b5724e9b8651","subject":"Merge \"rabbitmq: update to 3.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333fb35c79d2912a10b70001f5d8b5724e9b8651"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-15 08:10:57.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffcc9b4f0fb403799b5ae2befccf0768fe4df715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffcc9b4f0fb403799b5ae2befccf0768fe4df715"}]},"branch":"refs/heads/master"},"38ca1c98a2603a4e902242d6cef2671e951fef22":{"kind":"REWORK","_number":7,"created":"2022-02-15 08:24:14.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/7"}}},"commit":{"parents":[{"commit":"333fb35c79d2912a10b70001f5d8b5724e9b8651","subject":"Merge \"rabbitmq: update to 3.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333fb35c79d2912a10b70001f5d8b5724e9b8651"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-15 08:21:36.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38ca1c98a2603a4e902242d6cef2671e951fef22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38ca1c98a2603a4e902242d6cef2671e951fef22"}]},"branch":"refs/heads/master"},"368b2251938ed834872308594e754e3706b1e046":{"kind":"REWORK","_number":8,"created":"2022-02-15 08:35:42.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/8"}}},"commit":{"parents":[{"commit":"333fb35c79d2912a10b70001f5d8b5724e9b8651","subject":"Merge \"rabbitmq: update to 3.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333fb35c79d2912a10b70001f5d8b5724e9b8651"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-15 08:35:02.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/368b2251938ed834872308594e754e3706b1e046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/368b2251938ed834872308594e754e3706b1e046"}]},"branch":"refs/heads/master"},"51c06fafe6048816ed7324e5dc4fd1a5d52bd948":{"kind":"REWORK","_number":9,"created":"2022-02-15 13:45:24.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/9"}}},"commit":{"parents":[{"commit":"333fb35c79d2912a10b70001f5d8b5724e9b8651","subject":"Merge \"rabbitmq: update to 3.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333fb35c79d2912a10b70001f5d8b5724e9b8651"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-15 13:46:15.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51c06fafe6048816ed7324e5dc4fd1a5d52bd948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51c06fafe6048816ed7324e5dc4fd1a5d52bd948"}]},"branch":"refs/heads/master"},"9f89c876ce9ebef36a39c3a83cd4d64db7517ecc":{"kind":"REWORK","_number":10,"created":"2022-02-16 02:12:00.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/10"}}},"commit":{"parents":[{"commit":"cabfd43f49bc2c08f8efc3e5a0c2be34e11d586d","subject":"Merge \"Add LetsEncrypt images for cert request/renewal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cabfd43f49bc2c08f8efc3e5a0c2be34e11d586d"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-16 02:14:11.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f89c876ce9ebef36a39c3a83cd4d64db7517ecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f89c876ce9ebef36a39c3a83cd4d64db7517ecc"}]},"branch":"refs/heads/master"},"f481401c6dbab91e52b89a3fdc9c748d3376df09":{"kind":"REWORK","_number":11,"created":"2022-02-17 02:08:59.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/11"}}},"commit":{"parents":[{"commit":"eb79ef827da4401f30d1944e3bba64560c472119","subject":"Updating python testing classifier as per Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb79ef827da4401f30d1944e3bba64560c472119"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-16 16:05:34.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f481401c6dbab91e52b89a3fdc9c748d3376df09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f481401c6dbab91e52b89a3fdc9c748d3376df09"}]},"branch":"refs/heads/master"},"66281bed28d456d0a5a1d8e5b12880210243ce7a":{"kind":"REWORK","_number":12,"created":"2022-02-17 06:01:13.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/12"}}},"commit":{"parents":[{"commit":"eb79ef827da4401f30d1944e3bba64560c472119","subject":"Updating python testing classifier as per Yoga testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb79ef827da4401f30d1944e3bba64560c472119"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-17 06:02:29.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66281bed28d456d0a5a1d8e5b12880210243ce7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66281bed28d456d0a5a1d8e5b12880210243ce7a"}]},"branch":"refs/heads/master"},"01d67d67690823207df94b22d37564f729760ed9":{"kind":"REWORK","_number":13,"created":"2022-02-25 02:17:30.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/13"}}},"commit":{"parents":[{"commit":"2cee929b21eb71a2dd3eacfb42592bb7c3536a6a","subject":"docs: make formatting of kolla-ansible consistent in release management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cee929b21eb71a2dd3eacfb42592bb7c3536a6a"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-25 02:16:46.000000000","tz":480},"subject":"Skyline: Add skyline dockerfile","message":"Skyline: Add skyline dockerfile\n\nAdd skyline dockerfile\n\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01d67d67690823207df94b22d37564f729760ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01d67d67690823207df94b22d37564f729760ed9"}]},"branch":"refs/heads/master"},"9609ac8155e2cda76bafec3edf13c1caa471e6a2":{"kind":"REWORK","_number":14,"created":"2022-02-25 07:41:33.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/14"}}},"commit":{"parents":[{"commit":"2cee929b21eb71a2dd3eacfb42592bb7c3536a6a","subject":"docs: make formatting of kolla-ansible consistent in release management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cee929b21eb71a2dd3eacfb42592bb7c3536a6a"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-25 07:35:24.000000000","tz":480},"subject":"Add skyline Dockerfile","message":"Add skyline Dockerfile\n\nImplements: blueprint skyline-apiserver\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9609ac8155e2cda76bafec3edf13c1caa471e6a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9609ac8155e2cda76bafec3edf13c1caa471e6a2"}]},"branch":"refs/heads/master"},"252c59e464e36d8a90c528c09bb91996d581d9cb":{"kind":"REWORK","_number":15,"created":"2022-02-25 10:18:07.000000000","uploader":{"_account_id":33196,"name":"conna","email":"yang.shaoxue@99cloud.net","username":"yangshaoxue"},"ref":"refs/changes/48/826948/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/15"}}},"commit":{"parents":[{"commit":"2cee929b21eb71a2dd3eacfb42592bb7c3536a6a","subject":"docs: make formatting of kolla-ansible consistent in release management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cee929b21eb71a2dd3eacfb42592bb7c3536a6a"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2022-02-25 10:17:41.000000000","tz":480},"subject":"Add skyline Dockerfile","message":"Add skyline Dockerfile\n\nImplements: blueprint skyline-apiserver\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/252c59e464e36d8a90c528c09bb91996d581d9cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/252c59e464e36d8a90c528c09bb91996d581d9cb"}]},"branch":"refs/heads/master"},"889bd12ad3d54780b233ac94d7b29505ba86e2d3":{"kind":"REWORK","_number":16,"created":"2022-05-27 06:43:57.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/16"}}},"commit":{"parents":[{"commit":"a51ebc9e7fc2163623cb3d9b7540c23ea984c859","subject":"Merge \"[CI] Move queue setting to project level\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a51ebc9e7fc2163623cb3d9b7540c23ea984c859"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"zhu.boxiang","email":"zhu.boxiang@99cloud.net","date":"2022-05-27 06:43:32.000000000","tz":480},"subject":"Add skyline Dockerfile","message":"Add skyline Dockerfile\n\nAdd skyline Dockerfile\n\nImplements: blueprint skyline-apiserver\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/889bd12ad3d54780b233ac94d7b29505ba86e2d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/889bd12ad3d54780b233ac94d7b29505ba86e2d3"}]},"branch":"refs/heads/master"},"be8c7cd2394ddbea1e17d45b254abec500cf142e":{"kind":"REWORK","_number":17,"created":"2022-05-27 08:13:11.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/17"}}},"commit":{"parents":[{"commit":"a51ebc9e7fc2163623cb3d9b7540c23ea984c859","subject":"Merge \"[CI] Move queue setting to project level\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a51ebc9e7fc2163623cb3d9b7540c23ea984c859"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"zhu.boxiang","email":"zhu.boxiang@99cloud.net","date":"2022-05-27 08:12:53.000000000","tz":480},"subject":"Add skyline Dockerfile","message":"Add skyline Dockerfile\n\nAdd skyline Dockerfile\n\nImplements: blueprint skyline-apiserver\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be8c7cd2394ddbea1e17d45b254abec500cf142e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be8c7cd2394ddbea1e17d45b254abec500cf142e"}]},"branch":"refs/heads/master"},"603195f48114c004346b7d6ed29539d858fb7b26":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2022-05-30 08:12:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/48/826948/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/18"}}},"commit":{"parents":[{"commit":"9d5bdbf36d477da38b0dd7d44b3fc6feb9325378","subject":"Use the new image naming scheme by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d5bdbf36d477da38b0dd7d44b3fc6feb9325378"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-30 08:12:45.000000000","tz":0},"subject":"Add skyline Dockerfile","message":"Add skyline Dockerfile\n\nAdd skyline Dockerfile\n\nImplements: blueprint skyline-apiserver\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/603195f48114c004346b7d6ed29539d858fb7b26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/603195f48114c004346b7d6ed29539d858fb7b26"}]},"branch":"refs/heads/master","description":"Rebase"},"9fa51cba9dd813c6f3e27501978a8b78559807d9":{"kind":"REWORK","_number":19,"created":"2022-08-19 10:12:34.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/19"}}},"commit":{"parents":[{"commit":"3e5369db541c60df6d5cddccc20b04f50a3d45ad","subject":"Merge \"Remove Keystone admin endpoint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e5369db541c60df6d5cddccc20b04f50a3d45ad"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-08-19 10:12:18.000000000","tz":480},"subject":"Add skyline Dockerfile","message":"Add skyline Dockerfile\n\nAdd skyline Dockerfile\n\nImplements: blueprint skyline-apiserver\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fa51cba9dd813c6f3e27501978a8b78559807d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fa51cba9dd813c6f3e27501978a8b78559807d9"}]},"branch":"refs/heads/master"},"53759f0c04859506085f1f720c9990166cb28e07":{"kind":"REWORK","_number":20,"created":"2022-10-13 06:56:55.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/20"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-13 06:56:31.000000000","tz":480},"subject":"Add skyline Dockerfile","message":"Add skyline Dockerfile\n\nAdd skyline Dockerfile\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53759f0c04859506085f1f720c9990166cb28e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53759f0c04859506085f1f720c9990166cb28e07"}]},"branch":"refs/heads/master"},"bf6467290fc3505cb6c83a84a44a2d0ccbe5bbec":{"kind":"REWORK","_number":21,"created":"2022-10-13 06:58:17.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/21"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-13 06:58:03.000000000","tz":480},"subject":"Add skyline Dockerfile","message":"Add skyline Dockerfile\n\nAdd skyline Dockerfile\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf6467290fc3505cb6c83a84a44a2d0ccbe5bbec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf6467290fc3505cb6c83a84a44a2d0ccbe5bbec"}]},"branch":"refs/heads/master"},"c75be0b41b3b490f87a37c3c53e8e6a533db42ed":{"kind":"REWORK","_number":22,"created":"2022-10-13 09:36:41.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/22"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-13 09:36:20.000000000","tz":480},"subject":"Add skyline Dockerfile","message":"Add skyline Dockerfile\n\nAdd skyline Dockerfile\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c75be0b41b3b490f87a37c3c53e8e6a533db42ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c75be0b41b3b490f87a37c3c53e8e6a533db42ed"}]},"branch":"refs/heads/master"},"81716b71284f7d4d259822694741ac4ed106c95a":{"kind":"REWORK","_number":23,"created":"2022-10-13 10:18:35.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/23"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-13 10:15:42.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console. And we will install\nthem into skyline image.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81716b71284f7d4d259822694741ac4ed106c95a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81716b71284f7d4d259822694741ac4ed106c95a"}]},"branch":"refs/heads/master"},"039d94602ce5dff063c9174df4579303f628dfa6":{"kind":"REWORK","_number":24,"created":"2022-10-13 10:21:00.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/24"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-13 10:20:44.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console. And we will install\nthem into skyline image.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/039d94602ce5dff063c9174df4579303f628dfa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/039d94602ce5dff063c9174df4579303f628dfa6"}]},"branch":"refs/heads/master"},"62c8ed66a01bcf7f8f7ced1d126bb18e8c16689d":{"kind":"REWORK","_number":25,"created":"2022-10-20 09:45:51.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/25"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-20 09:45:32.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console. And we will install\nthem into skyline image.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62c8ed66a01bcf7f8f7ced1d126bb18e8c16689d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62c8ed66a01bcf7f8f7ced1d126bb18e8c16689d"}]},"branch":"refs/heads/master"},"1b1f74f93b364c9af13b03d3ef3d304b6178c98c":{"kind":"REWORK","_number":26,"created":"2022-10-20 10:25:46.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/26"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-20 10:25:30.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console. And we will install\nthem into skyline image.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b1f74f93b364c9af13b03d3ef3d304b6178c98c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b1f74f93b364c9af13b03d3ef3d304b6178c98c"}]},"branch":"refs/heads/master"},"a72642360579262ec70822398d5a7c626b5c97d5":{"kind":"REWORK","_number":27,"created":"2022-10-24 01:24:59.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/27"}}},"commit":{"parents":[{"commit":"4a90585634c91ac78c8443b98578e8721db786f4","subject":"Merge \"rabbitmq/aarch64: use COPR for Erlang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a90585634c91ac78c8443b98578e8721db786f4"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-24 01:21:23.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console. And we will install\nthem into skyline image.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a72642360579262ec70822398d5a7c626b5c97d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a72642360579262ec70822398d5a7c626b5c97d5"}]},"branch":"refs/heads/master"},"6d5871130ec479d2e04297ac7a13f01173f83110":{"kind":"REWORK","_number":28,"created":"2022-10-24 05:30:15.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/28"}}},"commit":{"parents":[{"commit":"4a90585634c91ac78c8443b98578e8721db786f4","subject":"Merge \"rabbitmq/aarch64: use COPR for Erlang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a90585634c91ac78c8443b98578e8721db786f4"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-24 05:30:02.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console. And we will install\nthem into skyline image.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d5871130ec479d2e04297ac7a13f01173f83110"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d5871130ec479d2e04297ac7a13f01173f83110"}]},"branch":"refs/heads/master"},"4c68be80f74b7ae82b537798ee0782a8157bed18":{"kind":"REWORK","_number":29,"created":"2022-10-24 06:16:07.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/29"}}},"commit":{"parents":[{"commit":"4a90585634c91ac78c8443b98578e8721db786f4","subject":"Merge \"rabbitmq/aarch64: use COPR for Erlang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a90585634c91ac78c8443b98578e8721db786f4"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-24 06:15:48.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console. And we will install\nthem into skyline image.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c68be80f74b7ae82b537798ee0782a8157bed18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c68be80f74b7ae82b537798ee0782a8157bed18"}]},"branch":"refs/heads/master"},"08bd2c9709397519698678127903b035896f613b":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2022-10-27 15:21:11.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/30"}}},"commit":{"parents":[{"commit":"8fe9440f9511413c5aa7fa0be0a746dfc119dda0","subject":"Merge \"rabbitmq: use Erlang 25 on AArch64/CentOS/Rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fe9440f9511413c5aa7fa0be0a746dfc119dda0"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-10-27 15:21:11.000000000","tz":0},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console. And we will install\nthem into skyline image.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08bd2c9709397519698678127903b035896f613b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08bd2c9709397519698678127903b035896f613b"}]},"branch":"refs/heads/master","description":"Rebase"},"fe04b2de907d01baf7d61b5a652bfa47ac445263":{"kind":"REWORK","_number":31,"created":"2022-11-18 14:21:05.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/48/826948/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/31"}}},"commit":{"parents":[{"commit":"abfb5713ae24799bc7fc2d5b007348fab154c4fd","subject":"Add Opensearch image(s)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abfb5713ae24799bc7fc2d5b007348fab154c4fd"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-11-18 14:21:03.000000000","tz":60},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console. And we will install\nthem into skyline image.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe04b2de907d01baf7d61b5a652bfa47ac445263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe04b2de907d01baf7d61b5a652bfa47ac445263"}]},"branch":"refs/heads/master"},"39418eebcda26b6d5ad817e7d21acc1d3933381c":{"kind":"REWORK","_number":32,"created":"2022-11-21 01:54:51.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/32"}}},"commit":{"parents":[{"commit":"abfb5713ae24799bc7fc2d5b007348fab154c4fd","subject":"Add Opensearch image(s)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abfb5713ae24799bc7fc2d5b007348fab154c4fd"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-11-21 01:54:27.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console.\nWe will build both skyline-apiserver and skyline-console\nimages for skyline.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39418eebcda26b6d5ad817e7d21acc1d3933381c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39418eebcda26b6d5ad817e7d21acc1d3933381c"}]},"branch":"refs/heads/master"},"f9cc6b0d67af9fb86c7ebc38700413573a553c00":{"kind":"REWORK","_number":33,"created":"2022-11-23 02:54:11.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/33"}}},"commit":{"parents":[{"commit":"abfb5713ae24799bc7fc2d5b007348fab154c4fd","subject":"Add Opensearch image(s)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abfb5713ae24799bc7fc2d5b007348fab154c4fd"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-11-23 02:53:50.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console.\nWe will build both skyline-apiserver and skyline-console\nimages for skyline.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9cc6b0d67af9fb86c7ebc38700413573a553c00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9cc6b0d67af9fb86c7ebc38700413573a553c00"}]},"branch":"refs/heads/master"},"c27962261d95390fee0a30f5a6e1d0d662804183":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2022-11-23 05:37:26.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/34"}}},"commit":{"parents":[{"commit":"ef786de775dda0cad195fd75405e82e782f01451","subject":"Merge \"[docs] Add Kolla design philosophy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef786de775dda0cad195fd75405e82e782f01451"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-11-23 05:36:49.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console.\nWe will build both skyline-apiserver and skyline-console\nimages for skyline.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c27962261d95390fee0a30f5a6e1d0d662804183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c27962261d95390fee0a30f5a6e1d0d662804183"}]},"branch":"refs/heads/master"},"22819f3db3dc88a65881756491ba6e22b6b7b352":{"kind":"REWORK","_number":35,"created":"2022-11-23 06:59:57.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/35"}}},"commit":{"parents":[{"commit":"ef786de775dda0cad195fd75405e82e782f01451","subject":"Merge \"[docs] Add Kolla design philosophy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef786de775dda0cad195fd75405e82e782f01451"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-11-23 06:59:33.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console.\nWe will build both skyline-apiserver and skyline-console\nimages for skyline.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22819f3db3dc88a65881756491ba6e22b6b7b352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22819f3db3dc88a65881756491ba6e22b6b7b352"}]},"branch":"refs/heads/master"},"3e2063db65543d590e467d9b490548c96e35d91b":{"kind":"REWORK","_number":36,"created":"2022-12-01 05:48:07.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/36"}}},"commit":{"parents":[{"commit":"ef786de775dda0cad195fd75405e82e782f01451","subject":"Merge \"[docs] Add Kolla design philosophy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef786de775dda0cad195fd75405e82e782f01451"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-12-01 05:47:49.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console.\nWe will build both skyline-apiserver and skyline-console\nimages for skyline.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e2063db65543d590e467d9b490548c96e35d91b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e2063db65543d590e467d9b490548c96e35d91b"}]},"branch":"refs/heads/master"},"5b4a7337a423e38cebc945fbda569d69612e8c42":{"kind":"REWORK","_number":37,"created":"2022-12-02 01:57:27.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/37"}}},"commit":{"parents":[{"commit":"6be0068f376b0ae67bc81b50a97e042a88317d28","subject":"Fix plugin builds with sources using type\u003dgit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6be0068f376b0ae67bc81b50a97e042a88317d28"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-12-02 01:57:10.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console.\nWe will build both skyline-apiserver and skyline-console\nimages for skyline.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b4a7337a423e38cebc945fbda569d69612e8c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b4a7337a423e38cebc945fbda569d69612e8c42"}]},"branch":"refs/heads/master"},"aa50cb253bf3dbed4f5406e3722ada387205d99e":{"kind":"REWORK","_number":38,"created":"2022-12-08 03:15:58.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/38"}}},"commit":{"parents":[{"commit":"69e8c4b6d9904dc18425a9e46ee12204217154a2","subject":"Merge \"base: add support for all Rocky Linux 9.x minor releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69e8c4b6d9904dc18425a9e46ee12204217154a2"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2022-12-08 03:14:57.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console.\nWe will build both skyline-apiserver and skyline-console\nimages for skyline.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa50cb253bf3dbed4f5406e3722ada387205d99e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa50cb253bf3dbed4f5406e3722ada387205d99e"}]},"branch":"refs/heads/master"},"cfc6297ec1a991718515c0c1340a50c147b15ec7":{"kind":"REWORK","_number":39,"created":"2023-01-30 01:44:18.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/48/826948/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/39"}}},"commit":{"parents":[{"commit":"1371e4fc5b9d3d55f5bb48540753f740f1d1af02","subject":"Add util-linux to centos base packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1371e4fc5b9d3d55f5bb48540753f740f1d1af02"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Boxiang Zhu","email":"zhu.boxiang@99cloud.net","date":"2023-01-30 01:41:53.000000000","tz":480},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console.\nWe will build both skyline-apiserver and skyline-console\nimages for skyline.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfc6297ec1a991718515c0c1340a50c147b15ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfc6297ec1a991718515c0c1340a50c147b15ec7"}]},"branch":"refs/heads/master"},"9d4736820f9c3dc589915acd6b0aaedfeaa68912":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2023-01-30 23:40:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/48/826948/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/826948/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/826948/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/826948/40"}}},"commit":{"parents":[{"commit":"8f0d9226a80cea7b333dbc52f1d0c58f56f7e272","subject":"Merge \"CI: add ubuntu opensearch experimental job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f0d9226a80cea7b333dbc52f1d0c58f56f7e272"}]}],"author":{"name":"yangshaoxue","email":"yang.shaoxue@99cloud.net","date":"2021-09-24 02:03:16.000000000","tz":480},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-01-30 23:40:42.000000000","tz":0},"subject":"skyline: Add new dashboard","message":"skyline: Add new dashboard\n\nSkyline is a new dashboard for openstack. It contains\nskyline-apiserver and skyline-console.\nWe will build both skyline-apiserver and skyline-console\nimages for skyline.\n\nImplements: blueprint skyline\nChange-Id: I68a562f15bf4fdda82b86c446e14fc8b5594413a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d4736820f9c3dc589915acd6b0aaedfeaa68912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d4736820f9c3dc589915acd6b0aaedfeaa68912"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
