)]}'
{"id":"openstack%2Fkolla~827532","triplet_id":"openstack%2Fkolla~master~I53e2cf4fba9762867cf16faa12380eef18907d24","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-05-03 16:51:33.000000000","reason":"Change was abandoned"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-07 09:28:43.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I53e2cf4fba9762867cf16faa12380eef18907d24","subject":"Fix Ironic direct deploy \u0026 iPXE on Ubuntu","status":"ABANDONED","created":"2022-02-02 17:20:47.000000000","updated":"2022-05-03 16:51:33.000000000","total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b9b6124d76a07e0fa5d7fe0f320331354c9c8f25","_number":827532,"virtual_id_number":827532,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-02 18:49:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":-1,"date":"2022-03-05 14:51:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-02 17:56:34.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-02-02 18:49:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-05 14:51:18.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"3a7f92b0f2d4f8ba0c06a94aa9368035a5619c8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 17:20:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c42f61056b98feff1a7bf18cf6b0e0da262751d0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-02 17:56:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1def152f6b1ab871ca3fc11f6d0b7196370eb9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 18:49:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4fed363678264a5b9cf7c5a86306acd1 : FAILURE in 17m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dce54842f7c541d3bc5328b2e9da1a34 : FAILURE in 39m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/edd6cfb6478345aaa61fb03e44239ef8 : SUCCESS in 1h 08m 05s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/91e9452ce467474ea2759d03ebfeef16 : FAILURE in 21m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e9e473b161fa448a827ab48810647547 : SUCCESS in 1h 21m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/064e44d2330443c5a2eaacf84b80d102 : SUCCESS in 34m 16s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/62fa890d7d2d467dabc761176cd0d6e9 : SUCCESS in 18m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1da782e5d4bd4929ac9d53681b1d4338 : SUCCESS in 57m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/21180e05c94043e6bf0a7a2e38f77fdf : SUCCESS in 1h 21m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e94e12a51604942af04dd1a2b931471 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eb5fa506a114b03a5b1372c939babec : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2998ab6f5066411aab8580ee232aa5de : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07858a5187024a3a85a37b2e5e07b610 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67b9ad15d6fb46cea93a41f27e2c7504 : SUCCESS in 4m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4478401eb5a4c709ce5c8999d1cf9ce : SUCCESS in 27m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e889ed9ea464eb2b5c01d41306356f5 : SUCCESS in 33m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ab177ab180b24b50a09931077995a733 : SUCCESS in 52m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/be569b6396df48aba2511e2ce8a8156d : SUCCESS in 56m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f0c559abdf624125a8f70e31192d8490 : SUCCESS in 1h 08m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0422770509af15bd6ff8bef01d541fafbf3a995c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 19:15:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c2f6d4e04a4b43db89b31d8f81417188 : FAILURE in 1h 50m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a4359e59c9d24ddda2e1fa4ca3f8b109 : FAILURE in 51m 46s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8281b2b954d4e5999bcb3db4c5a31fc : FAILURE in 20m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e798b592a02424a91f202b705efa5fe : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/893387e7d15a450690dd4137d50d32e0 : SUCCESS in 5m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/85d3865cbcb24d4a8540b7cb84a8a286 : SUCCESS in 1h 36m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77c7afc62be83a94f9678964298e3688f070e2cb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-05 14:51:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"422ce94207159db95e9a803fcae6b2913d64ef89","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-05 14:53:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bb3550ae4e2e9fd8cf094a7e6d585efee34780a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-07 09:14:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d34990ca06841a7b23eb4744505ef235609b58a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-07 09:28:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9b6124d76a07e0fa5d7fe0f320331354c9c8f25","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-03 16:51:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"502b50c83bf82664552f432b07a14d5d572577f7","revisions":{"502b50c83bf82664552f432b07a14d5d572577f7":{"kind":"REWORK","_number":1,"created":"2022-02-02 17:20:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/827532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/827532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/827532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/827532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/827532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/827532/1"}}},"commit":{"parents":[{"commit":"d84cd121cad5ed30fc87f97d5332f357b7f81e3a","subject":"Merge \"Ensure set_configs sets execute bit on directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d84cd121cad5ed30fc87f97d5332f357b7f81e3a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 17:17:29.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 17:20:27.000000000","tz":0},"subject":"Fix Ironic direct deploy \u0026 iPXE on Ubuntu","message":"Fix Ironic direct deploy \u0026 iPXE on Ubuntu\n\nIronic fails when deploying nodes using the direct deploy interface or\nthe iPXE boot interface when using the Ubuntu container images.\n\nThis happens because the www-data user running apache cannot read the\nimage files created by the ironic user.\n\nThis change fixes the issue by adding the www-data user to the ironic\ngroup in the ironic-pxe image.\n\nCloses-Bug: #1959835\n\nChange-Id: I53e2cf4fba9762867cf16faa12380eef18907d24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/502b50c83bf82664552f432b07a14d5d572577f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/502b50c83bf82664552f432b07a14d5d572577f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
