)]}'
{"id":"openstack%2Fkolla~827656","triplet_id":"openstack%2Fkolla~master~Ifbaccc54896bab096a6d16dc7733941dd600cf39","project":"openstack/kolla","branch":"master","topic":"827656-patch3","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-11 11:03:15.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-02-11 11:03:15.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-11 09:23:22.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-02-11 11:03:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifbaccc54896bab096a6d16dc7733941dd600cf39","subject":"rabbitmq: update to 3.9","status":"MERGED","created":"2022-02-03 12:56:19.000000000","updated":"2022-02-11 11:05:09.000000000","submitted":"2022-02-11 11:03:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827656-827656-patch3","meta_rev_id":"6640d3afc18e46bad62df1b7f4a8282345ad15f1","_number":827656,"virtual_id_number":827656,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-11 11:03:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-02-11 09:23:22.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-02-09 17:27:06.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-02-11 09:23:22.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 14:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 09:16:09.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-02-09 10:10:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-09 15:18:36.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"36e5a5d64f6cc80f02ad48f657619fba06254050","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-03 12:56:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6134eabd21cc58828c509260a4461fdd133f5c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 14:28:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a5d2e58f48a64355879a4f78a47674b1 : FAILURE in 16m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/26dc91ea5e7949d6ad00e07b468f6720 : FAILURE in 39m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c533fe27c71245bba16e0110bbaad032 : SUCCESS in 1h 11m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/be47d657e54c48da8d90c5d7ce0aa96b : FAILURE in 24m 47s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f15f9b6784bd448e945b960c32a150ea : SUCCESS in 1h 29m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2b92842414a648f6bf09bf592b283642 : SUCCESS in 36m 57s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0a2fe4ef31d54e11aa0c8c5aad35343a : SUCCESS in 32m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4956c5004b6641ad8a71c4f50b7dd254 : SUCCESS in 56m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/af0dbed9e3ae40258ec965b97aa413cb : SUCCESS in 57m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7f71453dee42c9bc1c8689555344df : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95ae18017b07418fbcf7b9481f841b1c : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc349f7d2d644652ae6ec15dabcc1e16 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d480f6444d624a69a31553001a64ff91 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59aed07695524765ab0546a6c7b59497 : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c90b6d9471294da7a157b49e2b051ee9 : SUCCESS in 21m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/24e97fd9f5b947848192235d9236e85e : SUCCESS in 32m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf4107f8b9ca47378d3173a62f8525da : SUCCESS in 56m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d143480b19b44d9ab2b861f9f549ec1 : SUCCESS in 54m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1f955187b1f8419aa03873b8774388c0 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":1},{"id":"916b00fc440a10758b7a43a26fa1917b5cae3d8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 14:34:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd1807c126434231becb4348ca2fefa0 : FAILURE in 1h 14m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7053ad08ec49478c9451998c873d42d1 : FAILURE in 49m 50s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f22e077b8a764d2482430e0d0c45606d : FAILURE in 38m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72585c548303465c8ae2e82ca0121865 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4dcda0bccaf4c7aa618f0f40e5bb0b2 : SUCCESS in 7m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf37b52705ba4dd99ddf0d2025d165d3 : SUCCESS in 1h 29m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53cc04318fefadb601c49348d0fb6af1244a04db","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 09:16:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da70cff3b666286e12f3d7c62037064ad65b9975","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 11:23:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5d8fb00859579d1a87bc7f0cf389cc3dd75d146","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 11:44:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fb74d9d6df90137a982f292f1c2bf38ec25c623f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:21:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/195b35bb854f49beb7165b1ac5b61c40 : SUCCESS in 30m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4a392b2834a24dcf8ab4a28e1394a9ac : SUCCESS in 42m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ca3d324106174cd4bac84c90966aaf92 : SUCCESS in 1h 14m 57s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4b8102f8734b482b954215a042c1464d : SUCCESS in 1h 14m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d995641c63094461b64d29b26abc1c99 : SUCCESS in 1h 28m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9f6daf8ed170458280b719d6b9b2b05d : SUCCESS in 33m 24s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cf57dc2cc6b643fea7828a22e5dc07b8 : SUCCESS in 17m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8d33257c1924423ead5fb583f2e27994 : SUCCESS in 56m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/09465992a2ac4500848d0356c7b8cfef : SUCCESS in 1h 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3345b60c34314a149648dd45328b2db5 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44d11870ecb4452daa316fe1ecab5078 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eaaa88fb80ad409b941f43dbee1a2f24 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d586156c983741dfb4f8ce8d3f5794a6 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/946621810a1d4dc69f5d99831f207759 : SUCCESS in 4m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/08e61010c375435da58675798c8e05c7 : SUCCESS in 24m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce9c25f16b014647b3dc91898369b9fc : SUCCESS in 32m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/037859fe92be43079aab7dccd63e8f51 : SUCCESS in 51m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f2075b12e8584c1e8220fb3f5ed3b225 : SUCCESS in 55m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/93aefd1ed3a448a6ad31d7dd0cae1b97 : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":3},{"id":"7f500f0aee699f9886cca1e3eec2850503248d79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 03:52:51.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f8b97d18e705465c9f81e2bdad83eae2 : FAILURE in 2h 55m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e6820b91063e4e2892f37a0fed3561a8 : FAILURE in 1h 33m 35s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c0d3f098a1d453abe41c2b71b2c44e1 : FAILURE in 43m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93524ed2d31e4cc488c5195500d64b70 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78adcd84bf1e4069921c5f099d2b7081 : SUCCESS in 7m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8228d996ced43fe8dc9807473ea5faf : SUCCESS in 2h 19m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9718d7e3bc6da306ed4d425ab71fb3a469264bd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-05 09:38:43.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"fe39638e8a0f9d7bb08b9b2223c3523370f0d76b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 10:49:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/76aeb758253f41d5833387c15d12f361 : SUCCESS in 28m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ad6177b919734c53b95a0157d236ecf6 : SUCCESS in 35m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a96e83fd57d64a2d9503f3a2240ddf3f : SUCCESS in 50m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/557ddba1399340528c6dc3872e10656b : SUCCESS in 1h 10m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4629087ca26e4d22a0160060f43a4fa4 : SUCCESS in 1h 00m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1b40673477c049e4b8b3c13d8ac813d6 : SUCCESS in 27m 00s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8c0096b4cd784b84aa24262e2864f99b : SUCCESS in 17m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dd7a4d53bcdf46a691c3ecb8ab418d4c : SUCCESS in 53m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/53f631905af544b1a0c493e52341739f : SUCCESS in 53m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5072546b8944d25acb72b0852fc3e3f : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d0095cf4e384f93ad535c0adb7241c1 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b76f637cb534b91854e94cc662478b7 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0abffdd2f6416689754ab5be635559 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cfdfeeb86534a9580252093158c7c22 : SUCCESS in 3m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18507d1257a44c168520ea54cc0ae88d : SUCCESS in 15m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d63b2f75c1104c72bb70663ff3300a41 : SUCCESS in 31m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e956e15a8d5243918932a48446246ff5 : SUCCESS in 51m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/39a8d88c844d49fc9bafcec2c95edbb1 : SUCCESS in 40m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0061a25b0fbb42d5b661c91ee2a22f92 : SUCCESS in 47m 03s","accounts_in_message":[],"_revision_number":4},{"id":"77deb7d91f64c638a316c25f87c54b02daaf0e80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 12:43:09.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a7820bce5b7043c08f8f8353cdf7c73b : SUCCESS in 3h 00m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/25be94e859a148a399400a8d47161817 : FAILURE in 1h 17m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/af7b2ceef5f64a71b147cec811486fa9 : FAILURE in 49m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2aa9613d1cb6404ebb18b29cc5693399 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb38b6d59fea4bcf9fb8572f11d72139 : SUCCESS in 9m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1dd519c467384ec7945e88be549a98af : SUCCESS in 1h 59m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7672e181fd9e1e55b3fc133cbe33b7de7deb5a34","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-09 10:10:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6c248c6c47672e0093b8e976ed341b8f17f2954d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 15:18:36.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ef8fcb0e86aee749181ed7bb6932909e35263a88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-09 15:23:44.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"25b6da35b79e94bc29bd6fde59482e75e273ef49","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 15:31:40.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e93ac592aa8b7b98ad43d5f7babf09a5d3a7ea0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 17:16:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/46c8ba654d924f5bacac509a1ade4522 : SUCCESS in 39m 28s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4377d503b10549ea9dd61d58f49f46d6 : SUCCESS in 47m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/03e4531662294915a361bedb83844826 : SUCCESS in 1h 09m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ee77cce80ea84f3582c45ee6aba9d69e : SUCCESS in 1h 14m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/27fbe88667754bc19c4758e988d3fcbd : FAILURE in 1h 36m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a71b6b7a861b49a0b37f582e19a7d093 : SUCCESS in 37m 54s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f78ca7d35ab74aa98990ffb8cbe0b769 : SUCCESS in 18m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3fff27cc9b6b45b78098447b2bcb45d7 : SUCCESS in 1h 01m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2e976b850f284451a1673467810ee276 : FAILURE in 1h 08m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51e06e4e32734f98847dd09778924b8a : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e049d3c70ffc4fa48fb2a728a33e3a2d : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e39e7eeb732439aa8be3e13a9bbd101 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f10d64ef394952bd241fd1055a33cc : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/161c93f7d4b64591b11bce07cad02f3a : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ee3159154cf0426bb06d3386a19e4620 : SUCCESS in 21m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb509d34d05d4a6781d519cd2324667f : SUCCESS in 35m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/96df3787946e4e38859d271ea30e2542 : SUCCESS in 51m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/24b6c9768e404c4ab02740bec2262cd8 : SUCCESS in 58m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bf2da337e6f845dd9974c6165222ef40 : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":5},{"id":"0ef5eabf04c892ae49abd1bcd480e16704e9b648","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 17:26:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ce9ff3b4d59a92ff11bb237da66ce97e9fde2310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 17:26:39.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"78062baa2809e7cbb821d948d0877626aac84a68","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 17:27:06.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f489f8e3e956c53e94f8a965150b0bbaaf525ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:22:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/706b816e45024d488e4333faf9c34a55 : SUCCESS in 29m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/967e2569de3d40128d1299d464ca6b95 : SUCCESS in 37m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b6dd6236db2044c98df1b1dbdadb4151 : SUCCESS in 1h 03m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b57f7e9d0cbc4b1f82f0b958fb337c08 : SUCCESS in 1h 02m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f93c438499884dc5b869697e48748512 : SUCCESS in 1h 15m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1c0203b06a6843bfaa4bec34dd9db946 : SUCCESS in 24m 59s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1f099d41ed1d41f1834d7a76698289c0 : SUCCESS in 19m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d1c051ad5c9c4e0a8efeab3b678f0082 : SUCCESS in 49m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/dc9533b526e74ec1993b7e2be2af54cb : SUCCESS in 59m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2603f5a47e2f480cbf575cae59fb4ae0 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4bd47449b2b4cd1b8a18f2b1e4298ba : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a48c376850e94d29b4b2b874cfd77ab7 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a613e2c39454490a57fa24bb10c6383 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e472e33bb2bf4ea7aeb323b6ca030349 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a81cb34b9b9541f1b6dbb0bb84becfaa : SUCCESS in 20m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/52e4f9908c0b47b4a402c528cc04e90a : SUCCESS in 31m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c84ea961312d4285b4b92f3fffb08c83 : SUCCESS in 50m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ea3421dfdc54e1d9f35e2aa851634f6 : SUCCESS in 50m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f1e2ca598f3341289031f23bb51678c7 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":6},{"id":"8ce8bd70e32264199d0210e3315cbf1667c3ee4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 22:26:20.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e779a0d528a54c73816d7a3ba350ad09 : SUCCESS in 2h 29m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/43bfbbe86d1a4b7ab3d079bc5ecc53ff : SUCCESS in 2h 17m 29s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9fc777219d374a47ac6f25509b9f2333 : SUCCESS in 2h 30m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be21e1e25ce2409d94bbe2b3ce00b1dc : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0ad5cb8e9fa47ed90ac3bb4f985fe6b : SUCCESS in 9m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/88cc473c2af14786baa9936a0c589eb8 : SUCCESS in 1h 51m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3f532a19051a98bd7d11b6cf8b7ec92a1de0a892","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 09:23:22.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e6ce08f3d0895de070aacbfbdbae598070a1f930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 09:24:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"348ce18374a3121450e0bb3fb5fffd9f8fb3c457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 11:03:15.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dc29bc4dd79b4db8b0b7dedece8b29e6 : SUCCESS in 43m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2ae704f1bcc94cbda5bd85af982926e4 : SUCCESS in 1h 24m 15s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/37f0982e8d8f43c68213ed5c3e2e6df3 : SUCCESS in 1h 30m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ef0e4ec99dd24e00b6a496dabfb5bed8 : SUCCESS in 28m 31s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/83d45eaf3b2942e0997146c481fed682 : SUCCESS in 56m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7330de493c2f4e3bbd8e9e381878fbb4 : SUCCESS in 1h 01m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9096763c443c40e78d496eff6cd1dcd7 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca166446e55642ba9197f8e9a3bcd630 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/743ea2c63eb94ae8b3c87168469b9db2 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46e0f5d217d94efd93e879a8dff41d7d : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83f91bc336db4fbbbbc2170bd50c4fa8 : SUCCESS in 5m 17s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eab0d3068531416398c1d4eb80606268 : SUCCESS in 27m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b33d9168b5a54511b48db8064881a891 : SUCCESS in 54m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2cc2bf205bcb4888a5a5f9c002a1cca2 : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":6},{"id":"22d5f27ddc4a972aaa0050e675b0d17dbdaac20f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 11:03:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"6640d3afc18e46bad62df1b7f4a8282345ad15f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 11:05:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a760aa5b792412ea6aa70de4d07e461 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d10d81075600483aba861887232d8034 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"732860302cebbccc7344d2e4c345d26259ab979d","revisions":{"355d16f6b1464bfa82c927b0a6738d3099b56954":{"kind":"REWORK","_number":1,"created":"2022-02-03 12:56:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/827656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/827656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/827656/1"}}},"commit":{"parents":[{"commit":"d84cd121cad5ed30fc87f97d5332f357b7f81e3a","subject":"Merge \"Ensure set_configs sets execute bit on directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d84cd121cad5ed30fc87f97d5332f357b7f81e3a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 12:55:59.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 12:56:16.000000000","tz":60},"subject":"rabbitmq: update to 3.9 on x86","message":"rabbitmq: update to 3.9 on x86\n\nChange-Id: Ifbaccc54896bab096a6d16dc7733941dd600cf39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/355d16f6b1464bfa82c927b0a6738d3099b56954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/355d16f6b1464bfa82c927b0a6738d3099b56954"}]},"branch":"refs/heads/master"},"bd5fec9ea0f271c024a2a0c64b2995554219f2f2":{"kind":"REWORK","_number":2,"created":"2022-02-04 11:23:00.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/827656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/827656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/827656/2"}}},"commit":{"parents":[{"commit":"73e2fd0b38a135f5fe4ad4eeba6c4441e31bf406","subject":"erlang: use packages from Erlang Solutions on AArch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73e2fd0b38a135f5fe4ad4eeba6c4441e31bf406"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 12:55:59.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 11:22:47.000000000","tz":60},"subject":"rabbitmq/centos: update to 3.9","message":"rabbitmq/centos: update to 3.9\n\nAlso update Erlang to 24.*\n\nChange-Id: Ifbaccc54896bab096a6d16dc7733941dd600cf39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd5fec9ea0f271c024a2a0c64b2995554219f2f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd5fec9ea0f271c024a2a0c64b2995554219f2f2"}]},"branch":"refs/heads/master"},"54ea8728bbc2d56045a91cf9e2403df843fa3e91":{"kind":"REWORK","_number":3,"created":"2022-02-04 11:44:43.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/827656/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/827656/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/827656/3"}}},"commit":{"parents":[{"commit":"73e2fd0b38a135f5fe4ad4eeba6c4441e31bf406","subject":"erlang: use packages from Erlang Solutions on AArch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73e2fd0b38a135f5fe4ad4eeba6c4441e31bf406"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 12:55:59.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-02-04 11:44:11.000000000","tz":60},"subject":"rabbitmq: update to 3.9","message":"rabbitmq: update to 3.9\n\nAlso update Erlang to 24\n\nChange-Id: Ifbaccc54896bab096a6d16dc7733941dd600cf39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54ea8728bbc2d56045a91cf9e2403df843fa3e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54ea8728bbc2d56045a91cf9e2403df843fa3e91"}]},"branch":"refs/heads/master"},"0bf259e997be864d1832161374ddbb55acad32b0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-05 09:38:43.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/827656/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/827656/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/827656/4"}}},"commit":{"parents":[{"commit":"f943c7d773a280b832dde9edf9d615540b259b5c","subject":"erlang: use packages from Erlang Solutions on AArch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f943c7d773a280b832dde9edf9d615540b259b5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 12:55:59.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-05 09:37:41.000000000","tz":60},"subject":"rabbitmq: update to 3.9","message":"rabbitmq: update to 3.9\n\nAlso update Erlang to 24\n\nChange-Id: Ifbaccc54896bab096a6d16dc7733941dd600cf39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0bf259e997be864d1832161374ddbb55acad32b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0bf259e997be864d1832161374ddbb55acad32b0"}]},"branch":"refs/heads/master"},"063b7801c1260be2d5969cb575e66d9492f3bf13":{"kind":"REWORK","_number":5,"created":"2022-02-09 15:23:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/827656/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/827656/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/827656/5"}}},"commit":{"parents":[{"commit":"f943c7d773a280b832dde9edf9d615540b259b5c","subject":"erlang: use packages from Erlang Solutions on AArch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f943c7d773a280b832dde9edf9d615540b259b5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 12:55:59.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-09 15:23:44.000000000","tz":0},"subject":"rabbitmq: update to 3.9","message":"rabbitmq: update to 3.9\n\nAlso update Erlang to 24\n\nChange-Id: Ifbaccc54896bab096a6d16dc7733941dd600cf39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063b7801c1260be2d5969cb575e66d9492f3bf13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063b7801c1260be2d5969cb575e66d9492f3bf13"}]},"branch":"refs/heads/master"},"732860302cebbccc7344d2e4c345d26259ab979d":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-02-09 17:26:39.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/56/827656/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/827656/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/827656/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/827656/6"}}},"commit":{"parents":[{"commit":"f943c7d773a280b832dde9edf9d615540b259b5c","subject":"erlang: use packages from Erlang Solutions on AArch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f943c7d773a280b832dde9edf9d615540b259b5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 12:55:59.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-09 17:26:39.000000000","tz":0},"subject":"rabbitmq: update to 3.9","message":"rabbitmq: update to 3.9\n\nAlso update Erlang to 24\n\nChange-Id: Ifbaccc54896bab096a6d16dc7733941dd600cf39\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/828594\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/732860302cebbccc7344d2e4c345d26259ab979d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/732860302cebbccc7344d2e4c345d26259ab979d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
