)]}'
{"id":"openstack%2Fkolla~827740","triplet_id":"openstack%2Fkolla~stable%2Fxena~I5f446fc276e245c1e6974ceeb6ff3dccd84ee502","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-04 11:56:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f446fc276e245c1e6974ceeb6ff3dccd84ee502","subject":"Remove missing collectd packages","status":"MERGED","created":"2022-02-04 06:24:06.000000000","updated":"2022-02-04 20:33:13.000000000","submitted":"2022-02-04 11:56:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827740","cherry_pick_of_change":827679,"cherry_pick_of_patch_set":2,"meta_rev_id":"0e4e9d5855f7c4c1de580fb5ee6d98174bceaf58","_number":827740,"virtual_id_number":827740,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 11:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-04 08:50:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-02-04 20:33:13.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-02-04 10:18:48.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-02-04 10:18:48.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 08:00:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 10:18:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-04 20:33:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"dea2b3560e6ff5df0e7945a9e4c062903418a738","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-04 06:24:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eb221f8efe7c802266abbde1fe76d1635a0d508a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:00:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/260c0d1d04c64d4ea0665c1168215a0f : SUCCESS in 23m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/29a1cc084a864fca8c28ee791030f538 : SUCCESS in 32m 19s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/14ff6744b6e5477197a90143bf18fa79 : SUCCESS in 1h 13m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/328c94817c8b47ccaeb0b51c174cb65b : SUCCESS in 1h 16m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b52f977f425b49ca9155ae36fdc0d97d : SUCCESS in 1h 33m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/088dc5d163ec49128aaf84d763e3f628 : SUCCESS in 25m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7507f01922cb4e66bb1cd09110b33d9c : SUCCESS in 17m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0daf5cc730e9423a8f45480b36d91e90 : SUCCESS in 1h 07m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af941ad7ef344f8bd09679d95fe50af : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a97f590cc3a14887a5d840439f84bba6 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7abc656b488b4e658a89fa66ad874def : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a8fb99ec778428c901e32e0a3d5f16a : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22aa488ea3aa459caf01317d989b63f1 : SUCCESS in 6m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c7e773bc070464d85ae88cc629bc6b3 : SUCCESS in 19m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec949905688e4709bd3305025cf51b20 : SUCCESS in 23m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a92308dfd1e42f4af8cb253ba9256be : SUCCESS in 53m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b74b1b320bfe433eb3310f624220f762 : SUCCESS in 46m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e35a50b3d3324cdb9f6dc6d139f34c17 : SUCCESS in 1h 05m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9544b0c337a215cdd727c1d3b87748452f9d7efa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-04 08:50:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"112bd0469d2d85afb422187a72c45fdc09c2fff2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 09:28:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/055099907afc46329907a205afcf3d3e : FAILURE in 2h 40m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7a7b271e8b5f4f74acb22c6518ad3c48 : FAILURE in 34m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/672f0722b3c842848ed5427d49e3a368 : FAILURE in 40m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab4c5eefa37a44c6ba32d9ef395473a7 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1f41df443394afb8d1c331e38b07871 : SUCCESS in 7m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d6a40485e8654a2d893d9d6b1138966e : SUCCESS in 2h 27m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fbd8c845b493624b3e6f4742c49c6b7a5d396dc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-04 10:18:48.000000000","message":"Patch Set 1: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff6d5654429f5b1f52398b72231c4a176f67027c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 10:19:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bdc734095758a2626ce66073968253e4849bb93c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:56:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/47d76c8870174e9f83d0eca20c8f8e3a : SUCCESS in 45m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4e85b6e3aa0c4d23ba0787f6ee839f3a : SUCCESS in 1h 21m 04s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d491cf5426d5489886a1095477a455d5 : SUCCESS in 1h 32m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70b3b8b319f94e1c8321f4b3df7c2802 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82bdd934d8b94b4a9d493c0513adadcf : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b574ede2e3b44fd7bb482804ecd8c474 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a248151b0d774234a0a99d21f45e605a : SUCCESS in 5m 47s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca00480442c647768a54ab34a5b8da73 : SUCCESS in 32m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0e6148e4880421c83f55450fe247cba : SUCCESS in 54m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/193effc22a374899a24afefdae4e86cb : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3b526aa1578ae9e8b97b526a6078dd53521198ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:56:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"78024cea91d2f7e7afa799ffadf5a264de0ca7c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 12:00:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f27bdbf788104cf08792e42d04fc1dbd : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0e4e9d5855f7c4c1de580fb5ee6d98174bceaf58","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-04 20:33:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6a972e5dbbfa357e45e196215faa1519f5a3167","revisions":{"c6a972e5dbbfa357e45e196215faa1519f5a3167":{"kind":"REWORK","_number":1,"created":"2022-02-04 06:24:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/40/827740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/827740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/827740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/827740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/827740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/827740/1"}}},"commit":{"parents":[{"commit":"2171f0a7babe1e25381a430a079198e079202b67","subject":"Use distro provided GRUB efi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2171f0a7babe1e25381a430a079198e079202b67"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 14:52:19.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-04 06:24:06.000000000","tz":0},"subject":"Remove missing collectd packages","message":"Remove missing collectd packages\n\nSince CentOS Linux 8 repository retirement from\nCentOS mirrors - we need to use CentOS Stream 8\nversion of Opstools repo.\n\nListed below packages are missing from the repo,\nand according to the maintainer dpdk_telemetry is\nnot going to show up at all (it\u0027s basically dead).\n\nThe rest of the packages might show up at a later\nstage - but it\u0027s not a priority for the maintainer.\n\n- collectd-dpdk_telemetry\n- collectd-libpod-stats\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n\nChange-Id: I5f446fc276e245c1e6974ceeb6ff3dccd84ee502\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6a972e5dbbfa357e45e196215faa1519f5a3167"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6a972e5dbbfa357e45e196215faa1519f5a3167"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
