)]}'
{"id":"openstack%2Fkolla~827741","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I5f446fc276e245c1e6974ceeb6ff3dccd84ee502","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-04 14:49:26.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-04 11:00:32.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I5f446fc276e245c1e6974ceeb6ff3dccd84ee502","subject":"Remove missing collectd packages","status":"MERGED","created":"2022-02-04 06:24:25.000000000","updated":"2022-02-04 21:11:54.000000000","submitted":"2022-02-04 14:49:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"827741","cherry_pick_of_change":827740,"cherry_pick_of_patch_set":1,"meta_rev_id":"63300cd50fe70589666c269e2d4507ecbb195b5f","_number":827741,"virtual_id_number":827741,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 14:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-02-04 13:22:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-02-04 13:25:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-02-04 13:25:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 07:27:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 10:20:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-02-04 11:00:32.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-04 13:25:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"582468c8ef68a0fdc4fbe3102dc37e0da357346c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-04 06:24:25.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"bcdc5886f5f633cb1a0e6311136d51869ff37454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 07:27:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f4dcc9bf307e421a95018235575461e4 : FAILURE in 23m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4610f2ec8d6c4828898129cbc13bb5e2 : FAILURE in 37m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0ee92dd8c7414faba3a6282786f0bb35 : FAILURE in 19m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8937c9cb5c3142178ddb3cba156d5ebf : FAILURE in 21m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4d3336bcd2c44ad4a2d5721c3d2271c3 : FAILURE in 19m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/185bd3df99614dd58c1cf9d3b4ad8d90 : SUCCESS in 25m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c696d73f22334f7e823244d0212deb15 : SUCCESS in 16m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/117e8f219c2e4e0a93b0a2d4950d4019 : SUCCESS in 47m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed1826d05d684172abbf38930be5f74e : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1571c2146b18440d9d7e4fbaf22cfd88 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12131b50b783447a97aff621f1452c65 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccaa0b39be604a3f9a56ca3cfb0fc1d5 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca0a0ab731c46c984eb873efbce9150 : SUCCESS in 6m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/611f276f03c94cc88bc640e2d8b7a601 : SUCCESS in 20m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c384669f38d3482a85ac88c569b4be9f : SUCCESS in 26m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/330a7f5cc42d498ab7c3123d66867674 : SUCCESS in 55m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/57f87438eec54f6c9990c0a9c912b814 : SUCCESS in 42m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/10a0d801d7af4314a5fe1504a11ec393 : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b7b31a17d94da718258bc40194f8ae980eeeeed6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-04 09:02:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"acb7777c68c65501a3f8e61730821dfc0d7f51d2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-04 10:20:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5ca17d0943a39e01d4a592970e9dd3eb2d144e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 10:28:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7e16758c605040ca9fb09a18198c8a73 : SUCCESS in 24m 30s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e3b3cbbade644b12ac1c9074edf4867a : SUCCESS in 44m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ec619e55ee094af0875af02131908c6b : SUCCESS in 1h 01m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/90fa109a14d94f838d90f2b382f9bdba : SUCCESS in 1h 13m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fbd42535d59d4b3580cd3994eb78cf9c : SUCCESS in 1h 23m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8cacdc29c56741b58fd80ff8c5f36609 : SUCCESS in 25m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0cff69e57e9e49e080d2e3446d890d7f : SUCCESS in 16m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9e7aeacb8ac3482484c52052aa6a09a8 : SUCCESS in 51m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6322d90267bf49d48825e97c243aed5f : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50aa29f71e824630ad9ed060729cdfa0 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bef18774be2c4d8e9587ab81a876e322 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28bac0cf26bc4a8b89e2e144fb80094a : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ccb857b1094425acfda5e475b6868a : SUCCESS in 6m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b92c71402e0842859f9d06cbaeded26c : SUCCESS in 29m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca564520203a44419c4978e59733b88f : SUCCESS in 33m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a37d693da2be4e85adc6c37a8e4bf50f : SUCCESS in 1h 04m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/31db7fda629b4773bfd1f0ce9ef1d27c : SUCCESS in 54m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9af5bb0f07734b719aa169ae297c9b3a : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":2},{"id":"55510689c61d69863e6f9e802736c866065392a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-04 10:40:57.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3421b808df6da5f6ca88eb74d7fd7aef93a37b0d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-04 10:41:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2fed6fc08a95ac0d0019fa7396b7d280479c6c39","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-04 11:00:32.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"497b0770e0a29bb0c77b7edeb2b3a225cd63b8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 12:16:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f4c8ff3a019249b0ab51f3e60375d378 : SUCCESS in 26m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b632c581f0c842998d101f3a55ff25ce : SUCCESS in 41m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5b5dc439b71641fe8a92e8d3909ae0bc : SUCCESS in 1h 10m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1df12463b16243a79ed96e6d82390d9e : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fbe10fcd8a8b462fa2391e55601d524d : SUCCESS in 1h 25m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dec1c97ac91149658ffb53908a50f72a : SUCCESS in 32m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c150207c25fa41ffb8dbb64bb37c814f : SUCCESS in 19m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/496c03c0e136474b8a6433b207be54e5 : SUCCESS in 52m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d429e57ca734dc3a3c6d7d8081aeefe : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26d10334b1fb4feb8ee4854a8d47db0d : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d9b69696c0a4950bbe8abc794884fee : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78e42b5702d946c48658eca34f7f0643 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce7e4e059f249b1be9851b6b32f5839 : SUCCESS in 9m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b95a3e72cc014736a746793702bfc272 : SUCCESS in 22m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d589382eca54b02833b0871e1379a5a : SUCCESS in 41m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d55af75b6ae74d1990ccaecef48396b5 : SUCCESS in 51m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/84052fe7750e4c8e81cb873af0b861e9 : SUCCESS in 55m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/80e1f9b2794d4b39a503d0f6a4d7a94c : SUCCESS in 1h 05m 11s","accounts_in_message":[],"_revision_number":3},{"id":"1c41a8c1d806fcd3d46d9062f1b506453e2819e8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-04 13:22:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8873c7dc58af0ba6d91023bbc657c7a65ed49a6b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-04 13:25:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"24b86e824b3fd3b9c693db3d367b034fcf34ca2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:27:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"58d2958e92425f9a30287866572b6bdb954211fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 14:49:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/14c8d6aa667343898b4b645962939346 : SUCCESS in 40m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b27482d097b24c068a004681a8373bbb : SUCCESS in 1h 02m 12s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e754202e3dc64485a33e9c7005e36be0 : SUCCESS in 1h 20m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09c76f810f4647c8b7043286e44776a3 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5991a52ef984a14bae6c4a01dc996bf : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39826444501f4fa28e9f555d7cba301a : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62bdf95944524d4aaf4a2f69bba1769d : SUCCESS in 7m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f0474d35a08456fb27cb9b1f5090081 : SUCCESS in 26m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e30225834cb341799a82cf07b9749939 : SUCCESS in 58m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/78398372b45f40559b70d62068e3ba51 : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":3},{"id":"bb55a8d69df45ab97f0aab41cc415f9a41c87797","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 14:49:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f5b3299d77f87c0fa35fb5b5712ab3937b6063a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 14:52:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa006f70efb54a77b9a28b8175d12eae : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3},{"id":"63300cd50fe70589666c269e2d4507ecbb195b5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 21:11:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/493f72b8e4054ad0975b4afdbbfb3c6f : FAILURE in 2h 27m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c570f2c5719c4ac6afee7fbac86063f5 : SUCCESS in 2h 27m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d912a5336e764004b8cea0f6ceaa5bd2 : SUCCESS in 2h 33m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ee39085397cb4bd8b305b65c51dbd586 : SUCCESS in 2h 26m 31s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f439afa42e9f5074e4680987e6185c4bd93aa090","revisions":{"46fc5de8516119dcebd62c25ad266ab723164cb4":{"kind":"REWORK","_number":1,"created":"2022-02-04 06:24:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/827741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/827741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/827741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/827741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/827741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/827741/1"}}},"commit":{"parents":[{"commit":"67e4f50bf7cd7166b27dcd38611215327b722cb8","subject":"Use distro provided GRUB efi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67e4f50bf7cd7166b27dcd38611215327b722cb8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 14:52:19.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-04 06:24:25.000000000","tz":0},"subject":"Remove missing collectd packages","message":"Remove missing collectd packages\n\nSince CentOS Linux 8 repository retirement from\nCentOS mirrors - we need to use CentOS Stream 8\nversion of Opstools repo.\n\nListed below packages are missing from the repo,\nand according to the maintainer dpdk_telemetry is\nnot going to show up at all (it\u0027s basically dead).\n\nThe rest of the packages might show up at a later\nstage - but it\u0027s not a priority for the maintainer.\n\n- collectd-dpdk_telemetry\n- collectd-libpod-stats\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n\nChange-Id: I5f446fc276e245c1e6974ceeb6ff3dccd84ee502\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46fc5de8516119dcebd62c25ad266ab723164cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46fc5de8516119dcebd62c25ad266ab723164cb4"}]},"branch":"refs/heads/stable/wallaby"},"951c9f24ca9f406435496da1d4e51cc42298b102":{"kind":"REWORK","_number":2,"created":"2022-02-04 09:02:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/827741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/827741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/827741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/827741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/827741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/827741/2"}}},"commit":{"parents":[{"commit":"67e4f50bf7cd7166b27dcd38611215327b722cb8","subject":"Use distro provided GRUB efi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67e4f50bf7cd7166b27dcd38611215327b722cb8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 14:52:19.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-04 09:02:20.000000000","tz":0},"subject":"Remove missing collectd packages","message":"Remove missing collectd packages\n\nSince CentOS Linux 8 repository retirement from\nCentOS mirrors - we need to use CentOS Stream 8\nversion of Opstools repo.\n\nListed below packages are missing from the repo,\nand according to the maintainer dpdk_telemetry is\nnot going to show up at all (it\u0027s basically dead).\n\nThe rest of the packages might show up at a later\nstage - but it\u0027s not a priority for the maintainer.\n\n- collectd-dpdk_telemetry\n- collectd-libpod-stats\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n\nTemplate override used to temporarily use Ceph Nautilus\nrepo from RDO.\n\nChange-Id: I5f446fc276e245c1e6974ceeb6ff3dccd84ee502\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/951c9f24ca9f406435496da1d4e51cc42298b102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/951c9f24ca9f406435496da1d4e51cc42298b102"}]},"branch":"refs/heads/stable/wallaby"},"f439afa42e9f5074e4680987e6185c4bd93aa090":{"kind":"REWORK","_number":3,"created":"2022-02-04 10:40:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/827741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/827741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/827741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/827741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/827741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/827741/3"}}},"commit":{"parents":[{"commit":"67e4f50bf7cd7166b27dcd38611215327b722cb8","subject":"Use distro provided GRUB efi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67e4f50bf7cd7166b27dcd38611215327b722cb8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 14:52:19.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-04 10:40:54.000000000","tz":0},"subject":"Remove missing collectd packages","message":"Remove missing collectd packages\n\nSince CentOS Linux 8 repository retirement from\nCentOS mirrors - we need to use CentOS Stream 8\nversion of Opstools repo.\n\nListed below packages are missing from the repo,\nand according to the maintainer dpdk_telemetry is\nnot going to show up at all (it\u0027s basically dead).\n\nThe rest of the packages might show up at a later\nstage - but it\u0027s not a priority for the maintainer.\n\n- collectd-dpdk_telemetry\n- collectd-libpod-stats\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n\nTemplate override used to temporarily use Ceph Nautilus\nrepo from RDO.\n\nChange-Id: I5f446fc276e245c1e6974ceeb6ff3dccd84ee502\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f439afa42e9f5074e4680987e6185c4bd93aa090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f439afa42e9f5074e4680987e6185c4bd93aa090"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
