)]}'
{"id":"openstack%2Fkolla~827835","triplet_id":"openstack%2Fkolla~master~If1844434ff0b7c3b35d0a8e742c7e2fad0dc8758","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-02-08 11:00:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If1844434ff0b7c3b35d0a8e742c7e2fad0dc8758","subject":"stringify docker-squash version","status":"MERGED","created":"2022-02-04 10:48:42.000000000","updated":"2022-02-08 11:02:56.000000000","submitted":"2022-02-08 11:00:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827835","meta_rev_id":"06305d445a1a08f8bef63218df712cafe67e487c","_number":827835,"virtual_id_number":827835,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 11:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-08 07:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-08 09:38:13.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-02-08 09:38:13.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 12:29:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-08 07:11:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-02-08 09:38:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"e8bd1307d54800c8b6f7582be7a751da6d92a207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 10:48:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7645c28e46c400d720703add11ce6282025ecb31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 12:29:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/679ff31d953c491697b0c4f88d666dc7 : SUCCESS in 30m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/804bc733cf1a41bc9524cfc43022d066 : SUCCESS in 43m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/96e53427a0374894898505bf45001500 : SUCCESS in 1h 17m 53s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/404c2049fc794b4295b69d0bc1434070 : SUCCESS in 1h 09m 12s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9b732a8a20bf4ecbad9e059d521e7aae : SUCCESS in 1h 37m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6d523c9045854195b466b7f456ace40f : SUCCESS in 31m 43s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/065f09f8faed440ea60c2d970693fd54 : SUCCESS in 19m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e17f6a605d8641d3bd396676865e9307 : SUCCESS in 59m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4054cd3d9d734a0a8a98afea7139a89c : SUCCESS in 1h 07m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6793363fdd454d8c8e648e01e72c3736 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5023735a2da44e63a55bd9c38dd27637 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd464fd4dbee44f59305d3631cd85eba : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f635e91fa4340af834f2b4b6ecb1ea5 : SUCCESS in 6m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3bf44d16f4fb4302abb1d0370c664bbe : SUCCESS in 20m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7a4f0dfc54e4c90a293987ca3b8a8d6 : SUCCESS in 30m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0c8c0e8193144b78eff94a7416c3547 : SUCCESS in 54m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e411c5e8de1463a8ed8cb4a8e818891 : SUCCESS in 48m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eb8e891844df4695922a6985b3f44bc1 : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1f04b9f420289176497458e7782c5e72dca854b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 07:34:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a6463509d5a468fb8bc60c045e3df2a : FAILURE in 2h 24m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2d0d7d3809d642c6820893053f7d80ff : FAILURE in 1h 07m 40s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4110e45e2b7544b48791adc917998eb4 : FAILURE in 38m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/896b62938a644d5398f47f00fc13bfed : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/64fd826dc95c44038443365dbf6ca798 : SUCCESS in 6m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/05ad45fc610441668a97ae4aac1cb5a8 : SUCCESS in 2h 01m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad201d1cdcd74dbb723be6eadc6d42275997b09f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 07:11:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb61a83606dd6c6a721e4bcc363535ab2b975405","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-08 09:38:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53554d7d75b40906df60ba6a325cea3979636a95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 09:38:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c556feb28b5154937759a1066aa9c81256590e4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 11:00:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/44067e4c11244084b07c8e8bc3f9cfeb : SUCCESS in 39m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b74506a4a0c64529a699ddeaece0d62b : SUCCESS in 1h 14m 02s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e0c92d66ea1141389d008cd6cd86ed1b : SUCCESS in 1h 17m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8db426b81b764676aaacd70cb58d1842 : SUCCESS in 31m 27s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/38f9590fe9c645cabb707ba42b5e8aa2 : SUCCESS in 52m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bca537ba461f425b88bd926ccc88b0dd : SUCCESS in 1h 00m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eef5b5aa2a844ee9cec0e4f988ec034 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9feb4aff4bd448a4bff1009acb96bd0b : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1aea4fa4d2b74c7b800a0a5f6e9c718c : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f94a1449bd5473ca837960f87432d3c : SUCCESS in 6m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/621f9292658045a0ba5108924ce39a1d : SUCCESS in 28m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6fe4d4c28cd40f3814d222377b9e596 : SUCCESS in 54m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7809ba59683148339d178ea5f8a1a5d1 : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":1},{"id":"321425af7aaeda944eee03a785744a74a060e980","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 11:00:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"06305d445a1a08f8bef63218df712cafe67e487c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 11:02:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce5f5833fdb483c89ad9fea40bac1c8 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bbb1f692cdb2cdf5ddfb8b4d7012e7dee52787d8","revisions":{"bbb1f692cdb2cdf5ddfb8b4d7012e7dee52787d8":{"kind":"REWORK","_number":1,"created":"2022-02-04 10:48:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/35/827835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/827835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/827835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/827835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/827835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/827835/1"}}},"commit":{"parents":[{"commit":"405fca90f2c8d3763ac56ce8b925611d136bba6c","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/405fca90f2c8d3763ac56ce8b925611d136bba6c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 10:47:29.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 10:48:36.000000000","tz":60},"subject":"stringify docker-squash version","message":"stringify docker-squash version\n\nBefore:\n(kolla) 11:46 (s) marcin@puchatek:kolla$ ./tools/build.py --squash\nINFO:kolla.common.utils:Image squash is enabled and \"docker-squash\" version is b\u00271.0.8\u0027\n\nAfter:\n(kolla) 11:46 (s) marcin@puchatek:kolla$ ./tools/build.py --squash\nINFO:kolla.common.utils:Image squash is enabled and \"docker-squash\" version is 1.0.8\n\nChange-Id: If1844434ff0b7c3b35d0a8e742c7e2fad0dc8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbb1f692cdb2cdf5ddfb8b4d7012e7dee52787d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbb1f692cdb2cdf5ddfb8b4d7012e7dee52787d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
