)]}'
{"id":"openstack%2Fkolla~827848","triplet_id":"openstack%2Fkolla~master~I8c740655c7f7803d9ab36dac79d30e35e65dd586","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-02-08 08:33:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c740655c7f7803d9ab36dac79d30e35e65dd586","subject":"pin out some package from Debian OpenStack Team repos","status":"MERGED","created":"2022-02-04 11:36:29.000000000","updated":"2022-02-08 08:36:36.000000000","submitted":"2022-02-08 08:33:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827848","meta_rev_id":"b86c6a7190a5c9d6731266380576fd691034b38c","_number":827848,"virtual_id_number":827848,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 08:33:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-08 07:10:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-02-08 08:36:36.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-08 07:10:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-02-08 08:36:36.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-08 08:36:36.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 13:23:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 20:35:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-08 07:10:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ab08eb847cc56c405951793aa177b5f763041224","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 11:36:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f7a62069d68c1e10cb94230da29a9a6d97265dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:23:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bde9044ec17e4e44ab66fc3150df5347 : SUCCESS in 29m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1756ec985f7946feae19d0380e6da9ef : SUCCESS in 38m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dbcc355246104db5bd54d781096198ec : SUCCESS in 1h 09m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/87ca1d0bc682460fae86bb061963c3b9 : SUCCESS in 1h 04m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b2a7341a0afa43e884dd2f58c03db701 : SUCCESS in 1h 33m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4a1a09190b5d41bbb733029a7944d5ba : SUCCESS in 27m 11s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9c358bb5cc094184ab23babbcd8d3aaa : SUCCESS in 18m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b76b55fc767f41e190c8532ef603273f : SUCCESS in 59m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2a2baada16684d1489dadf2bee549e5c : SUCCESS in 1h 33m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3828d9c5f7d448c79dc3fb07d0cc3623 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7398fe79b48428b8f98f42557f6dec9 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d8f9dff86cc423384efed5a5c4d856f : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db9e9e6379e44a2a8326c7352323c4b : SUCCESS in 8m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f3ab27625a8d40b395c637693c63ed3f : SUCCESS in 19m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f66f2f62722b468191b429c61c400c6f : SUCCESS in 34m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/318b65f014f5433a8179db31a59de324 : SUCCESS in 50m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/efede1dcc6614bed9e4b3e70dc0bccef : SUCCESS in 45m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac68352a1df54775acf2c4930a86bd0f : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d10decdda3eec61cf8b38491cb1446a442aa61b5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 16:18:47.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"537ad13edc114d579b01140a9a93654635951ff8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 20:09:06.000000000","message":"Patch Set 1: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9cd6c85cffe94fb8411024dd618864253f37b495","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-04 20:35:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06198c06bacb78db12e732c71cdacc1cddd7e289","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 00:55:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf9148acb99a4f5b8b0a5a4ff935f7d8 : FAILURE in 2h 21m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/99c3581a27954b91923941363110e461 : SUCCESS in 2h 21m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/90fad1ca6fda44c9ae808ac7f5b222cf : SUCCESS in 2h 48m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1a2519a5f0d4cd7bdbbb028f1343d9b : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/023d1958ee1c432ca8797c96f7788cac : SUCCESS in 6m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/29e5325a1c11465e87e73498b9bc8ed5 : SUCCESS in 2h 05m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"611f9c3a669e52ed11b692298b0688afc3e965a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-05 09:30:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4d4afb32f28a9b59f38287707da399ed09082841","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 11:02:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/dcf2100320cb42019f296666e828a71d : SUCCESS in 31m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a5ab43fe156c4d75b6d45df694771c38 : SUCCESS in 44m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c744491f5caf47fdb89ef99b45be193a : SUCCESS in 59m 49s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9d286d01cf0743babc214de7d42641e1 : SUCCESS in 58m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/966ed34a9efc42c68dc5ff16ad3a6a3f : SUCCESS in 1h 08m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/47f7a736483a442d8006fff881267642 : SUCCESS in 22m 28s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f508d1047231488eaa5e67b280c10edd : SUCCESS in 17m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/403e07b908714bdaa59ecfec6ef6ab1e : SUCCESS in 59m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/22cf57cfe74745bb8bcc2e545792376e : SUCCESS in 1h 26m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c483c3680d8416594a90a98389e3265 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ec9aa7b987b443ea8b4630d64ae07c8 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49c9eb4797b84ae7a6d5107c3c424515 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cbeccc81bc84b6c9b6967a45331562c : SUCCESS in 6m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f28e7941d55b4abea933c8d957cdf5d8 : SUCCESS in 27m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/395b77c25d15463da2c08e316acf2b81 : SUCCESS in 34m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f63dc7d4f3c43899d505136fe7bda93 : SUCCESS in 56m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d4050999bd9d41c8bec918a7f4a74ce2 : SUCCESS in 45m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ceef67edaaf4726ad1b8510538605f5 : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":2},{"id":"d85c8b970dd1675e8c9a4bd0d64de27b2f483598","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 22:10:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b4d22890dee4adc87828d02d6408b43 : FAILURE in 1h 57m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f1ec6a8c7e9493aa762635d558c7cfc : SUCCESS in 2h 38m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/04d249eb2bb8480cbffee45721cc228d : SUCCESS in 2h 41m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fcc655891bfc45a0b106e255103888b8 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7633e0d5219f4e1a938e82c718f59b4c : SUCCESS in 8m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c3feec5b483640d188e187607da18a2b : SUCCESS in 2h 52m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"becb498c367b1b53739678b3508015c1d5d16d95","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 07:10:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f83deba87f044f63f77fb894eca6bf98f2710d4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 07:10:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"740bbb2a86f0739b001c81ab332e6e40fe6c5eab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 08:33:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/56768782cb4e494caa308a4db06f8abe : SUCCESS in 43m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1aeee725ea9c46d5a239e2ee8da51c98 : SUCCESS in 1h 10m 06s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e477d40246b34a20b7bd7a9da10eda98 : SUCCESS in 1h 21m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2befc8b2e62a4796a0700255b83191b2 : SUCCESS in 33m 27s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c0de447ede8244b885d10bb915d651f6 : SUCCESS in 56m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9aec24d7828a4f23ba8a83f44ef11134 : SUCCESS in 1h 07m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc2cb74faaa645d1b438113aece62dab : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c424583bfd954864878c83a71d226156 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fd78be8f24044ecafef55a04d757c8c : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/581074082b4d41909551f0f1778c1683 : SUCCESS in 7m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fbe53d764c9f4a0baf7d1499e4719e51 : SUCCESS in 28m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4540ea9c63ba4b20968ea2be5b88622f : SUCCESS in 50m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/43a904db82da43c89035e122c3f17dd6 : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":2},{"id":"36bdcf26bfa5ed1e098e91230cd9c14d182df5b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 08:33:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"20c467461503c2750ff098cd5e21fa5d7ae20b37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 08:36:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f5d06e17b84b738fdb288743fb1506 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2553c1845c7c4cf04d720b820e4bb13519eca4c9","revisions":{"8a3d4cb49716ddea092707b4e06b353d54544274":{"kind":"REWORK","_number":1,"created":"2022-02-04 11:36:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/827848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/827848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/827848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/827848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/827848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/827848/1"}}},"commit":{"parents":[{"commit":"405fca90f2c8d3763ac56ce8b925611d136bba6c","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/405fca90f2c8d3763ac56ce8b925611d136bba6c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-02-04 11:34:32.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-02-04 11:34:32.000000000","tz":60},"subject":"pin out some package from Debian OpenStack Team repos","message":"pin out some package from Debian OpenStack Team repos\n\nDebian OpenStack Team backports many packages to get their stuff done.\nSometimes something goes wrong:\n\nINFO:kolla.common.utils.kolla-toolbox:The following packages have unmet dependencies:\nINFO:kolla.common.utils.kolla-toolbox: git : Depends: git-man (\u003c 1:2.30.2-.) but 1:2.34.1-1~bpo11+1 is to be installed\nINFO:kolla.common.utils.kolla-toolbox:E: Unable to correct problems, you have held broken packages.\n\nSo let pin out problematic packages as we are fine with bullseye version\nof git.\n\nChange-Id: I8c740655c7f7803d9ab36dac79d30e35e65dd586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a3d4cb49716ddea092707b4e06b353d54544274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a3d4cb49716ddea092707b4e06b353d54544274"}]},"branch":"refs/heads/master"},"2553c1845c7c4cf04d720b820e4bb13519eca4c9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-05 09:30:41.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/827848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/827848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/827848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/827848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/827848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/827848/2"}}},"commit":{"parents":[{"commit":"941d021e091567d2709bbb868768c6dc4a3372e0","subject":"collectd: pcie-errors is x86-64 only now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/941d021e091567d2709bbb868768c6dc4a3372e0"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-02-04 11:34:32.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-05 09:30:41.000000000","tz":0},"subject":"pin out some package from Debian OpenStack Team repos","message":"pin out some package from Debian OpenStack Team repos\n\nDebian OpenStack Team backports many packages to get their stuff done.\nSometimes something goes wrong:\n\nINFO:kolla.common.utils.kolla-toolbox:The following packages have unmet dependencies:\nINFO:kolla.common.utils.kolla-toolbox: git : Depends: git-man (\u003c 1:2.30.2-.) but 1:2.34.1-1~bpo11+1 is to be installed\nINFO:kolla.common.utils.kolla-toolbox:E: Unable to correct problems, you have held broken packages.\n\nSo let pin out problematic packages as we are fine with bullseye version\nof git.\n\nChange-Id: I8c740655c7f7803d9ab36dac79d30e35e65dd586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2553c1845c7c4cf04d720b820e4bb13519eca4c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2553c1845c7c4cf04d720b820e4bb13519eca4c9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
