)]}'
{"id":"openstack%2Fkolla~827883","triplet_id":"openstack%2Fkolla~master~Iad7e7ebb1fbf4d14f20960d6257345db71d0e214","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-02-04 23:29:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad7e7ebb1fbf4d14f20960d6257345db71d0e214","subject":"collectd: pcie-errors is x86-64 only now","status":"MERGED","created":"2022-02-04 16:15:45.000000000","updated":"2022-02-05 07:38:02.000000000","submitted":"2022-02-04 23:29:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827883","meta_rev_id":"94200bdde110ba70f95bf42b1bdac7e3d8702869","_number":827883,"virtual_id_number":827883,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 23:29:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-02-04 20:39:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-02-04 20:39:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-05 07:38:02.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 16:45:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-02-04 17:39:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 20:38:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"f27710e79c3793e4399d7aa11e54c2a8a8d11572","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 16:15:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc4d3f8b85730120b2ec41b477c65352b31139bd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 16:16:10.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d666dd4899fa641e1d96861ea7f0441a076cf3d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-04 16:45:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"53ed2c5e8a0a0bf0f24e04424816f32754d8f19d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 17:39:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fd7c72c7e09240829208dd04edc2ac18 : FAILURE in 31m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b5cfd90f36e54aaa82c8a92b328b58ea : FAILURE in 42m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3591b78e95204354b0f56225fc57942d : SUCCESS in 1h 06m 54s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b024706a950044df8f82405903340b12 : SUCCESS in 1h 06m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/269a43cec7e544b0b7fe32c0c6488a9d : SUCCESS in 1h 17m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/890521ebea26405e8088cb9db04e91e1 : SUCCESS in 33m 29s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9e2a76a028704462a3d5649405a6d522 : SUCCESS in 18m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bf622ea905c549b1bcd09dd9287ca352 : SUCCESS in 1h 08m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/095f4de1633e42338cdedd71867d2f94 : SUCCESS in 1h 05m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ba10025668b46328adfcbab0203c889 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef3e5bce02af43eb8e3dba661a8dc636 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba93d338c7cd40ebb555d9d790c91c7f : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2701b78df86c4c18ba1e5e8ee2faceb8 : SUCCESS in 7m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0efe49a5515e4326b9d466f9c79ec9ef : SUCCESS in 21m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/41c391cc52034b22beb8fa8e0cd7d3ca : SUCCESS in 36m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f62a9ede80d4c79aec68f981a6046f0 : SUCCESS in 55m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94fabcef7fc945d89f4cbbb6a7ae9e3b : SUCCESS in 52m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0d7c360943d74cb7b120ad73f2dd0e79 : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7c026adbfbee7e189ee0fd4b63fe04ad53da45b0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-04 20:38:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35cd70ac59e7a30d706e60d0f3beeeb39cac9fa6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-04 20:38:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6768620306c59d5643a6490af809e4af254c41e5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-04 20:39:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af846da46446def897a4ef69649c6d57bde5ccad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-04 20:39:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6e846c60244d3a01d7147f06ecf1c36d425d02f1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-04 20:39:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"830356be303cb6883f8f8a363fb80af0a14c677f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 22:07:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4f918a9bc84943808c87efe9ffcd46d3 : SUCCESS in 31m 03s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/38532c42934d47bb8992504cd13dda03 : SUCCESS in 46m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ca73238680c949e09aa9c41014882187 : SUCCESS in 1h 15m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/973e1193dd2e4bfaac8f787fe775599c : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4abfcf05446145169076fe7944a21732 : SUCCESS in 1h 24m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e1ac338feb294def99ca5ecbd125523e : SUCCESS in 38m 28s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/35d503c1886644dcbb877c3d25e424ad : SUCCESS in 22m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6cb5b73ca15d4709b777ce5b5c8a164d : SUCCESS in 1h 03m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/52215f0c942c432292dbf086e879c1cb : SUCCESS in 1h 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7735a90e53b4de0a12f83ed988fd929 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86f50ebe4125444e9a54b81d024b955f : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36026ebc4a724668bf13e0ba1a142ebe : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/704201c7711847b591c75fad121bf5e6 : SUCCESS in 9m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e23e7feae3c432a8453963b84de32d4 : SUCCESS in 19m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cadad4c8bdc745b6899f9347e9c8666b : SUCCESS in 33m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a62e8959636f402984d3f9aebb85c6d7 : SUCCESS in 52m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eaca2863038441458a71b56e403f860a : SUCCESS in 57m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/775078938e7f474489f74bd50ee114a5 : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":2},{"id":"85e37aae27915a6aca028e7911aba5cff71606ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 22:07:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c198edfeaf69ada739e5ca36475590074a09094b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 23:29:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/95c14aaf992a45079a3b8fc19fc551c3 : SUCCESS in 41m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b0e419c9375341efa11d688b63aa0a0e : SUCCESS in 1h 10m 19s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d656b68581eb456c8f9e77b7e1fe600a : SUCCESS in 1h 17m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b5e7818e8b5b48ce956a162e21abfcd4 : SUCCESS in 36m 40s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/df1007fe397d4c43ac26f7b79b1a113e : SUCCESS in 1h 04m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bff416b6a4804edeada5ffd193dd309f : SUCCESS in 1h 05m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9fc694d28be41d396cd4cd73c8624bc : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4330ed22996e4519bb446137ef83fe06 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f31559856fc431784780cd073433312 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c452874e3934944b20f6b7c59a6dfb0 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3ecea46b3934f43b5b22a023529cb7b : SUCCESS in 27m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64eeb7853eb3496586ec9b72c00e1eae : SUCCESS in 49m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/68a8fad6b6de4849a48f0f9cb9c0be8c : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":2},{"id":"7a6247c190e8cd47f9f3ca50646a40d88bd1626d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 23:29:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b06d6773f7eb400bffc6109373dcdd0730373f88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 23:30:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709b790beed34800ba107f2a77f297a8 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2},{"id":"0405ac367943f991060644b03761a81ff456c2a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 07:38:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c63c58f8eb24f46bebc871529c49b48 : FAILURE in 2h 52m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/881a9928ccb3411bbcaadeb71cae3e75 : FAILURE in 56m 29s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/acbedce081c147c689b48aff8e2c182b : FAILURE in 49m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e99f5abe3a54a1597355e8162511ebf : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06009471b80b4aeeb49121cb0b906513 : SUCCESS in 6m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/767e0a2c8ae3498b9baaa9da74d56719 : SUCCESS in 1h 49m 51s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"941d021e091567d2709bbb868768c6dc4a3372e0","revisions":{"1a6f55a00dbb83aaa5ec3b45387f2b7b8d1aaa16":{"kind":"REWORK","_number":1,"created":"2022-02-04 16:15:45.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/827883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/827883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/827883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/827883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/827883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/827883/1"}}},"commit":{"parents":[{"commit":"57f56e70623a5cb9bec3a97a1f05b087c99f1ccd","subject":"Switch to use Yoga binary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f56e70623a5cb9bec3a97a1f05b087c99f1ccd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 16:15:27.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 16:15:27.000000000","tz":60},"subject":"collectd: pcie-errors is x86-64 only now","message":"collectd: pcie-errors is x86-64 only now\n\nChange-Id: Iad7e7ebb1fbf4d14f20960d6257345db71d0e214\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a6f55a00dbb83aaa5ec3b45387f2b7b8d1aaa16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a6f55a00dbb83aaa5ec3b45387f2b7b8d1aaa16"}]},"branch":"refs/heads/master"},"941d021e091567d2709bbb868768c6dc4a3372e0":{"kind":"REWORK","_number":2,"created":"2022-02-04 20:39:17.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/83/827883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/827883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/827883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/827883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/827883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/827883/2"}}},"commit":{"parents":[{"commit":"57f56e70623a5cb9bec3a97a1f05b087c99f1ccd","subject":"Switch to use Yoga binary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f56e70623a5cb9bec3a97a1f05b087c99f1ccd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 16:15:27.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-04 20:39:17.000000000","tz":0},"subject":"collectd: pcie-errors is x86-64 only now","message":"collectd: pcie-errors is x86-64 only now\n\nChange-Id: Iad7e7ebb1fbf4d14f20960d6257345db71d0e214\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/941d021e091567d2709bbb868768c6dc4a3372e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/941d021e091567d2709bbb868768c6dc4a3372e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
