)]}'
{"id":"openstack%2Fkolla~827885","triplet_id":"openstack%2Fkolla~master~I3e599f780e8f84ea131988f4a9a45f17055c8ab2","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-02-08 08:38:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e599f780e8f84ea131988f4a9a45f17055c8ab2","subject":"nova-compute: \u0027ndctl\u0027 is available on all archs","status":"MERGED","created":"2022-02-04 16:58:49.000000000","updated":"2022-02-08 08:41:16.000000000","submitted":"2022-02-08 08:38:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827885","meta_rev_id":"e528efd2bb69391e0e658ea74f64f02144cbea1e","_number":827885,"virtual_id_number":827885,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 08:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-08 07:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-05 14:21:06.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-08 07:10:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 19:01:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-05 14:21:06.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-08 07:10:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"a2e2d82d559e260d01bd799dc44310457ba47ad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-04 16:58:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e56d90423ad187e4cb54b0448b9b782dcb7bda2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 19:01:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/11a75ae0dcde48e6bbb409dedbaf3613 : SUCCESS in 28m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/092979784fa24f7bbddd182c0ca05abd : SUCCESS in 39m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1230847cb7cd41b3a556f928d92619a6 : SUCCESS in 1h 01m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/13ef5da70d1046478335c2433b52d3c0 : SUCCESS in 59m 54s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/425b589548e44a4faa96d3b5efcb0187 : SUCCESS in 1h 52m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e43f72973a6a46009be2257cdab48257 : SUCCESS in 32m 00s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/681dd5d7f698402986806b58484c60df : SUCCESS in 19m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/09f91e4b757a4668adb7394a0fef33d1 : SUCCESS in 54m 11s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/71f847846f374b9e8cf1809f9a8e5be2 : SUCCESS in 1h 22m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49eceef0d6eb455fb5ff4e9e0eb47033 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dee9fda239f346619b2e35f5695f9616 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a2f3a627e76431ba0768f5651f6d67e : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/185d033526424729a23d8351b812e28d : SUCCESS in 7m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2602898a4a01443782359cc54fa1b4e0 : SUCCESS in 19m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/04dff914534b40038c01526295b10cb4 : SUCCESS in 29m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/67a6b01d070a4a0596ff16c38917a9ab : SUCCESS in 55m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9b533b6f21848a1a562ad32f74cd83c : SUCCESS in 46m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee6c801dd657406597cd38d5bf5e5a42 : SUCCESS in 1h 09m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7da3738cffbc117f638c2db38edc2bdd650992d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 07:03:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c814762416dc4cc3976c2ca317aa78ac : FAILURE in 2h 53m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b691d4dfe85e4e0caf72e0134abdff11 : FAILURE in 1h 02m 14s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5745894eb6d543eb8d9dd825f0821dba : FAILURE in 54m 24s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df0c9b10b0474bccb67112758db99ad8 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ba0d5f6de3145e6b9ddc2b4362a72f4 : SUCCESS in 7m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/658c3e7494e9424b9af9eda06de6fb85 : SUCCESS in 2h 06m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37e370d8057d3dc48a537d78bbf2d7fcfc4f882a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-05 09:32:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"29b237984176c713498a56e80292199815ac5bc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 11:00:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c66e6a7d54cd48e6875e3d43b1aa3145 : SUCCESS in 25m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/be32f4b8c28443098630e500071cafb1 : SUCCESS in 55m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ec9188e6c8444bbca34c4a4fd49ad111 : SUCCESS in 1h 17m 01s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e8e529aae0e3406e949e2033b2942276 : SUCCESS in 1h 11m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8ae324bb4d0046fdbbd0e61d2607ecf7 : SUCCESS in 1h 26m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6293d5d1bb4c4e979231450869b2ac4a : SUCCESS in 31m 33s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c9a773cec3e84de2bd04c6c180a13039 : SUCCESS in 15m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4ed1e571b1054d79872214deffce338d : SUCCESS in 59m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/191e9fe67a8e4aeab38aeafdd50e87ef : SUCCESS in 1h 07m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4980fe2074db48139f62075a7bdf5d75 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac1070e4a28e4c5e886a2451d2ee5a1c : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a75cd8b93fa452b97031beeaa7af65a : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544fbc9239584b98b2acb32268a95371 : SUCCESS in 6m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/915cd9629b3c41c2afe57867b24a0903 : SUCCESS in 28m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/69540d78aab345b6aad11478ae349e37 : SUCCESS in 35m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b6a6a8f8b724f1a862d85a925323ef6 : SUCCESS in 1h 05m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c671fa92e0d8444a912a7464bf44ce57 : SUCCESS in 51m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f9889dff1754b49ac5dae1abaae539d : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":2},{"id":"fb1e13ae4a3fed0158dd33cfb1d3cd7ae061d17b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 12:44:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef16f2af83f2497299b98e6dae0de144 : FAILURE in 3h 09m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8300b4c7d664bb984aa9220f2838345 : SUCCESS in 2h 56m 46s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8b1ecdbd04c4092b73fd004dc6b37e5 : SUCCESS in 2h 55m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9df7ed78edb64ccf84ea03702b741235 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47f543e19e0a4566b072c3580d78e8f6 : SUCCESS in 6m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/490a12ad87d04a058d7d3e3ccd450dff : SUCCESS in 2h 49m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ff8b8e70a2f75d963710f99dd10a0d9e8166aab","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-05 14:21:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c4d32bd5d26b2a3a933d6d554d35c3a270f24091","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 07:10:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5f40bd3d9a91dfa241be2d0ab613a34734338e3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 07:11:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a59537f701f5695cab2d9c5f469c1dbf92e0a9a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 08:38:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b01e389e88c9495db22cfe0ab83748f2 : SUCCESS in 44m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/53a246bb452a47c790cb975f06419df4 : SUCCESS in 1h 11m 31s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/419a41f150544785a45ebc270f7793c2 : SUCCESS in 1h 22m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ff3353e4b2524e49a1f4dd5e3a1167c9 : SUCCESS in 30m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f766eedb119c46c4a3fcbb868a7480d3 : SUCCESS in 58m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8c17751382dc418cb44836a970d9441f : SUCCESS in 1h 08m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdd6d4fae6a44e868ac063f05c823f67 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d510acfe115410a91801089fd62f31b : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d4bd0d5910047e0938057a35acb96c2 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/962470c5ebbc49049854a4487b882aef : SUCCESS in 6m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ea51475c51a04920902522e97d530035 : SUCCESS in 31m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b3e3da95645483db7c861963978a72d : SUCCESS in 54m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cde82a0e03bb47d989f50d90d7b99eba : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":2},{"id":"30e099d5a9c14fac5c5061b1c1586316400ce5c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 08:38:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e528efd2bb69391e0e658ea74f64f02144cbea1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 08:41:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8cfb9089514ba8aa42cd824b652040 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c638e768a84b2544924293fb382fe326c16a0022","revisions":{"bf7c79bd6ada715b79b599d3b879ef79633b231a":{"kind":"REWORK","_number":1,"created":"2022-02-04 16:58:49.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/85/827885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/827885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/827885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/827885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/827885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/827885/1"}}},"commit":{"parents":[{"commit":"57f56e70623a5cb9bec3a97a1f05b087c99f1ccd","subject":"Switch to use Yoga binary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57f56e70623a5cb9bec3a97a1f05b087c99f1ccd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 16:57:41.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 16:58:39.000000000","tz":60},"subject":"nova-compute: \u0027ndctl\u0027 is available on all archs","message":"nova-compute: \u0027ndctl\u0027 is available on all archs\n\nIn past \u0027ndctl\u0027 was x86-64 only in Ubuntu. This changed in 19.04 so time\nto adapt.\n\nChange-Id: I3e599f780e8f84ea131988f4a9a45f17055c8ab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf7c79bd6ada715b79b599d3b879ef79633b231a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf7c79bd6ada715b79b599d3b879ef79633b231a"}]},"branch":"refs/heads/master"},"c638e768a84b2544924293fb382fe326c16a0022":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-05 09:32:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/85/827885/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/827885/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/827885/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/827885/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/827885/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/827885/2"}}},"commit":{"parents":[{"commit":"2553c1845c7c4cf04d720b820e4bb13519eca4c9","subject":"pin out some package from Debian OpenStack Team repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2553c1845c7c4cf04d720b820e4bb13519eca4c9"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 16:57:41.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-05 09:32:15.000000000","tz":0},"subject":"nova-compute: \u0027ndctl\u0027 is available on all archs","message":"nova-compute: \u0027ndctl\u0027 is available on all archs\n\nIn past \u0027ndctl\u0027 was x86-64 only in Ubuntu. This changed in 19.04 so time\nto adapt.\n\nChange-Id: I3e599f780e8f84ea131988f4a9a45f17055c8ab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c638e768a84b2544924293fb382fe326c16a0022"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c638e768a84b2544924293fb382fe326c16a0022"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
