)]}'
{"id":"openstack%2Fkolla~828228","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I86aa697b8def34e0cfb11d2f5ad903dd54e975ef","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-14 16:48:01.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-14 14:46:22.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I86aa697b8def34e0cfb11d2f5ad903dd54e975ef","subject":"base: Drop usage of Ceph Nautilus from RDO","status":"MERGED","created":"2022-02-08 08:12:38.000000000","updated":"2022-02-14 16:50:04.000000000","submitted":"2022-02-14 16:48:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"828228","meta_rev_id":"76ce93987809971688768d84c672ffaaec8d8801","_number":828228,"virtual_id_number":828228,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 16:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-02-14 14:46:22.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-02-14 15:00:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-02-14 15:00:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 08:39:18.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-08 09:22:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-14 13:23:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"4717387e853bd73236351ea47d6659aeb9e7df71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 08:12:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0b2150c70b217f1a5328081c1a029caffbb9ce8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-08 08:39:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd8283083d5456283ae876bf84f12a1cf97dbca2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 08:41:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e310af71cf21da3944b56a2513e4b274da922a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 09:22:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3800fe0ebbac4b9b823ffdc089a62396 : FAILURE in 27m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/98773ddba7e345b0adb0ade6b041c834 : FAILURE in 37m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1a34fdd7f8a34e069164b8843e31ad51 : FAILURE in 22m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/72ef53a4907f453eb728c7493c0cfb2f : FAILURE in 22m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bc111dce6ab24707a95b485e6257261f : FAILURE in 20m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9295a235deba40c4b0ef9c09081007ec : SUCCESS in 33m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ac100b2cb02e4d60bc6126efe13ee069 : SUCCESS in 19m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9ea93622a8a74b3783eb643c093be003 : SUCCESS in 50m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79cbc4f9669a417b81c2dec8e672f930 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30683059b0a64f09b538ad585aef891c : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2656454ce55747a4b8d0273b605182c8 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cbe52faec7e42b5855a3b69d5a9046f : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b4e1a753294b4c9c34152082ce8c18 : SUCCESS in 6m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/11376de6f89e4445bbc03380b88eae47 : SUCCESS in 20m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9bcbf67f821469f9a1e042f65150ce3 : SUCCESS in 37m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/594051f6a7e5437ba8d2a1af50a83dee : SUCCESS in 51m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc93eb07a68b45a588391d43eb740075 : SUCCESS in 55m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/25e2676a9ed94ddda6f7a4063b9de105 : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b134181179e19e072d984ec75f7723d85cbe9a61","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 11:09:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba47c7d572a2302ccfa7920b2b08c235c4f8a622","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 11:13:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed28c7eca5494fdeb3618ad156012665 : FAILURE in 2h 03m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/08ae5bbbe3e54260b9a84f9568e000f3 : SUCCESS in 2h 01m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b42407fe077a41b4a6eb407170243491 : SUCCESS in 2h 21m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a82d38c9d6ba4f48a5a71c5009ae74a2 : SUCCESS in 1h 48m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"818bc88347a4cae4812f729d8085d20db2caf99e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-10 15:06:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a890ee2d36020d73fdc55accdcc25192aac89b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 16:16:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9b004a7e53a34f3bb433ea24edfc62a6 : FAILURE in 23m 05s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0a9a33e16d454096942dc504273336df : FAILURE in 41m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fc49f284044b48ad915f12e13008e3ca : FAILURE in 22m 04s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/50528d4d44234bb68781a9cf28325620 : FAILURE in 26m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f71b7f4ab1794547bcf11aae3e456070 : FAILURE in 19m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8b236a158d5c4affaa1c51a2c6761182 : SUCCESS in 40m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c8267d4fca6e4b2db56aef2c48e76bf8 : SUCCESS in 18m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/76f1307c46384f788782e382a082894c : SUCCESS in 1h 03m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dffdf331fcd41f8aba073f8a673d847 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5089e52d4554e16ba5d1f1c2decee26 : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10790fc073dc443eacdf5a04d716e4df : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e898317b35904fe4acbe0d449be28d0a : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0c0752df314156adb94ca35603a993 : SUCCESS in 6m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/820077a52a3841c1a348c319ba80f132 : SUCCESS in 20m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/34bb4dfb46684d3492cb8f5478e2c86b : SUCCESS in 33m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e34b7d5ff90e4efdb59e14b8d5e44c3f : FAILURE in 38m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3b5487da3de84915abd7e38fb40298be : SUCCESS in 59m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f874e3912414dba93e7dc23375312b8 : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2992889513dccbd4c785b2a799adcb23ac82ddf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-10 19:03:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"87c592b34ce7cbd926eeab04716421fb0161806e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 20:01:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bc1d49b7cbe8435b895fbb93aeef57a0 : FAILURE in 5m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2ef91d3e25a3464593e4ecec314f938d : FAILURE in 6m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/822461b9353040be9e11fb70300f8a75 : FAILURE in 10m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/72d4087400404738a9be827ea89a8997 : FAILURE in 10m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/291f4293d88548388701a808ac9c40de : FAILURE in 10m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fde12d369a734803b9c665441ca0f906 : SUCCESS in 22m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/41a1193a773b4e299daa101e04d4f596 : SUCCESS in 15m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/173648647a92436ba955e0c5ccc02d6b : SUCCESS in 44m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ba9b785dfd4496a86cf23a4f8b61cda : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c85f706e8fd4bee823cdbfdb0a74c2f : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7315f6455bbb4d709d82af89df2d7dc6 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fac45e1d7ebd4286a432a31e4d66c7d2 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b33409c7ac8941ee80467571560aa375 : SUCCESS in 5m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d6a6e7a7f2844adea297ee4aae8543da : SUCCESS in 23m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3dde7fa1144f4052a14add61d5a93de2 : SUCCESS in 36m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e7aae50db1e42528d74a3f453af8f57 : SUCCESS in 49m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/687eb3d6b3704b4a95750b55ee48dbb1 : SUCCESS in 48m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b3217fd3d39d4c878060dd2d2fcd2ed1 : SUCCESS in 52m 25s","accounts_in_message":[],"_revision_number":2},{"id":"2efb0010bfaeb84c657a6c461d15190c2ba5d01b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 05:00:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/43ae9a9dd45d422a834971d2aa1e9138 : FAILURE in 11m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ac7261236db04958930549163f6a0106 : TIMED_OUT in 6h 02m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/122689cd879b4275b5479488c9f5b546 : SUCCESS in 2h 31m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0cf02db3d1c84cdba6edf63d53fae7af : SUCCESS in 2h 02m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7d31967731a879a7159682d17a39558697bb5100","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-11 06:04:07.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0c9dd9e5e5112bc18c994fec77684d0f3845c0ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 07:06:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f60ff67d64bf4ad4b39de1a75119e61e : FAILURE in 26m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/71b5c52e204043d4a406499c10d8c480 : FAILURE in 29m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6000947d0f4d424c92efd7e9a411f293 : FAILURE in 32m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e55f81eb236d480ab91a120f8611cb59 : FAILURE in 20m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7f6e67d4427645898eab307686fa2bde : FAILURE in 23m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/06b18d7667a548c79cc30b97f7966f54 : SUCCESS in 28m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/282fcbd88568420d9e3ef2754a26f682 : SUCCESS in 21m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/693ea0ea255a4768b5a3d63d8ae74b3d : SUCCESS in 52m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6150e604067f40feb94ed9402f79d184 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/521e62f244694ca18fb6967ce9c55e7f : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/822a71aeecbc40c0b53905f3168625cc : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b83f2b1f211c4b1eb269342397fb5dde : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9922784ccef644e1b66351729291db02 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3608424d00b460b9611c0d17bca1bec : SUCCESS in 19m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6943207416c54ea79368aff1ed0b6195 : SUCCESS in 31m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a73cf8605f24403871b6939e472377e : SUCCESS in 54m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3184bf9b19674677aecbcb1bed7a5671 : SUCCESS in 57m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/030560f7671040a29fd959aeb0efe508 : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":3},{"id":"30cef315cb5d8b7036bb59e7a899ca7de50b8f16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 09:34:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b8777df7c4c448e92a0c6a31c807598 : FAILURE in 2h 07m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/392c6975d5fe457f89e9043fc3ce20dc : SUCCESS in 2h 19m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/99aa73a159b840bd968a1becca1bf94a : TIMED_OUT in 3h 02m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c66fa80f059c46baa8694148a481d023 : SUCCESS in 2h 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"872cef8118e342e4065de43b361a253d37cd5a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 12:04:38.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ce8cce931d21c0a42f46980fd8ca71c5c3831a3d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 13:23:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5133156d188a2567aeafea9b07b5eb8959c8380d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 13:25:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ca03af7a76ba4868a903504e948060b3 : SUCCESS in 28m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/19205391209e4cfda198a7d4f8319fc0 : SUCCESS in 36m 30s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f047f10b67ce435c97b2ad0aba1e9921 : SUCCESS in 1h 05m 02s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8aaa7bdb5fed458fa7a23584ade281e9 : SUCCESS in 1h 08m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e8a910586fa640a2b73adf33e360bc0f : SUCCESS in 1h 18m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d7dd5fd8e1124cbca1ba7f0d5e2a809a : SUCCESS in 33m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1e8603bdde7643e98b1e61222c7cc2d0 : SUCCESS in 20m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b38f4057286641959435a49a6732c37d : SUCCESS in 57m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0612ed7c4813485697ecac69a2f5c4f5 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22fffc519e15452d8597efb3f321a14f : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d6ca65bbeda43f0b1cbd576c688996d : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7da58404017a4579a4eb677d8fe1ae7f : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a630868493654c518f52a32d0533f613 : SUCCESS in 7m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ce1bbe000bf840cc9feda3613bbcaca5 : SUCCESS in 20m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/64e212898d3441b3ad9ebb43fb4428a2 : SUCCESS in 27m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ed58ff8519a455c83e7366d4018f722 : SUCCESS in 52m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/508ed688c0c74ecb8ecb155ca77d6c0f : SUCCESS in 51m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5eb1d153cc4d4882a40d3f3a119fcf4a : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":4},{"id":"2f9e7e9a7669817d3a89a3f58ebb4d1a61b5391a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-14 14:04:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8ba928782fa411b1100a67a356c9f6c7c06bbd03","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 14:26:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"de21b78565c4249f6128175c2330d0ac6ccfb090","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 14:41:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/500d942da9d240998ee8c56c127e0395 : FAILURE in 2h 34m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1080852745614bdb91ffaf533ed8d163 : FAILURE in 34m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/12bf88ea10f741e6a27a2d02bf520337 : SUCCESS in 1h 17m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e19c4eb1b04486080c5e2e56b08eb37 : SUCCESS in 38m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"14eb1cebffebd882e3c283bf25a7b9bdc8af1121","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-14 14:46:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e8b0f7292247c1f3a900447d410af36a68e8319a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-14 15:00:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"24add27e4970b1f013a4ffa3aa041a529dafc998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 15:01:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b19d2ce35af9107015d717a4cfe6e11612a5bcbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:48:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/077b569fe6d74540bfc3978cbddd80bc : SUCCESS in 39m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/137523bfc1ce4ff1a079699e02f13a2e : SUCCESS in 1h 01m 32s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6f20a414b8f14863a483c174a40c9878 : SUCCESS in 1h 24m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbbfd72a4ed443a7a02b8d352ca5947d : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/285b36b98d974323b609084c28fbde66 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bf2a9d1d01c4c69ba10f6a650660b65 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25a8d7fed1643d4aef266901257472d : SUCCESS in 7m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/db3080a2fd454538a45a95472b6ee449 : SUCCESS in 37m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a0004c6780e4effa8fc52e0fce7981a : SUCCESS in 1h 00m 08s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e98dacc528246b8a53f7cf768584bc2 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":4},{"id":"bc4f037015f5d2d0fba6903baf9b4603d16a1453","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:48:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"76ce93987809971688768d84c672ffaaec8d8801","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:50:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7cc330af0a54bfab76ad17dbb521342 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7f38bce812ccb621a089755526f79bd792f4e071","revisions":{"57e602f5e0fd5ccbcb2089e145f55b6ef0f01c4c":{"kind":"REWORK","_number":1,"created":"2022-02-08 08:12:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/828228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/828228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/828228/1"}}},"commit":{"parents":[{"commit":"f439afa42e9f5074e4680987e6185c4bd93aa090","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f439afa42e9f5074e4680987e6185c4bd93aa090"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 08:11:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 08:12:35.000000000","tz":60},"subject":"base: Drop usage of Ceph Nautilus from RDO","message":"base: Drop usage of Ceph Nautilus from RDO\n\nThis was temporary due to missing Ceph Nautilus\nrepo on CentOS mirrors (after removing CentOS Linux\ncontent on EOL date).\n\nChange-Id: I86aa697b8def34e0cfb11d2f5ad903dd54e975ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57e602f5e0fd5ccbcb2089e145f55b6ef0f01c4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57e602f5e0fd5ccbcb2089e145f55b6ef0f01c4c"}]},"branch":"refs/heads/stable/wallaby"},"74ae566f587ee2bb9764b135af647e5c6ef3dd65":{"kind":"REWORK","_number":2,"created":"2022-02-10 19:03:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/828228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/828228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/828228/2"}}},"commit":{"parents":[{"commit":"f439afa42e9f5074e4680987e6185c4bd93aa090","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f439afa42e9f5074e4680987e6185c4bd93aa090"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 08:11:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-10 19:03:57.000000000","tz":0},"subject":"base: Drop usage of Ceph Nautilus from RDO","message":"base: Drop usage of Ceph Nautilus from RDO\n\nThis was temporary due to missing Ceph Nautilus\nrepo on CentOS mirrors (after removing CentOS Linux\ncontent on EOL date).\n\nChange-Id: I86aa697b8def34e0cfb11d2f5ad903dd54e975ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74ae566f587ee2bb9764b135af647e5c6ef3dd65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74ae566f587ee2bb9764b135af647e5c6ef3dd65"}]},"branch":"refs/heads/stable/wallaby"},"2fc241516dc01f9b260ccb115d1c5db3d4aeb653":{"kind":"REWORK","_number":3,"created":"2022-02-11 06:04:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/828228/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/828228/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/828228/3"}}},"commit":{"parents":[{"commit":"f439afa42e9f5074e4680987e6185c4bd93aa090","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f439afa42e9f5074e4680987e6185c4bd93aa090"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 08:11:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-11 06:04:07.000000000","tz":0},"subject":"base: Drop usage of Ceph Nautilus from RDO","message":"base: Drop usage of Ceph Nautilus from RDO\n\nThis was temporary due to missing Ceph Nautilus\nrepo on CentOS mirrors (after removing CentOS Linux\ncontent on EOL date).\n\nChange-Id: I86aa697b8def34e0cfb11d2f5ad903dd54e975ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fc241516dc01f9b260ccb115d1c5db3d4aeb653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fc241516dc01f9b260ccb115d1c5db3d4aeb653"}]},"branch":"refs/heads/stable/wallaby"},"7f38bce812ccb621a089755526f79bd792f4e071":{"kind":"REWORK","_number":4,"created":"2022-02-14 12:04:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/828228/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/828228/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/828228/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/828228/4"}}},"commit":{"parents":[{"commit":"f439afa42e9f5074e4680987e6185c4bd93aa090","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f439afa42e9f5074e4680987e6185c4bd93aa090"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 08:11:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 12:04:38.000000000","tz":0},"subject":"base: Drop usage of Ceph Nautilus from RDO","message":"base: Drop usage of Ceph Nautilus from RDO\n\nThis was temporary due to missing Ceph Nautilus\nrepo on CentOS mirrors (after removing CentOS Linux\ncontent on EOL date).\n\nChange-Id: I86aa697b8def34e0cfb11d2f5ad903dd54e975ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f38bce812ccb621a089755526f79bd792f4e071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f38bce812ccb621a089755526f79bd792f4e071"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
