)]}'
{"id":"openstack%2Fkolla~828242","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I95ea4a66a80bf035f8bd1180790c7e0c85d759ed","project":"openstack/kolla","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-09 11:43:47.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-02-09 10:37:32.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I95ea4a66a80bf035f8bd1180790c7e0c85d759ed","subject":"Drop CentOS Linux 8 image builds","status":"MERGED","created":"2022-02-08 13:23:52.000000000","updated":"2022-02-09 11:53:47.000000000","submitted":"2022-02-09 11:43:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828242","cherry_pick_of_change":827608,"cherry_pick_of_patch_set":10,"meta_rev_id":"de9c6f3af5cbb606d25fa96a703cd1bc9421b321","_number":828242,"virtual_id_number":828242,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 11:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2022-02-09 10:31:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2022-02-09 09:42:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-02-09 10:37:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-02-09 10:37:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 13:25:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 08:14:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-09 09:42:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-09 10:31:00.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"9f40e61022079579f250091a1c617558c7389a8d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 13:23:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.\n\nThe following files contain Git conflicts:\n* .zuul.d/centos.yaml\n* doc/source/admin/image-building.rst\n* doc/source/support_matrix.rst\n* docker/base/Dockerfile.j2\n* docker/collectd/Dockerfile.j2\n* kolla/common/config.py","accounts_in_message":[],"_revision_number":1},{"id":"222102fcb9a1a2cf1b0002068c7296a6056ea847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:25:50.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/train.  The error was:\n\n  while parsing a block collection\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/train\", line 2, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/train\", line 5, column 1","accounts_in_message":[],"_revision_number":1},{"id":"4bbe1d28882516ccc92725b9258c6dadf7993728","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 14:01:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3c5267c3a9447d6ad3dcfd8d82dff1589f4be20e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 14:31:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0ad56a9c94f74862be4dc2b00cb78fd5 : FAILURE in 5m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6364a39754c94a49825814909f074e3b : FAILURE in 5m 20s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/036f4d21698d4608b3076d6c82863d9c : FAILURE in 6m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9c2236c0cb1245978ccdcb3b6e09dad1 : FAILURE in 5m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/42ae480b9d744ec28f6cffcbd9a16433 : FAILURE in 9m 11s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/baf7f043b4e24fb1b8f638fff86641e1 : FAILURE in 9m 08s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f9e3be5b17c84876b6943893a6c2a359 : SUCCESS in 19m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bf3af1c9716349c4b0eca02ca9d23165 : FAILURE in 6m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0dafa1ed44cc49c4bceb95b6d860d361 : FAILURE in 4m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7b3107495f154f4796ddf0df2c0be222 : FAILURE in 7m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e03616990824961a3dca9b5dcccb45b : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd0b48f7be904b2482abdf4cd5c3570f : FAILURE in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45013c2b298940a1a01efb1d9efe72ab : FAILURE in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2b34d879c8e4cc0a7a1c11f8fd3f966 : FAILURE in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e8a6eae209a4e4fb5860ff080aac25b : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d9917bbbf5b406c8f175cfd17636df8 : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/674f2696e8f34e61bbe1029ae4711a0c : FAILURE in 5m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8adce2cfece94cb3ab7e6a5bf355df8e : FAILURE in 4m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d0456e5a5084a489804439e9b433642 : FAILURE in 9m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/46b4e838599342aaacd810a0db8f915b : FAILURE in 8m 01s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"cf34a031a824bf08f2686ed8c354acdd0d945bfb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 14:39:20.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"45e5328bdd09aec04435658cb9066b2c44c4d27d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:24:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c69bb60b60d8416988175063210a93ed : FAILURE in 6m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f5373082cf6d4c9c9be47c4b4be52923 : FAILURE in 6m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8ae6cee004964153a778929a86b40798 : FAILURE in 10m 33s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/321cc7da0dc24852907e7dc3581460b1 : FAILURE in 9m 19s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c362b5fcc52a43a5a25f0c5789e0526b : SUCCESS in 21m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b6809e6321e247d39bea7beacc041056 : FAILURE in 9m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7b58f252a79f47309a6b7a555f03dd16 : FAILURE in 5m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a49ac96f6dae4cdbb5658a96473c3b82 : FAILURE in 8m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2192d7c10c8041ecb9d0aaa10112dc61 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d47e44ad1d9498293f39e1f6fa28089 : FAILURE in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf44339d71b944a7b59c1a0f1afd579b : FAILURE in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ffc2fe8553cd433a90722e1d1950a08e : FAILURE in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fa5a64c283242c5bd284ec8f65b0d40 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51400836645e4125b861f9a34ed9a3e9 : SUCCESS in 5m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bd6995cc2f37449886728d5643688e78 : FAILURE in 5m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d367c7aed68e4014bae48a2c37d94d52 : FAILURE in 6m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7701cae780ed4ce3b3de76b6e6f95773 : FAILURE in 9m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e34b8f3506d24dddb8b0c737b97513f1 : FAILURE in 9m 52s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"1a7928b35c178c149961a143954b528f79b92eac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 17:09:01.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"481e5a4f1c6ac91d21f0655337344c695225e699","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 17:10:03.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"1f07f7008268a185dd4efb5a8d1634b36ec65fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 17:37:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/820f5ef3663a461e931bea9a3dc14926 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d67fff8aecf44800bc12e0fb3b54172d : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ec0a264d36742918bf945d2b177293a : SUCCESS in 6m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf6862108d464ab3863393519494284f : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc120f2b9ff34706ac2125573a8ab3d3 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74c0d291c1564d4bb302eecc4ccea003 : SUCCESS in 5m 51s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"f09e8231b296186d3d8fd6f8344e7b58d2534571","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-09 06:27:54.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"2a100beb634a191930eaaa450a368ec1e2248925","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 08:14:49.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cf82dc2b4fa32c9e0a6ea1a2169401d7b033d48f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-09 08:21:25.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"36d84f777c5fcf6fc4ef8d73ecb23db6b600eeb9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-09 08:22:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0a9f60c84ece41acf987cc8620b5dad6d06c6b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 08:37:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed2bc13ab08e44f9aa6192e0326cc0d2 : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09f7993f79ca403ba1505c1c7224614e : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2541e9f6fe4400bacaf89f3986e46bb : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/280cebe80ed54b8587a1c13278c7ce5f : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a27e93d0086406290f35a740cf87076 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f1a971dfdde4a9495794b1c07120def : SUCCESS in 3m 57s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":6},{"id":"78ba97eb42a0dc11c36f9dbf5a3b3aff111ca7a7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-09 09:42:37.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f02c32e9a7c13cf78dbcc5a42bc98e52fed23cb1","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2022-02-09 10:31:00.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"56a05517122367e010cef2f1b394b02a7186e8a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 10:37:32.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"572c21715d4fe1c4b3ac1e6199826330cb32d820","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 10:41:54.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ab94c1d4bf9dd795e853a712e05a1ef110615e06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 11:43:47.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/617b84fcf58840aa8be2967d0bb3ecf7 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/835217e9015948b4aabd04f708520e8a : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f19eb30e17d64abb8787a9f059f78fd0 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d66666fad10d4128a6cd3b0c5f329f21 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c699fe6bc2f342d88b94355d3002ecf5 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0ef01d287684135ace8798004257945 : SUCCESS in 6m 34s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":6},{"id":"f54b21090d36d0c4a52f446b4066a3085c40e4a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 11:43:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"de9c6f3af5cbb606d25fa96a703cd1bc9421b321","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 11:53:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/118bd5f77bd14a21a5b0d722135c466e : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93c03ee65eb345cf974ca2b025dec527 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3c8bee583171f6c6a334e039cf16d176f95e46fe","revisions":{"da3282e1d45505948b350b86bf0020ff1e3224eb":{"kind":"REWORK","_number":1,"created":"2022-02-08 13:23:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/828242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/828242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/828242/1"}}},"commit":{"parents":[{"commit":"b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896","subject":"CI: make it work again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 09:19:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 13:23:52.000000000","tz":0},"subject":"Drop CentOS Linux 8 image builds","message":"Drop CentOS Linux 8 image builds\n\nIncluded patch:\n\nRemove missing collectd packages\n\nSince CentOS Linux 8 repository retirement from\nCentOS mirrors - we need to use CentOS Stream 8\nversion of Opstools repo.\n\nListed below packages are missing from the repo,\nand according to the maintainer dpdk_telemetry is\nnot going to show up at all (it\u0027s basically dead).\n\nThe rest of the packages might show up at a later\nstage - but it\u0027s not a priority for the maintainer.\n\n- collectd-dpdk_telemetry\n- collectd-libpod-stats\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n\nChange-Id: I95ea4a66a80bf035f8bd1180790c7e0c85d759ed\n(cherry picked from commit 07f519996381789c1f21850cfc9b27f3b86a6f4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da3282e1d45505948b350b86bf0020ff1e3224eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da3282e1d45505948b350b86bf0020ff1e3224eb"}]},"branch":"refs/heads/stable/train"},"56f36a3dbfab086cb416ba36643dc43624f77973":{"kind":"REWORK","_number":2,"created":"2022-02-08 14:01:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/828242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/828242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/828242/2"}}},"commit":{"parents":[{"commit":"b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896","subject":"CI: make it work again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 09:19:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 14:01:41.000000000","tz":0},"subject":"Drop CentOS Linux 8 image builds","message":"Drop CentOS Linux 8 image builds\n\nIncluded patch:\n\nRemove missing collectd packages\n\nSince CentOS Linux 8 repository retirement from\nCentOS mirrors - we need to use CentOS Stream 8\nversion of Opstools repo.\n\nListed below packages are missing from the repo,\nand according to the maintainer dpdk_telemetry is\nnot going to show up at all (it\u0027s basically dead).\n\nThe rest of the packages might show up at a later\nstage - but it\u0027s not a priority for the maintainer.\n\n- collectd-dpdk_telemetry\n- collectd-libpod-stats\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n\nChange-Id: I95ea4a66a80bf035f8bd1180790c7e0c85d759ed\n(cherry picked from commit 07f519996381789c1f21850cfc9b27f3b86a6f4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56f36a3dbfab086cb416ba36643dc43624f77973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56f36a3dbfab086cb416ba36643dc43624f77973"}]},"branch":"refs/heads/stable/train"},"0bcba3e083e014a3082faadd9f2e1a317a2522a5":{"kind":"REWORK","_number":3,"created":"2022-02-08 14:39:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/828242/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/828242/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/828242/3"}}},"commit":{"parents":[{"commit":"b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896","subject":"CI: make it work again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 09:19:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 14:39:20.000000000","tz":0},"subject":"Drop CentOS Linux 8 image builds","message":"Drop CentOS Linux 8 image builds\n\nIncluded patch:\n\nRemove missing collectd packages\n\nSince CentOS Linux 8 repository retirement from\nCentOS mirrors - we need to use CentOS Stream 8\nversion of Opstools repo.\n\nListed below packages are missing from the repo,\nand according to the maintainer dpdk_telemetry is\nnot going to show up at all (it\u0027s basically dead).\n\nThe rest of the packages might show up at a later\nstage - but it\u0027s not a priority for the maintainer.\n\n- collectd-dpdk_telemetry\n- collectd-libpod-stats\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n\nChange-Id: I95ea4a66a80bf035f8bd1180790c7e0c85d759ed\n(cherry picked from commit 07f519996381789c1f21850cfc9b27f3b86a6f4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0bcba3e083e014a3082faadd9f2e1a317a2522a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0bcba3e083e014a3082faadd9f2e1a317a2522a5"}]},"branch":"refs/heads/stable/train"},"f41eb2462f0665156471151b5f68ef27dc156002":{"kind":"REWORK","_number":4,"created":"2022-02-08 17:09:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/828242/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/828242/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/828242/4"}}},"commit":{"parents":[{"commit":"b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896","subject":"CI: make it work again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 09:19:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 17:09:01.000000000","tz":0},"subject":"Drop CentOS Linux 8 image builds","message":"Drop CentOS Linux 8 image builds\n\nIncluded patch:\n\nRemove missing collectd packages\n\nSince CentOS Linux 8 repository retirement from\nCentOS mirrors - we need to use CentOS Stream 8\nversion of Opstools repo.\n\nListed below packages are missing from the repo,\nand according to the maintainer dpdk_telemetry is\nnot going to show up at all (it\u0027s basically dead).\n\nThe rest of the packages might show up at a later\nstage - but it\u0027s not a priority for the maintainer.\n\n- collectd-dpdk_telemetry\n- collectd-libpod-stats\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n\nChange-Id: I95ea4a66a80bf035f8bd1180790c7e0c85d759ed\n(cherry picked from commit 07f519996381789c1f21850cfc9b27f3b86a6f4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f41eb2462f0665156471151b5f68ef27dc156002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f41eb2462f0665156471151b5f68ef27dc156002"}]},"branch":"refs/heads/stable/train"},"d848f7e027cb46cca67e25d2b26441d0c9f445be":{"kind":"REWORK","_number":5,"created":"2022-02-08 17:10:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/828242/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/828242/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/828242/5"}}},"commit":{"parents":[{"commit":"b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896","subject":"CI: make it work again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 09:19:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 17:10:03.000000000","tz":0},"subject":"Drop CentOS Linux 8 image builds","message":"Drop CentOS Linux 8 image builds\n\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n\nChange-Id: I95ea4a66a80bf035f8bd1180790c7e0c85d759ed\n(cherry picked from commit 07f519996381789c1f21850cfc9b27f3b86a6f4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d848f7e027cb46cca67e25d2b26441d0c9f445be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d848f7e027cb46cca67e25d2b26441d0c9f445be"}]},"branch":"refs/heads/stable/train"},"3c8bee583171f6c6a334e039cf16d176f95e46fe":{"kind":"REWORK","_number":6,"created":"2022-02-09 08:21:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/828242/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/828242/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/828242/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/828242/6"}}},"commit":{"parents":[{"commit":"b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896","subject":"CI: make it work again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b22f8f63a33b5a7c0ee43fc8ee346d636bb2a896"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 09:19:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-09 08:21:25.000000000","tz":0},"subject":"Drop CentOS Linux 8 image builds","message":"Drop CentOS Linux 8 image builds\n\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n\nChange-Id: I95ea4a66a80bf035f8bd1180790c7e0c85d759ed\n(cherry picked from commit 07f519996381789c1f21850cfc9b27f3b86a6f4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c8bee583171f6c6a334e039cf16d176f95e46fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c8bee583171f6c6a334e039cf16d176f95e46fe"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
