)]}'
{"id":"openstack%2Fkolla~828560","triplet_id":"openstack%2Fkolla~stable%2Fxena~I3045786e863b098d7339d1066aef6c857aa5f97f","project":"openstack/kolla","branch":"stable/xena","topic":"fix-fluentd-es-stable/xena","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-14 14:53:27.000000000","reason":"Michal Nasiadka replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-16 09:10:26.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-02-16 10:32:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3045786e863b098d7339d1066aef6c857aa5f97f","subject":"Unpin td-agent and cap elasticsearch gem","status":"MERGED","created":"2022-02-09 19:31:59.000000000","updated":"2022-02-16 10:34:39.000000000","submitted":"2022-02-16 10:32:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828560-fix-fluentd-es-stable/xena","cherry_pick_of_change":821695,"cherry_pick_of_patch_set":14,"meta_rev_id":"884d71e52a1e809c00f50159b6d3ab041883bd1e","_number":828560,"virtual_id_number":828560,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 10:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-14 14:53:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-02-16 07:12:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-02-16 09:10:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-02-16 09:10:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 19:31:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-02-09 19:39:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-02-09 19:39:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-09 21:19:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 07:12:40.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"2e68798250b33e8769d65b4e59e49ad15e43f7e1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 19:31:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/fluentd/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"cfd097965846d78b44c90d8ea582372836a92ca8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 19:33:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"777e16486123d0c60725270cb92c177a4c7048a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 19:35:37.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"960ce41773af2a4c07314a05ed7a7ad7cba5c90e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 19:36:23.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"204d8b57b57083bc925a2534b17a445f4713375d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-09 19:36:54.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"86eda5a7ffa6de9f20d8d95f425657f76a5bf5c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:19:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/aa4114a9418146d3a9bc132c6127a42d : SUCCESS in 31m 14s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c0dd59d3f54d4938b9baca35d91f235d : SUCCESS in 39m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/44a9916cebaa469591eb9d59d4ac0cc2 : SUCCESS in 1h 09m 58s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/36216f14eb804777aeb23d8f95ae869a : SUCCESS in 1h 21m 37s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c05d9fa8c67c4791a0ad1ba7b4977f82 : SUCCESS in 1h 21m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cfe038e7805f4855a7e3b030c5de6dc7 : FAILURE in 47m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/436166377f6747dea5076157b4493f19 : SUCCESS in 19m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/32b7a3a0bd424521aa41939a97c65b95 : SUCCESS in 53m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/056650baf28b4ba6a6afec5573ce4ad8 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0969dfdf86074ea79facfae500527a18 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a69bfd51086a4e47a91b2d812ccd093f : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da4e21eda2944b4e87947da5e123e232 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be2daa94073e47e4b1b46bfc21e9c1cf : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b390a22048f14c61a1c688814f14537b : SUCCESS in 6m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6dc7a813a933406d86edb265b0773d7e : SUCCESS in 21m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/76fe8ecf06fb4e6d918e481e73bfb2c8 : SUCCESS in 38m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/552f221c1f684535afc05117f5699942 : SUCCESS in 56m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/59c60407b0254bbba86b7471593db977 : SUCCESS in 55m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/11dddfb50770499980ff416a1f101b20 : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":4},{"id":"277b3eaaa71dc6724fefcf22c54458d8932fca1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 00:51:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/44a4d943d49a4efdaaa26ac6bdfd2abd : FAILURE in 2h 19m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/63f29f66e49f4b34b888c505dc0f27d6 : FAILURE in 52m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3e5d478aa00041fbaef6e8d955c93f65 : FAILURE in 41m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b49c9e1d53d454389c5a523345216e9 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c350b7103ea464e9b471582a492f8dc : SUCCESS in 7m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/833bdc7ab99344a3bae04f1481e9d14a : SUCCESS in 2h 22m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f281ef082bdf49676194a232e0f5d2309bfcc1bf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-10 07:05:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab62865b188455d7d6f950793b7848bf69c29d45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 08:31:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5d70eb6f1db54446aa61f3ac1c961a80 : SUCCESS in 28m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bbc6a11c725c4325bf42e10c583799b5 : SUCCESS in 38m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f34c44fc1b5041ce922bd33b7864c270 : SUCCESS in 1h 08m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3a443733f2d0425aa7514e057defccd5 : SUCCESS in 1h 12m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d49325b01ca4489ba8321316a937201d : SUCCESS in 1h 20m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/47a50132828f44cbaaae8286f6785984 : SUCCESS in 26m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b93295791e654310b300698da84cd0b5 : SUCCESS in 18m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fa682b7d69b2430c99cb2b482577cc92 : SUCCESS in 57m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a03ffed8610d4d24b54e61c0693427d1 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d782be8d87664382a8c3f56f30e3aa36 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/812fc3abba9f4f999471ecd68dc505e1 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fe292d8b255438a8375499003e24d54 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7543f45cffd4ed8a5467fc84adc19bb : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/820f8b2a16de4d3e8b57bb74468d2bfd : SUCCESS in 3m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/182f84745723441f89399447adc46346 : SUCCESS in 18m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e19333688424450d921b05c37eee371d : SUCCESS in 31m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0277f1bce9064851870104ed8ac20e3d : SUCCESS in 48m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5a85cc0ccb0d424198f563884c4f7a68 : SUCCESS in 48m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7672b152062547aba59ba94591957701 : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":4},{"id":"1c9beed8f5171bc39bdef78909c1d7265fab177d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 10:09:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4daf54eb111341e182df400e54a51a53 : FAILURE in 2h 09m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/81b3689c2e4d4382827908c7fc2c08e0 : FAILURE in 39m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a97566e3c2554c5f95b231937faf7f55 : FAILURE in 43m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9fd49db8889744ea90cae379f9d30cdb : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05be051994624dacb1fad81a83cee827 : SUCCESS in 5m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c909b11a7f941509367281be576f99b : SUCCESS in 2h 04m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3dc060178602e629a46911346f842862b4f90bd2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 14:53:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b31a3090b394b70ba048c4ecd55f357046836686","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-02-16 07:12:40.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3cf670b8bb6ca42186e4669a0bcd259335935558","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-16 09:10:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ab0f0b5b261a3f2c0bbaeb6ed745e9f09b80f84a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 09:11:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e05289112cfa9717daae98ac3f7b1c4f83eb9118","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:32:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a99612a952eb417286a40e546e95951d : SUCCESS in 37m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1deed29e74af4ada8ea6c74be053ed5f : SUCCESS in 1h 04m 15s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/539b887532b749518b5cc3c19f6a4215 : SUCCESS in 1h 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a96b9ea9ee4ad389457db26cd50f7a : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7b999d30fb84683b57929bf121f0b73 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e192cd18d994cf88b5fc99ed3eaa9e8 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120aa01d9dbc4834815b89c633f9442e : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/816602ad7d574a61ac8c9ed0c6360258 : SUCCESS in 4m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d62fc39334d4ce592808ed788a8b31c : SUCCESS in 31m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/74c4d448eb8c4888ae5a2ad985e69ec7 : SUCCESS in 55m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2c33b0ac40f1419f9d9c7e1a1d24f55f : SUCCESS in 1h 06m 10s","accounts_in_message":[],"_revision_number":4},{"id":"859bfe83079a39d3c461a0f0b81abc5a5ba8b23d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:32:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"884d71e52a1e809c00f50159b6d3ab041883bd1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:34:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b8347898ff8475d937b80c32ab8599f : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23bb1ae4c85c44cc880b908c36e4613e : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5c6eb173973a7781c8955115a6573ae59e45a479","revisions":{"2e9021b09f818cb64925bd5b5ace022754283c41":{"kind":"REWORK","_number":1,"created":"2022-02-09 19:31:59.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/60/828560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/828560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/828560/1"}}},"commit":{"parents":[{"commit":"c6a972e5dbbfa357e45e196215faa1519f5a3167","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6a972e5dbbfa357e45e196215faa1519f5a3167"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-09 19:31:59.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by downgrading the version\nof the elasticsearch gems.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/823155\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n(cherry picked from commit a8e6924beaea3e268435156f6bbc34812a69a01c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e9021b09f818cb64925bd5b5ace022754283c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e9021b09f818cb64925bd5b5ace022754283c41"}]},"branch":"refs/heads/stable/xena"},"c6b8a37c613e157acf99471e041f828155261cd8":{"kind":"REWORK","_number":2,"created":"2022-02-09 19:33:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/60/828560/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/828560/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/828560/2"}}},"commit":{"parents":[{"commit":"c6a972e5dbbfa357e45e196215faa1519f5a3167","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6a972e5dbbfa357e45e196215faa1519f5a3167"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-09 19:33:11.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by downgrading the version\nof the elasticsearch gems.\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/823155\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n(cherry picked from commit a8e6924beaea3e268435156f6bbc34812a69a01c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6b8a37c613e157acf99471e041f828155261cd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6b8a37c613e157acf99471e041f828155261cd8"}]},"branch":"refs/heads/stable/xena"},"d199eb45d2d576b8ba0efb76df69d6d432f8d921":{"kind":"REWORK","_number":3,"created":"2022-02-09 19:35:37.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/60/828560/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/828560/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/828560/3"}}},"commit":{"parents":[{"commit":"c6a972e5dbbfa357e45e196215faa1519f5a3167","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6a972e5dbbfa357e45e196215faa1519f5a3167"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-09 19:35:37.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by downgrading the version\nof the elasticsearch gems.\n\nBackport includes 711ec0829a95a12ae9e290aeb3837b457e1dd215 [1].\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\n[1] I353ef9b6b8da0e726c17ad2e06995d3690731e1d\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n(cherry picked from commit a8e6924beaea3e268435156f6bbc34812a69a01c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d199eb45d2d576b8ba0efb76df69d6d432f8d921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d199eb45d2d576b8ba0efb76df69d6d432f8d921"}]},"branch":"refs/heads/stable/xena"},"5c6eb173973a7781c8955115a6573ae59e45a479":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-02-09 19:36:23.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/60/828560/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/828560/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/828560/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/828560/4"}}},"commit":{"parents":[{"commit":"c6a972e5dbbfa357e45e196215faa1519f5a3167","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6a972e5dbbfa357e45e196215faa1519f5a3167"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-12-14 12:10:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-09 19:36:23.000000000","tz":0},"subject":"Unpin td-agent and cap elasticsearch gem","message":"Unpin td-agent and cap elasticsearch gem\n\nWith [0] the version of td-agent being installed was pinned, which isn\u0027t\nsustainable in the long run, so we drop the pin.\n\nLatest version of the elasticsearch gem no longer works with older\n(OSS) versions of Elasticsearch. This is fixed by downgrading the version\nof the elasticsearch gems.\n\nBackport includes 711ec0829a95a12ae9e290aeb3837b457e1dd215 [1].\n\n[0] Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\n[1] I353ef9b6b8da0e726c17ad2e06995d3690731e1d\nRelated-Bug: 1930867\nCloses-Bug: 1954759\nCloses-Bug: 1960444\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nChange-Id: I3045786e863b098d7339d1066aef6c857aa5f97f\n(cherry picked from commit a8e6924beaea3e268435156f6bbc34812a69a01c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c6eb173973a7781c8955115a6573ae59e45a479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c6eb173973a7781c8955115a6573ae59e45a479"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
