)]}'
{"id":"openstack%2Fkolla~828716","triplet_id":"openstack%2Fkolla~master~Ice6467086bd292af086322afc3fc4e869d89eefa","project":"openstack/kolla","branch":"master","topic":"refactor/engine-adapter","attention_set":{},"removed_from_attention_set":{"35104":{"account":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"last_update":"2022-10-11 09:04:59.000000000","reason":"\u003cGERRIT_ACCOUNT_35104\u003e replied on the change","reason_account":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"}},"34113":{"account":{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"},"last_update":"2023-02-27 12:57:29.000000000","reason":"Change was submitted"},"29858":{"account":{"_account_id":29858,"name":"Maciej Siczek","email":"m.siczek@samsung.com"},"last_update":"2023-02-27 12:57:29.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-02-27 12:57:29.000000000","reason":"Change was submitted"},"29636":{"account":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"last_update":"2023-02-27 12:57:29.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-27 12:57:29.000000000","reason":"Change was submitted"},"34501":{"account":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"last_update":"2023-02-27 12:57:29.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-02-27 09:40:16.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},"24713":{"account":{"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"},"last_update":"2022-09-23 07:33:57.000000000","reason":"\u003cGERRIT_ACCOUNT_24713\u003e replied on the change","reason_account":{"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-02-27 12:57:29.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-02-27 12:57:29.000000000","reason":"Change was submitted"},"33708":{"account":{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true},"last_update":"2023-02-27 12:57:29.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2023-02-27 12:57:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice6467086bd292af086322afc3fc4e869d89eefa","subject":"Add a way to add other container engines","status":"MERGED","created":"2022-02-10 14:03:16.000000000","updated":"2023-02-27 12:58:40.000000000","submitted":"2023-02-27 12:57:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":144,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"828716-refactor/engine-adapter","meta_rev_id":"6056f728f3a9a6980ab51347a07b2a797a9fe047","_number":828716,"virtual_id_number":828716,"owner":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29858,"name":"Maciej Siczek","email":"m.siczek@samsung.com"},{"value":0,"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-27 12:57:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":29858,"name":"Maciej Siczek","email":"m.siczek@samsung.com"},{"value":0,"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-02-27 09:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-02-27 09:40:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":29858,"name":"Maciej Siczek","email":"m.siczek@samsung.com"},{"value":0,"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-02-27 09:40:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29858,"name":"Maciej Siczek","email":"m.siczek@samsung.com"},{"value":0,"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29858,"name":"Maciej Siczek","email":"m.siczek@samsung.com"},{"value":0,"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2023-02-27 09:40:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"_account_id":29858,"name":"Maciej Siczek","email":"m.siczek@samsung.com"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true},{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"}],"CC":[{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 15:25:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-11 14:25:18.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-02-11 14:25:18.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-02-11 14:25:18.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-02-11 14:25:18.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-11 14:25:18.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-11 14:25:18.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":34113,"name":"Martin Hiner","email":"martin.hiner@tietoevry.com","username":"hinermar"},"state":"REVIEWER"},{"updated":"2022-02-11 14:25:18.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":33708,"name":"Ivan Halomi","username":"halomiva","inactive":true},"state":"REVIEWER"},{"updated":"2022-02-11 14:25:18.000000000","updated_by":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"reviewer":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"state":"REVIEWER"},{"updated":"2022-06-18 08:23:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-07-04 15:15:35.000000000","updated_by":{"_account_id":29858,"name":"Maciej Siczek","email":"m.siczek@samsung.com"},"reviewer":{"_account_id":29858,"name":"Maciej Siczek","email":"m.siczek@samsung.com"},"state":"REVIEWER"},{"updated":"2022-07-06 01:59:40.000000000","updated_by":{"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"},"reviewer":{"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"},"state":"REVIEWER"},{"updated":"2022-07-21 07:31:05.000000000","updated_by":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-07-21 07:32:15.000000000","updated_by":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"reviewer":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"state":"CC"},{"updated":"2022-09-23 15:37:42.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-09-27 11:42:21.000000000","updated_by":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-09-27 12:50:55.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-10-04 08:59:59.000000000","updated_by":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-11-16 11:21:37.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"fde8ba0af01fee53d690fcc6f63d8c483343cf50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-10 14:03:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"634717e5cab6217a8b8204bf0dbe0150ae248d91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 15:25:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9b6d8c874a714db897cf6e4286f03ca4 : SUCCESS in 46m 30s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7e890660ba2e47ba8c521691d2b1481a : SUCCESS in 59m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fee7448fa5684a0794b6788c0c6c9119 : SUCCESS in 1h 10m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e7439e355abc42968a3cddd21dac196b : SUCCESS in 1h 09m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/708a23124d1844eaa5c294f0f6317c01 : SUCCESS in 1h 19m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8be9515f8b674faeaa3e45fa64fc0010 : SUCCESS in 26m 08s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/72e6eef108764d56a0bd9c91ad692990 : SUCCESS in 18m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3870fd4b822745809e11bfac1b490490 : SUCCESS in 53m 13s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f007118c287d4ef292467bcd77f7cf3d : SUCCESS in 1h 03m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36f3c770b00e43539baf2c33d2f42340 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/854be6d9392e4064b642308896fb2ce8 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcc02ddb97b64da186923e9cb3480203 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c92bbb48ee7a44a4ad01d85c9eddf9b4 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f7bc2da335546ffbea357a872b64e31 : SUCCESS in 18m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f8dd7fe7f70445afa82849bb473fb4d0 : SUCCESS in 32m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/adc901d8799e48f7b1669cc2966d75b6 : SUCCESS in 54m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bed69711840645f58630fce72be33168 : SUCCESS in 44m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/43b4f25e471e45909a02ada9f95ca233 : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":1},{"id":"914b0fc52d39fcba7ea39f155de7baedf15fd6d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 18:00:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d1720eb804cf4d6291c32183c690cd46 : SUCCESS in 2h 21m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0834490fb4774c5a932a52b246c732ab : SUCCESS in 2h 20m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9df2129d1b654426b88ac701fedec043 : SUCCESS in 2h 34m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24b026705a6d40b7bc955b21055e9e1c : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42daac05968d4357abec3192cc53c588 : SUCCESS in 7m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea5b869f9a554294ac5ee966750edadb : SUCCESS in 2h 40m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58016024a49f1d81dc6f6589eb8348d373276fc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-02-14 14:36:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bf60a917e4e4f8ab850628120640037742392341","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:01:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ab1a46d02e6f47fb8e06f2745cc3f059 : SUCCESS in 28m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/aa78995bc38f4fbe8a8b3b1e42d1c520 : SUCCESS in 43m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0e168be378c14432902fbcc4b7453942 : SUCCESS in 1h 17m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5efd192482664791918a84193923f472 : SUCCESS in 1h 00m 02s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6f8b1bb0973643cb9fd8d8348f23c3dc : SUCCESS in 1h 23m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/629f99be54714c62b82800416de159ee : SUCCESS in 28m 04s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d022d9b1ffa54f309d990419a82da309 : SUCCESS in 20m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e3eaebb88a484456a337f15b5ffed2f7 : SUCCESS in 1h 01m 08s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ea26c34c1d044d62af0383fa8cf332ff : SUCCESS in 1h 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1d9c41033db409e9535118d36d23dcb : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/014d9bfc01644b949d450697b0ceec4d : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dd9a92ef901469bb39928cbe71e5f66 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d58fdf50a154c5fa9d15d6d9fe378fe : SUCCESS in 5m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1926c120ba6c43ec92141462d353e748 : SUCCESS in 20m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/45dc9d866cda49d8803968c441e37b09 : SUCCESS in 34m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ed5a55713e24b4ab07b73b04f98a23c : SUCCESS in 43m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/389e80a1b3dd4e3eb3922be6d7e2e268 : SUCCESS in 49m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ab23f5c9b9d641ce96015f2dbfea4d1a : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":2},{"id":"4923bbd2c990cd1961c38e58b7bdc40b45c9c61f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 18:32:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2a361326418c497ab60bf979739c3307 : SUCCESS in 2h 38m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/387bb34ce84b41028cee90130929013d : SUCCESS in 37m 33s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f639fe2565b4c3596b7afe8f8be8d33 : SUCCESS in 2h 26m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb5fa3d410bb40989e9dcbacbb6d88a5 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b847047d150476ebc394d0cf6f01712 : SUCCESS in 6m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f7c95ffab0c6411cb1ffd6b42cd598bc : SUCCESS in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dd67d1c46f44652c2a7dd50ff35006dc937c656e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-01 11:56:59.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6218b0df4fec20506da4e5b2f6cf14b51cdabbeb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-01 11:58:57.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"a491b0c393f2de70ce48e90f4ce1c7a0307d44dd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-07 10:58:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4631e503fc34259939d75e600a1fadc39c208928","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 10:59:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"53e9958bd4c23d37c3ab8e7dfa596ff885f1beab","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-07 11:16:57.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"709a16d85501b42d5f1d1b3ea97271463846af30","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-07 11:34:02.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"e7ec98feb54f5a443f02f26d53e7e24e8a632f7e","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-07 11:34:07.000000000","message":"Removed Workflow-1 by Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"dc710f54354a1292b71da51d79c7298610855919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 12:19:11.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4ebe49445ebe4eac93dd4eb0ac817cec : SUCCESS in 38m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/470171d30d304ffba2739900e7d8886b : SUCCESS in 1h 14m 05s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ab17ecbe23b54e3b95962573877c67d9 : SUCCESS in 1h 05m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/81e04d6262a648ce980f9c41cd713724 : SUCCESS in 34m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c7d1b0ad7e05453584a1437011a95c15 : SUCCESS in 53m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/31d1cb7a647e4bfba47c2294e89eb1e2 : SUCCESS in 1h 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03ccb94390a14973a213afdab5750f7c : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfef3f3f86d744a3b0ad69fdacdc9e18 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/205e718df26a4eeb819b612496ead299 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab502a74ec64e2fa5c1b27ecda5285a : SUCCESS in 9m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b5c04ad961c4b2a97e7584a7afeae59 : SUCCESS in 30m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/887a597d44ff49beaa69839ed80d6e52 : FAILURE in 15m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/046c1220b88f4b42ab133a871ff489cc : FAILURE in 15m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9762bc01b30875568f27139f189d4bd538f3c5b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-03-08 11:38:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"acd3975c3943d8e1c101d4beff8ccb15dbb409cd","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-03-08 11:51:00.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b094e29a95e057c7832a7e5b8e4cde15783407e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 13:24:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/52d6a7c9319c405fa0dcaba6302a0910 : SUCCESS in 32m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/df93e75ceb0242b78d92111fd79d0d3a : SUCCESS in 44m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/eb63b20a5d1a48228b76652cc4f8d0f3 : SUCCESS in 1h 18m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/007be54f6a0b4679a864daedb531342e : SUCCESS in 1h 09m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f94c1885144c403a83d9cc9b2d92c22e : SUCCESS in 1h 30m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d7565969e85143e79053ab753c4e3417 : SUCCESS in 37m 12s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7d941c0ee0ab4a9d8a0eb001a06efdbf : SUCCESS in 22m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/439a01d180bf4ca0999f39af7e169963 : SUCCESS in 1h 10m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2d6ca2da61ae49d2aa44eb8f4662e36f : SUCCESS in 1h 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3acfb0e8e6814300a5f4c4d3d8c0b780 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6da0e0d617f946bdb9d5fd72410e21d0 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f85b920d6a04a818489a6d2f159fff3 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d6665e18bf48789c8d78ab474fce08 : SUCCESS in 7m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a5abc361e8e4459ba6827797bafe3ffc : SUCCESS in 22m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/30872dde1ea44349b73249cb1e80135a : SUCCESS in 24m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/56eb5429db374f55a8229fb8c767a0e7 : SUCCESS in 45m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5683bded62884b0e9a58f4a31969cfd7 : FAILURE in 41m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8434ca0ef1a84aa68729d9f9ec419ee6 : SUCCESS in 51m 16s","accounts_in_message":[],"_revision_number":3},{"id":"3c965d95b62bba0b70dc92701c8b2310ee642c0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 14:31:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/469bd456694b4f818564ac138849d9c2 : SUCCESS in 2h 09m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f4382d04400415b89cd4b6c1f818913 : SUCCESS in 1h 51m 25s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d473a94270a34105a833a6307bb64657 : SUCCESS in 2h 20m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39c6ff2144824eddb9cf58b32d02048b : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a05ca6242a045f9997607cdf489265e : SUCCESS in 8m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f5f05e0ccea483c8ac1a9a3bebc5473 : SUCCESS in 1h 46m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5eaca8660ea7b92eb69a60bc3fc0c5c4cf0978b8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-14 21:14:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fd9e728d439df9a7805a10997d1cc9fb17581719","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-14 21:16:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e14dd786bc72decb39ea867066a882ed1345c794","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-14 21:17:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9f3552e9eb9d892ac720bba3c6175301def2ac26","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-03-15 08:37:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"384a8cf8b4b41a719c82b9145c8874a2959838c3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-15 08:52:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7d0741116a45db594ea68cf6a39ad5166fe07156","author":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"date":"2022-03-17 09:32:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d95f7639121b71bd32f43a0047a2a56077dfc344","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-15 07:26:26.000000000","message":"Patch Set 3: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f6723b5976e7062a77c22379eb14e493b9705beb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-15 08:52:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f79413cdbae74d53435e16de9c615b580ddb9d8e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-15 08:53:41.000000000","message":"Patch Set 4: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b1f75492f541748e00c0e032d7654903abc2b710","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-15 08:54:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a71f1ceac143721d8f489059645337cf71c9b491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 10:18:05.000000000","message":"Patch Set 4: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a3be128b41d74a8ca8f3d2e335d7c270 : SUCCESS in 29m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8d1021f52dd94cd1b6a3f3cffde66259 : SUCCESS in 1h 06m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d426d39f377f4d93a08511afc80662b9 : SUCCESS in 1h 20m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98c544ef2077406d90c3c9cf35f4470f : FAILURE in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16e51b47938640debfb2e851e10b5660 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28b25167ded94767a2bad814e9200ae6 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff544178cd3640c2a4c6bde20fddd9b5 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79ceae042ee44925bb1df6ec1819e35e : SUCCESS in 6m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/26928a1390204894989dc54ab437f5c6 : SUCCESS in 29m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eeb77a5af3c4463782393f20d4e8f430 : SUCCESS in 50m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a4088294d4484f58b7251ed3b2c5d75c : SUCCESS in 1h 08m 58s","accounts_in_message":[],"_revision_number":4},{"id":"cecdfd647955ccbc0a396c9a18edcb13a14ba66f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-15 10:51:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"15fee9a6fdb4412d6ac0ae391423b0d341f5c29e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-15 10:54:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"16883f8da7fbcdaa42310ecaa44c2a84b6198cc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-15 11:02:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"77adbc06d599c61d95fc286ac217f8d6ca67bd64","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-15 11:03:09.000000000","message":"Patch Set 5: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"68fd2c3067ee27eefe9aea535c3a018b0f086dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 12:43:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/70587425828c4bbc998b7b70d82bd33a : SUCCESS in 22m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1b4e156e38f24e17b718f59d1a8d064a : SUCCESS in 1h 04m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9c77d44379b5486c95370f072d03d2d2 : SUCCESS in 1h 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72222083ee93480bb9caca226fbf5f34 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea27ea221f844cb194e46e7e0bf2757b : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73de079b895747e9bcfbbdc00e9c005c : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34d0df148be547b591fc2fc47ea01b8a : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af33e326fb904d359c0521736bdd91f1 : SUCCESS in 6m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/046b53b8cc6041129bc3f877883ad4e5 : SUCCESS in 1h 16m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f5cf6c6a9a414370aff05a56ded802af : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d398583391ed4feba2ceacf5e513d189 : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":6},{"id":"9adbbd128cd4345cdf7e1ac72eea199828430192","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 17:21:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/93933c89ce254c989925cc35a8903af8 : SUCCESS in 1h 49m 37s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f55d05e1cc144b948a1a97888fcd7f12 : SUCCESS in 1h 37m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29daa2b8213c43efb9e7a3a0905fae97 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6176b75e1a8e44c7a40ee8db0088acfb : SUCCESS in 4m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/51befbaa55294ccfbb74bb3a30051e1d : TIMED_OUT in 6h 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c632099d69af6b4e0145d30372c53aa298b6fb00","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-18 08:23:11.000000000","message":"Patch Set 6: Code-Review-1\n\n(20 comments)","accounts_in_message":[],"_revision_number":6},{"id":"645ed054562db7b9aea3d097568877e49d09a42a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-20 07:48:04.000000000","message":"Patch Set 6: Review-Priority+1\n\n(14 comments)","accounts_in_message":[],"_revision_number":6},{"id":"96222217be39fe2a197b40186f1fc7f58079022d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-20 07:48:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c753572f3ad6fcfd81973f174e09e479cc52e3fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 08:58:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e49503885b2d4480894f62a4d3b739a4 : SUCCESS in 19m 26s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/25b07623c8f74ae28049644de0a8f624 : SUCCESS in 49m 50s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bfedb74b5d7942ea81cb68fd316eafb8 : SUCCESS in 1h 05m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e715cdc1d9464c63a62fd02db9e289de : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e537c6cbd78d425b8d6ee05dd3dfbffd : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f05a7cd38d1426ca2a3e66ea688ab7b : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5dabd19e8cc54949ada6396f9f52c8fe : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc4b352c8ca49459306eedd8d9a6eb1 : SUCCESS in 7m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e081e26975f1422597aecc70f11b770d : SUCCESS in 27m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bab4b2093e674460b7903b70c0c35edb : SUCCESS in 47m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1a409ae4ba424bdfab639fe971743e91 : SUCCESS in 1h 08m 22s","accounts_in_message":[],"_revision_number":7},{"id":"c8b6c9fefd8abac54aa33bcc839748fcafed1005","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-20 09:01:37.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b531b280a26d67979abc94163c574c6a0028fdad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-20 09:15:50.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"de185b55324bc12bb9463297b85fb69479492805","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-20 09:15:54.000000000","message":"Patch Set 7: Review-Priority+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c42989967535e736e30d5894e8e0ce2442ae13df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 10:32:56.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ed3e7c0cc137447e8f511943d21fc1e8 : SUCCESS in 25m 44s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/66a750c8a08946cfa11ab5688fa9a57f : SUCCESS in 59m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f623e2a415264186be7fd0dcb95cc1b9 : SUCCESS in 1h 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b7b9dcbb0834b8bb0cb9f80674de8fb : FAILURE in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3801aea192bc4e989518987ff0182f5b : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8aa18072e2f54832817567c54b69ac43 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/042709d73388471f8bc2a41d2d08569a : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e15f967497be4038a77907d3d0ebb082 : SUCCESS in 5m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/574aea7be2b94211a55778cd367d6162 : SUCCESS in 17m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e88644591ba44f62b12bd35ab11bd5cb : SUCCESS in 55m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d22e51a0a7a04892bf2590b4a6166553 : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":8},{"id":"92e3661b93946da8e32dd5a2aed867717ad0823e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 11:23:45.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b4acd3208f0c4f6d8f453f896fd95d00 : FAILURE in 47m 12s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c01e963e4184db98dafead51c4bf772 : SUCCESS in 2h 05m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26aa05cbe2c14d579e49fd8ac10835dc : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d93afec8a63a4614908b7f12fa707fee : SUCCESS in 7m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/64a4deeb27c642d28876c37ba6f30002 : FAILURE in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d12638943c1180508208ca00b7a1b678431d7c09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-21 12:46:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"235ef09ad5d251ecd82df973a89f6c9ea4c96c84","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-21 12:47:55.000000000","message":"Patch Set 9: Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a6a0f44b7e0e922549a6ae75b7717b10450d47d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 14:03:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4429af5db4964159a8031459d225343b : SUCCESS in 27m 26s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ac3c28633c1545f6bc702ad7f1d50fed : SUCCESS in 1h 11m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9d81b71b67b144d0bcb959da0dc47a1b : SUCCESS in 1h 06m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/850e9fe62c404fd891bb5d99183fca87 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9f31852e7ec4a00824edecae8d7816f : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41dd58f538bd480d93f520c3f017908e : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69566c5ee57f418394e7854e782b254b : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2663fa151c354636b1ecf087d6394a31 : SUCCESS in 5m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d651a3e7bf5046b3b0741caffabb2b21 : SUCCESS in 16m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d50d7a8c6354e6aa1f3d7907574f2e0 : SUCCESS in 45m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7013cca8110643468a0284588b7b4b67 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":9},{"id":"6c6a6a9d5c9a26d0bb2d3e691209ac0e99dd75d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 15:54:17.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2f8ed0702d184495a44c52b91a44c82b : SUCCESS in 1h 17m 05s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e52d3f11f5848ca8fb3471164cf78db : SUCCESS in 2h 17m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3118ba4ffc014a2aac9d832784da5009 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c7b6d9f30ac4e5f86193fbd20a6b70c : SUCCESS in 4m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/65bd9a28cb6c4db4879a9716c293f0b2 : FAILURE in 1h 26m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"612e79126f6a2e1f92b54f3baa41294bdf35316f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-22 05:04:29.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7a070ef30687c030c607d89fee42fb1a5eb907b0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-22 11:04:38.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7e34b74104c8c2aa5fbf81d490bdaa2c61d08139","author":{"_account_id":29858,"name":"Maciej Siczek","email":"m.siczek@samsung.com"},"date":"2022-07-04 15:15:35.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"620ed105e552e8a6da285e7480d919034b994223","author":{"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"},"date":"2022-07-06 01:59:40.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"17b081b6e3b918cd6a0deb4f4283b73e862b26f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-07-21 07:31:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9eefc3a9397e8641572bb0e9fec98f32da252285","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 07:31:27.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 828716,10","accounts_in_message":[],"_revision_number":10},{"id":"0055e248cd1658ed26eeaba4b03b5b1f75267d4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 07:31:27.000000000","message":"Patch Set 10: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 828716,10","accounts_in_message":[],"_revision_number":10},{"id":"561c2a742884417b4d44fdeb9176122d2d8313a0","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-07-21 07:32:15.000000000","message":"Patch Set 10:\n\n(10 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7d7dfcca21328180cfc300513d7cca9b805a738f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-07-21 07:54:32.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1e82b8733b87bd90e7a7dbab48a4091d9f9d73b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 09:06:54.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/594e11747cd64dc28a28a6dab5603bca\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/00b80248a3e5442795a4efe0fa807a57 : SUCCESS in 21m 24s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d68d9967db7c4e15868f43ac39c6e085 : SUCCESS in 57m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/873304fbe31d49828e2144330f046358 : SUCCESS in 1h 06m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a58855ad2540afa0c4bca077434791 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0757c904a87044ccbfee13d1743927ab : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d046f478f274e5d9c3a5d32391beadf : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9efeb840a75c4bdc8fae884ff23e251d : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b76b8db9c3d441a81da2598c9fa7e2e : SUCCESS in 6m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5fc9c22896ef4445b7e01e9213627635 : SUCCESS in 33m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a4b5cac9a594a82afdc29022c4a32c8 : SUCCESS in 1h 04m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1194b5280a1643ce91b4c20c1037c0b4 : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":11},{"id":"3887e5dd086f926a53d368623353c46e1f149aeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 09:54:38.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/189587610e27406ebc6f49daca46c63b\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/218d62197dce4133835e4e1032e7d7c5 : SUCCESS in 1h 57m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/66631e5c395240df8f759eb65e54d97a : SUCCESS in 1h 29m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8be393a05b84be882d6701ed456a5e1 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85288411c09844e69b17c29ae67a9dfc : SUCCESS in 4m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/256de19e5b6148e38a9f99a7bd2d26b0 : FAILURE in 1h 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a88e92c9cfd9273b9b58ecc7318b69519a315db7","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-08-23 07:27:04.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"b376fe1e714be621408a27ed5638defd4f7b9f92","author":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"date":"2022-09-21 10:33:09.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"88de2a6e5a73e0757cd2bb11f5bcfac5fa65ced1","author":{"_account_id":24713,"name":"Wonil Choi","email":"wonil0522@gmail.com","username":"wonil22"},"date":"2022-09-23 07:33:57.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"f081ec6e20a5cc35f5677240a1de7157692998d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-09-23 14:36:50.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"079eea865735cbd45a4caa90c1045a57293863a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 15:34:53.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6085e6499c343e085180e13453ff88d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a66bbcb980a44a43ab76dd4ac6410cd1 : SUCCESS in 16m 16s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5af59bbe52ea45bbba5ee20c95c0e0bc : SUCCESS in 45m 10s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5d98a9fe05ef401fae087704adb6cb73 : SUCCESS in 56m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a11e4263784b9b98e29b3c2d4e69ac : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/268f14f456df4be2965927bb43089e22 : FAILURE in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17bc379af5754a58b9f7d885c8ff19b6 : FAILURE in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15f9c7ca98654e188b766209d0a69f35 : FAILURE in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78efa5f83d134dea912a04ffce345c97 : SUCCESS in 6m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/223dfc5c1af844818312a4cb41caecd1 : SUCCESS in 22m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b600e576ffbd4917b60ae585fa30cf2f : FAILURE in 15m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/451c0c857a734d69b273c835cd91d45b : FAILURE in 12m 28s","accounts_in_message":[],"_revision_number":12},{"id":"9afd57837f4f8b103317d1c2ace3e91c3a1db038","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-23 15:37:42.000000000","message":"Patch Set 12: Code-Review+1 Review-Priority+1\n\n(12 comments)","accounts_in_message":[],"_revision_number":12},{"id":"7b059169be6f4f9d8d6446f24a6e247b92e31289","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 16:55:20.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fee8cf8a30bc4a43849513d5b1836e9a\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/079849d2747e4402b5bfae539798488b : FAILURE in 1h 33m 26s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/127bfebd7a9b4fa28dfae0af7ef44914 : FAILURE in 53m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c970ff1d0c84241b876aa802e6a2493 : FAILURE in 18m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3bac6d33403b4b6982812dfd0a51bf34 : FAILURE in 7m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/98221ad1c42940de8ef758cdc65c906e : SUCCESS in 56m 01s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"edf57bd4909c9ebfed38c72b685066a2e7235bfa","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-09-26 11:01:39.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"23abc9554bb524eb2dd1855205e3be4a3be2f224","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-09-26 11:02:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7025f6c707b894ef95ae77aef65225624fcc8a0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 11:02:40.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 828716,12","accounts_in_message":[],"_revision_number":12},{"id":"883741aa73c9f03f9c9b6ba67baf43fc966480f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 11:02:42.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 828716,12","accounts_in_message":[],"_revision_number":12},{"id":"d537a38407b3b73ff2e3e17753f69bf01dd64030","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-26 11:14:27.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"ad4fcd6c52b9507299416f26e5c35804a393b9e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-09-27 11:42:21.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"04ef5bd7d88a8afcd2b85f9946210b4c35d96770","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-09-27 11:50:24.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"556b4176c9e934472fcb9a5334f9a3a659a5fdaa","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-09-27 11:50:50.000000000","message":"Patch Set 14:\n\n(4 comments)","accounts_in_message":[],"_revision_number":14},{"id":"1f9d192547ae6dff9b8ac0ba8c53c5bc48364e72","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 12:50:55.000000000","message":"Patch Set 14: Code-Review+1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"410dda7b95357bd6f2d719b287e7fc55ce69638b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 12:58:51.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"5d316a285b3f1c85afb325b594c2ace43c4f16d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 14:21:50.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acafa55366b74bdeb4044b63a6e9a32b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9f6bf8074a9e4b0a9c8864804d99d268 : SUCCESS in 17m 47s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/fafc0b49d1044d8ea1f13484a95cfbc4 : SUCCESS in 1h 18m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd17bf4109ce4b4292236719cc4c01b7 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77ad7850a42d4e54b70970bb8a190d67 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f976886844540df8debe00f55aa2772 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5159de722db94fb6bfed486c3799015d : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8dce28ee225445f83e97ed10eef8707 : SUCCESS in 4m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bb906570e7234949bea14830ef2be0a1 : SUCCESS in 24m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2db1b5df4c8841faa99271cd49d86589 : SUCCESS in 57m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1c703dbf185e44a2a686df000c7433d0 : SUCCESS in 1h 14m 32s","accounts_in_message":[],"_revision_number":15},{"id":"c841c34030c51c7272b1acc097974b2347cb0295","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 15:01:13.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e085dc473a5a44a6b5221c204264a296\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/39b80cfed84d4d93ae10ffaa7a6d5c69 : FAILURE in 1h 59m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa2cd6570c694f83948b26371e636a22 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1581783d89824e53b53097647e54b7d5 : SUCCESS in 9m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/931deaa5fe284b0f8cfe14dbc049a7cb : SUCCESS in 22m 46s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0ceb35c3b433f7947138880858129980dd1a5c71","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-09-28 09:04:48.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"6ffdf28ce5c371689cb2699404522c659d31bd75","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-28 10:48:45.000000000","message":"Patch Set 15: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":15},{"id":"87e135f573f37137724848e3fdf599a46a994401","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-10-04 08:59:59.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"7978be604347bda57fd3837ffa150b28fde849f3","author":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"date":"2022-10-04 09:03:42.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"9639fab75d9c5ec695d0175fc6d499d3341d42a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 10:05:47.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be9ffa8c544f4bb1810271ca3a81546d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c2755f850e464f31a115f723de627848 : SUCCESS in 53m 39s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c71cc4bf3a624c87bfc384ecc8a7d6bd : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62d458a374d044a4b4901502cbd55015 : SUCCESS in 7m 32s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/539ceba55cab44ab926928a3a1929b6f : SUCCESS in 47m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5268f18fc4bf47f7981385b146225345 : SUCCESS in 1h 03m 44s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"bbbbdfabec63e5812e774a02e06ddb5c6c510ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 10:16:32.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0dc9d5a075b4e4fb4ea533ade4c2143\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/77c722926e254169a677771b1c648436 : SUCCESS in 24m 11s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/69d9f5162482466281d7e7f107e68dcb : SUCCESS in 54m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/520062cb027e462aa0b1a22420c71501 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65f71cf6ea864e5da7eb6901d08272bc : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/122dffb2af274dac9afbc4935e838750 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4fde5e7dc8f4ec68321e816b4ca65cc : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cd3e8bd28db4f90965e7e9cabca3bbf : SUCCESS in 4m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6a89043eefa247e08463899da889f155 : SUCCESS in 23m 52s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1f93d5b3618d4fc885901c2e71995c2f : SUCCESS in 1h 10m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5e95b11e42aa4f6f9b3e2e424e3e8426 : SUCCESS in 20m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/28bacaa6a4a14f4c9197713589f3edde : SUCCESS in 58m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/63e965415d5a4a2fb7ab8e1b7af0a6a6 : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":16},{"id":"1bc15e01eb52e2528c784735ca92442658853fb9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-05 14:29:22.000000000","message":"Patch Set 16: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"a206c54e5109cf4332bae250edd04566546f796c","author":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"date":"2022-10-05 14:41:17.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"cdfe7be07c1646eb9d4f6a1e0268ced16e921ea1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-05 14:57:47.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"a3c0b9643f9a6ac89bb8d695715203cd148a35e5","author":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"date":"2022-10-06 08:04:22.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"2eb6a2741bdf1b2ec9b89649aa006353635db544","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 09:57:44.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"5f77c9d5e07f6668c6a52421edd83404f3312337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 10:21:13.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"dcae6f25fb86423a4677695a9210128b9632ea60","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 10:22:38.000000000","message":"Patch Set 17: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"7b4fb69cd1f17bb8428602f3ce7034aa48e198a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 10:44:38.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04856594a86f49a5b2d54199de3e2feb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b8550429fd044d8189892e2fd05e82c0 : FAILURE in 4m 17s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/bd46fec3ddd74f47ad11dbfe5535c6e8 : FAILURE in 12m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e658a0da7dc74cd6bd74f75464bbf0f9 : FAILURE in 4m 18s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/25b3ce06e7c648839e650c0bf0c1da66 : FAILURE in 10m 29s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8806eecac6a94873b1afd0d00ea831ab : FAILURE in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d3ae10316f74f0281b12a678300a834 : FAILURE in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/201910f53dbf4812b898100b2821cb02 : FAILURE in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5eec9a614e23427989a82fc1ac1b6b33 : FAILURE in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66aca2463c134d929957e6db278fb40d : FAILURE in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccd5244448764083a84f4cf279ff7678 : SUCCESS in 5m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f5c5f90db5754417a4534c9c342c98ca : FAILURE in 4m 14s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ba0ce2b3abca49f5a5663a5ff4343e9f : FAILURE in 12m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/952609c00d9d4f15a2734c8ce49b2249 : FAILURE in 2m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd65e8e60f9f421fb004906083f52821 : FAILURE in 10m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1d69c7fde57246f9b15c64cd99072868 : FAILURE in 9m 24s","accounts_in_message":[],"_revision_number":17},{"id":"5453e9b4147837f51afdc3454321917b2ec1d7f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 11:09:11.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5079670cfd1c432589ad07e08872b4c7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cbbc282fb40945db93d235b39d54cf56 : FAILURE in 8m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79af5f12b314422f852632a25f1cc406 : FAILURE in 11m 23s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a08b1f033b5547eaa66421d06aceb4ce : FAILURE in 20m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/882d2590f84c45a78469a2b0ee08f83d : FAILURE in 6m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11a89a5072444fc6ac998dd3c4946518 : FAILURE in 8m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d24034da7c5f49f3af51efe311f3ab5e : FAILURE in 10m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/96077aa332a04cc5ae47722b0d180fab : FAILURE in 7m 04s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"3f79d27db366ea17e882b99a26fb2f5df391d85e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 11:14:31.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"42f9d91ee02f5862bf8e7f9862a4dedca19bf9e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 12:32:41.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dbf01777fd642aab2807a8f8d75e8b9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/976676669eeb49809c8fd0f253af89c1 : SUCCESS in 23m 45s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/89ae5f656674441d8165b87099bbf0e9 : SUCCESS in 1h 07m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c2ae65d92f734c2991d9840de9dbc909 : SUCCESS in 22m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/efa24055ce6649bb8b28f34a732b9631 : SUCCESS in 49m 03s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/06872ead24d64ba6a8b184cc113447be : SUCCESS in 1h 08m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3759f89414544f4db9a078dacf1ffef6 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d5114ef20854b829128dfef18c368f1 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63330301e70247fc986ed9ff9e564d02 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5298401350df44f8969457252a0c2ca8 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7246a33b83b443889bcbcbe200658ba : SUCCESS in 5m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4ab36779c21b44b285fc03c4e07d6e24 : SUCCESS in 26m 36s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/2fbfc33975394d6ca76a2c98db9cc900 : SUCCESS in 52m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a82a825b8c934e3eac17f28ca7d221b8 : SUCCESS in 20m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a167ef87dbab4990b66719505b02bec5 : SUCCESS in 44m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b968eb658816411ea7fecd7682e882f3 : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":18},{"id":"e1dbee8e135c18c9f31844791c9de94196894789","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 12:57:59.000000000","message":"Patch Set 18: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":18},{"id":"8d295fac58aae563cecd822b40dc6ccde879d679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 13:03:28.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"4ad7dc3d1ae6862315ce07f2711b77ec41c58302","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 13:03:33.000000000","message":"Patch Set 19: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"1c63295e660d7361a944330bafdd10ffad1a1558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 13:04:03.000000000","message":"Patch Set 19: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 828716,19","accounts_in_message":[],"_revision_number":19},{"id":"13896c41d1ba7a7078b6b61acb6d5e107a0ecce8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 13:04:03.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 828716,19","accounts_in_message":[],"_revision_number":19},{"id":"3d4941a925d767e86cb8f235a86b4c33b8f74499","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 13:06:10.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"211556274f640f970d4412283db5670ff1383d31","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 13:06:42.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"35a957848063cfd835936f22fd2fe9b54ad11598","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 13:06:54.000000000","message":"Patch Set 20: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"b84540a5f98661408b782a68fe7f478c70f922db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 14:26:46.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76be611780bd402d9423ec69a7f9ccac\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4dcef004a8a94f3d952323a5c94ae355 : SUCCESS in 31m 41s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/d36f481453704c999d488074dbbd35d2 : SUCCESS in 1h 14m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ed7702fff1624daf81f97d30f2a3afcf : SUCCESS in 28m 59s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e5a53b87b89e4cc2b84023f23e3021e7 : SUCCESS in 1h 00m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7a090216feba44439d39edb00760be9c : SUCCESS in 1h 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44014586d8b84c36b6434f5c148b7b9a : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e73dd37696d84da086471a0177453ccc : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be5a3089d04d4a23823535f60baef775 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0fd71cb7586416eaf36dd4af566b2fa : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76346d914c0c43b28b9b1d7d91cf42b6 : SUCCESS in 5m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9ebedc53eae342ef8df9137c9c380f1b : SUCCESS in 20m 31s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f3337d6a67764e5bb0381d43d20ef3ec : SUCCESS in 1h 04m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6dd632e557244476b603a3738b95e5de : SUCCESS in 23m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2675ae29e3284b9e92ff5f7e0d9c2a07 : SUCCESS in 52m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b57cdc3939d044eda0fdc191e50fac72 : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":20},{"id":"a5b6a21dcb2e5e81eeaa78d29ab0dbc89e2ada19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 15:52:51.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b5feb902f034bce8f00a3553415cebf\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2b4c2b003b9a4811b518b87f95b86ccc : SUCCESS in 1h 37m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7208f5183d3a4baa9521081cd4f1d911 : SUCCESS in 2h 05m 05s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ee0b4d5e021542508379854c50fe8ece : SUCCESS in 2h 03m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d519c0896094b938948d6ed9eb2e0ce : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b60a039938ff4ad3b1c6ad9dd13c5a0f : SUCCESS in 8m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6f99952ef2394ac1bc86c7cd752e88b7 : SUCCESS in 1h 28m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/716e26b808b84113a17b53262f884cd0 : SUCCESS in 1h 27m 15s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"27112b441671ad44e9462c98f903e462859b3ccf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 16:27:01.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"e5478757a422b35f6b250a626ee1ff432a2a9a19","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 16:31:57.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"043fc82826eb158079ec845073558b5e6dbe545e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 16:32:16.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"02d63f044fa22e93303ea4f6edc32d8ef10ecf4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 17:46:41.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a88033e41844385bcb65ffdcbec7fe0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cdfca0ec1c584c70a390a8f7cd26653c : SUCCESS in 22m 03s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/4228149140664cac84df0cd73073e566 : SUCCESS in 1h 03m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5ce29b9e2fd94e91b3e32d32d845b89d : SUCCESS in 22m 08s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/48c3f5d012534817a2b55d95430f863a : SUCCESS in 1h 07m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c3eb8b032c8543fda0019086d2365e4f : SUCCESS in 1h 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d49e6b7570e248e38e8e5292228d013d : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e15ef2bd3de4b2c98aed9740db7eb6d : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91520284d2104fe0aa4dd4835da68cb5 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43e30cb7e8ae441e8d0266ea14069e70 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02be5d510ead453d90f514119961b1ef : SUCCESS in 5m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2b326206e5804c4b837af7401ba02653 : SUCCESS in 41m 29s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/e3a82b880d834b16a4648b816c2229ac : SUCCESS in 1h 16m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/017ed2b059ed42f8925736440aa23e94 : SUCCESS in 24m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2be7a2e2cab34d348fc8c27a964166bc : SUCCESS in 51m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/db28bc4324b4409ba5ea6e6ae4bdbffb : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":21},{"id":"52e0588d3dd4414822a9b7e8221eaf3d51e7d6aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 18:46:05.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e91be611304041a2ac7b0358c7c50ecc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/05cf597a25a54259b218a852e5700c07 : SUCCESS in 1h 24m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f12d3688943d4a618ae299f4463b0454 : SUCCESS in 1h 09m 01s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b118a1c95eec4e8b86dfafb0b96006ba : SUCCESS in 2h 07m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0124bf240dab4c8faa060bcd5baa7f98 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c122df0515540bdbb5c97dabdbe949d : SUCCESS in 9m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8f568a4da7464d3384ad8e35e650e592 : SUCCESS in 1h 08m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/276337c6172e4b53abc1236d930f62f2 : SUCCESS in 1h 24m 40s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"36fe29e795cf046b248ef5cae30ef134c63c7516","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-07 04:42:51.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"c7459574c2a2146d83b5156049eed17db5444540","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-07 06:55:09.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"6d907cc59e30aa4f69aeab46821d81b372a345fc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-07 07:00:15.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"66eab3f19189324aa98c52a2ce927888f17b8941","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-07 07:23:42.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"63da2cfef3c3943509d3a49fd8cd81398afca972","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-07 07:27:58.000000000","message":"Patch Set 21: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"30d277dfdf68d5b177b968aa4baf8b180ce6f6ac","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-07 07:29:47.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"482b00d04ab83223134fef355683a4cc87db5498","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-10-07 07:59:17.000000000","message":"Patch Set 21:\n\n(5 comments)","accounts_in_message":[],"_revision_number":21},{"id":"f522d54845babcca9c3a56b14b35f709d369d8a9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-07 08:12:33.000000000","message":"Patch Set 21:\n\n(4 comments)","accounts_in_message":[],"_revision_number":21},{"id":"a6eb31b2b6316f00d9eb438d97ccf5bc11ce24a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-10-11 09:04:11.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"232de6b7658d4e606406c241260c4b3c4e273320","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 09:04:32.000000000","message":"Patch Set 22:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 828716,22","accounts_in_message":[],"_revision_number":22},{"id":"2b9b46e037f263d0c25a340a4635576012cfd371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 09:04:35.000000000","message":"Patch Set 22: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 828716,22","accounts_in_message":[],"_revision_number":22},{"id":"9c7752d0723a93dfc637a18cd51f7fd71e415693","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-10-11 09:04:59.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"460df078ea6e3ad9facf6adcccac1b6f005082d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-11 09:16:18.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"afe7fd94d9c503b5d0df76f3e1c9a662cfe65eed","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-11 09:16:39.000000000","message":"Patch Set 23: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"38b992bd06506b73b6f971ec3274c6a647d11b7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-11 09:39:47.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"b921f57792193f725dcd390b9881534394194b30","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-11 09:40:10.000000000","message":"Patch Set 24: -Review-Priority\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"cdf0d9c6a406dcf0d5700936f1d45e4a9f34771f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 10:59:12.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0652ccea24fe47af83159be9a40dd133\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e944b6674511465e91f99d7c539a9c46 : SUCCESS in 21m 45s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/260c62c7432c443981d966ab46db3540 : SUCCESS in 1h 17m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fa45176406444940b112758e6715ca5a : SUCCESS in 24m 32s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fe833531d22948c5b0b979817256fa97 : SUCCESS in 1h 06m 54s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2eb7e0bfde234ad788e94c83300752f7 : SUCCESS in 1h 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75cef9467a7840f1a00d77baca70d4b4 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8de545a18ee5410c9ada4da727776754 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55f1020008344c21b533d31b7373df9f : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ddb9bda74434f529ea1196f2a472033 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7107760d01446ea8c1f9ac993becd84 : SUCCESS in 6m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e41401f323164616b8b46838d2e86c4f : SUCCESS in 24m 57s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/c7d3d178f04444f087424c34ee3256b3 : SUCCESS in 1h 11m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/444fdd91e17e4865a0f2f355c4cbe696 : SUCCESS in 23m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/be7513cc09e24fe4aee8f754446e1d14 : SUCCESS in 51m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/91c72814efe7426ca81adcf5cf6dfd12 : SUCCESS in 1h 09m 47s","accounts_in_message":[],"_revision_number":24},{"id":"72b1c99d2ed5447dc9137514141a9c2a07404bde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 11:20:49.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55a4b41220f14c9884efce1cf54c9c34\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ed8a0909c15c4af39f18dadabeda65c7 : SUCCESS in 40m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c8aea9f5a8a146db9a8c1992358ed03d : SUCCESS in 39m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf74e06e2a9a4d1eae7f0f0ab676cd52 : SUCCESS in 1h 29m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d121f01d0a5f4a44b0e79b6ddd2bcaf7 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9705aee937940e6b9da3cfd027fbee7 : SUCCESS in 3m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ec2324c3b348478b94a096664169bd09 : SUCCESS in 33m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0292ff0afb7843c6b54a654a5df279db : SUCCESS in 38m 14s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"0f0a53652fb4f86dfba8d9115b9d2eb1cae7380e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-11-08 22:03:56.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"c6aaea59b6b5d2e5d8137ee2f2cbf6ba5f95533b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 23:33:01.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a0d58cbb81c497f8b8f8eca33527a53\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d8cb2c31940a462bbde430e0cf575367 : SUCCESS in 24m 41s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b9c55fa645bf45bea23910e8e2ed6a45 : SUCCESS in 1h 06m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/09e97f24441c4e228f587555b0dc63ca : SUCCESS in 18m 31s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e0290b18f5cd4c798668b9968e61a178 : SUCCESS in 58m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1ede0fd4ad0b4da48b623cb6fb9c2cef : SUCCESS in 51m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26d6e7e75fc34433b5348fe1d236f4bd : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e72760fcbee4cca8f4ed9725757361d : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e0302789487460c88ad01c58b58bb03 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0be591505b114bffa36e8c47c5122830 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8d85c024bfb46a0807714236bfe59a3 : SUCCESS in 5m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5d9e841a4ccc4af39d43b9d7423adcbe : SUCCESS in 24m 07s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/3e51fc1fd5bf44fdb6d2bbd84c6d39c7 : SUCCESS in 1h 26m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/af955aab85df4ad1bf9cc558654a7c6a : SUCCESS in 17m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fddd20550c284137886fff5250ef3b93 : SUCCESS in 35m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ed66ecc0c6624a059bcfc7a54afd1a5b : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":25},{"id":"f546fea979eae58c8f2f061d22a7099773e6f707","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 23:35:46.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d63609f1d2e4f61906aff8915cd344d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fa356d17113842d7a3e4aa6528fda609 : SUCCESS in 33m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fb85b408b19f415aaa0ea4ba6c5d8d70 : SUCCESS in 31m 05s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d53196544c7c4870957f9c2eaa397047 : SUCCESS in 1h 21m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b6d02b2433740a7b773d355d5ce9fcc : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fafd0e0464c047d5b1b36548225edd9f : SUCCESS in 3m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/01bb43098c6e4e1b96f7aaf2c92a23f0 : SUCCESS in 32m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/91a949bc52ec4b59bdad0a978c649106 : SUCCESS in 32m 00s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"a86ea863429dc3417780fe1c61adc0a6eb4dcb74","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-11-16 11:21:37.000000000","message":"Patch Set 25: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"2228bbb454064de3c4c37bad9834fa918cd103b7","author":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"date":"2022-11-21 09:16:55.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"8b315865d97cf857f71e67153a0354cf536fc59a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-27 09:37:33.000000000","message":"Patch Set 25: Code-Review+2","accounts_in_message":[],"_revision_number":25},{"id":"b49245805d0a5f172d0bc13bd6d7a10726d123fd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-02-27 09:40:16.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"41ce8ccb25e000a3616f376fc7b3d62b32b7c8cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 09:40:44.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"6b41ff2fd3fd8c9cd8033d6d379f9f3ef0ba5e46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:57:27.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcd69b6413734576a2a7c0e8dc33364e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6430b187cda44a79b32e07d1694f392a : SUCCESS in 15m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ff014c8901f24ecf8de658e1fd76f42b : SUCCESS in 1h 00m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2478054b830640b998902f9c39cb6675 : SUCCESS in 1h 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd1e5419736e4a708a793e0ec8c9c8ad : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f0549ef773e44368365a9ffffd06c78 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bae8b5e13564c968c3e2401c7b59ba2 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25a137d81a1492ba8fab504f7135be4 : SUCCESS in 5m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f9bf92d63a08434cbc87c023e480ff7d : SUCCESS in 19m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8eb75ba6ce2b4c60b1bafb0645a98afc : SUCCESS in 1h 11m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/03cf1793bd2e4186a80f9e2791e76afe : SUCCESS in 1h 18m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6e22a262543044ffbc8aebc8de7dc75a : SUCCESS in 15m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6c5bbfd5a3ba4c77add2564f16608375 : SUCCESS in 49m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/257aea9c5d7e4e3f97f135b7e2305137 : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":25},{"id":"bcb51324d45b135f7dacdb53dc62379c70e5e443","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:57:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":25},{"id":"6056f728f3a9a6980ab51347a07b2a797a9fe047","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 12:58:40.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/964cc5d40e4041c7b6fe0a0e59d479bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d37b2cd45c14b33b2cb12c4487aeb22 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"a2854da1b3e12987f7fa7590fca3bb74ac19c53e","revisions":{"336c921374c7c7376130c1497a7a0c22087cfb00":{"kind":"REWORK","_number":1,"created":"2022-02-10 14:03:16.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/16/828716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/1"}}},"commit":{"parents":[{"commit":"f7ef6c243af0d648e2d32254c52d20e9d0343ed1","subject":"Merge \"Clean MariaDB/MySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7ef6c243af0d648e2d32254c52d20e9d0343ed1"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-10 13:37:33.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-10 13:37:33.000000000","tz":60},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/336c921374c7c7376130c1497a7a0c22087cfb00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/336c921374c7c7376130c1497a7a0c22087cfb00"}]},"branch":"refs/heads/master"},"99c9f8c14a191c34f1557213774a037761a64742":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-14 14:36:19.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/16/828716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/2"}}},"commit":{"parents":[{"commit":"333fb35c79d2912a10b70001f5d8b5724e9b8651","subject":"Merge \"rabbitmq: update to 3.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333fb35c79d2912a10b70001f5d8b5724e9b8651"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-10 13:37:33.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-14 14:36:19.000000000","tz":0},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99c9f8c14a191c34f1557213774a037761a64742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99c9f8c14a191c34f1557213774a037761a64742"}]},"branch":"refs/heads/master","description":"Rebase"},"29dcb719dd827d77fb79db0c078086701b63f438":{"kind":"REWORK","_number":3,"created":"2022-03-08 11:38:43.000000000","uploader":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"ref":"refs/changes/16/828716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/3"}}},"commit":{"parents":[{"commit":"333fb35c79d2912a10b70001f5d8b5724e9b8651","subject":"Merge \"rabbitmq: update to 3.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333fb35c79d2912a10b70001f5d8b5724e9b8651"}]}],"author":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-02-10 13:37:33.000000000","tz":60},"committer":{"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","date":"2022-03-08 11:38:24.000000000","tz":60},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29dcb719dd827d77fb79db0c078086701b63f438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29dcb719dd827d77fb79db0c078086701b63f438"}]},"branch":"refs/heads/master"},"e0af8472acecd4bf53c0acf9730ebd0c281fff2b":{"kind":"REWORK","_number":4,"created":"2022-06-15 08:52:20.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/4"}}},"commit":{"parents":[{"commit":"09f6999cdcbae9f773cb7a7d84ece3d50becf2ad","subject":"Merge \"Add documentation to create monthly stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09f6999cdcbae9f773cb7a7d84ece3d50becf2ad"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0af8472acecd4bf53c0acf9730ebd0c281fff2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0af8472acecd4bf53c0acf9730ebd0c281fff2b"}]},"branch":"refs/heads/master"},"aaca359d0ff6d0b4cdcacd84b35f3da14ae6a402":{"kind":"REWORK","_number":5,"created":"2022-06-15 10:51:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/5"}}},"commit":{"parents":[{"commit":"09f6999cdcbae9f773cb7a7d84ece3d50becf2ad","subject":"Merge \"Add documentation to create monthly stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09f6999cdcbae9f773cb7a7d84ece3d50becf2ad"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 10:50:58.000000000","tz":120},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aaca359d0ff6d0b4cdcacd84b35f3da14ae6a402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aaca359d0ff6d0b4cdcacd84b35f3da14ae6a402"}]},"branch":"refs/heads/master"},"864c92c53158ef641253c50d92ecd193625b7b20":{"kind":"REWORK","_number":6,"created":"2022-06-15 11:02:53.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/6"}}},"commit":{"parents":[{"commit":"09f6999cdcbae9f773cb7a7d84ece3d50becf2ad","subject":"Merge \"Add documentation to create monthly stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09f6999cdcbae9f773cb7a7d84ece3d50becf2ad"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 11:02:45.000000000","tz":120},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/864c92c53158ef641253c50d92ecd193625b7b20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/864c92c53158ef641253c50d92ecd193625b7b20"}]},"branch":"refs/heads/master"},"2ad8035a6bd1e66ca482ff68f3c19582f5a9a42e":{"kind":"REWORK","_number":7,"created":"2022-06-20 07:48:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/7"}}},"commit":{"parents":[{"commit":"e8083e1def34294aa455c4d820ca7e9c9ec44fc9","subject":"Merge \"Switch OPENSTACK_RELEASE back to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8083e1def34294aa455c4d820ca7e9c9ec44fc9"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-20 07:48:18.000000000","tz":120},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ad8035a6bd1e66ca482ff68f3c19582f5a9a42e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2ad8035a6bd1e66ca482ff68f3c19582f5a9a42e"}]},"branch":"refs/heads/master"},"a206bc86bb26f8924f5f8021d3e9e3920125823a":{"kind":"REWORK","_number":8,"created":"2022-06-20 09:15:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/8"}}},"commit":{"parents":[{"commit":"e8083e1def34294aa455c4d820ca7e9c9ec44fc9","subject":"Merge \"Switch OPENSTACK_RELEASE back to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8083e1def34294aa455c4d820ca7e9c9ec44fc9"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-20 09:15:50.000000000","tz":0},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a206bc86bb26f8924f5f8021d3e9e3920125823a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a206bc86bb26f8924f5f8021d3e9e3920125823a"}]},"branch":"refs/heads/master"},"1c44cc2091ae7743fd57d9d611e674dcd0200cfa":{"kind":"REWORK","_number":9,"created":"2022-06-21 12:46:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/9"}}},"commit":{"parents":[{"commit":"9931f839136c28cd8cdd7a0e649b0efe1eb561ea","subject":"disable monasca for AArch64 on all distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9931f839136c28cd8cdd7a0e649b0efe1eb561ea"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-21 12:46:00.000000000","tz":120},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c44cc2091ae7743fd57d9d611e674dcd0200cfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c44cc2091ae7743fd57d9d611e674dcd0200cfa"}]},"branch":"refs/heads/master"},"351e81c96a673657c19f7daa2cf87939993caf83":{"kind":"REWORK","_number":10,"created":"2022-07-21 07:31:05.000000000","uploader":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"ref":"refs/changes/16/828716/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/10"}}},"commit":{"parents":[{"commit":"9931f839136c28cd8cdd7a0e649b0efe1eb561ea","subject":"disable monasca for AArch64 on all distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9931f839136c28cd8cdd7a0e649b0efe1eb561ea"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","date":"2022-07-19 14:46:10.000000000","tz":0},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/351e81c96a673657c19f7daa2cf87939993caf83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/351e81c96a673657c19f7daa2cf87939993caf83"}]},"branch":"refs/heads/master"},"33d5b432f56fe613f0eec111191a45e952776a34":{"kind":"REWORK","_number":11,"created":"2022-07-21 07:54:32.000000000","uploader":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"ref":"refs/changes/16/828716/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/11"}}},"commit":{"parents":[{"commit":"ea5d95d3f26c5d95a8084d59233692ce9cdd65f3","subject":"Merge \"Change kolla_version LABEL to git sha-1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea5d95d3f26c5d95a8084d59233692ce9cdd65f3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","date":"2022-07-21 07:53:50.000000000","tz":0},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33d5b432f56fe613f0eec111191a45e952776a34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33d5b432f56fe613f0eec111191a45e952776a34"}]},"branch":"refs/heads/master"},"4bebff5a596ab5ede019d1a9f6a0775f8ddc4844":{"kind":"REWORK","_number":12,"created":"2022-09-23 14:36:50.000000000","uploader":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"ref":"refs/changes/16/828716/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/12"}}},"commit":{"parents":[{"commit":"0040c6d3f64b9d0a821697557204b35eee11ca0f","subject":"Add missing labels blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0040c6d3f64b9d0a821697557204b35eee11ca0f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","date":"2022-09-23 14:36:23.000000000","tz":0},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4bebff5a596ab5ede019d1a9f6a0775f8ddc4844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4bebff5a596ab5ede019d1a9f6a0775f8ddc4844"}]},"branch":"refs/heads/master"},"cfc36319a4a5bc8a40fb6420c4cd0c7f1932965d":{"kind":"REWORK","_number":13,"created":"2022-09-27 11:42:21.000000000","uploader":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"ref":"refs/changes/16/828716/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/13"}}},"commit":{"parents":[{"commit":"542bbcdb6b2827812e529c29f4a44f542ec1f801","subject":"grafana: ignore SHA1 gpg key on CentOS Stream 9 (and family)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/542bbcdb6b2827812e529c29f4a44f542ec1f801"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","date":"2022-09-27 11:41:43.000000000","tz":0},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfc36319a4a5bc8a40fb6420c4cd0c7f1932965d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfc36319a4a5bc8a40fb6420c4cd0c7f1932965d"}]},"branch":"refs/heads/master"},"95d42eb03ba9adf7e7c839a06e94ce9f08ff13e2":{"kind":"REWORK","_number":14,"created":"2022-09-27 11:50:24.000000000","uploader":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"ref":"refs/changes/16/828716/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/14"}}},"commit":{"parents":[{"commit":"542bbcdb6b2827812e529c29f4a44f542ec1f801","subject":"grafana: ignore SHA1 gpg key on CentOS Stream 9 (and family)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/542bbcdb6b2827812e529c29f4a44f542ec1f801"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","date":"2022-09-27 11:49:57.000000000","tz":0},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95d42eb03ba9adf7e7c839a06e94ce9f08ff13e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95d42eb03ba9adf7e7c839a06e94ce9f08ff13e2"}]},"branch":"refs/heads/master"},"4a99e80d2d52dcc443632276767b71646adb9682":{"kind":"REWORK","_number":15,"created":"2022-09-27 12:58:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/15"}}},"commit":{"parents":[{"commit":"542bbcdb6b2827812e529c29f4a44f542ec1f801","subject":"grafana: ignore SHA1 gpg key on CentOS Stream 9 (and family)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/542bbcdb6b2827812e529c29f4a44f542ec1f801"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 12:58:51.000000000","tz":0},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a99e80d2d52dcc443632276767b71646adb9682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a99e80d2d52dcc443632276767b71646adb9682"}]},"branch":"refs/heads/master"},"4fc4b767971a9848569938fb5485188d93d7b2c3":{"kind":"REWORK","_number":16,"created":"2022-10-04 08:59:59.000000000","uploader":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"ref":"refs/changes/16/828716/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/16"}}},"commit":{"parents":[{"commit":"a9b0b63fedb5140ea6381c509c4dabde9772d5c2","subject":"Merge \"proxysql: Use almalinux for centos/rocky9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9b0b63fedb5140ea6381c509c4dabde9772d5c2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-15 08:50:16.000000000","tz":120},"committer":{"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","date":"2022-10-04 08:59:32.000000000","tz":0},"subject":"Refactor image module","message":"Refactor image module\n\nRefactor image module: divide main file kolla/image/build.py\ninto multiple smaller ones to divide responsibilities\nand make navigation in files easier.\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in\nall other modules.\n\nSigned-off-by: Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fc4b767971a9848569938fb5485188d93d7b2c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fc4b767971a9848569938fb5485188d93d7b2c3"}]},"branch":"refs/heads/master"},"9a3a1b90e7287c6da5049cc3783fe5e6a61f8e34":{"kind":"REWORK","_number":17,"created":"2022-10-06 10:21:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/17"}}},"commit":{"parents":[{"commit":"caf1be72dff56987bc30d801892a6cfe3976d00a","subject":"Move KollaWorker stuff to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/caf1be72dff56987bc30d801892a6cfe3976d00a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:40:23.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 10:20:50.000000000","tz":120},"subject":"Add a way to add other container engines","message":"Add a way to add other container engines\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other\nmodules.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a3a1b90e7287c6da5049cc3783fe5e6a61f8e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a3a1b90e7287c6da5049cc3783fe5e6a61f8e34"}]},"branch":"refs/heads/master"},"8678d80f019856862769b68f049adf8cde8202a9":{"kind":"REWORK","_number":18,"created":"2022-10-06 11:14:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/18"}}},"commit":{"parents":[{"commit":"2af88e979123085cda59e3cc0432c0d6308bc3b3","subject":"Move KollaWorker stuff to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2af88e979123085cda59e3cc0432c0d6308bc3b3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:40:23.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 11:14:17.000000000","tz":120},"subject":"Add a way to add other container engines","message":"Add a way to add other container engines\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other\nmodules.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8678d80f019856862769b68f049adf8cde8202a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8678d80f019856862769b68f049adf8cde8202a9"}]},"branch":"refs/heads/master"},"6fe7342a28fdf1e74a8a8cc3ba64f067a085fa80":{"kind":"REWORK","_number":19,"created":"2022-10-06 13:03:28.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/19"}}},"commit":{"parents":[{"commit":"2af88e979123085cda59e3cc0432c0d6308bc3b3","subject":"Move KollaWorker stuff to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2af88e979123085cda59e3cc0432c0d6308bc3b3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:40:23.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 13:03:01.000000000","tz":0},"subject":"Add a way to add other container engines","message":"Add a way to add other container engines\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other\nmodules.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fe7342a28fdf1e74a8a8cc3ba64f067a085fa80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fe7342a28fdf1e74a8a8cc3ba64f067a085fa80"}]},"branch":"refs/heads/master"},"e2ee4c3ff4ed8caf207395b16114775e6211a960":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-10-06 13:06:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/828716/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/20"}}},"commit":{"parents":[{"commit":"145d096bf7d93e200cd567f8177f09fd8492de61","subject":"Move KollaWorker stuff to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/145d096bf7d93e200cd567f8177f09fd8492de61"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:40:23.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-10-06 13:06:10.000000000","tz":0},"subject":"Add a way to add other container engines","message":"Add a way to add other container engines\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other\nmodules.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2ee4c3ff4ed8caf207395b16114775e6211a960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2ee4c3ff4ed8caf207395b16114775e6211a960"}]},"branch":"refs/heads/master","description":"Rebase"},"9465ae7985956f0d8335b73b090fc59d498f4700":{"kind":"REWORK","_number":21,"created":"2022-10-06 16:27:01.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/21"}}},"commit":{"parents":[{"commit":"e2c061ee63a701284bc2f6b1470d9d33a58e724c","subject":"do not play with PROJECT_ROOT in build.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2c061ee63a701284bc2f6b1470d9d33a58e724c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:40:23.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 16:26:47.000000000","tz":120},"subject":"Add a way to add other container engines","message":"Add a way to add other container engines\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other\nmodules.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9465ae7985956f0d8335b73b090fc59d498f4700"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9465ae7985956f0d8335b73b090fc59d498f4700"}]},"branch":"refs/heads/master"},"0d1574ee91215d7a2d1b3a0d5233b9551c0892cd":{"kind":"REWORK","_number":22,"created":"2022-10-11 09:04:11.000000000","uploader":{"_account_id":35104,"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","username":"k.yarovoy"},"ref":"refs/changes/16/828716/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/22"}}},"commit":{"parents":[{"commit":"e2c061ee63a701284bc2f6b1470d9d33a58e724c","subject":"do not play with PROJECT_ROOT in build.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2c061ee63a701284bc2f6b1470d9d33a58e724c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:40:23.000000000","tz":120},"committer":{"name":"Konstantin Yarovoy","email":"k.yarovoy@partner.samsung.com","date":"2022-10-11 09:03:43.000000000","tz":0},"subject":"Add a way to add other container engines","message":"Add a way to add other container engines\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other\nmodules.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d1574ee91215d7a2d1b3a0d5233b9551c0892cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d1574ee91215d7a2d1b3a0d5233b9551c0892cd"}]},"branch":"refs/heads/master"},"f4e58f1e9b8a8cb0cd5c6fb64d8a7ba472472a98":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2022-10-11 09:16:18.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/23"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:40:23.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-11 09:16:18.000000000","tz":0},"subject":"Add a way to add other container engines","message":"Add a way to add other container engines\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other\nmodules.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4e58f1e9b8a8cb0cd5c6fb64d8a7ba472472a98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4e58f1e9b8a8cb0cd5c6fb64d8a7ba472472a98"}]},"branch":"refs/heads/master","description":"Rebase"},"b0857b5d6e60db05dc26348e5ebfca517b24ae0f":{"kind":"REWORK","_number":24,"created":"2022-10-11 09:39:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/828716/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/24"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:40:23.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-11 09:39:35.000000000","tz":120},"subject":"Add a way to add other container engines","message":"Add a way to add other container engines\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other\nmodules.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0857b5d6e60db05dc26348e5ebfca517b24ae0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0857b5d6e60db05dc26348e5ebfca517b24ae0f"}]},"branch":"refs/heads/master"},"a2854da1b3e12987f7fa7590fca3bb74ac19c53e":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2022-11-08 22:03:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/828716/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/828716/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/828716/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/828716/25"}}},"commit":{"parents":[{"commit":"e275875c2587696025cb6dd5851cb42b010c56ed","subject":"Merge \"Update support matrix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e275875c2587696025cb6dd5851cb42b010c56ed"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:40:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-11-08 22:03:56.000000000","tz":0},"subject":"Add a way to add other container engines","message":"Add a way to add other container engines\n\nAdd engine adapter module to allow additon of other container engines,\ncreate one common EngineClient that would be called in all other\nmodules.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: Ice6467086bd292af086322afc3fc4e869d89eefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2854da1b3e12987f7fa7590fca3bb74ac19c53e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2854da1b3e12987f7fa7590fca3bb74ac19c53e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
