)]}'
{"id":"openstack%2Fkolla~829036","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I86aa697b8def34e0cfb11d2f5ad903dd54e975ef","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-15 08:00:09.000000000","reason":"Michal Nasiadka replied on the change"}},"hashtags":[],"change_id":"I86aa697b8def34e0cfb11d2f5ad903dd54e975ef","subject":"base: Drop usage of Ceph Nautilus from RDO","status":"MERGED","created":"2022-02-14 17:52:18.000000000","updated":"2022-02-15 09:40:46.000000000","submitted":"2022-02-15 09:38:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829036","cherry_pick_of_change":828228,"cherry_pick_of_patch_set":4,"meta_rev_id":"574ada03472b9ff140ace1d46c318ec2775297e7","_number":829036,"virtual_id_number":829036,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 09:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-15 08:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-14 18:14:18.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-15 08:00:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-14 18:14:18.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-14 19:31:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d7790a80d75b3102e4f6ebb5357a7242911c98c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 17:52:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"33925ff277a6f83a21e457c0274c2163578aed07","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 18:01:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"762725ceaa9634ab35426f08dc0e423152764d9b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-14 18:14:18.000000000","message":"Patch Set 2: Code-Review+2\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"225cf04ddb969f6d5a8a08021535fa6baef513c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 19:31:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4b73ecf58d3e43739ed537417d10cbf3 : SUCCESS in 31m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dfad8ff013904530a8bea4ce2c31f749 : SUCCESS in 40m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d3a3498758c64f3f9ac05c8628582544 : SUCCESS in 1h 14m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7038ba1a0931431b8a91d96d84856f89 : SUCCESS in 1h 16m 59s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3893dfec991941c88a0ed7158a32f4cb : SUCCESS in 1h 17m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/824689cff3424e8eb76c0cd8cd88f208 : SUCCESS in 15m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a08d4bed6a874d5eb60e896613930131 : SUCCESS in 21m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5473c22d601543498c0741530c950d23 : SUCCESS in 47m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a827075e2ff64b2e8f265efdc8b6280f : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1195dc649ea948f18ee62886bf6a708a : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61438aa59a99476c9fe022754a7cbcf4 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0a285dd8df4b319a00bf373c04c867 : SUCCESS in 6m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2d828e37edf640fb8fa94e591ee1c191 : SUCCESS in 23m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e13e3021f9b04ab18208eef3435f79e3 : SUCCESS in 18m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/87f91c4397504b7cbc45032f49d57752 : SUCCESS in 45m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d116af44177840879aad67a43be28d3e : SUCCESS in 40m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cf9f2163970e46c5b4120fb4155f562f : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":2},{"id":"abb989304682e21b16c87d9d34c01055e3decfd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 21:19:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8c1b5fd617574466bfeccf46a99080a9 : FAILURE in 2h 42m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e1a7738c28a4fb381ed7e4521fd1c68 : SUCCESS in 2h 21m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef8d1e36e50047918c3b6769d9edd62a : TIMED_OUT in 3h 03m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf0faaf51b414d878fc1a6124dcb4fad : SUCCESS in 2h 31m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d3013da625707fc7c1859f714a1721421f11d9e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 08:00:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a0cf879956fc3ef3c22f41831a7bdf513f8ba905","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 08:01:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"add37653e8b268230d8c498a97531abdfa0a0448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 09:38:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/29976911742149b6965b0cbae45175b2 : SUCCESS in 29m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b93c931c6408450480cb09f7c4755890 : SUCCESS in 1h 05m 36s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/443bf5d5bbfd4ba7b21fbfd0ae96b07b : SUCCESS in 1h 31m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39a872806224fe0a44c81c7556956a9 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2953e1c412c54c62b53624edb9fec935 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fbec9192fb4472582d01991f94a5ccf : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ceccc5a4bec44a8897cf357ac6cd14e : SUCCESS in 7m 10s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/93306327a015433a9876904868a42b5b : SUCCESS in 27m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/50d941d8631149119d9a19a3275d70f8 : SUCCESS in 58m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/89b0a632e781472f8ee3fa4a016cb8be : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d787b59763e64d76097e70dd425146c8b48768f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 09:38:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"574ada03472b9ff140ace1d46c318ec2775297e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 09:40:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28579de2b0c14037a59e8ddd78108673 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"07a4e434d3c04108e49a8185a3803298d53e1367","revisions":{"fd63571039bcdf10d601e1c521fb5d5761d95200":{"kind":"REWORK","_number":1,"created":"2022-02-14 17:52:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/829036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/829036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/829036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/829036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/829036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/829036/1"}}},"commit":{"parents":[{"commit":"77d683caaaed46149d2323b5a94c0e73bef3241c","subject":"ironic: Fix UEFI \u0026 iPXE bootloader filenames","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77d683caaaed46149d2323b5a94c0e73bef3241c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 08:11:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 17:52:18.000000000","tz":0},"subject":"base: Drop usage of Ceph Nautilus from RDO","message":"base: Drop usage of Ceph Nautilus from RDO\n\nThis was temporary due to missing Ceph Nautilus\nrepo on CentOS mirrors (after removing CentOS Linux\ncontent on EOL date).\n\nChange-Id: I86aa697b8def34e0cfb11d2f5ad903dd54e975ef\n(cherry picked from commit 7f38bce812ccb621a089755526f79bd792f4e071)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd63571039bcdf10d601e1c521fb5d5761d95200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd63571039bcdf10d601e1c521fb5d5761d95200"}]},"branch":"refs/heads/stable/victoria"},"07a4e434d3c04108e49a8185a3803298d53e1367":{"kind":"REWORK","_number":2,"created":"2022-02-14 18:01:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/829036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/829036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/829036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/829036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/829036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/829036/2"}}},"commit":{"parents":[{"commit":"77d683caaaed46149d2323b5a94c0e73bef3241c","subject":"ironic: Fix UEFI \u0026 iPXE bootloader filenames","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77d683caaaed46149d2323b5a94c0e73bef3241c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 08:11:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 18:01:14.000000000","tz":0},"subject":"base: Drop usage of Ceph Nautilus from RDO","message":"base: Drop usage of Ceph Nautilus from RDO\n\nThis was temporary due to missing Ceph Nautilus\nrepo on CentOS mirrors (after removing CentOS Linux\ncontent on EOL date).\n\nChange-Id: I86aa697b8def34e0cfb11d2f5ad903dd54e975ef\n(cherry picked from commit 7f38bce812ccb621a089755526f79bd792f4e071)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a4e434d3c04108e49a8185a3803298d53e1367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a4e434d3c04108e49a8185a3803298d53e1367"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
