)]}'
{"id":"openstack%2Fkolla~829054","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I75373eb21c4269c20f72603ad4cf162a472de25b","project":"openstack/kolla","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-03-01 11:28:52.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-02-28 11:09:19.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I75373eb21c4269c20f72603ad4cf162a472de25b","subject":"Build CentOS Stream 8 images","status":"MERGED","created":"2022-02-14 14:22:12.000000000","updated":"2022-03-01 11:36:50.000000000","submitted":"2022-03-01 11:28:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829054","meta_rev_id":"a244bb0275e7ceae0515a291ddabec7f338f813e","_number":829054,"virtual_id_number":829054,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-01 11:28:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-03-01 10:17:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-02-28 12:14:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-03-01 10:17:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-14 14:22:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-14 14:22:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-28 10:46:35.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-02-28 11:01:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-02-28 11:09:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-28 11:15:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"ad22364c01a744147df3c252ae777eb5e574e841","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 14:22:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4504e0c16a19736394e8e3a7732d57fea64dea84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 14:22:39.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/ussuri.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/ussuri\", line 2, column 10","accounts_in_message":[],"_revision_number":1},{"id":"4b2084bfe0d8898d363401b87ad593917e950b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 14:22:56.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/ussuri.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/ussuri\", line 2, column 10","accounts_in_message":[],"_revision_number":1},{"id":"b86e6099a19af171ce2b2633417b82c66f5e6e77","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 14:30:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d86cd8e6b40d054d986ae574e834f97098d68a94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 14:31:20.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/ussuri.  The error was:\n\n  Job kolla-ansible-centos8-source not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-centos8s-source\n          - kolla-ansible-centos8-source\n      check-arm64:\n        jobs:\n          - kolla-build-centos8s-source-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8s-source\n      periodic:\n        jobs:\n          - kolla-publish-centos8s-source-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8s-source-dockerhub\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/ussuri\", line 2, column 3","accounts_in_message":[],"_revision_number":2},{"id":"3c4a8fcf868cbda6ec134b0e6e9ecdf061d22e76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 14:31:50.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/ussuri.  The error was:\n\n  Job kolla-ansible-centos8-source not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-centos8s-source\n          - kolla-ansible-centos8-source\n      check-arm64:\n        jobs:\n          - kolla-build-centos8s-source-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8s-source\n      periodic:\n        jobs:\n          - kolla-publish-centos8s-source-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8s-source-dockerhub\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/ussuri\", line 2, column 3","accounts_in_message":[],"_revision_number":2},{"id":"1f7f37c1a03c3572dbc33befdf9ba7c4111433da","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 14:40:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6c46f0da4baa459f9c6783ab51bac1b9cecd364c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:00:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7613f6eaecf146da94f5f887ee55dd0a : FAILURE in 38m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f9e3a7d6573f4d86829bdb886c5efe9a : SUCCESS in 21m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/780f6e36a14f4f8d9f055292921a1bcc : SUCCESS in 23m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bc3638e213f64485969201a7bb55c379 : SUCCESS in 53m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e164cd86351e474f964c17993d78c126 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37af81b437a04f5eab8e7d21c66cdb59 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1393f77c9b2432692c300c0f7deb220 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9000f762e08b465eb2a050b599d29781 : SUCCESS in 6m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bd574512604f4098b255485419f21d22 : SUCCESS in 26m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4734bff1270b4b35bfbd92315db2bef1 : SUCCESS in 37m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc4d1f67e3774627a98a7bda386f2c75 : SUCCESS in 56m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/300bdd7f3ec84b94a4d034b393cf92c5 : SUCCESS in 1h 01m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f011a6991cb742dc948f9a1692d99475 : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":3},{"id":"b24312c3c9b389eeb73b418edbfa1aed1fb6de17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 18:03:27.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8772190d7825bddd915f5f56c134ef7a91496b2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 19:12:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0c846a18cfb84b99bb65d37976c61ec1 : FAILURE in 48m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/34dec1981096455395253e62b34be28a : SUCCESS in 22m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4828dbd2317145039670e115a083303d : SUCCESS in 31m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4800040d23e14218bffb6a95697f2abd : SUCCESS in 58m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f11b36ea022a480cbc839b169060037a : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06f805e11f97480ca5dd71aa4a927a38 : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcee040367724edaa12895b6fa742fef : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b086768173f84242874f3af45863ad96 : SUCCESS in 8m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/06c57c13d96d4743aecf53bf49969b1f : SUCCESS in 32m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/67f4c7a81604421fb70b536e1b18f135 : SUCCESS in 28m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a9a802e6c234fe18a4c1e7c6aacd307 : SUCCESS in 53m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d6b054141454056853ee6e3d7988286 : SUCCESS in 54m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9a213d2f6a374243beb3427cf12d2bca : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":4},{"id":"30ff6b71f34057914b35829747b78ba6bb8738ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 00:06:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a21d784f27274789928e26c8f1e5aeab : TIMED_OUT in 3h 05m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/62819aab35394015b87726e4f1a300b6 : SUCCESS in 2h 25m 49s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8bd52e2334e9491dbc3b4ebfbc3f0a42 : TIMED_OUT in 3h 02m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/31a151ef633649cb819a051d78b1bb75 : SUCCESS in 2h 01m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c0b04c2b3d8e8acb9ccbf6fb0f5832b9054bfbc7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 06:17:34.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"bbc3292ccbef99d54eaa51000c85d32090318ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 07:32:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/77579aad601343b58c57237469d8dc95 : SUCCESS in 41m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/693956a076a2458caab788a69c5c48e4 : SUCCESS in 24m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a778cfc0c6ad429f97b834e5c4b30b6d : SUCCESS in 19m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/44ae1af9cb1445fbbccd3decb6a25d7b : SUCCESS in 53m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc65a1a312e468babbc776cd1f90833 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8fa74581c0e40f6946b373a8f0819de : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c89f9b0aa9a345d5ae5c8d36b95310f9 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5785944c509c43229d6b9527710777fe : SUCCESS in 5m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9c0db2f9f414497a49bca75d8bea743 : SUCCESS in 20m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/17dbf43cd49041f8ac0798b5f6300fed : SUCCESS in 33m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b144b7f5685e4b1bb6dc7639f93127f9 : SUCCESS in 1h 11m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ac6777c3794d45d3bd1188b41089488c : SUCCESS in 1h 10m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a653a88746504159833bf5c7e4232a2b : SUCCESS in 1h 13m 30s","accounts_in_message":[],"_revision_number":5},{"id":"fa23f1f52a7c82d19bd67bf04b44c64ae8e57155","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 09:14:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/20522648923c4de7bd5a9fbc81e24927 : FAILURE in 2h 31m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f40a37b5c54458fa7a03e5a552ffd0e : SUCCESS in 2h 25m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4038b35f919f4b0c9e4e102fb44f6853 : SUCCESS in 2h 54m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5dcd04591b914c07b2b303800c7ced08 : SUCCESS in 1h 48m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d575f32ce14a4985bbed7be2fa89382e7161e9b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 12:16:41.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"53bd533a9947475b420ccdf03b6ca2b9a99528de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 13:09:09.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"b2d21ac06131a73d88368f071fa4c26412e1b1dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 14:27:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/794dd9c1c1b14bc4a3edc99da7d81b1e : SUCCESS in 1h 12m 19s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/56e653a85b5547788ed46741452ade94 : SUCCESS in 31m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dae8f80698ba413bb8ca453dd8eb4e2f : SUCCESS in 22m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/72665197ca8f4115ada97eaabe082763 : SUCCESS in 14m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bbdede27b88c42328860324c129629c3 : SUCCESS in 55m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/135be846aa364adfb7d4edc21436e238 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6966e9cca6934a30893e5a1c99c72cce : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/414aa486fb7343909c6a86a1ca02a01c : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/928ff87ae71f46f887b02f415f554cbb : SUCCESS in 6m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/59160277551e4bcb838920213e38d5af : SUCCESS in 21m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/79a3537df2b546678b43ce663e91ca42 : SUCCESS in 30m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/43a32af0538f4f7ca2fa7e7c9424e9f3 : SUCCESS in 49m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3f100a58c9dc4411936e4d5b6c783bdc : SUCCESS in 47m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/527e8f29c53b4341b5beb913cde9b797 : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":7},{"id":"44fe52b0fedfc49ec2924837186ff01929836c9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 15:49:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e79fe3e6900b4d35b180e926b88af464 : FAILURE in 2h 34m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5569ced4b56e4f6c88abd1e6e4897cfb : SUCCESS in 34m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/71d6f388ca6f4143b2a1239870b61d46 : SUCCESS in 1h 24m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb1473faf616457d857d399033637d9b : SUCCESS in 37m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"51e241eea630fc212fee5a8913e763fa31bce439","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:04:12.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"5c4899dac01c5dc2943bd0d97b1da31735204da1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:28:39.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"4dd0d7d7663fc6c6816513a3093fd59d7eab7bae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:35:32.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"204fe3cdb8edb1fa9d324d5d413cc6a704ad8c79","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:37:56.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"c664ecf656d7b37c1c71b2514ab1faf2960efe53","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:38:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"411ec825a7bee61efa397f963cdbc6239482e443","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:40:28.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6e07114f3c690e84f548a4ef66ce5909ad694869","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 21:43:22.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"30e38b59bba8d4db1157cd46bbecb5905400dd25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 21:43:27.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"fa8ed9bc4a36f0a8202102d20f2def436d0cdf69","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:44:08.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a56387b3f9133cf634d1c5d11f91a8dc6075f62e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:56:31.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"60bd806d09321725cb181d1e9815f547165f0f11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 23:02:50.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b9b73470dc4149eaa37408d1877bba9b : SUCCESS in 1h 04m 40s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/210d3c14de4e4d4b9ab5139cd85f7672 : SUCCESS in 31m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/80ab9071e93a474db6305dd9f021b059 : SUCCESS in 31m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2467ba9fa2254fe481b7eb1ec08af887 : SUCCESS in 20m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6146868b0b0a4200bcf5ba67dd63af21 : SUCCESS in 1h 00m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f0ef337255a42f5b42570d6f71a88de : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c0471272c2041e687c86e3a6791dda6 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3ab03e0033f4f02abc56565fc41eec8 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fdea4fd4cfd47de98d4f6564fcf8e3a : SUCCESS in 5m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d99558b9288949959fc284d5308483ae : SUCCESS in 25m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b9cc61ba33a4ef1a3234ab3bf33b346 : SUCCESS in 24m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/345abb1d74cb48af9f0588b22b050901 : SUCCESS in 56m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/147be5e1978b4429b1151fd8371e9a00 : SUCCESS in 54m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/70241225902748a496c960e801388279 : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":12},{"id":"10408c2c6cbb7717ad819a257973b0f10632a1f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 03:04:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b264da522a724ab3bdf08c033e67b477 : TIMED_OUT in 3h 07m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/edd97f983e1d47089c42e7896a46762c : SUCCESS in 2h 34m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/37f248db5c0d43bd967f49bee827daff : SUCCESS in 2h 49m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ffccbe0b089a403bbc46027f8dd089e5 : SUCCESS in 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b4d82906eed5eec718e2047c5d0aca6bf0f8af25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 13:44:18.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"88663b3c9886e063a8fec9d33386d1e1bfadd99b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 15:07:44.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/babfc26a448c4a818daa10fadb031754 : SUCCESS in 1h 05m 24s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2fd39579945944e29685b8780cf45596 : SUCCESS in 32m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/50c802ba33f44add966a462153db3168 : SUCCESS in 20m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/424d11bfe4df4fbabfdf9e5e2bcb7bf8 : SUCCESS in 21m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9c7a61699e894e4d84998198cca79e31 : SUCCESS in 1h 09m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/048ff1c99b3b401ebc3ad7c0c594e84c : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc43ad9d6c0f4d9a92a0e897a28df728 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c42caea7ec404224957469f8976152bd : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ab4163b0e245b48106fc9439a1c3bc : SUCCESS in 5m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c904b3ae4f24cc182513aa99550d954 : SUCCESS in 24m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b56a51cf6464538a900445d618436dd : SUCCESS in 22m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ae5909c545246d5abd4d5899fc7a959 : SUCCESS in 55m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99267da3cbbf469f8b6b4b2aac0fea49 : SUCCESS in 42m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/08fb232e0db043c8b433080bac6e785c : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":13},{"id":"79ca62d659fb26d6ff3f927d82bccc7cf65b2166","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:38:40.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7bf5a70144cf4ca1bc6d18209ef892ee : SUCCESS in 39m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b99ba53537d244ba90afa29143ea72fc : SUCCESS in 2h 09m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e62b6294907a4b4eb96ae01d0cd17e41 : SUCCESS in 2h 43m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/22a38f70617842a2b7160eec89800c7c : SUCCESS in 1h 52m 17s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3ac009ae15d1c64bd19a12869b453d251c6ea113","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-28 10:46:35.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"c2fac70936bb3ebeb002499083f9c7eef0bf8398","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-28 11:01:49.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"da735c6c6ca02d78bee9c59ab58ae133fc008f55","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-28 11:06:21.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"2b7fcfffda63c492d25ef62f126b3eb025821fd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-28 11:06:41.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"968a7fb3dc03ba5c6c5d1b02aef7fe4f20a9877a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-28 11:09:19.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"9b1e4896e698ed9367ff8f3eb59b2a52571eaffc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-28 11:15:21.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"96a71a3bfa59d5fea842ff35da9620e71f41b6d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-28 11:56:46.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"568c24fb917f6e380f9dc7a2589556960b7eb529","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-28 12:14:22.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"0ddcd3ba0ecc86dc1f0f6416c73ac1b2c09c13b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 13:11:21.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/33930bfa62d943e8b4cbd06f58aae5f1 : SUCCESS in 1h 11m 48s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b4ffd8f1b92d4eb3bed649eec9708734 : SUCCESS in 37m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ad6cdc455e224e058674dc220cd13e5a : SUCCESS in 25m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/be605dc549704570b7e840749b192972 : SUCCESS in 19m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f195a6159308409db75bc018d74a3c54 : SUCCESS in 58m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2384ec337ac4f84afc9565b9fe31fdb : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44300bb3b33d44acaa8e34444d3dc03d : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0be11ac0ed914d6db5a6ad413b7076f4 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a2db9a1054493c80f11ab5adcb54aa : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6650f0de63ca47f0bb6f9e29b2d9544b : SUCCESS in 5m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f7b9f2794c684e0f9f0393e84e021aac : SUCCESS in 23m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8ea5b59d4c4466ca6812b242824909e : SUCCESS in 29m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b6bc6a63890407d809ec69718fecd9c : SUCCESS in 54m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aa7dbd6b3a384c2ea017d53480f8b8cf : SUCCESS in 55m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/38179f8043e9442bb2c9622298608604 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":15},{"id":"d2236438b967e6bf4f0466861c45d9902b1f6901","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 18:41:42.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ee36d98555314245a92fde1955077a7e : SUCCESS in 2h 57m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d9b02c8f7d544ef1988d2222e26e333c : SUCCESS in 2h 04m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8abd42a55d11404e908a3648de13c1d6 : SUCCESS in 2h 28m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3ab3560732949d88cf976d223bc1a66 : TIMED_OUT in 3h 36m 37s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ad0a52d768522d14b0cf5d786f65540c5b915ea3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-01 10:17:09.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"9daec7b2a46368eb412193c0366a7aeff6776f38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 10:17:50.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"f6aac42fc29cc722c169195cb689ce93951b7486","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 11:28:52.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/59be4330c4c1439480eade56a03a8fa7 : SUCCESS in 1h 09m 24s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0c39943757a940549a7f46c38020b23a : SUCCESS in 32m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/824aa3e4bcd948d3a261ef83bcdd2af0 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb0827b81c524f3d9e00e91671f3fb5d : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/669939d3a4d341d6bb296b4f0594dd45 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff930e08dabc4271bcf5bbef55d045b0 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3185233a6cda4b21ab2a02f88dfd66fe : SUCCESS in 5m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d4243a677394b4e8c14cbde03b92c20 : SUCCESS in 30m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/62a164bee2714e21819b03d031f9f26a : SUCCESS in 46m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/125bc4c74e524a3fa09d53b03a644350 : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":15},{"id":"36785dc126071863bb6aa5143666048a9cf91b73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 11:28:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"a244bb0275e7ceae0515a291ddabec7f338f813e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 11:36:50.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa9e0f23866349df83419835a7110460 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39a5ddb291b6431d884e3ace48392f7e : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"67d6d28b6208cc004a50f19be1386d88c2404bc4","revisions":{"9c4dc17b067771184c94a0bf64af5ca18898cce5":{"kind":"REWORK","_number":1,"created":"2022-02-14 14:22:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/1"}}},"commit":{"parents":[{"commit":"7ba238516f9552f2b053d339cac75acaa2014794","subject":"Drop CentOS Linux 8 image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ba238516f9552f2b053d339cac75acaa2014794"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:22:09.000000000","tz":60},"subject":"WIP: Build c8s images","message":"WIP: Build c8s images\n\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c4dc17b067771184c94a0bf64af5ca18898cce5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c4dc17b067771184c94a0bf64af5ca18898cce5"}]},"branch":"refs/heads/stable/ussuri"},"0ed88941badab30291833debbb2336cd997c41c9":{"kind":"REWORK","_number":2,"created":"2022-02-14 14:30:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/2"}}},"commit":{"parents":[{"commit":"7ba238516f9552f2b053d339cac75acaa2014794","subject":"Drop CentOS Linux 8 image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ba238516f9552f2b053d339cac75acaa2014794"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:30:46.000000000","tz":60},"subject":"WIP: Build c8s images","message":"WIP: Build c8s images\n\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ed88941badab30291833debbb2336cd997c41c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ed88941badab30291833debbb2336cd997c41c9"}]},"branch":"refs/heads/stable/ussuri"},"7ce4bf4486d0bb0146954684c4e4cf5456f14e8b":{"kind":"REWORK","_number":3,"created":"2022-02-14 14:40:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/3"}}},"commit":{"parents":[{"commit":"7ba238516f9552f2b053d339cac75acaa2014794","subject":"Drop CentOS Linux 8 image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ba238516f9552f2b053d339cac75acaa2014794"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:40:14.000000000","tz":60},"subject":"WIP: Build c8s images","message":"WIP: Build c8s images\n\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ce4bf4486d0bb0146954684c4e4cf5456f14e8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ce4bf4486d0bb0146954684c4e4cf5456f14e8b"}]},"branch":"refs/heads/stable/ussuri"},"7bbd454d8f3ba5660cbfb20e71ac150ed4f9bf07":{"kind":"REWORK","_number":4,"created":"2022-02-14 18:03:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/4"}}},"commit":{"parents":[{"commit":"7ba238516f9552f2b053d339cac75acaa2014794","subject":"Drop CentOS Linux 8 image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ba238516f9552f2b053d339cac75acaa2014794"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 18:03:27.000000000","tz":0},"subject":"WIP: Build c8s images","message":"WIP: Build c8s images\n\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bbd454d8f3ba5660cbfb20e71ac150ed4f9bf07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bbd454d8f3ba5660cbfb20e71ac150ed4f9bf07"}]},"branch":"refs/heads/stable/ussuri"},"8c36bdba93e07351a78601c86f001405e3f92ec9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-02-15 06:17:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/5"}}},"commit":{"parents":[{"commit":"c5f2482fa55c54a107d9fbe493745fe191a15163","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5f2482fa55c54a107d9fbe493745fe191a15163"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 06:17:34.000000000","tz":0},"subject":"WIP: Build c8s images","message":"WIP: Build c8s images\n\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c36bdba93e07351a78601c86f001405e3f92ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c36bdba93e07351a78601c86f001405e3f92ec9"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"},"4340ea39b5f4beee6f458db91748d4e5a0d1d906":{"kind":"REWORK","_number":6,"created":"2022-02-15 12:16:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/6"}}},"commit":{"parents":[{"commit":"c5f2482fa55c54a107d9fbe493745fe191a15163","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5f2482fa55c54a107d9fbe493745fe191a15163"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 12:16:41.000000000","tz":0},"subject":"WIP: Build c8s images","message":"WIP: Build c8s images\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/829202\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4340ea39b5f4beee6f458db91748d4e5a0d1d906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4340ea39b5f4beee6f458db91748d4e5a0d1d906"}]},"branch":"refs/heads/stable/ussuri"},"12043a985d22103d56b8ca8a16030aa0176a7143":{"kind":"REWORK","_number":7,"created":"2022-02-15 13:09:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/7"}}},"commit":{"parents":[{"commit":"c5f2482fa55c54a107d9fbe493745fe191a15163","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5f2482fa55c54a107d9fbe493745fe191a15163"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 13:09:09.000000000","tz":0},"subject":"WIP: Build c8s images","message":"WIP: Build c8s images\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/829202\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12043a985d22103d56b8ca8a16030aa0176a7143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12043a985d22103d56b8ca8a16030aa0176a7143"}]},"branch":"refs/heads/stable/ussuri"},"071b9d9a249bcef998eaa13368603f922022b9dd":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-02-15 21:04:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/8"}}},"commit":{"parents":[{"commit":"db271c559809c6df6683d3661063b1555ae708ff","subject":"collectd: pcie-errors is x86-64 only now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db271c559809c6df6683d3661063b1555ae708ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:04:12.000000000","tz":0},"subject":"WIP: Build c8s images","message":"WIP: Build c8s images\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/829202\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/071b9d9a249bcef998eaa13368603f922022b9dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/071b9d9a249bcef998eaa13368603f922022b9dd"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"},"0c8bb6a4fedd648267c96bba87ead2d94059c289":{"kind":"REWORK","_number":9,"created":"2022-02-15 21:28:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/9"}}},"commit":{"parents":[{"commit":"db271c559809c6df6683d3661063b1555ae708ff","subject":"collectd: pcie-errors is x86-64 only now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db271c559809c6df6683d3661063b1555ae708ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:28:38.000000000","tz":0},"subject":"WIP: Build c8s images","message":"WIP: Build c8s images\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/829202\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c8bb6a4fedd648267c96bba87ead2d94059c289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c8bb6a4fedd648267c96bba87ead2d94059c289"}]},"branch":"refs/heads/stable/ussuri"},"06ab94d15ab3c9264f87bb371b72e4d3a8215057":{"kind":"REWORK","_number":10,"created":"2022-02-15 21:35:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/10"}}},"commit":{"parents":[{"commit":"db271c559809c6df6683d3661063b1555ae708ff","subject":"collectd: pcie-errors is x86-64 only now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db271c559809c6df6683d3661063b1555ae708ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:35:32.000000000","tz":0},"subject":"WIP: Build c8s images","message":"WIP: Build c8s images\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/829202\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06ab94d15ab3c9264f87bb371b72e4d3a8215057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06ab94d15ab3c9264f87bb371b72e4d3a8215057"}]},"branch":"refs/heads/stable/ussuri"},"b9b21a0d636b68047a81b4fdae221da5d683225e":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2022-02-15 21:37:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/11"}}},"commit":{"parents":[{"commit":"db271c559809c6df6683d3661063b1555ae708ff","subject":"collectd: pcie-errors is x86-64 only now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db271c559809c6df6683d3661063b1555ae708ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:37:55.000000000","tz":0},"subject":"Build CentOS Stream 8 images","message":"Build CentOS Stream 8 images\n\nThis is part of the effort to allow a migration path\nfor operators using CentOS 7 and Train today.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/829202\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9b21a0d636b68047a81b4fdae221da5d683225e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9b21a0d636b68047a81b4fdae221da5d683225e"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"},"597a7148b77cedf307f40dce3c726bbee5b2ac43":{"kind":"REWORK","_number":12,"created":"2022-02-15 21:56:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/12"}}},"commit":{"parents":[{"commit":"db271c559809c6df6683d3661063b1555ae708ff","subject":"collectd: pcie-errors is x86-64 only now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db271c559809c6df6683d3661063b1555ae708ff"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:56:31.000000000","tz":0},"subject":"Build CentOS Stream 8 images","message":"Build CentOS Stream 8 images\n\nThis is part of the effort to allow a migration path\nfor operators using CentOS 7 and Train today.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/829202\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/597a7148b77cedf307f40dce3c726bbee5b2ac43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/597a7148b77cedf307f40dce3c726bbee5b2ac43"}]},"branch":"refs/heads/stable/ussuri"},"bb84d2c3bd4acefa7c367a7b98928aeeb42de0c1":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2022-02-16 13:44:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/13"}}},"commit":{"parents":[{"commit":"6ef3644d74e04709ebb087010abb2707c8141994","subject":"erlang: use packages from Erlang Solutions on AArch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ef3644d74e04709ebb087010abb2707c8141994"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-16 13:44:18.000000000","tz":0},"subject":"Build CentOS Stream 8 images","message":"Build CentOS Stream 8 images\n\nThis is part of the effort to allow a migration path\nfor operators using CentOS 7 and Train today.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/829202\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb84d2c3bd4acefa7c367a7b98928aeeb42de0c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb84d2c3bd4acefa7c367a7b98928aeeb42de0c1"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"},"175791496df6eb32f93c7b566aff11382a0fa292":{"kind":"REWORK","_number":14,"created":"2022-02-28 11:06:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/14"}}},"commit":{"parents":[{"commit":"6ef3644d74e04709ebb087010abb2707c8141994","subject":"erlang: use packages from Erlang Solutions on AArch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ef3644d74e04709ebb087010abb2707c8141994"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-28 11:06:38.000000000","tz":0},"subject":"Build CentOS Stream 8 images","message":"Build CentOS Stream 8 images\n\nThis is part of the effort to allow a migration path\nfor operators using CentOS 7 and Train today.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/829202\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/175791496df6eb32f93c7b566aff11382a0fa292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/175791496df6eb32f93c7b566aff11382a0fa292"}]},"branch":"refs/heads/stable/ussuri"},"67d6d28b6208cc004a50f19be1386d88c2404bc4":{"kind":"REWORK","_number":15,"created":"2022-02-28 11:56:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/54/829054/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/829054/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/829054/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/829054/15"}}},"commit":{"parents":[{"commit":"6ef3644d74e04709ebb087010abb2707c8141994","subject":"erlang: use packages from Erlang Solutions on AArch64","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ef3644d74e04709ebb087010abb2707c8141994"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-14 14:21:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-28 11:56:44.000000000","tz":60},"subject":"Build CentOS Stream 8 images","message":"Build CentOS Stream 8 images\n\nThis is part of the effort to allow a migration path\nfor operators using CentOS 7 and Train today.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/829202\nChange-Id: I75373eb21c4269c20f72603ad4cf162a472de25b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67d6d28b6208cc004a50f19be1386d88c2404bc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67d6d28b6208cc004a50f19be1386d88c2404bc4"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
