)]}'
{"id":"openstack%2Fkolla~829264","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Iad7e7ebb1fbf4d14f20960d6257345db71d0e214","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-16 08:40:02.000000000","reason":"Michal Nasiadka replied on the change"}},"hashtags":[],"change_id":"Iad7e7ebb1fbf4d14f20960d6257345db71d0e214","subject":"collectd: pcie-errors is x86-64 only now","status":"MERGED","created":"2022-02-15 21:05:11.000000000","updated":"2022-02-16 10:10:07.000000000","submitted":"2022-02-16 10:08:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829264","cherry_pick_of_change":827883,"cherry_pick_of_patch_set":2,"meta_rev_id":"218c12d8660059c96f042ba1c580f1d8b1b3fee6","_number":829264,"virtual_id_number":829264,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 10:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-16 08:40:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-16 07:56:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-16 08:40:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 21:05:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-02-15 22:27:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 07:56:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"18cb64f8133f0f80e2ada21939326a11e8c4c778","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:05:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/collectd/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"225702745935be888748b32826c50eeb6b409ebe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:05:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"794320472ef1e9850a737f9357b4e77861c3a420","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:05:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"bd588f82cc86d5914b4429d32791a4187b8b41d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 22:27:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f9c3e737563e419c85876c97f09d6b4f : SUCCESS in 24m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/101b65131c704e0a9ec0eeb250749630 : SUCCESS in 36m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/eae2e0380ce04a57963b6787c0a74772 : SUCCESS in 44m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1927951733d640859f2681f05f6ee7b0 : SUCCESS in 54m 13s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a196bfa0cef14e87a139e8eff099a899 : SUCCESS in 1h 02m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/66600896c930484d88de4643644cc501 : SUCCESS in 29m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1a92d8f5f0374aaea035e7d8f6f28549 : SUCCESS in 18m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/064fa70b0713461195a0a26e524a52a7 : SUCCESS in 52m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f83455d1cbab46708ffea6c1f78d8bdb : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb9d62ea3b694e89bf584542f63c2955 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab525126cee5438e9c8e33f384155c54 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7193b6062c1442c7b969180b592b8c76 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1700f994ce6b409fa350b4183f5ab419 : SUCCESS in 6m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3357a2d4b8054ba7b137a387ddfd25ab : SUCCESS in 20m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1539be2c3934b05a2605fe4047769c1 : SUCCESS in 32m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/99a7d3482a03456ea917d711045ded67 : SUCCESS in 58m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b3033aebb5047b4bab2d020d8fa3af3 : SUCCESS in 55m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/76ff7329e68a421c9d22f977f1b5728f : SUCCESS in 1h 06m 07s","accounts_in_message":[],"_revision_number":2},{"id":"bc462d9b46262207bdc02d5c7c13e6b000e66f28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 00:11:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dac1916650384780b3ccc73d02be9934 : FAILURE in 2h 40m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8ab8e69983f47e9aa48228ac41f02d4 : FAILURE in 2h 57m 24s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc13b6b8144c417bb747819d39855e5b : SUCCESS in 1h 28m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/30604916e8144f0f9c9d34ad66385fbc : SUCCESS in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df2ab5ead9fbb985679d9d0386a0724d3d82ae7c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-16 07:56:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f52634b1f4a71e24449979621c088bf390e1f908","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 08:40:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f37ca201e76b93e5361e001215c4d3f7b3980e75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 08:40:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a136e1da1ed1bdeafa727db904b2ca3a3a0d4e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:08:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2c60249da47643bcbe3d93b8c67cb22c : SUCCESS in 38m 38s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7b4d16fd45884df186f2ee6ce4b327b3 : SUCCESS in 1h 01m 25s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7f1ede69ad1b40deaaefcee06000b350 : SUCCESS in 1h 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b5156810cec4457898169dc4c78f18b : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17a7c45e092946b4b614e9a33c7c5a03 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e4701b9a4fe46acbfdb346bff7208ef : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd9b8b2ac9b435fa92f9241bfa998c4 : SUCCESS in 6m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/01e8100e375d4883a71568c554f78412 : SUCCESS in 30m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/475512ccea864b1aaf3db2d4408503ef : SUCCESS in 57m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/139eb617b63f456684ebeb0159ec4d8c : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":2},{"id":"df500305e7c99a30b4edc6861b1a7c01edcb54f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:08:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"218c12d8660059c96f042ba1c580f1d8b1b3fee6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:10:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40e66d470c042a9982f80e8520d3c49 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a29648baf8030d754eaf3764fe456146db320bff","revisions":{"aaabc547237a6206add9a5b3b1179b6475e49756":{"kind":"REWORK","_number":1,"created":"2022-02-15 21:05:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/829264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/829264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/829264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/829264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/829264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/829264/1"}}},"commit":{"parents":[{"commit":"1edda3945fc69f678e63f7cdd6196f59cf656dfb","subject":"Merge \"base: Drop usage of Ceph Nautilus from RDO\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1edda3945fc69f678e63f7cdd6196f59cf656dfb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 16:15:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:05:11.000000000","tz":0},"subject":"collectd: pcie-errors is x86-64 only now","message":"collectd: pcie-errors is x86-64 only now\n\nChange-Id: Iad7e7ebb1fbf4d14f20960d6257345db71d0e214\n(cherry picked from commit 941d021e091567d2709bbb868768c6dc4a3372e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aaabc547237a6206add9a5b3b1179b6475e49756"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aaabc547237a6206add9a5b3b1179b6475e49756"}]},"branch":"refs/heads/stable/wallaby"},"a29648baf8030d754eaf3764fe456146db320bff":{"kind":"REWORK","_number":2,"created":"2022-02-15 21:05:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/829264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/829264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/829264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/829264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/829264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/829264/2"}}},"commit":{"parents":[{"commit":"1edda3945fc69f678e63f7cdd6196f59cf656dfb","subject":"Merge \"base: Drop usage of Ceph Nautilus from RDO\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1edda3945fc69f678e63f7cdd6196f59cf656dfb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 16:15:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:05:31.000000000","tz":0},"subject":"collectd: pcie-errors is x86-64 only now","message":"collectd: pcie-errors is x86-64 only now\n\nChange-Id: Iad7e7ebb1fbf4d14f20960d6257345db71d0e214\n(cherry picked from commit 941d021e091567d2709bbb868768c6dc4a3372e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a29648baf8030d754eaf3764fe456146db320bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a29648baf8030d754eaf3764fe456146db320bff"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
