)]}'
{"id":"openstack%2Fkolla~829265","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Iad7e7ebb1fbf4d14f20960d6257345db71d0e214","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-16 08:40:10.000000000","reason":"Michal Nasiadka replied on the change"}},"hashtags":[],"change_id":"Iad7e7ebb1fbf4d14f20960d6257345db71d0e214","subject":"collectd: pcie-errors is x86-64 only now","status":"MERGED","created":"2022-02-15 21:05:46.000000000","updated":"2022-02-16 10:23:31.000000000","submitted":"2022-02-16 10:21:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829265","cherry_pick_of_change":827883,"cherry_pick_of_patch_set":2,"meta_rev_id":"ded606a2311db8e1179e31b0d842aeb9f5a236e6","_number":829265,"virtual_id_number":829265,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 10:21:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-16 08:40:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-16 07:56:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-16 08:40:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 21:05:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-02-15 22:42:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 07:56:31.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"a3c313b60dc985eb5459b5be9ea0eac8411d4722","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:05:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/collectd/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"bb9c4d8da2ac7a83f916eb68c7c397f432be1e3b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:06:03.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"456d137c18e856b035699556cb5f9d779187b304","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:06:07.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"af9691057241328808e4fc266f6353e50c1f4422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 22:42:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/181dd566eda642ed8f574dbad0445923 : SUCCESS in 40m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0af623316c6e4cedb5f7cc4bb8360e23 : SUCCESS in 46m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/12302e5fe75b449d92d5e17829a1b523 : SUCCESS in 1h 05m 29s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a9cd0dad79604bbd9b29873afe56a375 : SUCCESS in 1h 05m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8c8bec3251ae4ad0b3210041a28dfcd1 : SUCCESS in 1h 10m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/75fcbe008a2d49e7a52d0b16bac41dc5 : SUCCESS in 20m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/da88102211734ae6b042e571f6b267b2 : SUCCESS in 17m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b9cea840ad59410280d14616a9aa4bf8 : SUCCESS in 1h 00m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aac318005725473789bc590750ea7026 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf66a4273aeb4704b49b25fa40bd6255 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/019dfe9558c54d68aed0626b228f44ab : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64ba79e08854da8828f9d8f15045e15 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/145edf96fd1a4eafb68fe27018aaf6dc : SUCCESS in 23m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e5b2fbb244a421b98ac211289901c3b : SUCCESS in 26m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e03a8d1ae8a4ab7adec068179f35422 : SUCCESS in 47m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/996558b6e23e49a38ffaa512b723dc31 : SUCCESS in 52m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/24ebf5c641254171be47c5ef6228397a : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":2},{"id":"7c6a2df23748a8410146c12298d83cd3d6901975","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 00:35:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e15e43b310074eb6ad846d3a1b55e064 : FAILURE in 41m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e72b631d71d4e41b192123a25aaa0fe : TIMED_OUT in 3h 03m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/df6318f0a15d47b092148b78006feda8 : TIMED_OUT in 3h 04m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8c196ab14b0416e88f0a3ce7722eeea : SUCCESS in 32m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f15c7fa441d59a67217ce14bfdc35ad8dcaec373","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-16 07:56:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"da9f0d1d82254cf11edfc439f447d5cf1538abae","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 08:40:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a544abed40f525f0c23d29353d25dddbdb1f641f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 08:40:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"83670f832e1d002967aff6ac7a688e85d73c966a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:21:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4efbd29425d945c28ebda1c2a331e56f : SUCCESS in 28m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6aa03df90be14726869ec796146ffbdf : SUCCESS in 1h 06m 35s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d7aff0eda2174419802b38962a3ecab7 : SUCCESS in 1h 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5de3ddc7ba7748dd9d8ce4750efaf7a0 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/531d99eaf5bb46319d2fe05696fcaab6 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6c0b6a168284b6a910669f8f1841e1e : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce52cb93bc464e32befd80ae16a7b430 : SUCCESS in 6m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad4f522e9440414a91f0dc0259a6a351 : SUCCESS in 27m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/13f52c6197794ed1b5c24e40b997a493 : SUCCESS in 42m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d7f99541a5984fa6bd062d8ebca5d83e : SUCCESS in 1h 02m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d0aceeed452f9ab04575f85a2c82526bb50b394c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:21:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ded606a2311db8e1179e31b0d842aeb9f5a236e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:23:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f24abb4c764683bb5ac66253dd5482 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"545fed783d5f72fce41a4a54669c4518f7f8ac63","revisions":{"c2e82d9ca92e97fdc88637110646d3be1389f326":{"kind":"REWORK","_number":1,"created":"2022-02-15 21:05:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/65/829265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/829265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/829265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/829265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/829265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/829265/1"}}},"commit":{"parents":[{"commit":"07a4e434d3c04108e49a8185a3803298d53e1367","subject":"base: Drop usage of Ceph Nautilus from RDO","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a4e434d3c04108e49a8185a3803298d53e1367"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 16:15:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:05:46.000000000","tz":0},"subject":"collectd: pcie-errors is x86-64 only now","message":"collectd: pcie-errors is x86-64 only now\n\nChange-Id: Iad7e7ebb1fbf4d14f20960d6257345db71d0e214\n(cherry picked from commit 941d021e091567d2709bbb868768c6dc4a3372e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2e82d9ca92e97fdc88637110646d3be1389f326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2e82d9ca92e97fdc88637110646d3be1389f326"}]},"branch":"refs/heads/stable/victoria"},"545fed783d5f72fce41a4a54669c4518f7f8ac63":{"kind":"REWORK","_number":2,"created":"2022-02-15 21:06:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/65/829265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/829265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/829265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/829265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/829265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/829265/2"}}},"commit":{"parents":[{"commit":"07a4e434d3c04108e49a8185a3803298d53e1367","subject":"base: Drop usage of Ceph Nautilus from RDO","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a4e434d3c04108e49a8185a3803298d53e1367"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-04 16:15:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:06:03.000000000","tz":0},"subject":"collectd: pcie-errors is x86-64 only now","message":"collectd: pcie-errors is x86-64 only now\n\nChange-Id: Iad7e7ebb1fbf4d14f20960d6257345db71d0e214\n(cherry picked from commit 941d021e091567d2709bbb868768c6dc4a3372e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/545fed783d5f72fce41a4a54669c4518f7f8ac63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/545fed783d5f72fce41a4a54669c4518f7f8ac63"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
