)]}'
{"id":"openstack%2Fkolla~829266","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I667ccda5bb4edbfbfb97dfb0cea264837ed05cae","project":"openstack/kolla","branch":"stable/ussuri","topic":"827656-patch3-stable/ussuri","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-16 14:58:54.000000000","reason":"Michal Nasiadka replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-02-16 14:46:24.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I667ccda5bb4edbfbfb97dfb0cea264837ed05cae","subject":"erlang: use packages from Erlang Solutions on AArch64","status":"MERGED","created":"2022-02-15 21:08:27.000000000","updated":"2022-02-16 16:45:16.000000000","submitted":"2022-02-16 16:42:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829266-827656-patch3-stable/ussuri","cherry_pick_of_change":827663,"cherry_pick_of_patch_set":6,"meta_rev_id":"de061d6cf70525c71c22bf79e8d35e00b7bc9634","_number":829266,"virtual_id_number":829266,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 16:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-16 14:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-16 14:46:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-16 14:58:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 21:08:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-02-15 22:25:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 11:54:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-02-16 14:46:24.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"925fb1078c8fb5c877fc0749dd825e9509055bd8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:08:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2\n* docker/kolla-toolbox/Dockerfile.j2\n* docker/rabbitmq/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"846c7bc6656f54efe50be056b08da6acc8be3d6d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:11:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a96c8f038133320f76a022ec71fbcc5f5765f540","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:11:37.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"c29889f47cc2b54b8ce01982386e8e4fda78b636","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:12:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"03f2122a5d6473d8d0caa1ba904b65d75669efe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 22:25:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2189189610dc45528c0ecd90aaeaa019 : SUCCESS in 22m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9445f4f783bd4855847609f33c0026b4 : SUCCESS in 18m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b88fa76fdbb841c0a0e9104a4e8fd9d2 : SUCCESS in 1h 02m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec07e51ad1b04703934432f668e2ad35 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4803e0744cb4c7f9b60c2bf44f89104 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6566310f8beb40e988e3c9cf0de5d624 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/328995e779274d1aab4d349330e467f9 : SUCCESS in 7m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/67f066f90f7f4b65a7c1aa12f0590a92 : SUCCESS in 20m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba631b6714e54452a56ccecc53a0ec65 : SUCCESS in 31m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/86643b23b1c24789b559e9621553cd5e : SUCCESS in 1h 00m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fe56d27076054173a130c06359d9eec5 : SUCCESS in 42m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/72c1430106b54ce9a69963ac8863780b : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":3},{"id":"8d4310a094c70f594a840cde3a73741e09cd3e61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 23:37:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a0568618dbac41598702e4cef7cd6f1a : SUCCESS in 34m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9aaf8c53d6434ecb884a44392ee81738 : SUCCESS in 1h 23m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ad6fc1e781646f980c1d97fcc9011d3 : SUCCESS in 2h 05m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a9b5a4dfb8aa6aaa1e312f9829f0b72a625ee78a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 10:23:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4c00a4fa7649cf8b713fe20231abf53b8b620130","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-16 11:54:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f61ea5670a1a894c10b24a27d349e861a684a67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 12:15:00.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"178ea4542550fa57cc7f3afdc005650b13ee64b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 13:32:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1f1b20ceb4d847789ad65d9cb5d2fabe : SUCCESS in 32m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2072e9f712544dfd9976e699df848d1f : SUCCESS in 18m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/add4659786aa48c285ff81ccb11c3c02 : FAILURE in 16m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58d36ccd243240319f626ad01fda2043 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12888813e5e347f1ab1aac15aba23e8c : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a6272b911a44f0080fe60a20fa65df0 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b67a6db476842748a4592ef492681e2 : SUCCESS in 7m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/547115beda95443d9afef6e3a89c7a4a : SUCCESS in 21m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4bb6deaaf4bd475b8efd23d19e00ea0e : SUCCESS in 20m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3961995ba10c4b6db7125b3622d2b45b : SUCCESS in 43m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d40a78e038014c60ab2548f95b2b9475 : SUCCESS in 47m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f4b82078396a49c1afefa3a538518303 : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":4},{"id":"671c0c47759a223ddb569018d6603a2afce6eded","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 13:44:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b3490ffa1520aea41d45d1c2c1c5804c7284fe94","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 14:27:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7179cc451070d779528958edf17d8bb637692568","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-16 14:46:24.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8b20490528b495004c1807786b87767fe024471d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 14:58:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"444fe1a0a3ef581ed382a5139bcaf76d7ef41516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 14:59:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c48cdf70bd1738202a0a3b4ff6158034f6bd4aab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 15:10:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7376f0708c33437a9e0a29ea5910bafb : SUCCESS in 2h 32m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4bf9bcbc12e3470b889a6ec96732725b : SUCCESS in 1h 26m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4543c347da66434882088323d2fd9925 : SUCCESS in 40m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"14c2c7d3a380e842213e85633bd147012680ddd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:42:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/439616d1e7bd4f489a2cd539fbfb97f4 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cb1f825aed945bc877f513e2ebc7500 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8bad604a5df0459c8dca93a04d467ea1 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dd97d98f5ea4897a880303575370b61 : SUCCESS in 5m 40s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c1b859f0132b43c6b4384e2f8fa0bfe5 : SUCCESS in 27m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eaad874cf7c643a58b49f2cc8ee405c0 : SUCCESS in 52m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bf529dc19737462c89365b384455df8b : SUCCESS in 1h 27m 49s","accounts_in_message":[],"_revision_number":4},{"id":"6e526cd0f554d03a1b78e7194492eb8119e22915","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:42:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"de061d6cf70525c71c22bf79e8d35e00b7bc9634","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:45:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4f16b3664d54e7380f688d847e2d628 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6ef3644d74e04709ebb087010abb2707c8141994","revisions":{"a2af480e413085b3a37212c929713458af9d3b55":{"kind":"REWORK","_number":1,"created":"2022-02-15 21:08:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/829266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/829266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/829266/1"}}},"commit":{"parents":[{"commit":"c5f2482fa55c54a107d9fbe493745fe191a15163","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5f2482fa55c54a107d9fbe493745fe191a15163"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-03 13:14:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:08:27.000000000","tz":0},"subject":"erlang: use packages from Erlang Solutions on AArch64","message":"erlang: use packages from Erlang Solutions on AArch64\n\nRabbitMQ requires Erlang 23+ and soon will bump to 24+.\n\nOn AArch64 we used version 22 as there was no other option. Now it is.\n\nChange-Id: I667ccda5bb4edbfbfb97dfb0cea264837ed05cae\n(cherry picked from commit f943c7d773a280b832dde9edf9d615540b259b5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2af480e413085b3a37212c929713458af9d3b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2af480e413085b3a37212c929713458af9d3b55"}]},"branch":"refs/heads/stable/ussuri"},"6c2b43c58bbb75175a626aeab8fc9e7ce01cd1ee":{"kind":"REWORK","_number":2,"created":"2022-02-15 21:11:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/829266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/829266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/829266/2"}}},"commit":{"parents":[{"commit":"c5f2482fa55c54a107d9fbe493745fe191a15163","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5f2482fa55c54a107d9fbe493745fe191a15163"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-03 13:14:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:11:21.000000000","tz":0},"subject":"erlang: use packages from Erlang Solutions on AArch64","message":"erlang: use packages from Erlang Solutions on AArch64\n\nRabbitMQ requires Erlang 23+ and soon will bump to 24+.\n\nOn AArch64 we used version 22 as there was no other option. Now it is.\n\nChange-Id: I667ccda5bb4edbfbfb97dfb0cea264837ed05cae\n(cherry picked from commit f943c7d773a280b832dde9edf9d615540b259b5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c2b43c58bbb75175a626aeab8fc9e7ce01cd1ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c2b43c58bbb75175a626aeab8fc9e7ce01cd1ee"}]},"branch":"refs/heads/stable/ussuri"},"026256cb57581fd1ec4f94143cb2b8c95090c179":{"kind":"REWORK","_number":3,"created":"2022-02-15 21:12:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/829266/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/829266/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/829266/3"}}},"commit":{"parents":[{"commit":"c5f2482fa55c54a107d9fbe493745fe191a15163","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5f2482fa55c54a107d9fbe493745fe191a15163"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-03 13:14:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:12:11.000000000","tz":0},"subject":"erlang: use packages from Erlang Solutions on AArch64","message":"erlang: use packages from Erlang Solutions on AArch64\n\nRabbitMQ requires Erlang 23+ and soon will bump to 24+.\n\nOn AArch64 we used version 22 as there was no other option. Now it is.\n\nChange-Id: I667ccda5bb4edbfbfb97dfb0cea264837ed05cae\n(cherry picked from commit f943c7d773a280b832dde9edf9d615540b259b5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/026256cb57581fd1ec4f94143cb2b8c95090c179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/026256cb57581fd1ec4f94143cb2b8c95090c179"}]},"branch":"refs/heads/stable/ussuri"},"6ef3644d74e04709ebb087010abb2707c8141994":{"kind":"REWORK","_number":4,"created":"2022-02-16 12:15:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/829266/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/829266/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/829266/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/829266/4"}}},"commit":{"parents":[{"commit":"c5f2482fa55c54a107d9fbe493745fe191a15163","subject":"Remove missing collectd packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5f2482fa55c54a107d9fbe493745fe191a15163"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-03 13:14:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-16 12:15:00.000000000","tz":0},"subject":"erlang: use packages from Erlang Solutions on AArch64","message":"erlang: use packages from Erlang Solutions on AArch64\n\nRabbitMQ requires Erlang 23+ and soon will bump to 24+.\n\nOn AArch64 we used version 22 as there was no other option. Now it is.\n\nChange-Id: I667ccda5bb4edbfbfb97dfb0cea264837ed05cae\n(cherry picked from commit f943c7d773a280b832dde9edf9d615540b259b5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ef3644d74e04709ebb087010abb2707c8141994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ef3644d74e04709ebb087010abb2707c8141994"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
