)]}'
{"id":"openstack%2Fkolla~829267","triplet_id":"openstack%2Fkolla~stable%2Fxena~I667ccda5bb4edbfbfb97dfb0cea264837ed05cae","project":"openstack/kolla","branch":"stable/xena","topic":"827656-patch3-stable/xena","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-16 10:23:18.000000000","reason":"Michal Nasiadka replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-02-16 08:31:17.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I667ccda5bb4edbfbfb97dfb0cea264837ed05cae","subject":"erlang: use packages from Erlang Solutions on AArch64","status":"MERGED","created":"2022-02-15 21:12:39.000000000","updated":"2022-02-16 12:16:46.000000000","submitted":"2022-02-16 12:13:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"829267-827656-patch3-stable/xena","cherry_pick_of_change":827663,"cherry_pick_of_patch_set":6,"meta_rev_id":"a329b4e648e487e818b17cbe4a46f63bd16ff029","_number":829267,"virtual_id_number":829267,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 12:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-16 10:23:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-16 08:38:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-16 10:23:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 21:12:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-02-15 23:22:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 07:58:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-02-16 08:38:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"17451d4d332c76c8cd4df01432889c956db4e7a0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:12:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"be16e213ee712fcc497970747d1b9d76a177389c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:14:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6cd351ab98e98390ac0b9441b922e7aad970cf4a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:15:01.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"f8c74e56675ff6389ec3477922be61a990d83ae8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:54:18.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d46ceed22e37a26a6c13f8069bcd0e38c8e14ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 23:22:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a603358695d34962b00720707936a9e1 : SUCCESS in 33m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b290bddb27d74a5cbfc4f822b9b6e5ad : SUCCESS in 39m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4f46831c34d14ef8b491e63f09e97022 : FAILURE in 31m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3620420d9a8a4437805aaf7eae720886 : SUCCESS in 1h 10m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c673700a1666402dba501e05617fcef0 : SUCCESS in 1h 27m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a60c5ca3df75438ba9e3f2c0888d4786 : FAILURE in 29m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b4765e4f942d445a8d2291d8d09141dc : SUCCESS in 20m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f7f884f30bcb425888ff73d922ae3a4a : SUCCESS in 1h 01m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d006a60bf7413099434bc1a6d9e803 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c55db1e7b8046a2a1ff5009ecab9401 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cf708c2da3f40b59c4be6e847a6c30d : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec6d29e61bf34c8f8bda92576eb5b6af : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b8c4ab29ab4f14913bd7f200a7f413 : SUCCESS in 5m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/792a593ec705429cafeaefa746c5af1c : SUCCESS in 20m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/997cde97c347479dbadf74b51c86af9c : FAILURE in 31m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/22ef8b8d1ddd423b89f9aa6c4b79c7d9 : SUCCESS in 59m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c686c83dd840473ab192b3678417be32 : FAILURE in 21m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6782e88296cb4c438b09b718fe3a9089 : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":3},{"id":"567a19b5e5a6dc39f5f9a93462b2bd1b9113714e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 00:03:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e23d9901cc9462cbedeac4e82fc0bff : FAILURE in 43m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/acd28e248ce840d8bf770468c6af3745 : FAILURE in 14m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c23b984b861d4782a2f844fd0f9985cd : FAILURE in 28m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e6222f709214efb88184d418d70d244 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e01dd9fa4bdd4cb6b66e6a4cdad66fa4 : SUCCESS in 4m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc351d991f9d472e8687d20bd8e5f05b : SUCCESS in 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f0d91c8997b277f11e7c7075f64bfe77b7e28ba4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-16 07:58:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"52ef117a72e60949f1380bf1262e05486c29a194","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 08:22:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad30c8685cda43abc4707c9a0c96c9ac3ba9f82d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-16 08:31:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b74a272bc8828092b77617cec278fac2ece389d3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-16 08:38:53.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d612e9f6199f19be22c022993287a86f9417157e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:07:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4b389a07cc314d24bb5e49e776511886 : SUCCESS in 26m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/22599e5bccb24ad39d30f4fe4b394d86 : SUCCESS in 41m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/472e117eaf054517a0252edc3cf7d4db : SUCCESS in 1h 10m 30s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e763db9e068c42a88c2171c5446a6258 : SUCCESS in 1h 06m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/125c2ced077149a599218b2b39a1583c : SUCCESS in 1h 20m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8522863f2a774b6c850a35e19c708551 : SUCCESS in 28m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/de6c83c550a04ac793385557d69f2af0 : SUCCESS in 17m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/27425351c6c448fba08f614c42563fb2 : SUCCESS in 1h 04m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8bb0f8cc94946f2b8fdb1ca02017e6a : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2b0738e90334a509f94e203284b4929 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/845d1e6ce96b478b84fa8415c4ccade8 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a2091199afe45da84ce6f469fca32e4 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9cd57f6b1a140f8a851c671e419b7ce : SUCCESS in 5m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/234ac2dbce8b4634b59a86254b07b749 : SUCCESS in 18m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a9ee293cb02e4ba0a402207ad103501f : SUCCESS in 30m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8de5a51af0b24ef2a4f2b66435e6ea59 : SUCCESS in 49m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8a9fb6585133490bacc2ada8071f95f1 : SUCCESS in 49m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6d2b221e2ea64259bb77ee8aafc95b83 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":3},{"id":"0b5d44be81d3ca89ca3c82e94264456529833220","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 10:23:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a1f65cd4f6bd53c41a3653d020acefcf770444c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:23:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"db6e2339875f1d4d7a8ce8167b6af3858e0d44ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 11:26:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/243513a927a941f7a496b4584bc75bb3 : FAILURE in 2h 37m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8769a02fd5a2431ea4379a3700a50018 : FAILURE in 56m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/40e76716db2c4d79b3cd8d740e6e34bd : FAILURE in 25m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8395605541aa473585cc170c40b37eaf : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c68443c6fb6e4d68a2c01c1c04399813 : SUCCESS in 6m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b7b37f4c0f44aeb82b84b4160d99cfc : SUCCESS in 2h 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ed1549179aaeaebdafb6368ff3eef0c4652b23b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 12:13:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9f12024eed654454bb9e7c1c22ea8f2a : SUCCESS in 38m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a63737240bf54bd6a1ddad520dbd704d : SUCCESS in 1h 18m 39s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/db8bdedfa7624247a2eb9cfa05a2df07 : SUCCESS in 1h 45m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f70b0b13e2840d9b259fe8ed8cc6a83 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb7453b6eb064e54812fce1092bc7bea : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c6bfe80b12247cc8a95d959400299d8 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9294efaf2cc54d8e8d9ca8b15100b992 : SUCCESS in 7m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/03b0561e78d4425b8e8dd068e843f049 : SUCCESS in 36m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a73f8fe43a274c5b84c7675fdf7f4856 : SUCCESS in 54m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3586ff4461444f888e11bd81eb134cea : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":3},{"id":"30bd00f60864bb60717ff1be3224d37ef3e2c723","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 12:13:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a329b4e648e487e818b17cbe4a46f63bd16ff029","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 12:16:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808542dce07043288ef9d5f02d07bd08 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b38582b48c3716d5d6118535814bfd31cecd1416","revisions":{"dec209319e5e4f01ca4bff16d55a782526723cac":{"kind":"REWORK","_number":1,"created":"2022-02-15 21:12:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/67/829267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/829267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/829267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/829267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/829267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/829267/1"}}},"commit":{"parents":[{"commit":"af092df6aa66f58a8d3ad3b363db0d81f54c0944","subject":"ironic: Fix UEFI \u0026 iPXE bootloader filenames","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af092df6aa66f58a8d3ad3b363db0d81f54c0944"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-03 13:14:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:12:39.000000000","tz":0},"subject":"erlang: use packages from Erlang Solutions on AArch64","message":"erlang: use packages from Erlang Solutions on AArch64\n\nRabbitMQ requires Erlang 23+ and soon will bump to 24+.\n\nOn AArch64 we used version 22 as there was no other option. Now it is.\n\nChange-Id: I667ccda5bb4edbfbfb97dfb0cea264837ed05cae\n(cherry picked from commit f943c7d773a280b832dde9edf9d615540b259b5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dec209319e5e4f01ca4bff16d55a782526723cac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dec209319e5e4f01ca4bff16d55a782526723cac"}]},"branch":"refs/heads/stable/xena"},"865f6fae66be0d7c7627e1728c40d65e6384e793":{"kind":"REWORK","_number":2,"created":"2022-02-15 21:14:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/67/829267/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/829267/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/829267/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/829267/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/829267/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/829267/2"}}},"commit":{"parents":[{"commit":"af092df6aa66f58a8d3ad3b363db0d81f54c0944","subject":"ironic: Fix UEFI \u0026 iPXE bootloader filenames","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af092df6aa66f58a8d3ad3b363db0d81f54c0944"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-03 13:14:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:14:58.000000000","tz":0},"subject":"erlang: use packages from Erlang Solutions on AArch64","message":"erlang: use packages from Erlang Solutions on AArch64\n\nRabbitMQ requires Erlang 23+ and soon will bump to 24+.\n\nOn AArch64 we used version 22 as there was no other option. Now it is.\n\nChange-Id: I667ccda5bb4edbfbfb97dfb0cea264837ed05cae\n(cherry picked from commit f943c7d773a280b832dde9edf9d615540b259b5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/865f6fae66be0d7c7627e1728c40d65e6384e793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/865f6fae66be0d7c7627e1728c40d65e6384e793"}]},"branch":"refs/heads/stable/xena"},"b38582b48c3716d5d6118535814bfd31cecd1416":{"kind":"REWORK","_number":3,"created":"2022-02-15 21:54:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/67/829267/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/829267/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/829267/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/829267/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/829267/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/829267/3"}}},"commit":{"parents":[{"commit":"af092df6aa66f58a8d3ad3b363db0d81f54c0944","subject":"ironic: Fix UEFI \u0026 iPXE bootloader filenames","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af092df6aa66f58a8d3ad3b363db0d81f54c0944"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-03 13:14:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:54:18.000000000","tz":0},"subject":"erlang: use packages from Erlang Solutions on AArch64","message":"erlang: use packages from Erlang Solutions on AArch64\n\nRabbitMQ requires Erlang 23+ and soon will bump to 24+.\n\nOn AArch64 we used version 22 as there was no other option. Now it is.\n\nChange-Id: I667ccda5bb4edbfbfb97dfb0cea264837ed05cae\n(cherry picked from commit f943c7d773a280b832dde9edf9d615540b259b5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b38582b48c3716d5d6118535814bfd31cecd1416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b38582b48c3716d5d6118535814bfd31cecd1416"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
