)]}'
{"id":"openstack%2Fkolla~829268","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I667ccda5bb4edbfbfb97dfb0cea264837ed05cae","project":"openstack/kolla","branch":"stable/wallaby","topic":"827656-patch3-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-16 08:42:32.000000000","reason":"Michal Nasiadka replied on the change"}},"hashtags":[],"change_id":"I667ccda5bb4edbfbfb97dfb0cea264837ed05cae","subject":"erlang: use packages from Erlang Solutions on AArch64","status":"MERGED","created":"2022-02-15 21:15:25.000000000","updated":"2022-02-16 10:24:04.000000000","submitted":"2022-02-16 10:21:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829268-827656-patch3-stable/xena-stable/wallaby","cherry_pick_of_change":829267,"cherry_pick_of_patch_set":2,"meta_rev_id":"6f30af6d9a05c4eda7c059795c5a317e901af325","_number":829268,"virtual_id_number":829268,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 10:21:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-16 08:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-16 08:41:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-16 08:42:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 21:15:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-02-15 23:10:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 08:41:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"a05b5d7023492123d462f8fed7a0cae1dfea1b47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:15:25.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"2429f4af9ceaf46f501bcfbd705bf3fba17f8c57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-15 21:55:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"82359c6802964c4524644b678735b836748641e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 23:10:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1f2e1409e1aa4b30ae84e6ab5079c222 : SUCCESS in 28m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8fec30a91be54fb6b4e63732c527d905 : SUCCESS in 44m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c0a3cc6e3e9d4d3c87abb46d0eae19e3 : SUCCESS in 1h 01m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d36af25fc49840d7b2be892941bb9f4d : SUCCESS in 58m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b562629b109a420e95a4d69a33b6ccae : SUCCESS in 1h 09m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f8c22759e6fb4c8bb5413099526374bf : SUCCESS in 31m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/77d2cef0abbc486c87a6d213f6ba743a : SUCCESS in 21m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aef1b0d4bf814700813b3a5958670ea4 : SUCCESS in 59m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6d9127df27640bdbffff1e3e2a1173a : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f40932d39eea47f882220aecb7ac425b : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cff21ade712148009ecddc33184e13d9 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a15247d4007a42f6b73a5f8ae4472b6e : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59527f8873d741bb81b15c38b6676ec8 : SUCCESS in 6m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3abc112f73064e74a75070d58b97977c : SUCCESS in 21m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/558f96598fa34ccb816e017fff495d6f : SUCCESS in 31m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/164eff610cba4aa4819ecf01e1d0e1c1 : SUCCESS in 48m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f72d8763b8c142cbbcf1d130c3582ea2 : SUCCESS in 48m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/764924d9815f4ba99600fffcc549d6b4 : SUCCESS in 1h 04m 52s","accounts_in_message":[],"_revision_number":2},{"id":"78ca87dd75edb0fc7afbeb7684fb2c8c0c98d0c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 02:50:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1331468bd72f45b09ef54d865b1d4131 : TIMED_OUT in 3h 23m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ebb8ecd44c384473914203b40db14435 : FAILURE in 2h 56m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fe3c4699dcf142a99eb774ee8a97bb47 : SUCCESS in 1h 15m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/89a04eb02c414d3ca37656225ea8fb31 : SUCCESS in 36m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b77e7170e887d1e715fe5424bf41247080c7c5bc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-16 08:41:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5d4870cd37e6236d01ad2356027d2544979d05e7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 08:42:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e41303a4f16e47c94adeb59080758ccf6fc6ba3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 08:43:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c9d532fad174ca9875413e0c3d0bc6b162730797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:21:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5c2145eb098943c29600ff212a21b979 : SUCCESS in 36m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5fa8a055d02a4d4abb11035fcbc31d9d : SUCCESS in 54m 11s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4b5af0b00dea486d824019306614fd3d : SUCCESS in 1h 04m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f8726793554df58ab1ec3027240d34 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17239d0b139542a6a00494d1c22b06e8 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a375cbab26684d8ca35629e2abcf1eff : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cfc6300a3e74526a93a762d983ebd1e : SUCCESS in 6m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c0b6d616594d4f91a835c32145bc9b98 : SUCCESS in 34m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cfe1506c7fee4697ac3e9045abdee322 : SUCCESS in 58m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a5e3c1567754dc69d37ea9d507216ef : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":2},{"id":"a7bc4c338df08765494caca0c811955e53e5d0b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:21:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6f30af6d9a05c4eda7c059795c5a317e901af325","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:24:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ac1a96b0b714dceb883122f9d7fd45a : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"50b1f117a907cf092f2523fd561a3717e607e1d1","revisions":{"75944b675582cd824d1df9c6e4e5bf5416feeebd":{"kind":"REWORK","_number":1,"created":"2022-02-15 21:15:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/68/829268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/829268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/829268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/829268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/829268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/829268/1"}}},"commit":{"parents":[{"commit":"1edda3945fc69f678e63f7cdd6196f59cf656dfb","subject":"Merge \"base: Drop usage of Ceph Nautilus from RDO\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1edda3945fc69f678e63f7cdd6196f59cf656dfb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-03 13:14:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:15:25.000000000","tz":0},"subject":"erlang: use packages from Erlang Solutions on AArch64","message":"erlang: use packages from Erlang Solutions on AArch64\n\nRabbitMQ requires Erlang 23+ and soon will bump to 24+.\n\nOn AArch64 we used version 22 as there was no other option. Now it is.\n\nChange-Id: I667ccda5bb4edbfbfb97dfb0cea264837ed05cae\n(cherry picked from commit f943c7d773a280b832dde9edf9d615540b259b5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75944b675582cd824d1df9c6e4e5bf5416feeebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75944b675582cd824d1df9c6e4e5bf5416feeebd"}]},"branch":"refs/heads/stable/wallaby"},"50b1f117a907cf092f2523fd561a3717e607e1d1":{"kind":"REWORK","_number":2,"created":"2022-02-15 21:55:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/68/829268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/829268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/829268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/829268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/829268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/829268/2"}}},"commit":{"parents":[{"commit":"1edda3945fc69f678e63f7cdd6196f59cf656dfb","subject":"Merge \"base: Drop usage of Ceph Nautilus from RDO\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1edda3945fc69f678e63f7cdd6196f59cf656dfb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-02-03 13:14:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-15 21:55:06.000000000","tz":0},"subject":"erlang: use packages from Erlang Solutions on AArch64","message":"erlang: use packages from Erlang Solutions on AArch64\n\nRabbitMQ requires Erlang 23+ and soon will bump to 24+.\n\nOn AArch64 we used version 22 as there was no other option. Now it is.\n\nChange-Id: I667ccda5bb4edbfbfb97dfb0cea264837ed05cae\n(cherry picked from commit f943c7d773a280b832dde9edf9d615540b259b5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50b1f117a907cf092f2523fd561a3717e607e1d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50b1f117a907cf092f2523fd561a3717e607e1d1"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
