)]}'
{"id":"openstack%2Fkolla~829495","triplet_id":"openstack%2Fkolla~master~I7451528cd647e016834b98ff2397e74bac5689bf","project":"openstack/kolla","branch":"master","topic":"bp/support-openeuler-os","attention_set":{},"removed_from_attention_set":{"15054":{"account":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"last_update":"2022-03-07 02:17:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7451528cd647e016834b98ff2397e74bac5689bf","subject":"[WIP]Add openEuler Distro support","status":"ABANDONED","created":"2022-02-16 09:07:47.000000000","updated":"2022-03-07 02:17:06.000000000","total_comment_count":9,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"89d299a72ab93522bd5885d7d97296656d885da8","_number":829495,"virtual_id_number":829495,"owner":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-16 10:40:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":-1,"date":"2022-02-16 11:32:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":-1,"date":"2022-02-16 11:58:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 10:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-16 11:32:23.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-02-16 11:58:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"7e863820b064fbf43a894e20df463e691fb77d9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2022-02-16 09:07:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c53e0e099d609aba44db6bf594b1d60908a41810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 10:40:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/994d120fccca4f8085c6ce31cc2934b1 : SUCCESS in 36m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/811befd8b2dd42deabd792e93ca8e0c9 : SUCCESS in 42m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dad2d287339c46a1be19ca275f1eb0fe : SUCCESS in 1h 14m 28s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d648fef313594c0985332879aeaf0903 : SUCCESS in 54m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/433ea1ff6e2a474785dce1d4f49fdf18 : SUCCESS in 1h 26m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cd0cc3d18efe4cceac994f94125bf392 : SUCCESS in 33m 18s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/29a9388613264b468ef26c37e8a756fe : SUCCESS in 19m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fcaa3c64f9dc467ba4184a868d5078e6 : SUCCESS in 51m 48s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f7a11e09ad7e49f79c805dff6880f184 : SUCCESS in 1h 04m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d56bdc4ca6e4094b996c574f36a8700 : FAILURE in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b7021ebec75438493be1af01b004950 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b40f7ef0a4764084a04ea45213347b3b : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44fa99ce4224a4b902ecab7d7ca5488 : SUCCESS in 6m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ea48ceec053d4fc6b70e162f9776b2cb : SUCCESS in 18m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7ec2e01932748adb60643f722e3492e : SUCCESS in 32m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/50b1b7476b8c4bcfa4a50374201a9b16 : SUCCESS in 48m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1baccf17bfbc4550bef8faf8f7c9e35f : SUCCESS in 54m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ef1af365f60477c8e933561f221e96f : SUCCESS in 54m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bdd763a2092aa566d0f97ecdeed8cb9311a95489","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-16 11:32:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eee5e486bca401d4357566d896b6f6180829c5e3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-02-16 11:39:44.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e69b6d444f92c225962819bdcba60d6b355df044","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-16 11:58:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f8dce27890d89099b31a31f1398aea13c9ddcbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 12:08:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e7d2d0b5898a401bb9ad5959f4368ee3 : SUCCESS in 47m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d38beb5981464e75a094a2b410f408af : SUCCESS in 2h 02m 19s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/02b271e479b1424d80adb22ab8968c40 : SUCCESS in 2h 32m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1efb9acb84a8459580b2e7c60ddc295f : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96bb23823fa54650bfe2ec2a9f85c8ee : SUCCESS in 5m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8bfd3fa53a2430d9c9704cc359bc4c9 : SUCCESS in 2h 08m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89d299a72ab93522bd5885d7d97296656d885da8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2022-03-07 02:17:06.000000000","message":"Abandoned\n\nkolla don\u0027t want to add more distro support","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e69627d08347bc1f295adab20933d123c6a6215","revisions":{"3e69627d08347bc1f295adab20933d123c6a6215":{"kind":"REWORK","_number":1,"created":"2022-02-16 09:07:47.000000000","uploader":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"ref":"refs/changes/95/829495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/829495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/829495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/829495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/829495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/829495/1"}}},"commit":{"parents":[{"commit":"333fb35c79d2912a10b70001f5d8b5724e9b8651","subject":"Merge \"rabbitmq: update to 3.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333fb35c79d2912a10b70001f5d8b5724e9b8651"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2022-02-16 09:02:58.000000000","tz":0},"committer":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2022-02-16 09:07:30.000000000","tz":0},"subject":"[WIP]Add openEuler Distro support","message":"[WIP]Add openEuler Distro support\n\nopenEuler is the newest open source operating system, which is quite\nfamous and widely used in China. This BP aims to add the OS support\nin kolla.\n\nThis patch added the support for the newest openEuler lts version:\nopenEuler-20.03-LTS-SP3\n\nbp: support-openeuler-os\nChange-Id: I7451528cd647e016834b98ff2397e74bac5689bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e69627d08347bc1f295adab20933d123c6a6215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e69627d08347bc1f295adab20933d123c6a6215"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
