)]}'
{"id":"openstack%2Fkolla~829776","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Ic6e3ae4ff40c6ce5a5c0646ed309a2938903f6c0","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-02-19 10:10:48.000000000","reason":"Radosław Piliszek replied on the change"},"34076":{"account":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"last_update":"2022-02-19 11:26:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6e3ae4ff40c6ce5a5c0646ed309a2938903f6c0","subject":"Ensure set_configs sets execute bit on directories","status":"MERGED","created":"2022-02-18 11:03:23.000000000","updated":"2022-02-19 11:28:07.000000000","submitted":"2022-02-19 11:26:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829776","cherry_pick_of_change":826672,"cherry_pick_of_patch_set":4,"meta_rev_id":"b0fe72b65d2e6ef33fc7de2827ecf0b11b174c9e","_number":829776,"virtual_id_number":829776,"owner":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-19 11:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-18 11:54:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-19 10:10:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-02-19 10:10:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-18 11:54:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-02-18 11:54:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-18 12:15:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af878558b9451d91de5ec9481539d2e34691cd06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-02-18 11:03:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d693f51a082eb06efd0cc0de8807813ff4e30714","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-18 11:54:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2b2011d2a312d8dfb3932ef58e3d4bc361868ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 12:15:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/68a8bdac21da43288f480c6adcb13c97 : SUCCESS in 28m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d0475b9dc41c41e199558d87c4ce6d76 : SUCCESS in 32m 30s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c087ac3ff29e4951bbaf227d07ee0d93 : SUCCESS in 1h 08m 58s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/251e102cb0e947ab841583eee709ef76 : SUCCESS in 1h 01m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/58a4832612f7420dbaada78b2483c6c2 : SUCCESS in 56m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/52b624cab28947e4a15a919bc0a2de17 : SUCCESS in 23m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6a357465bc0d40b7915e234754bd1d1a : SUCCESS in 22m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a9c93a931c114a848a2a1091a33c3456 : SUCCESS in 51m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa0a4bb5ef849d89c09e108cf255dd0 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/864580215b5c46ac849adec059195953 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2e32300a8f947b09eb616bced5f81e7 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be482b736344ec9afa9d15bf5ec9b2c : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/939d69f12a8144348d086de2951b399d : SUCCESS in 3m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0eb6e47f3b0b4ae496868d6f4f747b77 : SUCCESS in 23m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb29149773b14647a0ea14c5cc790d14 : SUCCESS in 22m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/51333afe1e054f3f98e0176f60d020f9 : SUCCESS in 56m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/156cc2533d3345d4afec3a8340b4e5e6 : SUCCESS in 40m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/827371ad1250487ca0890adebece870d : SUCCESS in 49m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c27ff55f211abd5b1a1b3a2fdf4584a30a889deb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 14:13:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4863ad63a18f421ab597183ac3d00124 : SUCCESS in 2h 36m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80bca04fbabc4687bcea3401fce6dcfe : SUCCESS in 2h 06m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/603e37170bfd408a95836f1fb6781147 : SUCCESS in 2h 58m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a75cab64218c4f16a7d366ff50669057 : SUCCESS in 2h 10m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e3b9de09f28ea899b1d62179ab62c5a65fe2811","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-19 10:10:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb67c37c4328aec77c3c6a6745051f418898331b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-19 10:11:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9167a7e01aa97139906835f9e7cbcb04690ed4e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-19 11:26:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/077a400dfde840099d954f53ad5f03a6 : SUCCESS in 27m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/98788ef1c458407ebbb657dc8d411c65 : SUCCESS in 1h 04m 06s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/731e654e47be46e49703e5f824e78a1e : SUCCESS in 59m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fdf22b3c1bd446392a8dd8a1736a79d : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17f7c17063ac468c881f839f82edbcf2 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c623e683e92477eb743ef4482995c88 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aaa029a75714901b0fdc5c1d2a49ae3 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9803c6bf9d944f6a64624e6c0ecf4c8 : SUCCESS in 4m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e7152d70adf49acbb62a225d02664f4 : SUCCESS in 22m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/11e0c7b4f71a470abdb6dfda5a058140 : SUCCESS in 44m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/92c9e48d450a4275bf73e54ed007d9ff : SUCCESS in 45m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3b1f1d5587a1cbfae815f77a84d85af431bfe54b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-19 11:26:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b0fe72b65d2e6ef33fc7de2827ecf0b11b174c9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-19 11:28:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1273b5d7155743388a7207f3c2851421 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71622088810d4dd8afe402dd35bd425a : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b1ff36ac4cba497404852f329069f179659ed97","revisions":{"9b1ff36ac4cba497404852f329069f179659ed97":{"kind":"REWORK","_number":1,"created":"2022-02-18 11:03:23.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/76/829776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/829776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/829776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/829776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/829776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/829776/1"}}},"commit":{"parents":[{"commit":"f633f7bc320df01514a1379ef0173092c6405d69","subject":"Merge \"erlang: use packages from Erlang Solutions on AArch64\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f633f7bc320df01514a1379ef0173092c6405d69"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-01-27 13:39:54.000000000","tz":60},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-02-18 11:03:23.000000000","tz":0},"subject":"Ensure set_configs sets execute bit on directories","message":"Ensure set_configs sets execute bit on directories\n\nWhile handling permissions for directories, set_configs.py configures them\nsame as for files - i.e. 0640 set in config.json which works fine for file\nwill cause any potential subdirectories to lack traverse permission.\n\nCheck and permission change was added to handle_permissions function\nto add +x if +r is present for user, group, others.\n\nChange-Id: Ic6e3ae4ff40c6ce5a5c0646ed309a2938903f6c0\n(cherry picked from commit 1fe983df4fafa5a4a8eafe206e4abfaec2418da1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b1ff36ac4cba497404852f329069f179659ed97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b1ff36ac4cba497404852f329069f179659ed97"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
