)]}'
{"id":"openstack%2Fkolla~830858","triplet_id":"openstack%2Fkolla~master~I67adef1762d36f23b84ab8a2297e1f3385082b8a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-03-01 20:45:41.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-03-01 20:45:41.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-28 11:40:09.000000000","reason":"Mark Goddard replied on the change"},"30523":{"account":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"last_update":"2022-03-01 20:45:41.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-01 17:00:23.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I67adef1762d36f23b84ab8a2297e1f3385082b8a","subject":"Revert \"neutron: fix build in stable branches\"","status":"MERGED","created":"2022-02-24 16:01:51.000000000","updated":"2022-03-01 20:47:21.000000000","submitted":"2022-03-01 20:45:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":701880,"submission_id":"830858","meta_rev_id":"ce9fe4f65a89ae12efc5aeba4f355728355c4b77","_number":830858,"virtual_id_number":830858,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-01 20:45:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-01 20:47:21.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-03-01 17:00:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-03-01 17:00:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-24 16:01:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2022-02-24 16:01:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-24 16:01:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-24 16:01:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-02-24 16:01:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 16:01:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d0b971bc01d874f22c332beb11226ead81064a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-24 16:01:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42bb1b0cdfbb5c8e6c3ef01adf9e6be4fad8235f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-24 16:01:59.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"cd948bce4aacfae2bfbc20e50796045887bce4ef","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-24 16:38:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f86f6fabf97dbde83e22f22f21f15e4b137b8f12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 17:30:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a7e62b518b724fada8322e88eb843f5e : SUCCESS in 28m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f4404c761d2a4f56a03289e34ada88f1 : SUCCESS in 43m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6490bf0859674d67a6e03db31113d6be : SUCCESS in 1h 13m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2434e8d32d184aac8f36162e7129dcda : SUCCESS in 59m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a9afdb7b835a4ec4b09b5c7921a96c2b : SUCCESS in 1h 26m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6293ecc8620948ce9d9753a55dfeba4b : SUCCESS in 40m 54s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2e9aeace8b444be58dac27ef0d99397d : SUCCESS in 17m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f9ea1a22947d474d904056a35911d871 : SUCCESS in 59m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b35bc3a5b3d740caa02b1439759ef6f0 : SUCCESS in 1h 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5082f62efd4de8ac36b7d161f597cd : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b209cc4250c9485a85de806460d78ec8 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b783783370c441d8816aceec73bde45 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3eb38cf6464f0f95489654bdb4f39e : SUCCESS in 6m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7482142dbcd449adb7acb9ea2ad74cac : SUCCESS in 20m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a6ef97bcb9a4244821f9d8339cc0b9f : SUCCESS in 29m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd17a77d0a1d4279b53f94373c8625c0 : SUCCESS in 40m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/533d1a08b18b481e83c61731d70c0b9e : SUCCESS in 53m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec1343a660154864a68a446698f57a84 : SUCCESS in 57m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c32bb0a87855b4d6e0105d8d98e06922ef2130f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 18:49:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/24e7f125d2a44b10afc3b36fae8ecd79 : SUCCESS in 2h 44m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e2b7e618e4a24371a3702a269a5ea5ec : SUCCESS in 1h 52m 31s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fbc6c2794a734f118d2f5c930ce87957 : SUCCESS in 2h 17m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15a19103ed12493ca00d75ed1888317b : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f1fca572ba34cc0a5f3360db495a7f4 : SUCCESS in 5m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8de66578ba6e403fa54d66993ed51704 : SUCCESS in 2h 18m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e3b6477c73e6cc1cf68835856decde9b43b4b5b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-24 19:23:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"86f30de778945861cdbdf48c24833c7dbabb18d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-28 11:40:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c30de9ec4e70043c8e277f708ed7df041667486","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 11:41:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6b2f8c094096c9461eeca57887cc20a87f27c97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 13:13:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ecf2fcc1198b402cbcc79a488590d7c8 : SUCCESS in 40m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3a90cebdf7944856bcae6dbc6bad2c5d : SUCCESS in 1h 00m 49s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d27478bc83214694b4a28a97f37f8484 : SUCCESS in 1h 27m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/77e3aa5fdee44f3a9a4f91a7fc0e1587 : FAILURE in 7m 16s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/044c0beecd3b42118fb6537442c5c714 : FAILURE in 16m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7acec6cba388452d915512792e954fa0 : FAILURE in 21m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f23915bff1b4363b9a41847e7392108 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5239f14f87e544d188593a2fb64e5e3c : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c158894d6fd049af96234ba6d02d71b6 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce192a42bf54defb4b3240a9d78dd49 : SUCCESS in 7m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/254c1431f53f4b519c02b483c2319706 : SUCCESS in 32m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a000fa661d0440eebf9aa9373ea3d808 : SUCCESS in 55m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/46bd6c32ab174a3ba2afe27be03cb012 : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7d6d6c833f7e7aa39cf874e89de69f945665183a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-28 13:51:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16eaf2738972d991842123584eee06c0733c9ccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 15:13:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/97c5debec80e4ab4a678b8dc9998b578 : SUCCESS in 35m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cd6dc8d9a08c444d86eb51f3927ed67d : SUCCESS in 58m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bf4b35a83f284bf5a308416a7a17bb26 : SUCCESS in 1h 15m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2412446de31441a5863b5c01a585c138 : SUCCESS in 1h 06m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9ea6b8773e754cfd8d95a9c2ce261ce7 : SUCCESS in 1h 20m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0e5aab39f9c44f4ca8f156afa226e203 : FAILURE in 6m 42s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7ceba8dfa41447ddbf0acccfb6052984 : FAILURE in 6m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5bb6a7dfd80b4c80a91d3458c3eb6445 : FAILURE in 24m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9482e5697a1d4d14a82fa2e02d757f9c : FAILURE in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3e79fb2407243fe997acd78bdedb1e9 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/686afcb31e4f4f5999cb4b0946629fd8 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad4f362179804e5095eabff29073335b : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35e48bc623f1453284f37de2fa05281d : SUCCESS in 7m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cc5cf6659e6c45fd82a069eb3a260843 : SUCCESS in 22m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/22ea7ccff61441ad8f555091384f9076 : SUCCESS in 36m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bbc649bacd064365ab68fbaf935f3f86 : SUCCESS in 57m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/64c68b054b98402a8488c47ed4f7f5f1 : SUCCESS in 54m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/afa6be094277402ba3ecdff8a35cfe07 : SUCCESS in 1h 09m 29s","accounts_in_message":[],"_revision_number":1},{"id":"edb42fd3cc7bcda3841dade37da0144647cf17e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 20:45:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/08aa2bd9484f413997e608af85fed770 : SUCCESS in 2h 14m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8f8e8049e44845df8206119337d5a03d : FAILURE in 11m 22s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/19094bd2a4cb49d2874d0954124dfecd : FAILURE in 26m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4da2d198c2734965b41a16b7c1fabe70 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4904c806dd14713875d91a06cfa4e49 : SUCCESS in 8m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3efd4b42eec744cf8d074598d7455f5e : SUCCESS in 1h 43m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90e504ffc981742ffed9b303332d1eb3359e9e99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 17:00:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5deeebe05175befe63d75eb6e13cd5a515b0e151","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 17:00:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"407551745322eeafa6a102fdaf5da19c7554eef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 19:01:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ddc386816d114c8993f548c81d0d0bc4 : SUCCESS in 34m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f7da8d2711874b1885dc8b4d85ecf346 : SUCCESS in 48m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/148f65ec014e4a96802d27a3c58cf511 : SUCCESS in 1h 19m 01s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/91b408bc6114454eb608fe45c49a15bd : SUCCESS in 1h 10m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a5367d96ecd644e19dc608cdb0f2f3cf : SUCCESS in 1h 39m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eb4ed8260fd14b34b1f4df7e45f91c9f : FAILURE in 7m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0df7f7abaa134897b433498eccbaf010 : FAILURE in 7m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dc0e2b10380e437c8f227cbd2229ef6d : FAILURE in 16m 26s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8a2a0d9b13a2480ba897db9a93a241df : FAILURE in 19m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26807f6a40d44af9b716ae56ed7db294 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ca0f42fc5af4cf3ae1a718b42ee3a07 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ec1fe037e50457d87310d5cda2aecf7 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef8ecb2034b4db3950733a3e37c3339 : SUCCESS in 8m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bf0c6869784f463ab13149265a2d0ecf : SUCCESS in 21m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/120e1f129dcb495a95ad4cca607268dd : SUCCESS in 36m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/baaacf4bb28946f8b945f87abbaae42b : SUCCESS in 1h 31m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e1795bb6eba24daeb34a99d316fc142f : SUCCESS in 56m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b08590524f824d3e8e8f43d8aac267da : SUCCESS in 1h 52m 05s","accounts_in_message":[],"_revision_number":2},{"id":"55b1b0af77f106b11766b1ce132f0a850634a355","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 19:02:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87f4c5342e1633202e752eb56da8c4b259a00f53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 19:18:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7286e878605a42b0b3bacc376d628a6b : SUCCESS in 1h 56m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/27c1c08a35de429e9dd1b5674fd639ea : FAILURE in 10m 10s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/504d2309bad045ea807668b6e94cc29b : FAILURE in 21m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57d173f9c51a428681ee8ce9b473e35a : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d8eb02f83184dbf8419aa65cec07c7c : SUCCESS in 8m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/54e4d02e661142bdb93e989f7d7cf5a5 : SUCCESS in 1h 26m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc5550e89660fb988b6624a29e6b16e0058cbd92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 20:45:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/577198118cb749be8083323d0608d13a : SUCCESS in 43m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2e925c5879a145269092b71d3f0949b0 : SUCCESS in 1h 18m 19s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/419b02801af844fbab9fabdd8b8bec97 : SUCCESS in 1h 38m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aea35b733fe4a74b66b3caffedd662e : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/819974a5168b4e558d4885da1931d9b4 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/454519d332c24a3fb5cd485de04e5c52 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8b25bc487004837b7554313eacc254e : SUCCESS in 6m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/38281be6805447b38cd64a8f025c7ff9 : SUCCESS in 31m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/78ae7d7643f246c080de9ce2f19e0b57 : SUCCESS in 50m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/39fdc016d6274f19ae92b8e3e37fc0f2 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":2},{"id":"bdc2651402e08090151577f3a95aa4f7c43748c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 20:45:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fd044a9a2e6a5f3ac42378ba6b2e96339ded82ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 20:47:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c46c726a5bc41558f1934f6c5b54ef0 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ed3a2897445ad3ffefc2cff98218f7e207c34b9","revisions":{"872c0922459af7ae8431ab28291231e918d84333":{"kind":"REWORK","_number":1,"created":"2022-02-24 16:01:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/58/830858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/830858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/830858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/830858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/830858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/830858/1"}}},"commit":{"parents":[{"commit":"360d335e94246d7095672c5aa92b59afa380a059","subject":"neutron: fix build in stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/360d335e94246d7095672c5aa92b59afa380a059"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-24 16:01:51.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-24 16:01:51.000000000","tz":0},"subject":"Revert \"neutron: fix build in stable branches\"","message":"Revert \"neutron: fix build in stable branches\"\n\nThis reverts commit 360d335e94246d7095672c5aa92b59afa380a059.\n\nReason for revert: I don\u0027t think that\u0027s still needed.\n\nChange-Id: I67adef1762d36f23b84ab8a2297e1f3385082b8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/872c0922459af7ae8431ab28291231e918d84333"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/872c0922459af7ae8431ab28291231e918d84333"}]},"branch":"refs/heads/master"},"1ed3a2897445ad3ffefc2cff98218f7e207c34b9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-01 17:00:20.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/58/830858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/830858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/830858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/830858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/830858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/830858/2"}}},"commit":{"parents":[{"commit":"a9278d54378bc9d9a18ffa7d70f2e25aafdc16a8","subject":"CI: temporary disable voting of Debian jobs - gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9278d54378bc9d9a18ffa7d70f2e25aafdc16a8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-24 16:01:51.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-01 17:00:20.000000000","tz":0},"subject":"Revert \"neutron: fix build in stable branches\"","message":"Revert \"neutron: fix build in stable branches\"\n\nThis reverts commit 360d335e94246d7095672c5aa92b59afa380a059.\n\nReason for revert: I don\u0027t think that\u0027s still needed.\n\nChange-Id: I67adef1762d36f23b84ab8a2297e1f3385082b8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ed3a2897445ad3ffefc2cff98218f7e207c34b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ed3a2897445ad3ffefc2cff98218f7e207c34b9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
