)]}'
{"id":"openstack%2Fkolla~831033","triplet_id":"openstack%2Fkolla~stable%2Fxena~Ibd6c581d54d7048a0203be18828ba9b1ea7fb10b","project":"openstack/kolla","branch":"stable/xena","topic":"ci-octavia-stable/xena","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-07 10:48:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd6c581d54d7048a0203be18828ba9b1ea7fb10b","subject":"[CI] Add K-A Octavia jobs to the experimental pipeline","status":"MERGED","created":"2022-02-25 20:44:39.000000000","updated":"2022-03-07 10:53:07.000000000","submitted":"2022-03-07 10:48:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831033-ci-octavia-stable/xena","cherry_pick_of_change":830964,"cherry_pick_of_patch_set":1,"meta_rev_id":"c3c3ae2562c78205aa27f7665ab63a6232ec7b22","_number":831033,"virtual_id_number":831033,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-07 10:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-03 09:44:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-07 10:34:33.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-03-07 10:34:33.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-25 21:20:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-03 09:44:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-03-07 10:34:33.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"d756675e64eb9dc6773fc2f2c82339a29de949b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-25 20:44:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5566ec24cf7b110fbe07225f758f01fde4cfd4f2","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-25 20:44:45.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"7a7c60ad478ce3a8461138574f7c7cc12f315b79","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-25 21:07:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"baefe16a08668b31df81359a8cf20e730b987ea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 21:20:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4c65655d8e1408a99b7c781f42c20d2 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b4dd73e341e4c0b8b1f27b50daf5713 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a8ce2649c934aa8af6952bcc4214139 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a80a2f1d4ef4f4cac7c55c174cf5cb1 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72c7e278ee224028b4a8a6d60a3fd93d : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d99fd985e725f11da39ac260ae5ed08669602edc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 21:20:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db1d519b89e74293a908ac16cc60834c : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2eeeb5189eaa4c18a75d4a33611e8e99 : SUCCESS in 7m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d48f3b5ab043600be1a744cd8a93f5113b6d747","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 23:13:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-ansible-centos8s-source-octavia https://zuul.opendev.org/t/openstack/build/f0327f3fed4f467a8e0ea044cc7afba9 : SUCCESS in 1h 29m 33s (non-voting)\n- kolla-ansible-ubuntu-source-octavia https://zuul.opendev.org/t/openstack/build/00d4c512f2fb471d98f5b49123e44f7c : SUCCESS in 53m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4bbd730c5c1941d15398537b84741cbf57e8974b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 20:01:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99274962e323a731500207f0192cf222faac4a19","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 20:01:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"90e36af7de04e5d731bcdf8a75c032c905cbd956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 20:10:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd2d673c5f24267b1eccae130d9d652 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7d113f0e63d4412884ca84ce801bc7d : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d94863aaaf04f3493d6d7ab0e24281b : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f86df13293e446bead8296e3cb1b57fe : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d38c0f322ab84f9091e69aeeaff8825d : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":2},{"id":"1294c00dc1a533f5b150423cbe90a81a099075a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 20:10:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4625336a248479cb3beb35dd90abe37 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f349fac61b3c4741b51edc017597e1b7 : SUCCESS in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"261ce21a2d31299dcc5edba908f4048f5647ad62","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 09:44:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"98bd352238e5e3de0c70d4d60dd9a40f404fdfa7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-07 10:34:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"440d6104f37f2e8036291e8782fbc0c7e6d7bdfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 10:35:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b35633b07ccef8d1ce9d184563b79f9c6c220fec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 10:48:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/816b950787034105af305012d9bd0313 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d35baec6bf63445e8a3c526901fd8d1b : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23debb7d987d482fb9a4a224ad2c3ce7 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f72470387f4f5b863c1235ed182c0e : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":2},{"id":"4b673b449598fb7f540cc00bfdc7e16922705af6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 10:48:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c3c3ae2562c78205aa27f7665ab63a6232ec7b22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 10:53:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a335780a1f4289bf99b4762e130dae : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2b6785dfd067943ad0496f9d83d263ea4d385d0d","revisions":{"0efc0b5f4887b9a121440c1da3512265b7faa761":{"kind":"REWORK","_number":1,"created":"2022-02-25 20:44:39.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/33/831033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/831033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/831033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/831033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/831033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/831033/1"}}},"commit":{"parents":[{"commit":"b779128957fe8d082f0474810633535b84920d36","subject":"Ensure set_configs sets execute bit on directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b779128957fe8d082f0474810633535b84920d36"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-25 10:50:19.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-25 20:44:39.000000000","tz":0},"subject":"[CI] Add K-A Octavia jobs to the experimental pipeline","message":"[CI] Add K-A Octavia jobs to the experimental pipeline\n\nChange-Id: Ibd6c581d54d7048a0203be18828ba9b1ea7fb10b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0efc0b5f4887b9a121440c1da3512265b7faa761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0efc0b5f4887b9a121440c1da3512265b7faa761"}]},"branch":"refs/heads/stable/xena"},"2b6785dfd067943ad0496f9d83d263ea4d385d0d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-03-01 20:01:02.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/33/831033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/831033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/831033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/831033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/831033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/831033/2"}}},"commit":{"parents":[{"commit":"b779128957fe8d082f0474810633535b84920d36","subject":"Ensure set_configs sets execute bit on directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b779128957fe8d082f0474810633535b84920d36"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-25 10:50:19.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-01 20:01:02.000000000","tz":0},"subject":"[CI] Add K-A Octavia jobs to the experimental pipeline","message":"[CI] Add K-A Octavia jobs to the experimental pipeline\n\nChange-Id: Ibd6c581d54d7048a0203be18828ba9b1ea7fb10b\n(cherry picked from commit 4f0a9bf1022da289bb1a7d49f95525649981fc60)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b6785dfd067943ad0496f9d83d263ea4d385d0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b6785dfd067943ad0496f9d83d263ea4d385d0d"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
