)]}'
{"id":"openstack%2Fkolla~831411","triplet_id":"openstack%2Fkolla~stable%2Fxena~Ib79b8486f4d5064e4f249201d28cf5d6541c69ef","project":"openstack/kolla","branch":"stable/xena","topic":"qemu-img-stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-02 11:51:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib79b8486f4d5064e4f249201d28cf5d6541c69ef","subject":"Add qemu-img also in nova-libvirt image","status":"MERGED","created":"2022-03-01 16:40:18.000000000","updated":"2022-03-02 11:55:32.000000000","submitted":"2022-03-02 11:51:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831411-qemu-img-stable/xena","cherry_pick_of_change":830401,"cherry_pick_of_patch_set":2,"meta_rev_id":"a042f0cf28b1c7aacc412582812274faf555fcaa","_number":831411,"virtual_id_number":831411,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 11:51:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-02 10:13:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-03-01 16:57:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-03-02 10:13:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-01 16:40:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-03-01 16:57:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-03-01 18:24:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 10:13:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"eaeef789952cf2b49737d8446d2818135a7f14f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-01 16:40:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a5ee5e2a24d90ec2f3cd2099c5d4e19eeefc6e91","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-01 16:41:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"191ae8ca63b42ed120b252d3a29c6e3e91f64cc2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 16:57:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe2d0c8131df91747002c093facc8049a2abb642","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 16:59:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"488c78834814db39709b9db0beea76b26401d300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 18:24:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b5e58a37809747fda6871dcc929b155c : SUCCESS in 29m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b0729985ec7348219e01e7e5867e3024 : SUCCESS in 39m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a872ddd1788d431ca2ec2524436ad951 : SUCCESS in 1h 32m 02s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8047f6ee08124dc1aa9d07862e0d117e : SUCCESS in 1h 06m 00s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8c0a0848ac7a47b19f1d0b9081f4b15f : SUCCESS in 1h 36m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/461e74d4ff7f47f88d125bfe9cde7efa : FAILURE in 7m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a1522e0f8f034883ae61772fafbe4d53 : FAILURE in 6m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8403d498b9174431b3ad66257b945f81 : FAILURE in 15m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a67d30e73704789b51b9abd65ce946e : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f10e2d2eaa6b4714a50ec6768bee5265 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/693a077626f9455ca1d30837b6f651a2 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f0b23f79c854ba5b8cea62ab8ae1657 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b13153b9d2d4cd385ce621dc7db8d90 : SUCCESS in 7m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0082a6f34ed84f0abc7ec5b838f5c44c : SUCCESS in 19m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fea1bb93b07748b5ad109da820505473 : SUCCESS in 34m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/79200437cd384d469a2a98970188ecbc : SUCCESS in 56m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/236453729f1846a4bc5da8c8c89160b7 : SUCCESS in 55m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ffa943c11ace4eeeb12de49285d19cf4 : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":1},{"id":"16cab2f0997ea11e93c8ebe0e07713a61bff2abd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 18:38:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06b74a762cee73d8e38bbccd23550508883b3fcd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 18:45:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9be94d3ec7764dae93362c244e268c4c : SUCCESS in 2h 02m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d97ce49ae98b4e72b27b06ba6a3b852b : FAILURE in 13m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3d94105bd8c4379b257303deedfb334 : FAILURE in 21m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ebe0f3f2a4184896912b7f70070c83bf : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3740874356d948fb80427714c72f216b : SUCCESS in 7m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/150e5aea884249268a3ef9262703ef47 : SUCCESS in 1h 32m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4433a8c5bc5f8fe9648c736a58225558d3895e02","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-02 10:13:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed8dd8fe3fcf3a80a2dd15171f21cf2a9babe066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:14:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e4ce8a1e7067348f337790ca03c2e404d33ba05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:51:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1783fddfcff24a22b6ed3f6f7df671c9 : SUCCESS in 45m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/07762c3cb65d4f4cb271766d48f82bb1 : SUCCESS in 1h 09m 45s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/169cb468a37c46cfa9bd7f5516733306 : SUCCESS in 1h 32m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22690377b1f9491c848d73460b0f8d59 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2804d5fdd30544ce955a87112f248459 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c92c0c61b70744f89ce0ab4c64436776 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/850def4106f24fd9a10ab570e82f1fc6 : SUCCESS in 8m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0e68cf750694bc5ad55af62b6f74d44 : SUCCESS in 32m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e8f038988a884f5d873912e66dd52020 : SUCCESS in 51m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/520fe92b7a474c9c907ef104216a1c9b : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":1},{"id":"81321f54b97829d8c00cc9f4b975903e02c0ab76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:51:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a042f0cf28b1c7aacc412582812274faf555fcaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:55:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b63d632b4124f33bc0d81c3389b00fe : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df511582268f554701ff6f3954f558f86856ba6a","revisions":{"df511582268f554701ff6f3954f558f86856ba6a":{"kind":"REWORK","_number":1,"created":"2022-03-01 16:40:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/831411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/831411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/831411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/831411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/831411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/831411/1"}}},"commit":{"parents":[{"commit":"b779128957fe8d082f0474810633535b84920d36","subject":"Ensure set_configs sets execute bit on directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b779128957fe8d082f0474810633535b84920d36"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-22 12:14:39.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-01 16:40:18.000000000","tz":0},"subject":"Add qemu-img also in nova-libvirt image","message":"Add qemu-img also in nova-libvirt image\n\nThis is noop on CentOS (it was installed as a dep already - better\nbe explicit) but installs qemu-img (and other qemu utils) on\nDebian and Ubuntu.\n\nqemu-img may be used by libvirtd to create non-raw images as it\nhappens, e.g., when Kolla\u0027s libvirtd is used by tenks.\n\nChange-Id: Ib79b8486f4d5064e4f249201d28cf5d6541c69ef\n(cherry picked from commit 0236bc6d940ddb35c0fdbaac9dc55885b972d2b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df511582268f554701ff6f3954f558f86856ba6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df511582268f554701ff6f3954f558f86856ba6a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
