)]}'
{"id":"openstack%2Fkolla~831416","triplet_id":"openstack%2Fkolla~stable%2Fxena~Ideae0c81d7c3833c8b0e22fce8691e9e42949ec5","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-02 11:21:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ideae0c81d7c3833c8b0e22fce8691e9e42949ec5","subject":"Use python3-openvswitch from distro","status":"MERGED","created":"2022-03-01 20:02:25.000000000","updated":"2022-03-02 11:23:31.000000000","submitted":"2022-03-02 11:21:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831416","cherry_pick_of_change":830844,"cherry_pick_of_patch_set":11,"meta_rev_id":"97e99390039d7a78a2bfc164da93839c43774ad2","_number":831416,"virtual_id_number":831416,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 11:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-02 06:15:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-01 20:02:38.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-02 06:15:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-01 20:02:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-03-01 21:34:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 06:15:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"cd8cace5d1c2d28d0fb43d307a8b839d75f137d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 20:02:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"85229ad063373a866e2f73bb10525638c4220446","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 20:02:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da078adc5719c19410c3229359cd011328a909bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 21:34:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3eebbd77ab0645e487341ec7eb8cade7 : SUCCESS in 27m 56s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ce0540dfe26240a790b85910a9aaa982 : SUCCESS in 33m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2688922ca0f4411da99dc44f5cd4f8f0 : SUCCESS in 1h 17m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4d4298bafefd4438a2926bfaf74257c2 : SUCCESS in 1h 19m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9d667d2b5c27482287a8d687a2fd7d43 : SUCCESS in 1h 30m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9bd8db7a2c9c4335bd6dbcd828af231a : FAILURE in 6m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/21b15d059f8e467a8bf8fd82c6fb9d04 : FAILURE in 6m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9806cc70119c42fe82e1f5cd4ebfc46a : FAILURE in 18m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bec787afc0a547a7869617b558604039 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a87597e580c445bb8c0195f21177e3a8 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d90135793924a0f90cb3ed28a6e2342 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c1840a6b110410c85b4f82573173059 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db0f8f2c25d4ac7a1ca7209eb3252d3 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f04c3e16d76e451aa5349ccbcc4fadd4 : SUCCESS in 5m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ef9c22bf0215491d998daf9a5a510ff9 : SUCCESS in 18m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0da5e81ee10245e18262aa404c25d691 : SUCCESS in 31m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/79ae9ea70c0d4640a60e7f37f5edbbe6 : SUCCESS in 42m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c5234f353f5942d2a8f94cd36fb85803 : SUCCESS in 52m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b53e7ddea27147ed981ff7e83be0ec3e : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":1},{"id":"fa4c5e061a4160d82466e099949b8f9406bc433c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 22:05:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/69376a9dd8a644cd8f809cce1d619458 : SUCCESS in 2h 00m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6039bea4fdf54935b82b0f0d5773362e : FAILURE in 9m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/00d8e075b41c489999c3ce333a28d8db : FAILURE in 23m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52c5b18187864e3d8c930dc9df5df798 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8716f67faf342e7b43f2b1b73f65fba : SUCCESS in 7m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/32af2adfd23a4b32a5fede50dc9593d0 : SUCCESS in 1h 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"482ac1af5140ea2ecca4156427ff22b364aa864d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-02 06:15:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0970ef3a6b9458a986473df01b7d8e49d17c0bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 06:16:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"272943b7ae529606a26606d18ed78f1be3861584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 07:48:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ed6c7e7fc79c466aac011aaa825bea0f : SUCCESS in 34m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f854f03686f5497ab2b36b9f3743bede : SUCCESS in 1h 09m 08s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/059a37e0892a48b8af5c6287f1a28e52 : SUCCESS in 1h 31m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/836751520e164eb7b4d129b6c83bf456 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f27ac196bd724aa698201eda34a04489 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63a600d633f74218b36feb34a012ed1e : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6dfe566d3dd49e3977abc1b5c759649 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c57969b7a0884883a44409c698e7ff76 : SUCCESS in 5m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b9e224f53424c0594e2b203eb19b52a : FAILURE in 35m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/495fb4421d6545f395961484ffc2e48a : SUCCESS in 45m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/02ad9bba97314754b4df52280d6883bf : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"26bb446e915dd828e83ccf2a8eb9cdd90045b01f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-02 08:08:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e479912ab267aefc81e007ff001613ebe54c5fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 09:46:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6d0a3a6598ba47dbaae28e8ce5e4896f : SUCCESS in 32m 59s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5968d66d51f443f390bec81a3f488ccb : SUCCESS in 31m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5d4e6e2ed07f44fdb510960575bb7e75 : SUCCESS in 1h 14m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f38a0ea168924c3f8d354ad192c54d04 : SUCCESS in 1h 01m 37s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/52f065350d3946e191076a976150dc48 : SUCCESS in 1h 33m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b20124352c254506853c80de395a7e95 : FAILURE in 7m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8a72a89d54974b32b7e74f4390c7e2a9 : FAILURE in 9m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b12056d93cf24349866139bd9e1ddf75 : FAILURE in 17m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d8bd3b8cd5d49ee9e86eb60394406d9 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bddb047731346c4bfb2be7262f8332a : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5529e630d46d4c82866bde80b683b086 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a2ca4976154402b917e17a53d4b09e4 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da85d5bd8374060a485e60508190ce5 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33a69475abac44e5bfa4c2e976f2f76b : SUCCESS in 5m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f866951a1c024c24853dcf6f76543b76 : SUCCESS in 20m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf17e3b86e6b462e8308bd5e1770cdbd : SUCCESS in 28m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/235935a375a64b418e09521cf3a2df51 : SUCCESS in 44m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/115d0f898ae544c68d369c604a0edfac : SUCCESS in 49m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b0bb610ae0854f00856d02477dfaac4b : SUCCESS in 1h 06m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1ddf409d0de68f090d9444805fd4a192b3115d59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 09:46:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f2ad25505aa6b05558c4906c6fc54c145e21812","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:15:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/61342e86f25340f39e29b7bb26267fab : SUCCESS in 2h 02m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/98c4c30126b04a248ba3177fc73fb930 : FAILURE in 11m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80b92c9537ea405e95b74ef6f153e4fa : FAILURE in 20m 39s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1579e07849d94a40933fbd0e75e1f999 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70a1e84edd0b448ebb502cd19a2f2c50 : SUCCESS in 7m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d271d334fa04538a34f146599dd9638 : SUCCESS in 1h 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"384e68582f8a22c7445e552681aabe33e6f61b2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:21:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/84188feda1df40cf807d4948b123bb2a : SUCCESS in 37m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/363b6aa1b05e407e8ba75bee3e2a578b : SUCCESS in 1h 09m 01s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ceb2c09597e64b218d4806d8a6bea5f0 : SUCCESS in 1h 29m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b023ef04d94d418abecc48f470260434 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18f4b17289cd4ffbaecd92beb174d5da : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aa8dbc3e2b74275a94ff3ca45d4d900 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b5041bfc9e64f6e8315ad2efc5650d4 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1f1871820674404a29a73b794082b83 : SUCCESS in 5m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba984c74e0bb402ea7a78416c41550f1 : SUCCESS in 31m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/797a24bf5be5485f8e7175549a9078ed : SUCCESS in 49m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7298b11190bb413d936bac9a408504ff : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f20f0b7c90037fdd66c8d4a6b5d756385ef88316","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:21:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"97e99390039d7a78a2bfc164da93839c43774ad2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:23:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65ce179312ef464196033669d2931d8c : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fc1ea7ac3f8466fbf87869f73344b40 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6760c2a98c6a2c86e357aea1de6be2ca7526f23a","revisions":{"6760c2a98c6a2c86e357aea1de6be2ca7526f23a":{"kind":"REWORK","_number":1,"created":"2022-03-01 20:02:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/831416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/831416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/831416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/831416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/831416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/831416/1"}}},"commit":{"parents":[{"commit":"b779128957fe8d082f0474810633535b84920d36","subject":"Ensure set_configs sets execute bit on directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b779128957fe8d082f0474810633535b84920d36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-24 16:00:56.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-01 20:02:25.000000000","tz":0},"subject":"Use python3-openvswitch from distro","message":"Use python3-openvswitch from distro\n\nWe should use the same Python OVS bindings package\nversion as running OpenvSwitch. See related bug.\n\nCloses-Bug: #1961874\n\n(cherry picked from commit 8e8a499adc3ae0c4009bacde65cf3a141632ef3c)\n\nChange-Id: Ideae0c81d7c3833c8b0e22fce8691e9e42949ec5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6760c2a98c6a2c86e357aea1de6be2ca7526f23a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6760c2a98c6a2c86e357aea1de6be2ca7526f23a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
