)]}'
{"id":"openstack%2Fkolla~831417","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Ideae0c81d7c3833c8b0e22fce8691e9e42949ec5","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-02 08:14:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ideae0c81d7c3833c8b0e22fce8691e9e42949ec5","subject":"Use python3-openvswitch from distro","status":"MERGED","created":"2022-03-01 20:02:46.000000000","updated":"2022-03-02 08:16:12.000000000","submitted":"2022-03-02 08:14:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831417","cherry_pick_of_change":831416,"cherry_pick_of_patch_set":1,"meta_rev_id":"1ccc09e2b84004a3ebe5479e6654e27927848683","_number":831417,"virtual_id_number":831417,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 08:14:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-02 06:16:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-01 20:03:04.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-02 06:16:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-01 20:02:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-03-01 21:19:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 06:16:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9d79749936d7ece0a68e96a1abdcafc3fd30ac18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 20:02:46.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"7963579f7879a3068d31d016a12c03b1d60710e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 20:03:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"40f8afa5a41fccd3707995278a33e281f1af717f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 21:19:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2894fafe0f134d2abe6b7e9d50c97c17 : SUCCESS in 30m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bd95189e7ad94f67bb4a1d32817c1b8a : SUCCESS in 38m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0b3ce83803974b8982479b7fafa32268 : SUCCESS in 1h 01m 57s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8b4a5821341b400ba10724ef4c7add5a : SUCCESS in 1h 08m 12s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b277c9a2167444e99d2162877c0cad97 : SUCCESS in 1h 05m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/303c25ea1256475f941f1c67c88cb313 : FAILURE in 6m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/177c50488997461e9df4ac5fd153d4b5 : FAILURE in 6m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2f34f4bae5ba44bf9bf57761cc4ff2fe : FAILURE in 15m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c3195b69ee4a67b83ecae4d3449501 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98b1532cd5f24e85a96db404a24fb39d : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/855116905d4a4eea9beb431fa5c361c4 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34913cc4bf4b4b08a10718e4eb961a76 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/502011a0751242bea2d5200fd22fb4d8 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de04e61eaf8a4860b494b70893a2fb6b : SUCCESS in 5m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99207aa38a6d4e6c9f76cba8a8b704bc : SUCCESS in 21m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b1a0f60460e841fa8fcd67b68168366c : SUCCESS in 35m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa60491f03f14191ba79925bb46990b9 : SUCCESS in 53m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b46e4d00b694787ba95e827c6cfe70a : SUCCESS in 1h 10m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/be01219f350d4de183cbc03f40c507ce : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3d750804cc453c08a8f82dcff1b1a639ffa2890e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 22:10:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8d79d1f60f54451fbc33ff07b763b1c0 : SUCCESS in 1h 59m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/90d98f2d8cfc4b2897b44c698d168c1d : FAILURE in 9m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8f53aed1987a470684e65b2d36bce3f8 : FAILURE in 30m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cace23a3a7ab45d7b252d5835c5d122a : SUCCESS in 1h 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67edf17889323ca93fd982e03ab92a5527c1004b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-02 06:16:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1867ab9851675d3b0ea65f63256cabd09c9f2484","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 06:16:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37fd8333b55dbb9c060ddb601f4b61e010497ffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 08:14:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5e4908ab9b2c4a6c8c02a358eee98409 : SUCCESS in 41m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d0686c54da9d4f90a70175e193717c4e : SUCCESS in 1h 09m 41s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e13137f074b440ac99d6c52b0cdee92d : SUCCESS in 1h 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d67189f0d0284ce5bcdfa9d9ddd06e87 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a90ba23e593b4c6e961a8fdb834d9c1a : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ced5c1713cf14b878c008baea0cbae85 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b83123ed394f3583d12a4ddbb7d3e0 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88e101b1b1324ca79bbc0ab32cdd1c71 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7654bf0cfbde405faa23f93076a45f6b : SUCCESS in 35m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d71b0f2325142f9a5c5704cb55b7ae5 : SUCCESS in 1h 02m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fccf926ca0f34d1b9f630382cdb2ca8d : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9b68b6c24a68b28154c3ab9817bf1dc0d18d4c97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 08:14:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1ccc09e2b84004a3ebe5479e6654e27927848683","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 08:16:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b662932aa1624acd993e4f5725b4b2b0 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1adaedc9a964f8ebffd38f3cc445034 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3e2bcdc2008acdb6a8207fd015bc09dd181e66b","revisions":{"b3e2bcdc2008acdb6a8207fd015bc09dd181e66b":{"kind":"REWORK","_number":1,"created":"2022-03-01 20:02:46.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/17/831417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/831417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/831417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/831417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/831417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/831417/1"}}},"commit":{"parents":[{"commit":"94cecf86e1ad35b521e7bc5d526ca1dfa9edbd09","subject":"Ensure set_configs sets execute bit on directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94cecf86e1ad35b521e7bc5d526ca1dfa9edbd09"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-24 16:00:56.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-01 20:02:46.000000000","tz":0},"subject":"Use python3-openvswitch from distro","message":"Use python3-openvswitch from distro\n\nWe should use the same Python OVS bindings package\nversion as running OpenvSwitch. See related bug.\n\nCloses-Bug: #1961874\n\n(cherry picked from commit 8e8a499adc3ae0c4009bacde65cf3a141632ef3c)\n\nChange-Id: Ideae0c81d7c3833c8b0e22fce8691e9e42949ec5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3e2bcdc2008acdb6a8207fd015bc09dd181e66b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3e2bcdc2008acdb6a8207fd015bc09dd181e66b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
