)]}'
{"id":"openstack%2Fkolla~831418","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Ideae0c81d7c3833c8b0e22fce8691e9e42949ec5","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-02 08:23:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ideae0c81d7c3833c8b0e22fce8691e9e42949ec5","subject":"Use python3-openvswitch from distro","status":"MERGED","created":"2022-03-01 20:02:53.000000000","updated":"2022-03-02 08:25:21.000000000","submitted":"2022-03-02 08:23:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831418","cherry_pick_of_change":831417,"cherry_pick_of_patch_set":1,"meta_rev_id":"be2847821948cdee02d3211a14467b0b803ec3a2","_number":831418,"virtual_id_number":831418,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 08:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-02 06:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-01 20:03:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-02 06:16:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-01 20:02:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-03-01 22:38:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-01 23:11:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 06:16:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"19ca86ab615c069ebc9074e29814068e01851721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 20:02:53.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"e36870708c3b82ca03ad96dcdfddb2ad31ca8f9d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 20:03:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2afbfd1ec9638ebe291cf4e605cacf4b3e068953","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 22:38:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b52fece646d4fb58ad3c20589fe62ca : SUCCESS in 1h 46m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/066391ee4857457c84689a154091e431 : SUCCESS in 1h 26m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0de95f7ee1084bbcae88d71b4ff93701 : SUCCESS in 2h 16m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3032b14cc435428c9d9ab5de366ee68b : SUCCESS in 1h 26m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9224bf42fa6a56ec93d58dcda651b75d6654b2ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 23:11:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1e04c1efb1864233bb44a16610a2b55a : TIMED_OUT in 3h 03m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a19589dafe0d4b2fa26806efb73c2b97 : SUCCESS in 31m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e9b155723e7e40a099e163d0161983f7 : SUCCESS in 1h 13m 01s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4a3836fd3f094e618c38f37ad72cbad9 : SUCCESS in 1h 09m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d256e3820b7643888ff50a5393fefe0a : SUCCESS in 1h 15m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b537b9107ea54533b21f48a6e462b9c9 : SUCCESS in 22m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3e45df4b3cc84048b9aa76072df4a1a6 : SUCCESS in 20m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8481c2cdd42c4e869e5700276c4c8470 : SUCCESS in 53m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f23cd9423c4adfa2092aa720b4a613 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe1c39d33b864a749c5323c1cddafaaf : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffd43e5f623e4381805e86183612cd82 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/930ce314d7114d2daf37279c3e98219c : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5542d85a38094f82a562ded563cb9da2 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/813f3461f49d4acd8e6b33f9b47eae93 : SUCCESS in 21m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d698dc9add114c6e9dbe047dc4260694 : SUCCESS in 22m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/48d61cae464147679a260c997bcc97ca : SUCCESS in 49m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1f8638b8c5c24dbeb06faf88040cac60 : SUCCESS in 49m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec2db1e3ed7344d3a67b8927a72c8730 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"610a01e70019cda3ed27d60fdb367c0b3068dac0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-02 06:16:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c3dc5d5f8efe1b6275cf9be60606654321ce542","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 06:17:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55b399230bcf7131573653692e2208c59063a464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 08:23:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3f08eb66c946460b916095df94234ebd : SUCCESS in 33m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e062272394094a4db3d4654719c1be30 : SUCCESS in 1h 07m 43s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6b30e245acf94a4d8480a99c8c19c9b3 : SUCCESS in 1h 25m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28747580d0094c05b5f1170a7127c3d8 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edad88cab50a4793b595eddf71bf339a : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46e5a37428b840cb99ec45f6d3f81a10 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7246321833844085829065acbbee50f0 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd9c30c540c54b4c9f3ea2a1b35a4850 : SUCCESS in 5m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/55ae88a8e469487ca918200b0db88a23 : SUCCESS in 18m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7cbbc42e96224bc4af03e4c041046bd4 : SUCCESS in 46m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/07f993caaf9c4b7b919b17d38b5f73d6 : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":1},{"id":"90ee9768f3abdb1415218acbcab8a2f6146f16dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 08:23:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"be2847821948cdee02d3211a14467b0b803ec3a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 08:25:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d3fefc33424a6cab022d4052aa59ff : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/feecd32ddf89401989a49b678c99048e : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61f1705bce5628c6d19903640748c23c003f9fb3","revisions":{"61f1705bce5628c6d19903640748c23c003f9fb3":{"kind":"REWORK","_number":1,"created":"2022-03-01 20:02:53.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/18/831418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/831418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/831418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/831418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/831418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/831418/1"}}},"commit":{"parents":[{"commit":"9b1ff36ac4cba497404852f329069f179659ed97","subject":"Ensure set_configs sets execute bit on directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b1ff36ac4cba497404852f329069f179659ed97"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-24 16:00:56.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-01 20:02:53.000000000","tz":0},"subject":"Use python3-openvswitch from distro","message":"Use python3-openvswitch from distro\n\nWe should use the same Python OVS bindings package\nversion as running OpenvSwitch. See related bug.\n\nCloses-Bug: #1961874\n\n(cherry picked from commit 8e8a499adc3ae0c4009bacde65cf3a141632ef3c)\n\nChange-Id: Ideae0c81d7c3833c8b0e22fce8691e9e42949ec5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f1705bce5628c6d19903640748c23c003f9fb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f1705bce5628c6d19903640748c23c003f9fb3"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
